)]}'
{"id":"openstack%2Frequirements~879230","triplet_id":"openstack%2Frequirements~master~Iff1489f1b0abe5d4b01f2137aa61b889f48454cb","project":"openstack/requirements","branch":"master","topic":"cap-virtualenv-setuptools","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-14 06:53:11.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Iff1489f1b0abe5d4b01f2137aa61b889f48454cb","subject":"DNM: run all jobs on virtualenv uncap","status":"ABANDONED","created":"2023-04-02 12:18:52.000000000","updated":"2023-08-31 16:22:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a6a7ea820fea523a35d75219a745947804e4fee","_number":879230,"virtual_id_number":879230,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-02 14:23:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-02 13:10:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-02 14:23:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e4296c57f9ad05ee61edfd58065e97dc7a99dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-02 12:18:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c49533e76bc8d4dbeda87a45ab8a8e1c6eabfd79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-02 13:10:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b19ea9e213b84623a3481b6b926e7109\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6fde6fa0a9014cb28339379f3f6fdc7d : FAILURE in 6m 16s\n- build-wheel-cache-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/03a514f993d9420c8a293eb145b41ede : SUCCESS in 26m 20s\n- build-wheel-cache-debian-buster-arm64 https://zuul.opendev.org/t/openstack/build/a0ed033900a34e15b52f30b12e482c0a : SUCCESS in 18m 33s\n- build-wheel-cache-ubuntu-bionic-arm64 https://zuul.opendev.org/t/openstack/build/2be0b69e26444cefb6af73ba8a6c13b6 : SUCCESS in 35m 58s\n- build-wheel-cache-ubuntu-focal-arm64 https://zuul.opendev.org/t/openstack/build/156a97de9dcf45e7863dac2c65a837a5 : SUCCESS in 18m 57s\n- build-wheel-cache-ubuntu-jammy-arm64 https://zuul.opendev.org/t/openstack/build/f580fec6878e4f3f93524e4f8bcc124c : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"061b45dc0bdd52dc4ba71ff133a81f835e73085d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-02 14:23:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bcac10e05801429ca666f4156cd85bd6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6abea3a0d1448593587e229d8ebb60 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2445b9fe8f414f9c9c59e0b4796d1d18 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f329414d05d460ea32fd5044aed94b9 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f166b5de1e4345aeb4d5e869a2f0d2bf : SUCCESS in 3m 22s\n- build-wheel-cache-centos-7 https://zuul.opendev.org/t/openstack/build/1305d95f1f2f4b4d9e4bb95654e5daa9 : SUCCESS in 1h 01m 59s\n- build-wheel-cache-centos-8-stream https://zuul.opendev.org/t/openstack/build/a14847e0914b44f1a3aec11f6050f89e : SUCCESS in 1h 13m 28s\n- build-wheel-cache-debian-buster https://zuul.opendev.org/t/openstack/build/421cb58f357c4dad86c0a3e884e280de : SUCCESS in 1h 03m 55s\n- build-wheel-cache-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b462b9699c974c8ab8b0ca6e7551d0a1 : SUCCESS in 49m 37s\n- build-wheel-cache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b8a85e4965e4da28b3658568464d262 : SUCCESS in 45m 58s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fcb69bc426c4c19842c6b2ee61e99ed : SUCCESS in 52m 05s\n- build-wheel-cache-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/263c7f5af1f54d96970ab72d6d171ec2 : SUCCESS in 1h 56m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e5181d8589d44478a8b735c88c66b69 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f77f320341e4bfebce903f81ec77c6d : SUCCESS in 6m 25s\n- requirements-tox-bindep https://zuul.opendev.org/t/openstack/build/cabd9d8b69f04b4a894c056c914b80f5 : SUCCESS in 3m 38s (non-voting)\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eada739ebc5a4a9fa730483a3dc5bba3 : FAILURE in 7m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e0376e423a7449a693e9bcf4576ec2c2 : FAILURE in 11m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f6946686d2cf4eb48666094c43b26d7e : FAILURE in 11m 47s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6a5dd9ffb81f4895b7f054ade4e24e64 : SUCCESS in 6m 32s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8ed430a865fa439396bb9ee77fc76d94 : SUCCESS in 4m 45s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/27f335b30e1b4ecf9e4aa225c1b989b9 : SUCCESS in 7m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/350f6d6a4fb2424fac8167b915c0c9e3 : SUCCESS in 10m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f056251ed67b4cc0bfceb9f8d367da6a : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6bcd7bf690764b30b5d406a88314dbbe : SUCCESS in 4m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/94fbbe74a98d44b682b184e85dc27c55 : SUCCESS in 8m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/cea8f4d020a94051a7be12421b80ce5e : SUCCESS in 7m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76565026b6b547e7afc8b7ac3fa523b1 : SUCCESS in 9m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/422ae5519f8c4639b3abe2891da90b7a : SUCCESS in 8m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bbb03a631f2b4c57a299da18c66c4abb : SUCCESS in 9m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/78b7f37c52df4b18a7e029929b32a9ab : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/da27ae5898d948a8922a39481941470b : SUCCESS in 9m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/51bea961340244c7b735e35b2a049374 : SUCCESS in 4m 35s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c144141798c04732bd0821f5edd1c354 : SUCCESS in 8m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0608abf2bc5a40e4a20dc188f8497183 : SUCCESS in 29m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/919dacb11fe34ee8ad3a055921d55bd8 : SUCCESS in 12m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d05a7e03c4d34189a30a4d04e8766f8a : SUCCESS in 10m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3b835796684848dc8ca279eb89bb5163 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fa6320ec741b4caeb450c996a20d8bc0 : SUCCESS in 5m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a697bcf6f1e742a7834e5dd7e4e0dcd5 : SUCCESS in 23m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a4110ec8ce6a478f83070242017cf164 : SUCCESS in 4m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8fcd32affd5f4cd0bc4c1da2650c3562 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4dbe96abdcc44c6c9318bb3b0377a537 : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/20d262ed73af47ea8b1c6f02b13add80 : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5633e440c9904977be5113f314f91680 : SUCCESS in 49m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/746770b9dfb84f729aa549f396b9d416 : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"2a6a7ea820fea523a35d75219a745947804e4fee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 16:22:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7b63c587ae81b0f35192cba8d03dd37b865f47d","revisions":{"e7b63c587ae81b0f35192cba8d03dd37b865f47d":{"kind":"REWORK","_number":1,"created":"2023-04-02 12:18:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/30/879230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/879230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/879230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/879230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/879230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/879230/1"}}},"commit":{"parents":[{"commit":"2fac876afb95c7bd2be5562fcf9c01941df3a9c2","subject":"Revert \"Cap virtualenv (\u003c20.18.0) to unblock gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fac876afb95c7bd2be5562fcf9c01941df3a9c2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-02 12:12:33.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-02 12:18:37.000000000","tz":120},"subject":"DNM: run all jobs on virtualenv uncap","message":"DNM: run all jobs on virtualenv uncap\n\nChange-Id: Iff1489f1b0abe5d4b01f2137aa61b889f48454cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b63c587ae81b0f35192cba8d03dd37b865f47d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7b63c587ae81b0f35192cba8d03dd37b865f47d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
