)]}'
{"id":"openstack%2Frequirements~879231","triplet_id":"openstack%2Frequirements~master~If57cd13d1363af3124ab0ed36a5f571174a1d6fb","project":"openstack/requirements","branch":"master","topic":"salty","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-05-10 01:47:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If57cd13d1363af3124ab0ed36a5f571174a1d6fb","subject":"Bump salt to version 3006.0","status":"MERGED","created":"2023-04-02 12:37:20.000000000","updated":"2023-05-10 01:48:16.000000000","submitted":"2023-05-10 01:47:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879231-salty","meta_rev_id":"2c1a991e5695e0266e6b9d1a926efec6ce10bcaa","_number":879231,"virtual_id_number":879231,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 01:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-04-23 12:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-05-10 00:22:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-05-10 00:22:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-02 12:42:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-02 13:49:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-23 12:04:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-05-10 00:22:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5c96be9c074f0c15ba0d4a6d2ca794091670cc61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-02 12:37:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b80d09e45c1e585fe684de061f1712b3fff39397","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-02 12:42:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b33a85d3400d41e7bb067b1405437a56\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b03c1251647d46df8e61374b00e528d3 : FAILURE in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5e90beb1df83582cf61e8ed713d2cf0ff0898718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-02 13:49:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2919afd9a1864bdabecb65e923f8df7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8954521c4444527afec94107f8031a9 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d2314f920948138c8a522ae98710c0 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd412d8d4a264eadb2c00aed84b5dfef : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/220b7febebca4e1a9fe0e522d15ea4ef : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d5c43832a2245c89565fffbe91fad4d : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17756b13a6af4a57814374162211a775 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f283b20a5c0c49f9a8f007b331d8f1d0 : FAILURE in 3m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/af263863437e448fa27b55e1ea3a8fb5 : FAILURE in 4m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/05aabbc75d84427e8cf06302dde77eb0 : FAILURE in 3m 53s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e9982535d1234cb08b922f0da2af592c : SUCCESS in 7m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7c5a4d889f184eda91c237b5ca281dcd : SUCCESS in 4m 22s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f33f30df9dff481d97208c6e5db4666b : SUCCESS in 4m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/78ab540edf114616ad8d5b316ba338ff : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6acc50647e914f3887b6a1424b9873b0 : SUCCESS in 3m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/302ce509eac648fc84450d1fc6340214 : SUCCESS in 5m 42s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/718701cbfed042e6beea9d098740baad : SUCCESS in 7m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4b33ebba722342f587e82ff1d9afb207 : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb4d51f9ddde4a3c853f805555567da4 : SUCCESS in 7m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8616b75c655e482fa2a749eb63abbebb : SUCCESS in 7m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/68165fc6b66c4ee4a5c86aa7842c46f6 : SUCCESS in 10m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/36058f4508104693938b50ff772c8996 : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/52f415737d0141a9bf862c6dbc837a0d : SUCCESS in 7m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/22c84706d29e40c994e005033c9bcc77 : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a6404f5dcaf04c52877a6e366259e134 : SUCCESS in 8m 44s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/58668797c3af451b99450abb87b0e41e : SUCCESS in 26m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9f6cf2d34ff14d869cd56c08d4c0def7 : SUCCESS in 8m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/126b250017bd4a4cb2bc52594174eac1 : SUCCESS in 11m 15s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0d6f3077001b4302981212e46aab79cc : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/375f1b2416064d4cba1274b9169adb91 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6aa8768958e945b2b72aa1cbe6d58521 : SUCCESS in 17m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bbbee0fbbba0425dafd6cedcb0eca309 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/15156e64f00143e3a6a60fdfdcad371d : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c8899dc75584a9395c5a9e00a647288 : SUCCESS in 6m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/60ec130ccc7c49e38a568896ba7ee535 : SUCCESS in 8m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23c399238cd244168bcf6a7949033676 : SUCCESS in 42m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cead8af91287437784074bddaeb808a9 : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6d2e6aa71757046611e79a43237db88d9df5c400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-03 05:13:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6374af145325bb3afd45140447a3156c460fffb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-03 06:34:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a6a10ed0d729ade3718830d5be890382b70ef5c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 08:18:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52e4601c91de4a1e8bda4714b2b5dc56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e99958ebce4c388cf11aedb6b1f844 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8c36050fc144128b2f970bce42fbadd : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/603fe099c0494ce69c48e158ad806be8 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71e1e832087f49e1a4a6a2acf535e91b : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e92d5a7d14b4bf28fcbb2082528c6bb : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9b5c981450a94723b1e53696ca439ef4 : SUCCESS in 3m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bbd397815eae4580a62d9dcc8af79f50 : SUCCESS in 13m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fc67e7411f0445868ace09ca6178b3c9 : SUCCESS in 8m 20s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2af10e3d87854606ae2dce5747d088f8 : SUCCESS in 6m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cf503a7f903f4c2080f2883f51f4626b : SUCCESS in 4m 45s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c73de4b34fb74dc2af794dbf88188b9f : SUCCESS in 5m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/eb0af2eb69a1493081c1905dd688b9c7 : SUCCESS in 12m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d7d56081acf74383b5bc6d99f740ad0f : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4a041d7cb2114043a8ae012785aa4f14 : SUCCESS in 4m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c03b1b5cd8154f148ee9ad9252b8c61f : SUCCESS in 7m 21s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4417dfa4e09a4875beb3f1190e477ae4 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3012a3c46fbb4547b2a9bfcfa292c9c4 : SUCCESS in 8m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/11c9df2e597f4673b66a16fd0b527a9b : SUCCESS in 7m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/47d56e062fd342a98f49f5ea36925444 : SUCCESS in 15m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ccee051f285b4832ae53012c807d3f28 : SUCCESS in 3m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7882c0946ade4ef2bb273bbb94a63978 : SUCCESS in 16m 28s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5b60192fe4d24be6b23c4bcb1f33cbff : SUCCESS in 4m 53s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ba600d49866b4664b96daeec214f7c66 : SUCCESS in 10m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a93f886550b74ce0b42cc9416d72fa84 : SUCCESS in 32m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/efac38d23c0749d3a8522e5d8b7b2997 : SUCCESS in 10m 09s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d73f964971564323a792f1f78b891629 : SUCCESS in 14m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a8fe639537e248fead706ce26ce03355 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8221b7e04b0742efaa8ddf747b0cacc8 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5bd4a57b7f2e4368a9dd13932c3c0182 : SUCCESS in 23m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8dfeb5e915b041e49c9a18d8af874659 : SUCCESS in 5m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bcebbb96c8014b759fc63591d0e91f9a : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd15fd980abf480fa84a7d4908425dab : SUCCESS in 6m 16s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9709f56d00504945b37c05e0146a3528 : SUCCESS in 18m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c50972077624fcc83129696be97f097 : SUCCESS in 59m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7066b7a55be4ab0bd359b1ff07e24f0 : SUCCESS in 1h 37m 30s","accounts_in_message":[],"_revision_number":3},{"id":"81ca63c9d5fb1ddedea31aa80b141aa5841fc63b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-19 16:21:13.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"481ac8c88a895857a659b16026913ae3b6803547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-19 16:22:01.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"3c0ac02f77cb34f27dccbe3a9a6fa6fe5f3363ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:57:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa536bce990b406fadf2f90cb4c080ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c1b193fa16406cbd0da5d87a6dbde3 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb8b7237b46483994e5e53054f9ee46 : FAILURE in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d2db2ebb2904b71a66b3af3e997145c : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43e72c0d77b94b0eafcda82f7a333649 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07b32ae3b15b41d89af22568148fb892 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91d6cf9038d84488b0119fe27c760a30 : SUCCESS in 3m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1ed84c9e7b54400c9d43a57908cb0102 : FAILURE in 6m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9c78103fe09648ffabe7fcab2317af5b : FAILURE in 4m 28s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/99be781511a5427c8e62d41d3944e561 : SUCCESS in 6m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8302468f11da42d5902af3b212214b4d : SUCCESS in 4m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/83172317730549bbadf066275556bb6b : SUCCESS in 6m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/549b8d3774ce400284717169f0bfed5e : SUCCESS in 10m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e19319ca4cea4e7ca3c258da3352981c : SUCCESS in 4m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7a503ec9c248433c9db153c64990b88c : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6ead079651f0466e96f8a1456404a372 : SUCCESS in 7m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6c9f28a5b9234bd9856093be2bb7be1a : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98d1cf676ed64e8d973a913c3aa0fcab : SUCCESS in 6m 54s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6bbdbda07ad4497d84992e7983b052b1 : SUCCESS in 6m 47s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/30193af7cdd54d428476eb78c954546b : SUCCESS in 12m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d8ac3b53c42943d6ae5a1621b3b05a2e : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/49beeb0b90dc4133aea3902ceca06dc5 : SUCCESS in 10m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d772ec9124d946afbc54c06281ac6710 : SUCCESS in 6m 09s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4759ba4371eb43dfbf1e00d3aa48a147 : SUCCESS in 10m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f038d3a2c299410f91806f03a70382fd : SUCCESS in 24m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/db183a749f46461ab46dd4da639268bb : SUCCESS in 9m 30s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c27650ff952b4e56909df7514571ab65 : SUCCESS in 13m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f7a869ed09724aa3b5747f02e609b5fe : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0c4e47d53ec74025bc0ecd2ad3e85626 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/733d6431d41140f98d8f0cf9899ba6ea : SUCCESS in 23m 22s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2b9bd8ab867e4297a2e004fd3ac7b1db : SUCCESS in 5m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e080de1b5960420ca3082dba9159608d : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/60bc5101599b47e4987854e0ca6b5443 : SUCCESS in 6m 38s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8ed3b0bb5aa246b78733718863a5cc3b : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/deb29f703513487092b5da9bffb5f675 : SUCCESS in 43m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/714615f80e2944a2a4865d56d8713577 : SUCCESS in 1h 27m 41s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file update.py has only 1 lines\u0027\n  Comments left for invalid file openstack_requirements/cmds/update.py\n  Comments left for invalid file update.py","accounts_in_message":[],"_revision_number":5},{"id":"86e7ae116e313e5ee2c0e2260d4c9052a8ceddba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-19 19:05:10.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c4ac5d3ff46f5d07e0151e033c5b5e8d80326c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 20:36:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e967718251c42959ad04d0d6dc6612f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5327bad364d4a37bdfc986947cddddb : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af3ee1df4624692b78f3502d63c55e3 : FAILURE in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69e3296a0df64bad939a2fa2e831a3ff : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/321e388cd15b4f2a8f66c9a0e09190de : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f51c0e393fc14e32bb4d077c3d39666c : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8817bcdea424ce882a02547e947079c : SUCCESS in 3m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/db8bdab56af0452fa37f9fed2ba1d532 : SUCCESS in 14m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a934ea5f6ebf4ff98db1b44d47786331 : SUCCESS in 10m 07s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6ba6428adb90477797c306ea3d2e3f57 : SUCCESS in 7m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/68dd98dfd60a4ab8ac9c23f893bacc9a : SUCCESS in 5m 09s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/098d143ef92f4b529374d5cd733146de : SUCCESS in 5m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4a6ee891fba646e4b377fe86f5e3a9fd : SUCCESS in 14m 43s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3ea08c4a0e894e93af8827aefbce3d06 : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8e85de1267ce4ab9b4d9cb8def8003af : SUCCESS in 7m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7a584cac8f43405f97de806c8c908226 : SUCCESS in 8m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8d9127352cb24b6680aa649ff1a00c20 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1fe26fad2b9c43caba85c94e77153995 : SUCCESS in 8m 43s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/30b02aedad2643b595018b4f36ad2d2b : SUCCESS in 8m 25s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9181080ad9bd4c68a7d22c35e8648047 : SUCCESS in 12m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6222d0faf7814f0da44a414b7fa1f8e2 : SUCCESS in 3m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/dcd3225bfdef49b1b6e6f579d058356d : SUCCESS in 10m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3cad3b78aad1421ab47d93182994b8a1 : SUCCESS in 5m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/51927dea3f7841fb8da887372c745e6c : SUCCESS in 9m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/46c300f52b9840578df3e523dcb7d08e : SUCCESS in 25m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d5e66d5257c94950877e4eb5bb08c9a8 : SUCCESS in 11m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e744fb4aea0b4a5c9ce7d572eb64b23d : SUCCESS in 13m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1a0298d2c9024d5cad020897721e4284 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/937abf8eeeb54cd4a71d46e37956e187 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82ce3b584b674e619bc43b53ce30d3bb : SUCCESS in 23m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0ee1d6b384cf45b5acf1ecfd6dc2b59d : SUCCESS in 5m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2fa404b8c81742aa9902ad109ccb9557 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1bf53abdc21a4cf780ec1d90e1dedfc8 : SUCCESS in 6m 20s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/428c263923b74d7b8ef31ee4f1239e07 : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1664bcc3b4524ba287a7d282ee7ce434 : SUCCESS in 1h 07m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b24396f1f3a4c8885cef5930c345d63 : SUCCESS in 1h 18m 01s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file update.py has only 1 lines\u0027\n  Comments left for invalid file openstack_requirements/cmds/update.py\n  Comments left for invalid file update.py","accounts_in_message":[],"_revision_number":6},{"id":"2a714422ae45d49c43437ab7ea0882fda6f3c920","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-20 06:46:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"424ea476d86d37aa9c3ce1defc381269f126b5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 08:26:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb1b5fa476454902938ecf762e57852c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53f9046008a4ab9a9266c3f763fb121 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83bd0004e1c74275818753a03d95eb86 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c50a10031558413ebccf5940831b858e : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acf7488ca1b04068872329be3675947d : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/989af97e48c8475d86cbb51569ceed76 : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60ceb8fd5d854dd5b1e8add1ad08be8a : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/63e05d4db8b3454c9cedcde2af5aaaed : SUCCESS in 11m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c7dcd8ba0d0948fc9c7852ae6913114a : SUCCESS in 10m 33s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4ccb6698f33c4ebe8ccb73b2a00869d3 : SUCCESS in 7m 11s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/037d7daee98c4811adaa06f771449870 : SUCCESS in 4m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/67db43f53eb9435c92e17db424e0a07a : SUCCESS in 5m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b6c4f369fc3f4bbc9fe0f35de65cfddd : SUCCESS in 11m 32s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a2f4de229f084b18b333c86a7b4017c2 : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6b25549a085a472f99c1d2dd8df52ebd : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b0ffac006da04d268701c57ec209ca69 : SUCCESS in 7m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a831e0f50aa8444d99d66cda291d7c5d : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d62491b53ff41c99e34db1b96ab7a98 : SUCCESS in 8m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e80714040eba424ca38e9a5c7aa0446d : SUCCESS in 8m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/519817812f324e348a4eab5df3642e97 : SUCCESS in 11m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/66554fae407d441395182d90f14e0bb4 : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/eea80f5fbafc497fa025146f5fa365fb : SUCCESS in 7m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5fdd73be249d478d8dbd741cc33dbd29 : SUCCESS in 4m 49s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b3585df279b543729829f3a6520394eb : SUCCESS in 10m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6be63a3df5a243e280e09cde686abd0a : SUCCESS in 26m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b25927a3e10e497da96a889ea2fa2924 : SUCCESS in 12m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4ee3e2cfe90e4e3abc1d9271b23d1f01 : SUCCESS in 13m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ed6ebac7f53f4f5b8d4e37ac0541290a : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f6e76b046ed4797b3adc8e4c422af39 : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ef741b5beec4f2dbbc13766b89bfcbd : SUCCESS in 22m 40s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/36443e80e3cc48b280b32a46475eae36 : SUCCESS in 5m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c8c7550bc111491ebda7b2d376b195e3 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a18d521ef47e4124a2c71bea05d921bb : SUCCESS in 6m 55s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/056f97e9e23f403abbcd8d926f13dbdc : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fd862468e2f49ee9d94448b0b666f93 : SUCCESS in 1h 00m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/836f3af8e6a241078086ee33b8e9c099 : SUCCESS in 1h 31m 49s","accounts_in_message":[],"_revision_number":6},{"id":"c73ff8209c6163f559954174190bbcb5ebde8e35","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-04-23 12:04:22.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1f67de5dcdd3e8e4dfc7c7218d95d925e91e589d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-10 00:22:14.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9492edb5b81f2e75b7d5fc174d7d4fc95bae5e51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 00:22:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"880fd160ea03a2b5235de2474993a307a151b08b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 01:47:24.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8711fa635c8e4b5bb30745fee07005f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b016f31907c349de8b562e3017c47091 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/328788c8b3e14be199be85edaeab5f5b : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/774e06fe404f434897f65ccd87c73e0f : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afa799ae9fa343ffb9bf61a92d169a72 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f53233d098084b608b824b81926c18cf : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0ffdade04d8470383c184afb4bbce31 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9fcf64536aa4e7baa7419b57bd4b363 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c638dfccda1748d98eec781f0df9ca80 : SUCCESS in 10m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c90198dfda5946a28af7c402993fa22a : SUCCESS in 13m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/400fc41f25b14d0ba68c2bc22a5821c3 : SUCCESS in 10m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/544113dc755c4211a39dea4d3230c5b5 : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9a8b5a44805d4b76a7d7a43906b20acc : SUCCESS in 4m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/83757f12894540b2ad49f3b57a6bea99 : SUCCESS in 4m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/093f77a0c7934dad969d66a7f2d1fc1d : SUCCESS in 7m 55s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e21418866450472a872ba6b97fece09f : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5546c00699f14154b1c8063579a0d1b1 : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/22ac717bb73740f9942e0615104edec1 : SUCCESS in 7m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b5d3953a490d442dbcb13f8bee79a3c5 : SUCCESS in 12m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7894cb14fcb44e959f8f71bd32944476 : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/935e5e56e82d4693a15ce7f1210d74c9 : SUCCESS in 7m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f8904efa9e3044ce812667efb11dbe92 : SUCCESS in 4m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/27b84d2b92594893a316951147914b4a : SUCCESS in 30m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/23a9ab6a9bab437d82d45f596f4813f5 : SUCCESS in 9m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6626b04927744f6ba6433d617b24690b : SUCCESS in 10m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6b3dcbfaefb34697a595f4232a255c38 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fac41e6bd97d4ef3954d1f2859e9d82b : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65e0a0dea83c41209a402b721e2a4430 : SUCCESS in 19m 52s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fd2f7a41c30742f699682197a58b07c1 : SUCCESS in 5m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8f79b90a14604ac4b2e9a56bfc8fdd51 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56f4e607692d40f295278d5f4c735b5d : SUCCESS in 6m 12s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/016d9b55bb6241cbb369b6ed559f883a : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca5f354e8d7348df81a176e26d2e9197 : SUCCESS in 1h 04m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/356e8bc44ca9446cab8e80f6ab113f05 : SUCCESS in 1h 20m 03s","accounts_in_message":[],"_revision_number":6},{"id":"28b88e1c227d058bd032f0bfc80b7b12561945b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 01:47:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"2c1a991e5695e0266e6b9d1a926efec6ce10bcaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 01:48:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b796104f934dd28fc6fa6ee5b5b760\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b193b73a104c77a6054e5b7dad0f88 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"411ea1949aec3c4519bc19b1a1d13c2d8f8577df","revisions":{"8901038d24bd12b448320be596967a8c407b2d05":{"kind":"REWORK","_number":1,"created":"2023-04-02 12:37:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/31/879231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/879231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/879231/1"}}},"commit":{"parents":[{"commit":"52b2df622747e43d80431a7ceb1b3570d33e04a2","subject":"Add cross-nova-pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b2df622747e43d80431a7ceb1b3570d33e04a2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-02 12:36:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-02 12:36:55.000000000","tz":120},"subject":"DNM: Test salt 3006.rc3","message":"DNM: Test salt 3006.rc3\n\nChange-Id: If57cd13d1363af3124ab0ed36a5f571174a1d6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8901038d24bd12b448320be596967a8c407b2d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8901038d24bd12b448320be596967a8c407b2d05"}]},"branch":"refs/heads/master"},"7ccb4bf009ae508fb6d788a475b7c734338b2cc3":{"kind":"REWORK","_number":2,"created":"2023-04-03 05:13:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/31/879231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/879231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/879231/2"}}},"commit":{"parents":[{"commit":"52b2df622747e43d80431a7ceb1b3570d33e04a2","subject":"Add cross-nova-pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b2df622747e43d80431a7ceb1b3570d33e04a2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-02 12:36:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-04-03 05:13:23.000000000","tz":0},"subject":"DNM: Test salt 3006.rc3","message":"DNM: Test salt 3006.rc3\n\nChange-Id: If57cd13d1363af3124ab0ed36a5f571174a1d6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ccb4bf009ae508fb6d788a475b7c734338b2cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ccb4bf009ae508fb6d788a475b7c734338b2cc3"}]},"branch":"refs/heads/master"},"ad78c634826e8fe12b66eee9f8279e27b7c09be7":{"kind":"REWORK","_number":3,"created":"2023-04-03 06:34:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/31/879231/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/879231/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/879231/3"}}},"commit":{"parents":[{"commit":"52b2df622747e43d80431a7ceb1b3570d33e04a2","subject":"Add cross-nova-pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b2df622747e43d80431a7ceb1b3570d33e04a2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-02 12:36:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-04-03 06:34:43.000000000","tz":0},"subject":"DNM: Test salt 3006.rc3","message":"DNM: Test salt 3006.rc3\n\nChange-Id: If57cd13d1363af3124ab0ed36a5f571174a1d6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad78c634826e8fe12b66eee9f8279e27b7c09be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad78c634826e8fe12b66eee9f8279e27b7c09be7"}]},"branch":"refs/heads/master"},"f1e8683c556651a03873d07ba0a29f76abba8c7b":{"kind":"REWORK","_number":4,"created":"2023-04-19 16:21:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/31/879231/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/879231/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/879231/4"}}},"commit":{"parents":[{"commit":"52b2df622747e43d80431a7ceb1b3570d33e04a2","subject":"Add cross-nova-pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b2df622747e43d80431a7ceb1b3570d33e04a2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-02 12:36:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-04-19 16:21:13.000000000","tz":0},"subject":"DNM: Test salt 3006.rc3","message":"DNM: Test salt 3006.rc3\n\nChange-Id: If57cd13d1363af3124ab0ed36a5f571174a1d6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1e8683c556651a03873d07ba0a29f76abba8c7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1e8683c556651a03873d07ba0a29f76abba8c7b"}]},"branch":"refs/heads/master"},"8e11f792e52f2e13019074aefdb41de5d6ca9450":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-04-19 16:22:01.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/31/879231/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/879231/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/879231/5"}}},"commit":{"parents":[{"commit":"52b2df622747e43d80431a7ceb1b3570d33e04a2","subject":"Add cross-nova-pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b2df622747e43d80431a7ceb1b3570d33e04a2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-02 12:36:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-04-19 16:22:01.000000000","tz":0},"subject":"Bump salt to version 3006.0","message":"Bump salt to version 3006.0\n\nThis should fix failures with python3.11.\n\nChange-Id: If57cd13d1363af3124ab0ed36a5f571174a1d6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e11f792e52f2e13019074aefdb41de5d6ca9450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e11f792e52f2e13019074aefdb41de5d6ca9450"}]},"branch":"refs/heads/master","description":"Edit commit message"},"411ea1949aec3c4519bc19b1a1d13c2d8f8577df":{"kind":"REWORK","_number":6,"created":"2023-04-19 19:05:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/31/879231/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/879231/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/879231/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/879231/6"}}},"commit":{"parents":[{"commit":"52b2df622747e43d80431a7ceb1b3570d33e04a2","subject":"Add cross-nova-pep8 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b2df622747e43d80431a7ceb1b3570d33e04a2"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-02 12:36:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-04-19 19:05:10.000000000","tz":0},"subject":"Bump salt to version 3006.0","message":"Bump salt to version 3006.0\n\nAlso bump pyzmq as required by the new salt version.\nThis should fix failures with python3.11.\n\nChange-Id: If57cd13d1363af3124ab0ed36a5f571174a1d6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/411ea1949aec3c4519bc19b1a1d13c2d8f8577df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/411ea1949aec3c4519bc19b1a1d13c2d8f8577df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
