)]}'
{"id":"openstack%2Frequirements~879558","triplet_id":"openstack%2Frequirements~master~I87ef5cd2258a4f09786b5a7473735db3200cabc4","project":"openstack/requirements","branch":"master","topic":"bump-alembic","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-14 06:53:17.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I87ef5cd2258a4f09786b5a7473735db3200cabc4","subject":"Bump alembic to latest version 1.10.2","status":"ABANDONED","created":"2023-04-05 14:16:57.000000000","updated":"2023-08-15 21:25:15.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1005284d71c0b2fbb4165b0259b2c71f4a056823","_number":879558,"virtual_id_number":879558,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-05 16:12:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-05 16:12:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d46c917fd226d7e30d10cdcac8f61fb325b27371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-05 14:16:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bca06a0082db482dd95c77ac7992465158ae84d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 16:12:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a83fd3738a9b4e269526806d42901bed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6e71f7c8ad4b91b9e0667211ae56ef : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d85f22b4ca4c5e8dde3355062c3ba4 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6a24445f2b7409a956085e658a103a6 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5414ecb0db74423c9a641cbd3ebfa9b7 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd7880587ae846cd94734eddfab5610a : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77e78d5ffb514d059b679e93915aa4aa : SUCCESS in 3m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/780dff8965fb48228f644969cd833399 : SUCCESS in 13m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/40c3d6ce273844b09b204fb320d7f2db : SUCCESS in 8m 53s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ce00d898b80c42228c537ad6c6999ece : SUCCESS in 8m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/90e9a14c1f784fb4b2b2964591c57d3b : SUCCESS in 5m 42s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2d50d9e6e9664cafafe4c3cf298e6165 : SUCCESS in 5m 29s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2219b01129e644dab9951991cb122ebf : FAILURE in 10m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/172d9a443aaf42d7be68ae42b16d7bdd : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/76c419d8643e4932b2325d671b3c6931 : SUCCESS in 4m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1694898312c84cf18e3c4a2ff62f9fdf : SUCCESS in 7m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ea0f10674b02443995acbef2877080b0 : SUCCESS in 12m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5626ee2308ca4af6bc3a62bc8e267183 : SUCCESS in 7m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9038fb9be4a2422abc95af2ea920920b : SUCCESS in 7m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/655618fd3ce743b897d44c05000b026c : FAILURE in 12m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/585be11a2a3c46c8bcace7b9a188d82b : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/109005853e6c40d1afb8ced1b597c9c7 : SUCCESS in 8m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a4f173776ab948da9d50d7c6681399d4 : SUCCESS in 5m 09s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8f985957a3c2495981efa25c18c38354 : SUCCESS in 9m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e9f3aa1376a040d9bd6a1683f6bb0ca9 : SUCCESS in 28m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/95d825e604964c13b655ce3a4434fa1c : SUCCESS in 9m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/39571d5a25c341e89292bb12cf377686 : SUCCESS in 14m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/77be97df1c834668a7ba9d236d303e22 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2adb4ec742ff41b68e6ad764777ecac3 : SUCCESS in 2m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f577c34c6dfe40408affcc240ffa7581 : SUCCESS in 21m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4b5282d792904414a8a1d593f7e56882 : SUCCESS in 6m 18s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ab47ec348fc944b4ab0709f485f82118 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68154440f6244876b3fc2d0ad1fd9cc2 : SUCCESS in 6m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9dfc976df2124713b009d337b6b756a6 : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0f9b34746cc44e5bd264df72b920a59 : SUCCESS in 1h 06m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b08260fb747e4904970d8a3f3c81b75b : SUCCESS in 1h 48m 22s","accounts_in_message":[],"_revision_number":1},{"id":"99fa9fb887cf5a1703ef6484f7e4ef14f3e6f57a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-30 09:01:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1aa968462c7d7fdb264130527782ff4bf1577e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 11:03:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85b1e64c73e042da83e99ccc81b125e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d230492a0d54a12b0676ed51c46487b : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c612865131f47a1a882a1f8bc687d8c : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efe429e75c6244de9ce8e796db6ec1dc : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/146e05a35c4c4ed6bf1eaae78fc7a1a1 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/290fca98343d4759938773c9cecff62d : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0c69becd55b4f349c3348400152aecc : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38b0f310589c4272a886d9e50fd1f56a : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4f9f51753edd4335bc07a652d71f0580 : SUCCESS in 9m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2bf255f849f54f448ef6bb65246f9ca0 : SUCCESS in 14m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fc01d1e727984d2f930962804e8c7830 : SUCCESS in 9m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d03eac0f879442148576f284bc90d796 : SUCCESS in 13m 30s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e54a46168e56487a86cf1321edc5c376 : FAILURE in 6m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e6eaa852b07b4c7aba51a22a70394f6a : SUCCESS in 4m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/426f4e747d604b058d2b3bccb9e0cd4c : SUCCESS in 6m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/517ca9dcd4f34fababf7e59106babb26 : FAILURE in 10m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/80a6654dd3b6431586ad2a4e9fdbb2c0 : SUCCESS in 4m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/36edbf8129fa4a2cb2ec0b8c22c8797e : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b172fd5b590843b1be0a752e0eacad50 : SUCCESS in 7m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/99e049bb681541d6a4707c53fac3f9a8 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce1e5ff329de42b9b38236ad4cdcde2a : SUCCESS in 7m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a62fe45007b04ae98f4eed6243095898 : SUCCESS in 6m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/62a15e7d499046f88a30f2146493d847 : FAILURE in 11m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bce1355c163e4fd1bd9aea01e7b4fddf : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e0e5400f8395483799ad35c65a993df6 : SUCCESS in 10m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b2708ad8092f481eaa8c48c3c96db3ba : SUCCESS in 4m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/de9fd46b8dcc4dfeae9d1a8b6e4f293f : SUCCESS in 10m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8e3d314336f749648c4e132b82d36db6 : SUCCESS in 30m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4df94b17e34e4d5388313fcd1838375c : SUCCESS in 9m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bbb42917c16c4919a508380170481194 : SUCCESS in 13m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/acf5f80a9c734fb9983ac8bf26805cb0 : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/91cf59c21a324880ac0b030b95d3f089 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8868d012b9c46cfa5af2f3c8a717876 : SUCCESS in 26m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0c74e7198a4e48b5917090e310981dd2 : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9a7029d8674d45c886fbc00fb31db103 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0bc7a9057c9f42748800f5e2962f12e4 : SUCCESS in 6m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4d6c9652a5d94468abbf237522f9ce7e : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54cfaea78ae94203a3ed9b58a75f2627 : SUCCESS in 1h 02m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32004012c41c49e6bb10000de09874f6 : SUCCESS in 1h 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1005284d71c0b2fbb4165b0259b2c71f4a056823","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-15 21:25:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9a5a5ea31d09d6d9fb52c690a76e75b33b1aac4","revisions":{"b9a5a5ea31d09d6d9fb52c690a76e75b33b1aac4":{"kind":"REWORK","_number":1,"created":"2023-04-05 14:16:57.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/58/879558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/879558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/879558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/879558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/879558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/879558/1"}}},"commit":{"parents":[{"commit":"692e27ef26cac11e3a2cc14f8cd9e699a5ca0331","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/692e27ef26cac11e3a2cc14f8cd9e699a5ca0331"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-05 14:16:21.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-05 14:16:21.000000000","tz":120},"subject":"Bump alembic to latest version 1.10.2","message":"Bump alembic to latest version 1.10.2\n\nChange-Id: I87ef5cd2258a4f09786b5a7473735db3200cabc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9a5a5ea31d09d6d9fb52c690a76e75b33b1aac4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9a5a5ea31d09d6d9fb52c690a76e75b33b1aac4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
