)]}'
{"id":"openstack%2Frequirements~879743","triplet_id":"openstack%2Frequirements~master~I23af86739e699bdcb42e1b112861497d35edc42e","project":"openstack/requirements","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2024-05-08 15:51:51.000000000","reason":"removed on reply"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-05-10 01:19:01.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-05-09 12:31:38.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-04-24 15:56:02.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-05-08 04:59:20.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-05-14 13:06:08.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-14 13:06:08.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2024-04-30 15:12:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11816\u003e replied on the change","reason_account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-05-02 08:01:44.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-05-14 13:06:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23af86739e699bdcb42e1b112861497d35edc42e","subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","status":"MERGED","created":"2023-04-06 14:03:04.000000000","updated":"2024-05-15 13:20:35.000000000","submitted":"2024-05-14 13:06:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":54,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879743-sqlalchemy-20","meta_rev_id":"7830f3748bd0b5ec57c3741469f30fbc655aa3fd","_number":879743,"virtual_id_number":879743,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 13:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-05-08 15:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-05-09 12:31:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-05-10 01:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-04-24 15:56:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-05-08 04:59:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-04-30 15:12:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"date":"2024-04-24 15:30:38.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-05-08 05:58:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2024-05-02 08:01:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2024-05-14 10:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-05-14 10:08:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-06 14:34:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-12 14:03:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-07-13 14:19:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-09-12 11:18:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-10-24 13:52:21.000000000","updated_by":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"reviewer":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"state":"CC"},{"updated":"2024-01-12 22:16:58.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2024-01-31 06:21:50.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-01 07:18:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-08 22:46:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-04-09 09:19:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-04-09 09:19:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-04-10 13:36:47.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2024-04-12 10:41:05.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"},"state":"CC"},{"updated":"2024-04-12 10:41:05.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-04-14 17:00:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-04-19 08:29:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"CC"},{"updated":"2024-04-19 08:29:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2024-05-08 04:59:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-05-08 04:59:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2024-05-08 05:58:18.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-05-09 12:31:38.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-05-14 10:23:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"23c6c4db270ace36c9f25d7ad436f638a41ae798","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-06 14:03:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28496c2106095147b89b2035bb83b3c9b4e1e8a4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-06 14:08:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc1e1c24274be606305a1189049eefb6817f44b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:34:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3141cb0cdafd4c5aac2f52cae098cc5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81dc923e9f44b20bb08c93a38fca926 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f56fd3516473468cb76d521d440ca0a1 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c629a33742964c52b09ad1cf4ef992fb : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5092b96e97774cd08f3db9cdaa8ebac6 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90992c5209704d3ab4e31ce485cb514e : FAILURE in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09e32d5c3c0843918670f6bc0a914218 : SUCCESS in 3m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dcc27485c4954a13b18b83c3dca4dee4 : FAILURE in 5m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0c04d0a79e4c498e8b496882b4608d59 : FAILURE in 4m 14s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c3bf70ea56a543d38e40ee5157a1dfbc : FAILURE in 6m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a24b796d0d8d42eaacde3966bc31a451 : FAILURE in 4m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b573ad3b875847d0b1e5796fe9416c6e : SUCCESS in 6m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/64f4dc50876741dfb6288e3bb546b639 : FAILURE in 6m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/dc862fb3d8b240bea50caadc4db9d813 : FAILURE in 4m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0dadb7a3740d40cdaa0dc552d760a2ae : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a311222526a84d6694d6bf9fe93a7bf9 : FAILURE in 4m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c9dbd5d0a3754e919b7cdb47c5cafe66 : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d97123f543714e198e6ed3fffde3c003 : SUCCESS in 6m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cf1742b2368641529d4cc36943f606b3 : FAILURE in 10m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e3bc48c86c2044ebb5c2c5d9fbbcdbe0 : FAILURE in 4m 28s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/16fbe7afa32b4a5f92bcb346190169a1 : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/56b21cded9e64ec9999459e99adcb525 : FAILURE in 7m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c91046b5e8ea4c55bea8ea1a2f2bd028 : FAILURE in 4m 13s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e7e94aeca98b4398a9b3acc01e23d91c : FAILURE in 6m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d3817bb7958e4c13bcca54f54cf74421 : FAILURE in 10m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b3c7d82a7e2344ceaa07f76157a37e41 : SUCCESS in 11m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/69ccb838855f426290ba5f18fbb2b3bb : FAILURE in 5m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b8345d57b3bb442da516d1b34f79065c : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd2bfa4842fc49d78e03f56f7cf94357 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aafa93c3b2ad4dc4b7682a31d504619b : FAILURE in 5m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3d66c192a3884a8ca653b3415e810380 : FAILURE in 4m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7d2df4d1f59141e9b157d7e78a46b622 : FAILURE in 3m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be907b1bb8754283aae897ccade0d801 : FAILURE in 2m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b30b84f5965a4507990d309a55de9b5e : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fe1b11cc34b420691a86ca3d570732e : POST_FAILURE in 24m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7ceae2d931344ad9b8b6c83badd26c2 : FAILURE in 24m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ef4e6fd556c945147893256e05fdf4a1c215bd45","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 09:20:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e21ac152ce20b0b4eaa7a87b8d3df1c12cbbb068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 09:56:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/210fb7e969d0498e88ade2f502a4d76f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208a7456483642dd84f756617d1efe43 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2240a4c7527a420caa4f117cce5b7197 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47301eefbe2e4b9db0f4466ae0acb5c7 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c537a774e0f499ab5aa9f25a8d3c6b5 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08bb23fef28549c3af5e8db6f9fc113f : FAILURE in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/85a0a4a4689f48d58709506f610b2417 : SUCCESS in 3m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/68a06b8c66de48f98349d31cd40dd512 : FAILURE in 6m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bc53890c1ed9486f8d8167b59383854a : FAILURE in 4m 32s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d555e2bbc71246f2848463f2b886c9ef : FAILURE in 4m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f97becf6e33240edbdb4b1e5b5d03c57 : FAILURE in 4m 28s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3f841ae7def942e88abc7230300d34c2 : SUCCESS in 5m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/69413f6ff28547dda30dd64c4d67d581 : FAILURE in 4m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3549f7e0181443b9901160810bf86e13 : FAILURE in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b00c01669880484da9f73a2f9772551b : SUCCESS in 4m 33s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5e29d8a7762b4e72ab23fae1339ad7e1 : FAILURE in 3m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0844de6a0d07424bb075f9373c68191d : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0355eb4ec385449daf46ef893cba714b : SUCCESS in 6m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/00d3d0b297804ed8978c33ab569c0ede : FAILURE in 7m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f8bae8ee9e8b4cadad268d5b39442a0f : FAILURE in 4m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bbbe722191f4483baf1f05b699b60ab1 : SUCCESS in 3m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fa2d0c75e6e649259349dc759442c908 : FAILURE in 8m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4d0feb5602be499ba01fd529d6083fbb : FAILURE in 3m 43s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/aad64888073f4c3a9fa8d08c041fb74c : FAILURE in 6m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ee19c65d9c1d406cb2bfb03206a4fd42 : FAILURE in 11m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/98ad3708f0fc488fa6608fa4259baa05 : SUCCESS in 11m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/938dc7da91d94f03ac99af0e147b5d38 : FAILURE in 9m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1dcbb6ef463f413f8b78007030d87fa4 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dd392e0d0d5e4b72806aff339cd16fb7 : SUCCESS in 2m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5ccdfcc25b347ef90eccf69bead0374 : FAILURE in 7m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c54a697d4ab147f18dd4a5dc921e49c8 : FAILURE in 4m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ec4bbff446f049c18cc72c88e2bf59cf : FAILURE in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ad9dd1058b34700a44ecb79bc7f9191 : FAILURE in 3m 21s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1338490d745d43f99e862966f57c2893 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2510d7ba84c45c38fa9d662930565e3 : POST_FAILURE in 29m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee5a6bc5f0734a3bb57ea6e159d8d483 : FAILURE in 22m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6457a19c29060db5decbabbfdba0e129095ee6c0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-19 09:21:49.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6598c48d61e62f385a27832cf12e62a122813640","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-19 09:35:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"198fcb21f93ec6c364a4aab573c9062f9dfbd562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 10:10:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1847993841f948b8981f4428af4268c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e96bdab0a8427e953632795fdb5335 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd7225de64e413ebb7efbb7e21cfa76 : FAILURE in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c87344c5f28d4ace805b3d39affff216 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab5980cfd173403eb03d75b648699763 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3df686cd92ec46ce855c56f180cf4001 : FAILURE in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e709350e1284fb987e7820a602fbcd2 : SUCCESS in 3m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c9bd8e3a787f40c3ac1de5b06e7d1ae2 : FAILURE in 5m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1bd3e4b0ef56438f862eab4a4024286d : FAILURE in 4m 28s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/608fa2365d0a42a184870b6a02e2e654 : FAILURE in 5m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/012103db12da4b4db1816f78f67b1f16 : FAILURE in 4m 56s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a020a5ea7f6e449bacca7457ae3eb622 : SUCCESS in 6m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/956125c59e4a408a9af2371f09108023 : FAILURE in 5m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b22bd37d3e1849289596457a0d63f862 : SUCCESS in 4m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/27f64caaa0954f9dbbf236633acc41f7 : SUCCESS in 4m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3638f6eca1704a5f81e429209d4eb24e : FAILURE in 4m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/918408e3963045f781f3d34d8debb61a : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bfba7ab58ac14c8da83e28ffc38809f8 : SUCCESS in 8m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f26b2eb402a14a1bb7e2aeccfb3e3c41 : FAILURE in 8m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1b3782ce10af42e1910a95552c0a6dd7 : FAILURE in 4m 39s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4d1985f755f24ff2887dd50b8200a28f : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b8b6f394db5d4469a08f6ea8fdc466b5 : FAILURE in 4m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a0fdbf9137a8481aa1de37126bfa3b42 : FAILURE in 4m 49s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a63fc01c19984275b6f9e86b579c9519 : SUCCESS in 11m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f464cb69f84144508b3239602cfdd121 : FAILURE in 28m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/411425627f614e459220b44f020e437d : SUCCESS in 12m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1511426100684b6eb64293dc3f87a8a0 : FAILURE in 10m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/32c1fb643b4049a890a7219488dae714 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2d987c2fbdd344b3a4e7696338faac33 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f651e2027a24312b0ac0dfaa7b57676 : FAILURE in 8m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c18ede78bdd342b5a1bc35fdae6b25bc : FAILURE in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/99afbecb1de74a69856388648eeead9d : FAILURE in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fae45b96d4e54c3a99b76a8f1fce6d92 : FAILURE in 3m 51s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7a545a2a9b714f2db20006c50941b7ac : SUCCESS in 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d14b0313f6cf4f37a8e01c82a7108061 : POST_FAILURE in 12m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03c60a9193624a1a9a8ceddfe54f0c72 : FAILURE in 22m 14s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file update.py has only 1 lines\u0027\n  Comments left for invalid file openstack_requirements/cmds/update.py\n  Comments left for invalid file update.py","accounts_in_message":[],"_revision_number":3},{"id":"a36ebb31a180f93accf3a6fda02755b93ccef53f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-13 17:42:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ec535f76d9da2c6abb481febaa576044cebb0bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 18:17:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/39d8f837fec84166b1e7402fe56ca345\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fdab92110cb46368cf18b229407ca15 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/806cb3a88573492d8390985a6637f57c : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70001bf5d178433abf8d3a79161fc4b2 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baee70fb252f4e29b4b6c84d65362aec : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88c0a4241e374e7096b6a0bb2d33a4ca : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f70fb07094c4df29f77f68c0877e69a : FAILURE in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac30e01141da4476bfd69304aca9d37b : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8a66dd2130e47e8b7da3efb6c73d2e0 : FAILURE in 3m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/27ce9a619eda4f398962b32c711eb36b : SUCCESS in 9m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3df72f9a3ef4452a82bfd9a737fcae0a : SUCCESS in 7m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/202830e779e847f2a77af051f1525a1a : SUCCESS in 14m 40s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/71e1736708074367a6d6a706263f1d5d : FAILURE in 4m 29s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/aa1d6234eab649909013270fc916557d : FAILURE in 4m 14s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/da487740ad484c87bedf5091f9fe80da : FAILURE in 6m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/122d74d3b87e4ab3b7569c56923d2c6a : FAILURE in 11m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b4d991564fed46019b5b23dd568a9de5 : FAILURE in 3m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/886d14dd581e49419cf1b2f7a91aa5eb : SUCCESS in 5m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d9128f6ec1dc489d963a5b74e9bb6660 : FAILURE in 6m 53s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e018e3e9a4fa465aa31246d666a1cdbd : SUCCESS in 7m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5ed07d8aa444d9095a2e36ec233c4a9 : SUCCESS in 6m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c4320aa1352d45f9a90a4d1d284fffc1 : FAILURE in 6m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/52c45f715d33450b9b65743e16d777ec : FAILURE in 5m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c1c467effa6344bea9fb9098b5184210 : SUCCESS in 3m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f470e93794324c25ad6639c066cea48d : FAILURE in 4m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/08d9a3a90fdc41d8b0736e36412a17ff : FAILURE in 3m 52s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/5600a0eff5214c668ef4cbe6078d8422 : SUCCESS in 10m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/522944b57da64d779627cf99cb7e93ad : FAILURE in 22m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9fe8f8266530470ea6ea3e1c09475b98 : SUCCESS in 11m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c5ea54d46f584838a55131a97b061276 : FAILURE in 13m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e8643108ad514c56b9b6c530f00f3a8c : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8cf326675f16435e87acbc353b030577 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50c90e0f3b4f479e946e4ebd9e6cd49e : SUCCESS in 26m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b3c5c33ceea542bbb45ed4baab7cadc4 : FAILURE in 5m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/56545248cc0c4403aef92055549aaec3 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce13c25a60384825a6df3bb17a53a3d5 : SUCCESS in 5m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8bae7b492612497996dd6be54b0f5cbc : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d51f184349b4acb8f728c42e393159c : POST_FAILURE in 27m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d543690beb334246b42d855947cb1287 : FAILURE in 10m 47s","accounts_in_message":[],"_revision_number":4},{"id":"36d13cd23a44ab1515de8810adee4ecc10bd071c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-13 18:50:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0974e3abdf252523341405e0f8174634b5ace436","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-15 22:03:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a65ba294b62233722d963b18f5d3cd440d3c17a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 22:38:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2baffeb3ed784439bd8baefc5276077f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759a7351ec8c48a281ffbd0d4f89dda6 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754ceeee708e4fcb8711d7787bdb1303 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58dc2490a0544d3f9b35cd59326ffea9 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c75b932d5cc44f5787102c3623e28465 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6d02a70b76f4b4987ace8912c34c3d3 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1478c046044b44e59c195884ca4435ef : FAILURE in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56497d2bd349438394fc9f9db634efca : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ea1e085131c54caeaa73268c23ecc1a8 : FAILURE in 4m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1ca5c698994f42c48832ee892d613e94 : SUCCESS in 13m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c9949772d6ea4f649d9db995bdff58a2 : SUCCESS in 11m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/be26fbffdd404b418106ece44904a3b6 : SUCCESS in 12m 40s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a501325c52ae4fe8b5ce89fd0a7cf18b : FAILURE in 5m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9eb441ca2a0b47a2bd99458489ae4d8c : FAILURE in 4m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/013cdc59c40547e28ba49c3356736eae : SUCCESS in 5m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fda846ec50eb439384a0bbddec0f26d6 : FAILURE in 13m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/944a18f183e54f14a0c900f8c91b73a3 : FAILURE in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/47422329e2a54ef2b92c3b11ead5d037 : SUCCESS in 5m 00s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2ba95b19db614e79853725bd134711a3 : FAILURE in 8m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b570898337f2482eb6a86da15005e7b6 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46c81df6a44c4897b9e938f7d2c8a7e7 : SUCCESS in 6m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2d11f3a1589a40a38649263b9e5e5cb1 : FAILURE in 9m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e9229634cb7644b8bd7b731699d8514d : FAILURE in 4m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7771661647ed45f5a79140e9898ba3f4 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7cea0f737c0a42b799ffce5a31798453 : FAILURE in 5m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f16b643339e7464b9a0d377c28dce975 : FAILURE in 3m 48s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/bb7611b868794c90983b27b8d692197b : SUCCESS in 10m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1a073e58665647aca1ece66e5d5a449d : FAILURE in 27m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/89fade7640504cbc8554ab423a0bfd5e : SUCCESS in 11m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9e67749b9446481cba4358052f0fbaf8 : FAILURE in 14m 46s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/71d3b476b8604364906ffba8166ddf87 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20f1906eb3414f098148b5b650c40729 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5cd858e2901f4567b48216d2c79c1053 : SUCCESS in 27m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/beafa7ee8dbb4ff197df476a06d6de57 : FAILURE in 5m 14s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/42ef4633f60042daacf4881c8595f518 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b0700861baba42c8acb3440e3e7a59de : SUCCESS in 6m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f51d70957e2b4efc888982c4eda2057d : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09e2cb176faf48c2ae7ea5f8438292cd : POST_FAILURE in 12m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/396699b18d7b468398e0be24f23e650b : FAILURE in 12m 47s","accounts_in_message":[],"_revision_number":4},{"id":"ccdf533a7604587215350454405f5eb8f8d8c927","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-12 11:56:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"87d03e94db45b416ead8208e33ab677fb4f8b8c0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-12 11:57:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e60becf61978e63cc1a15f4959d2b661a753cb54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 12:28:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c959a76d1f64d769ba0e3ca434ad395\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482277e1a1de4257aba63d71352323fe : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23252464960c476eaa339059518c1f54 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa9fe480b7014d41996ccf5c598fd2ae : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dea388abcb3a4d5395bfb0e5e18e5575 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab7396c3025844729d15737812e5a0c4 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24c17854e7ff4067aff1d184dd300103 : FAILURE in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0050ff6c08446c685c409307b4532cf : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a1fec2f61a234f77b0788ad97755e954 : SUCCESS in 13m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/55492b046398466385a04bf80d4779a2 : SUCCESS in 15m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f1d373c836e240608be436360512ae4f : SUCCESS in 10m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e2e335ca17c24f429daa0077bd7f54a1 : SUCCESS in 11m 01s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/98cab1ffd2f242c19e640aadff0cefd0 : FAILURE in 6m 34s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/58627b0aef6a4ab3b1dc4d8511c3fe25 : FAILURE in 4m 38s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6fa937f035d0409c84c8fde57668957f : SUCCESS in 4m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3088d48b0d3042b7b59f6520297f6d98 : FAILURE in 14m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/401b38f516b74e8db450af93e2d27c98 : FAILURE in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8ebe5d782d2e46348eb082bd29ec7c5c : SUCCESS in 4m 32s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/87987b28623b4ab89cde55118740c308 : FAILURE in 8m 29s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e576fdfd40b74b2db910da12e6c97d22 : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c6ed4639bdd4ae69ae3ff211d098c6e : SUCCESS in 7m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bd00624d95064a11ac59c2a792c3272d : SUCCESS in 8m 20s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f8daeb245f8c4bc49abb43aaf453733a : SUCCESS in 12m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/13cf6c11978142ada1c8b014eb11bc5e : SUCCESS in 3m 54s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/09fa47ad94004a9abb940a6752db2706 : FAILURE in 4m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c3dab6a7ddf64d84987766868e8d135f : FAILURE in 4m 24s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c4f89e30dd624634af0f224aa56b8167 : SUCCESS in 12m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b86c934c290c43aa8ce47f768a0e7230 : FAILURE in 26m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/449566b5294347ba9fd36e01b9086986 : SUCCESS in 13m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/32c45fd87e304862bae76d0255fc28a4 : FAILURE in 13m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3192c3994e6b489fb6a7dcf5e9b27ad4 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e1259e44a923424d873f4b328d723859 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c9de52e36934c9e83c6196a4a7e1c45 : SUCCESS in 27m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/33434f27d67a4e048a2056356a107931 : FAILURE in 6m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bdc4e9cbc16d47cab07076083c4086ca : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0fb7944ad034705b8cb6c23f6abd908 : SUCCESS in 6m 20s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/1717f35b018e4cfa8dd87409b7da0d0b : SUCCESS in 10m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b3783875fb94808a2025c1808ac3c32 : POST_FAILURE in 21m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ecdc447241344cbeb5c3b2abdb865a13 : FAILURE in 18m 43s","accounts_in_message":[],"_revision_number":5},{"id":"547fc650528a07826ff3aadffddf34c6e571409d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-12 14:03:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aa9205dd5748e61427aeeb8ec0780cdd53d89686","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-12 15:39:57.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ca79cdd799f5286c3f2e6103644784dbe083d176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 16:34:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f10a97929134ae5ac45a6bafc06b2b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b85b297c0b674fbea92429409aebad2d : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5eb343f650f4656a36ffdcb97fdbfc0 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f087025d5e04f92a2729f7d60b518e1 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e0cffb35d134dcc9898ac109032abe7 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40784e570aef4e9c9ba91e9827432fd4 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/baba0f2c4d2e474e8adc465d3169929d : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0859aaf841b9459da225fb6e89311fb8 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/58508bb8292341cab16c2871ad8a8cf3 : SUCCESS in 9m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b3219ae41e2243cda6f83b91195bed66 : SUCCESS in 12m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7811f752da54474289867e2e13b0afff : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/81438353122e4004bbe573ee9a535aeb : SUCCESS in 13m 27s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/87f3b0bcfe954ae0a5ad9e01559687ad : FAILURE in 6m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5369dd0b6a774094b2ad620bb6b28355 : FAILURE in 5m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d255ae88a0b1459f9b4053c4475401c3 : SUCCESS in 5m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a0c4305e7bce4b7cbe6e60ffaecb77fe : FAILURE in 15m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f9d8f0fa45404c1499e91d66d2479474 : FAILURE in 4m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ae71cc083df9439fa647c41946c5d9d9 : SUCCESS in 5m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/99559716930044778995b8b2bd4fe2e4 : FAILURE in 10m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f9b5166ccb234527bf6ca47a6280e11b : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3654a2579dce40cd9f28af35b081c6ec : SUCCESS in 12m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ac11385c3865442f8fb9830d3e94e6a9 : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e95bf2229b0f49be96bb32f9466bec12 : SUCCESS in 12m 56s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7855d10a436e45e895838246361ffc66 : SUCCESS in 3m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6801509f1c6042e0aa05633041f2054a : FAILURE in 4m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f6dc775a76cc490d9a03bb7d743ee657 : FAILURE in 4m 28s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/242948ca7ee8408598e5e29e29d829fc : SUCCESS in 10m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4f2cf1c664db45e0865223a8453b5ad6 : FAILURE in 25m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a066e4ddfd4c46f68c604a1a1444aaa8 : SUCCESS in 10m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f8b668ea34de447cbecc1875c31389e5 : FAILURE in 19m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6bd9eb1a5adf4ef58159a8daa974497d : SUCCESS in 6m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ae8b76545e3544bf88c41c4ed5b66e5b : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/49e8f1a6a6614c5b9b712c3df9fc2984 : SUCCESS in 22m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6f3dab06ba7943b79828265fa1e1fdeb : FAILURE in 6m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3c3d06a7681d4b848ad18dd136a0e1c0 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a9c23868e23464093ead30ba61e6bbd : SUCCESS in 7m 59s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/bce387c43b834609a9ef5415963f324f : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cec0b417b50443388f2e9b899590daad : POST_FAILURE in 21m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66d56eb62c0f4b65a679663e1268c026 : FAILURE in 43m 45s","accounts_in_message":[],"_revision_number":6},{"id":"9a4eaccf8bdd12956ceb949cd32634f6152f8211","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-13 14:19:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4f4c130cedd9112b4eeed56b93a2a30779916d22","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-13 15:25:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8c54be1b882840adcbe8f135a684d3ae5656eaff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 15:48:27.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d39dadae8c5b48c1ac1f4644d7239a87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f25ae41912545ea845fd2994d0af506 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5059a82ca07483783907df1c572229c : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51d7e7fa223644c5a00b8f498721de25 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f6212fafb9b41a383c1412fc9b6d02f : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8173797a882430a9fad3039a6db99c7 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff82b00de42744bcbd2b2350586112ef : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/65cc647d992549ff9158e16b22964f6b : SUCCESS in 6m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4601ac90d5ee4ccd91e504b7f22ca032 : SUCCESS in 10m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/772148e300394b91a65c9aee6d8326d6 : SUCCESS in 13m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c7674c72195a4071b7a827756b87a247 : SUCCESS in 10m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/76edb85a232242f5853db94289fa74f3 : SUCCESS in 12m 59s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d5ffdf19d4404a96baa2eabdc3c5b4c3 : FAILURE in 5m 50s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/50f318c43f964668b6f1a4b3583c0bc9 : FAILURE in 6m 03s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/05b0bea0800448e182ee917dbc189d64 : SUCCESS in 5m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/21b1cbbfb30845a1b27a4035e6ac3921 : FAILURE in 14m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e96eb4e2885544f49ac030470d6a9a56 : FAILURE in 4m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1913bb7b68be4f59a81d058db001321b : SUCCESS in 6m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/af5013e38777447fbc003d91165a5b6d : FAILURE in 8m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8bfb5818ab214559a2208b0a67a4cbc1 : SUCCESS in 12m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0c85ddac2694af9819a7b4637d41b2b : SUCCESS in 10m 22s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7148ab1f6f95433d84ec0339ba62cbcd : SUCCESS in 8m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dde0af4666ef4273990f06407667c9f5 : SUCCESS in 17m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b2c04683f3d74aee9c8dfb0ced5efdbf : SUCCESS in 4m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d0233380a6ad436f906c4a77e325347a : FAILURE in 4m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/845be0e3a1274fc9b5ade958f4761378 : FAILURE in 4m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1f72cfd9b3fd4f319180a3aa37a185ec : SUCCESS in 11m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f81700835e0547c598da9fd5ce8a9bfe : SUCCESS in 27m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/46a0556654124c15b5881962c5a2ca2f : SUCCESS in 11m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/68a73dd43c904a92a814e3cd37e830b2 : FAILURE in 14m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/927d59ee10684ea7bb8254dc8659c22d : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/82a411a9c42d4c909daa8bca4c62b4ae : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6df23470d774cb29253f1fbc2fcd888 : SUCCESS in 28m 05s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/cfe34f1eff464718b8c6337ff509cf9c : FAILURE in 5m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/61537c4e638344eba40e1442a41e125c : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0fa9cc2bcf9d42c9980f72cf89f5936f : SUCCESS in 7m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ef66d1f835eb497ab08b8b068343b77f : SUCCESS in 15m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/594fe18599424a00bb18f6e68b061f0c : POST_FAILURE in 40m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8905e296a5a04e3cb628aaf222aac24c : FAILURE in 52m 29s","accounts_in_message":[],"_revision_number":6},{"id":"60fc6015f8fbe41e01ff89a3e61d7e004e913e46","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-11 17:22:17.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8b92417e346611ea5e3fcd00cf7331e4c7494333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 18:01:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cca6c5a657144018849c4e0b15c571b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fd4814ca81407aafbb366aaf81b011 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe83c1b7ebc342c7b20843bf7037aa8c : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30645d9f8d524aa9b623d1f6a60b0697 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfab3c17f6ef4676a6710b11ee77e376 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ab8eec46b8d4e3ea4d9c67b76a58db1 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/621d1a8b6f7e425fa2ed3f02c96373ad : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56f82f846b4e4f09941c27b247803cd3 : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f162348d5b9f481984e74c544a91767c : SUCCESS in 10m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/01cda7b8e8c64069be47322538431493 : SUCCESS in 15m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/89fa313071ed4bf39fb3880fe0654442 : SUCCESS in 8m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3ae02f104358404183c94a481c76b183 : SUCCESS in 14m 42s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a0ab7784926f4752a8b71c3e8ab7d8ff : FAILURE in 5m 00s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5affe37951b14179aaac8045c1029c0e : FAILURE in 4m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/97eac526c3164eccab88ef42c6ee4d69 : SUCCESS in 5m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/66f2cb882a8c4762b111a2cfb55e4b7f : FAILURE in 12m 33s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b60c518267014737956f75d2d9750084 : SUCCESS in 5m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4a2144e675e1431dbc00c69c671eca8c : SUCCESS in 5m 39s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ac3a65fcf5d245a8a380aa710309a951 : FAILURE in 9m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6f1f0af7994946e9851bd021cf942794 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed3bb63d5f23492cac429b0601cdc091 : SUCCESS in 7m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1322dd82756348639e893519f9bd4913 : SUCCESS in 7m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ce186a88e5e742f1babbfc565bb5ed81 : SUCCESS in 11m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/45a01a90efd44698ad4a089a806b1e71 : SUCCESS in 4m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/df9945b18a8a4a7288834c8376fad5ae : FAILURE in 5m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5027e72a79a14b309f2680bc5d4caf9b : FAILURE in 4m 05s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1194aa5f7dd84c51a0c129ff839aaa84 : SUCCESS in 10m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8cc136d3ff004125a2ef2ba425bce969 : FAILURE in 28m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0ebdf7ed64b843bb94538c32c96b3921 : SUCCESS in 12m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2a1ab52355e842999ceefc00db1c8380 : SUCCESS in 14m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0fd494db1d5842dcb1c7b4888d0bb236 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bc8e2cfc893b4f81a4d89accd79fd6a0 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4fd1dc31f8634ff79e99ba05a02fd77e : SUCCESS in 25m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f05318467d494523940c61bec982dba5 : FAILURE in 5m 51s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c9dc94c34d4440e6addecdc85729cc60 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c965a896077442db9b9798ff772cc9e7 : SUCCESS in 6m 53s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f9c88bec648643ef9d82ef491fb84295 : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e12231bbd144b9ba3226f655dc8401e : POST_FAILURE in 21m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d4440c8a9194bd8aee2699c3e952b3b : FAILURE in 30m 01s","accounts_in_message":[],"_revision_number":7},{"id":"fa844b1a98c6de2eaead1c087f33bc617e1ca5d1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-23 10:59:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c4a8de34d2402daae854edd835a7671d8ad41228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 11:48:58.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb8f284081454ea3ac4d27049043acdd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22dbe6bdf184e2b8d115078f48a8165 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a710cef75f4e548a9453b7836fee88 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8a847a0d8f640239991f2a04d3015c2 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ed4a2bd151143e0b8d9307cbd00f8e0 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0188d8c1249341e9bdda7f419fecea2a : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f5cb6d828b74b3ea2e20885bab1a291 : RETRY_LIMIT in 1m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c08844537f904f418c573ace1e244552 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/90f0798c367a4c4f8d76c7ed8b9dbb23 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d1b7eeeddd184d90b592e2296e7b3eb4 : SUCCESS in 12m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/86ccc77a244a4880b47e7598707d2b2b : SUCCESS in 15m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/21de4f65001040b9bd1cef7489bdc938 : SUCCESS in 11m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4b2694e1374e4db398873d5426ef5583 : SUCCESS in 13m 34s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1e875d46feca4f3a9b1c61c7409b7c1a : FAILURE in 5m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/290f9e15ed424b508132452529728906 : FAILURE in 4m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9da72a0bf0854d7a972cf37dc6f5c0c0 : FAILURE in 5m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1f1b9f9b89c947b9ae7f312f4714acd1 : FAILURE in 13m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2bef64c75dba4b758ebd6d8d46ee0833 : SUCCESS in 5m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c097794652094a789c51027b7a7cbd06 : SUCCESS in 5m 08s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8c1c26d89d564ed39eea111d47726301 : FAILURE in 8m 55s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c23e0b22fcd641079f1afb6bcf148aff : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e3d2a5338e844e083fe7a5e97ee1bd4 : SUCCESS in 7m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8bc27ba7bd004ab09fd88adb78acb0c3 : SUCCESS in 7m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ea92b99d37594d208a9e2fd7e3cb8e9b : SUCCESS in 14m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/98777d7da00449a586cbe253fa402b0f : SUCCESS in 4m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4a06b675600e45b9b9dd742f746c095e : FAILURE in 4m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8e2fcbbba8d5400a9317ca2bcc9478a0 : FAILURE in 4m 08s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/713e5b43160e4a888a31e924334ea26e : SUCCESS in 10m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b1995e56f0c9483382fef00c23bba4c2 : SUCCESS in 27m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e1b7cbcdf457418db4e946a061e44bb1 : SUCCESS in 11m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5dd0fd50e43341e39a282034d397dc7c : SUCCESS in 14m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fce808af57d945c4b908572f270b14c7 : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c9846b3464674d7fa40698281bd4978a : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6290dbe93da64ab1a6aac5a324774140 : SUCCESS in 25m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4bf78efb363d48b2b55150181c12179e : FAILURE in 4m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/288d4bcc77b4404f9a410adb70d417c9 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc9c763fb78a46058c2bf9b59ef7ac63 : SUCCESS in 6m 50s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6bf0a18695994d82a34dea638e161318 : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fb500ff18384db29be0d8b64a45ef9d : POST_FAILURE in 34m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ee22b2d1fd147b28dafa69aa7ba4082 : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":7},{"id":"cd8f6ecaa7e173fda8d8fa0758238a80dd6a2072","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-28 09:44:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"228baf72523ca76349351ebe84de1951708821c7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-28 09:59:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"52b1b0a94419471389b3a78ff44c2bc89108030d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 10:38:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a85c6c9e99f84c7b96688a4672a7eb6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6706ab0936014c3ca7e02f5d0f6d5384 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2af4561b364e8d992003903ed94ab8 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f01084a5e8a54b50b6ac8d37677f7df8 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c25ece0984c4618ae82aaeb19c0db2a : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e74427eb9be42a69ccbb4a75a073ea2 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e41c230457c44d1891bbdfaf5e978c10 : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2da89d57567643d1b97f0c0c5e5461fb : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c582561724b6441996e4d86c91bb6815 : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/218b749e9d49487fad0fedfa21247cda : SUCCESS in 8m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/35e72892281e47e5bbc9d57cf1ea08a8 : SUCCESS in 13m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7305622878c548e6bcfcac02969f7364 : SUCCESS in 8m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ee4eea030f934740886cfe1ff574f56f : SUCCESS in 13m 55s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4159ea84f1cc43e981e8bbba02ba6cd1 : FAILURE in 4m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/957d8baf0a614400aad090d9f0e745bd : FAILURE in 4m 48s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9e6e491e59e8463d9245747ddc61091b : SUCCESS in 4m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/92e30ef2eb454e20b55def86550f5412 : FAILURE in 14m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ea6ed4952bb04ba4be4bf986bf078eae : SUCCESS in 4m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/44cad9a3c81a4c51af7f28f9708a4ffb : SUCCESS in 5m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eecb4612d62d4e27bd6bc20485a23ebc : FAILURE in 8m 45s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b59769a5dd204c158a46bd8c8775075e : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9282e8e649414a1caa8b930d3e76b5d3 : SUCCESS in 7m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/258196dd13f642b8b7c2dd3200c6d0a0 : SUCCESS in 6m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/208ce8ec880b4d688586e38078785102 : SUCCESS in 13m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1ce9a740937a4132a53e6f98827595e5 : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/50a401dde43042ac80e3e867bb3a8ee5 : FAILURE in 4m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/872feaca619a413192565ca4817476e4 : FAILURE in 4m 09s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ad4c4624e92849f6816d085955277f6d : SUCCESS in 11m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9d7056615fab4ca9849ef23e407aa9c7 : SUCCESS in 28m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ccdc81f3b8544d0d8ef07526051b005c : SUCCESS in 11m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/44f2832892b14c0ab607d1856d803d37 : SUCCESS in 12m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d88c727b5ed6402fa2460c0e7d9e73f7 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c943a5a236c849da97267ccfdb6845c6 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce5a243e740649ada7e817ff8ca080bd : SUCCESS in 25m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3a4f67967fef4a919d8a4a5bebfb8f0e : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7121d0d2d6a44dc5be6ed955db492ee8 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c4fb25824f12475e9173ee8a01bd0b8a : SUCCESS in 6m 00s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/92b340adefb74497bd4be011533da14f : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b5b2bb788b645cfa0bc73ae7e86431d : POST_FAILURE in 40m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fe42a802d8c4f4c82154d22939e333b : FAILURE in 20m 11s","accounts_in_message":[],"_revision_number":8},{"id":"95a68cb0ce9c5247f37e32eb3f61c3d45812f3de","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-12 11:18:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f128c07c2b908fbeada2d1a349b2cbf176232a84","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-13 13:45:36.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"9733ced9f0e48fba5a65747bd58ac2a36c88c9e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 14:15:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20f255a5698d478a8437be893803df60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d08dafd7b54ed384b56d06edddd1e8 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de078100bbfe416296407b71844abfb0 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dae97fc1d0948778d29b45079416fc7 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b29155073044431ab9b37282decd9e1 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6908530ead3b48da88e006f69424d452 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d69dd0a3a4d43e4b275fe26f8b8f893 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce8c7613cfc744ac9cb2a556b32fce5f : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6bc6f67099d459f8a8259521c4c9d6b : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5046c97e8cd04742bb08b29ddd262b6d : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6b4c89efca574aeaacee2d9dd0f54ea0 : SUCCESS in 13m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0f970682b3b947169406f8657d47fcf7 : SUCCESS in 11m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/beb71f7139764b6b8a46ab028ab3d9c1 : SUCCESS in 12m 46s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ffd08e6f6a7b4a249ff67d84ff12fee6 : FAILURE in 6m 19s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f50834a9f4ee40669e7b0de93e71b10c : SUCCESS in 6m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/415d630a0bf24f77be8fcb6a5e5f2ad4 : SUCCESS in 4m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c75598485f7947b5b95528bcf87f3fe3 : FAILURE in 14m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0b912c8d756f4332a8a311d6f371f6cf : SUCCESS in 5m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4a7cc504e62941d3b42176b5ce1927c7 : SUCCESS in 5m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3d3e1a39ebb3483ea3fe9cbc80b96407 : FAILURE in 10m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a9778ffb0b35446c96b09c5a036c9505 : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a76789300cf44a0f99edfaee648eb2fa : SUCCESS in 7m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e2321827bb1548dea4c042cea64ec64f : SUCCESS in 8m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b2dce9edc0584932b8617a4749c4bfc0 : SUCCESS in 11m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fc30ea3eb27d4efba12b0178f1c00ac1 : SUCCESS in 4m 08s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/15a8fe3e01094622873361059c4731e0 : FAILURE in 5m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/aef2ccdec84f4847a15c82e31d8f7462 : FAILURE in 4m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/646ba08ce7004604ae8c7351fed4c53d : SUCCESS in 10m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bd80e30e25d847c5a01579ef611d8fc4 : SUCCESS in 26m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/79b1321838dc43d5a9d0e2f916e8d2e9 : SUCCESS in 11m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8651d8edd5a84a5187e65f6eb3d6fbb4 : SUCCESS in 14m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/926417868d2f4733a3c69930301ad2bc : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d6100d971ba44baca44c3006bdb9d283 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b32c044df8fd41b089428d9fb205319c : SUCCESS in 23m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4ba6c75c9d614e5abbe0e3a4a2288265 : SUCCESS in 5m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7ae249de282a4abaa2da31e78644d218 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c1853c75dba4b598d7baab82525b08a : SUCCESS in 7m 40s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ce2182c4339a4573ad8b8e9d2a155684 : SUCCESS in 15m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8aeacebeef8747aca92e15a0cc6ada85 : POST_FAILURE in 22m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df86880fff1c4d55ab508fd204e1737a : FAILURE in 20m 51s","accounts_in_message":[],"_revision_number":9},{"id":"fcf83ec1de7ebbd45f4e89fdb7d4f8bd1d968e97","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-13 14:18:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6db87b253814dcda6f98944fc56659bcc2d0c3df","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 13:27:00.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4d5be442019a19d8aa11bfee13b5502986f68aa0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 13:27:04.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"2fa58ddc36377d6221fe55ae9f5f9d106048bada","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 13:27:37.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1e718dc12863a94ba8663a855d050371dfda0e11","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 13:30:47.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"2fbb8403ae56615cbf181574f07e83fc16590a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 14:13:16.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80f0eb23189845a7b43ed1b64e934ab1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382e4cb3fecd4ebba6e1394199540663 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f57dbd18b66e44dfbf752d1f5d487338 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac9c22c8ed3645c0a54da2e7a02e4a48 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a967f18288b421cbd71c500dc0c4d24 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15833acdb01c410a9694d54a62d60b3f : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d39d0d20037415b90c5ab47a1de8136 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86e3444822b14ea68817f4ad3a7d2ec1 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ae61a84dc014e61ad5e84f20ac04c2f : SUCCESS in 4m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5b5656dfc1e14510b1627c1755113476 : SUCCESS in 12m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/68069a4778f248198e4978bec7c72863 : SUCCESS in 14m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ea27517bd6ad4bc0aa80ae6bd0fb76a6 : SUCCESS in 10m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8069774bcd60452baadf6bc6330f9653 : SUCCESS in 12m 36s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9886ecdb003c47be83cabf864b7b47e3 : FAILURE in 17m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e7821626fbba403990d1e5b5c18334f9 : SUCCESS in 6m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f3720787f20e4488b5a2e2e35082bb9b : SUCCESS in 6m 02s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/14938acaee6042729cd326684b151a30 : FAILURE in 15m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/00743fcfb03e45b48e48ac7bb69b56cd : SUCCESS in 5m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1b69f72a4d1e477fad4e5197866d803b : SUCCESS in 6m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6c3a081b9e614057a2aa1ad54b1fa77c : FAILURE in 9m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f2a2c478772847fd9455be90b6c63937 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f53d020129f499090f5dc9761ff0af6 : SUCCESS in 8m 58s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1a099342e6514a4da6df9ad9b218a424 : SUCCESS in 9m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a3433cefb4224b42af62af54981a3fdd : SUCCESS in 11m 30s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0f59f560c95349b6a38690e969407d83 : SUCCESS in 5m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0fa209a84f0947cc8c4ea0a390e903cc : FAILURE in 5m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3942a68414fb42fc9186759263b0da32 : SUCCESS in 5m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/48ab2c94bbb5491a8a3be778160233d4 : SUCCESS in 10m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/51f97cdd92f4403eae76212bfbea63f8 : SUCCESS in 23m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/45d4893e76854a77a7d906c6cc89ee30 : SUCCESS in 11m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7b2ba7f30caa46bd99bd732bafbc3818 : SUCCESS in 15m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/980839f5ccdc4bb3bd607c03dbd57aae : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e4a6d03d2d5423caeea0960d537199f : SUCCESS in 5m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ad6308440504b4c914caa18049ba411 : SUCCESS in 25m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/39575ba15618476b869e14b1f4964f90 : SUCCESS in 6m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0671eff61ade4407b1b8b1db543a6ae8 : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a6c88bfc2ad4830825669f0a3aa3f73 : SUCCESS in 7m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a6bdea90645c4c67a6ba45ea542ad700 : SUCCESS in 15m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa2551b6210b4c57b95b1319f0296a89 : POST_FAILURE in 24m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b2a830ad4f94006b2977b3d3d6050af : FAILURE in 33m 10s","accounts_in_message":[],"_revision_number":11},{"id":"5a7038db0eda9d8a711fe346f86c0ce4c6815e17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 10:37:11.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"ddcc7378615901907e75d61cb90674ced02bcc8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 11:26:55.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb0389818ab6411b9deb034edeed5599\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a3e8c8aee74427095b620c4c05c52ce : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8857844a0b54902b7771956a9b393b9 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d7480fe36a347a5be9e74b4c3755061 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f63e1f7961ea4bd88fb5af1adc6ba8a2 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f03809a610040e6910c5c87b15635c9 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2834f4bc6397482d8a34ebcd55bcbff9 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c419f2e604b94ca3a17763c6b498668d : SUCCESS in 5m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3020f9cb217a48c78f6f27d63ccc1eca : SUCCESS in 3m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/811035626ce741579eaf45cc01762061 : SUCCESS in 11m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/08ba3a9ec2024fd4b604e9505f0afcdf : SUCCESS in 15m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6e90b5496b654d6882e846191f7e43d5 : SUCCESS in 12m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/78815844c0b246c887652464bd9b1a7f : SUCCESS in 13m 27s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/12f135d3fc034b9a9ea28f70aff03748 : FAILURE in 16m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/44076f6c26c6477d9f1f965352548599 : SUCCESS in 5m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ca5907a3d11b4505acb0d00e548e7d3b : SUCCESS in 6m 29s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7c440f947d264b30bf734f74a7567544 : SUCCESS in 15m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7ba3d1386b5449789ea8b73088bc4a39 : SUCCESS in 7m 16s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/43fb3dfb7f96466a95b17cc720bf7478 : SUCCESS in 6m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6cc4e4d735134983b49942de2e918e7e : SUCCESS in 10m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/192362869d834c75a82a9ef3b19543d1 : SUCCESS in 13m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9569ff6aec3d47c69f31dc544a97a342 : SUCCESS in 9m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b105919a293d41edb878ae1eb8efe62b : SUCCESS in 8m 20s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2fd7ca4e4b704ac2b26f4f94186f6fca : SUCCESS in 14m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bca837e7584a4c9d849ff4dfe456760f : SUCCESS in 5m 12s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7575f051a9e64932ba2084b6af0fa1f2 : FAILURE in 10m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/094b98206c6d47749c5ef9283d6414e4 : SUCCESS in 5m 52s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/255565039ea74fb8ae5ceac8a0298b5d : SUCCESS in 11m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a59c7f15a0d0407dab4355cd208ec56a : SUCCESS in 29m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/70b5d92e5e9549c8a8906e1cbd56308f : SUCCESS in 13m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e5d18c321c5e44cda39171aff6dc19a7 : SUCCESS in 13m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6fd52944bb84424da8db86f370fda4e4 : SUCCESS in 6m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21cd55969ff7430089d411408dd2ae67 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3640317290604c93b4d83bd2823267c7 : SUCCESS in 25m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b08a37cf79a24587a687c89c4bef355f : SUCCESS in 6m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/739dd63b5e064d92bcff08a49846c712 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b3bf84a51444a1883c2cc3833ddea6e : SUCCESS in 7m 19s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/af7c546449ed42688a066a1ee33241ae : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00dc5b187d8943a681f15460cf9e03fe : POST_FAILURE in 35m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a52fbe8fc52246e1bc41ca189466cb0f : FAILURE in 33m 07s","accounts_in_message":[],"_revision_number":12},{"id":"6c10e75f9921939a9a4e4a1159b5b1eba8602310","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 12:37:31.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"5344718db0eef8590b8d77f9efc98aa2846cad10","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 12:37:52.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e6ad18a0b5faf7ebe58a174c06fc35e98b2373c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 14:16:17.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43ba1eba19e44f09b4022af54e71adc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003a9e0524f34468934e04e0e63cadad : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f76a77384b4669845ccf011cac7842 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8384741787eb4a9d9c8a42c98d2eb96e : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a130a3fc712346ed921a8b9fb808f0f2 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f42397f1f7944328e327148435a246a : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a14b3cf7c84b46ff872d5dc834c2b041 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7440f8dd62d04c1a876757413dba2001 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f76fce2a16c4c9ca022e33d2bbd0fd1 : SUCCESS in 4m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6267443279b44f3d9fd371407a834ea2 : SUCCESS in 11m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d92b5b7b0f774be19767111c1b5e0611 : SUCCESS in 15m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/54d4a5392f19401d9c8041bb94371ce5 : SUCCESS in 10m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0813ff32cd424734b48a7100349e7952 : SUCCESS in 12m 25s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9c171dfb446447349c2152bd1673c380 : FAILURE in 16m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/63d8163716d849b78ffb863f5420f66c : SUCCESS in 6m 18s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e427d5e4ae4a45d2a17dc0d632f7bb0f : SUCCESS in 7m 26s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/14e757dcd875413095a1ffcf6db01477 : SUCCESS in 13m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1a6b21d8690145d8a49c35260b28bad6 : SUCCESS in 7m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e813bf95e3ab4e1da80fc8c7eb5bd110 : SUCCESS in 6m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0de505862df94ffb8e18e68181eb208d : SUCCESS in 9m 02s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/92f2dc900ff54d99a783e491bf1d0113 : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f5c993c49e8449fd80cf9eb2cf283dc7 : SUCCESS in 9m 52s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e4aed8bb318e48338ff2e5c31fcc04cb : SUCCESS in 10m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/10abaab0a3614743b8496c3276bfe98a : SUCCESS in 11m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c1763b51565c40c89f8ea35bc68ba6c6 : SUCCESS in 5m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/088946e0b6c64c5b834eafb3190dbbac : FAILURE in 8m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ea2211c488a246a494c5b40686839ec3 : SUCCESS in 6m 13s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c54128c670904d95a9bcb46c3cca199a : SUCCESS in 13m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b3e03448e6e9496ea9a1ffb1fa6fcca1 : SUCCESS in 20m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d5f8b1f5c6a9441ba9e643c577a547c9 : SUCCESS in 12m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3b862cc794f64bcca179e35a519e3e63 : SUCCESS in 16m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/00f3c5e1038c4ccd90b14936dd25a004 : SUCCESS in 7m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c881390841f42e0ab641e2d6c8cf06d : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f2a913a15c5457cba63b67c3885230e : SUCCESS in 25m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1d5db281b1594aa78fa71943c9133ab4 : SUCCESS in 7m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e09615f28acd4ca59e75c8ab2d01511b : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b9d1b354e74d43e7b336b60f2096f697 : SUCCESS in 9m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3ae16a98d5f24b0d9702d5146eb36953 : SUCCESS in 9m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a942a1381f394025ab2f47955ce87079 : SUCCESS in 1h 06m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b74ae9febe74bdca952b541f0b11e2a : SUCCESS in 1h 30m 48s","accounts_in_message":[],"_revision_number":13},{"id":"5debc6e5475f840b528ee52567ba864479f71fc3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-27 17:54:04.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"9d92576021d3d562965afd3267d6784e58703f0a","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2023-10-24 13:52:21.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"90966284f1511493695562dd6f7012edebdc6531","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-06 12:40:46.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"d8976fd841e29fa45063c8bd938b92e9ec18e4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 12:41:15.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/888319 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"08096921812c551e81d606b9c2bdb2adb753cd88","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-12 22:16:58.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"80d1ea3a650d792bcd02d4b1d72b7c296fd98fc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 22:17:13.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/896097 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"e23d0b35baf012cb438f5c59534f29f0e4b63996","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-31 06:21:30.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"bfddc2fb203e03b47b645eaf9765697fe137e266","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-31 06:21:50.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 879743,14","accounts_in_message":[],"_revision_number":14},{"id":"c761f875c0c6f1a18cf604459d93281d059a1f0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 06:22:10.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 879743,14","accounts_in_message":[],"_revision_number":14},{"id":"7a38c415f0bf1a5cebb52b468929a8f1ae2f08be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 06:22:14.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 879743,14","accounts_in_message":[],"_revision_number":14},{"id":"c89c5fac51143fd14ed962733a6e80029708fc78","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-31 06:26:07.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"52d9020c246552315ce57c9c5ee9ccda12abca6f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-31 06:27:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4641aec113f081784a578012c2b2c57c6839e592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 08:03:22.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a23e5e12e6b048389d969128f3f20162\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee134b3ed58d4453bb958021fbf01f12 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b32ebb0393a4df6b8aa89052bedbc54 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97a1562132e645148eeffd889262df94 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/438cff315d7041788174dc72a0dcee29 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be3b4fbb48ad4b07bb163becd1fce8aa : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fea0dd7662a14b218d54eec34e63f11f : SUCCESS in 4m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/516932766d094d16a9073f2e253690df : SUCCESS in 9m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/38b73d196c07486fad55b3348ef2abaa : SUCCESS in 15m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/14fb1f9c4ec84d889454bdf3b6351c0d : SUCCESS in 10m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fa65464928a443f8a6bc4aec2d86f559 : SUCCESS in 18m 25s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ece321b52e53412ab027f26f7f63ddd2 : SUCCESS in 5m 31s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/afe1c4d548954ec19ccb4f07c3374007 : SUCCESS in 6m 24s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/af80344174064b0ba62a153b454164db : SUCCESS in 5m 56s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/194d608971d1482da327e6adddd2121b : SUCCESS in 12m 01s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b8ae046bb2eb4f2fb6f3e6e4f4283067 : SUCCESS in 5m 29s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/67c9207902964272800727d922469a52 : SUCCESS in 7m 16s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cca2471a5ab546a4b6ba3ae1259bbdf9 : SUCCESS in 9m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/37c7708df99d4ff798f8d3160efa0121 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/35bbbf4080ea420987932132c0a3cd30 : SUCCESS in 7m 32s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/dedd41f1e51d4361ba8b016a0fc8d4dc : SUCCESS in 7m 01s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f64f3e36891d4807a7920ba1a0c98c13 : SUCCESS in 10m 06s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/dcb9f1c425d24723a9b9bffbf5e5d7e1 : SUCCESS in 4m 07s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b10d99827c4e4f4b8212bbc34ea45afc : FAILURE in 7m 59s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0991483a842a41acb6e59a10e85384cf : SUCCESS in 6m 05s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/932e8ed4578b4444a90d72b5ff44b213 : SUCCESS in 15m 10s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0e34c8dbb5e04dceb71e6eb9e0431cdb : SUCCESS in 20m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a2c0da09489640cca0e1a8fe483a173a : SUCCESS in 6m 55s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f1c60aee0c244682930b9ebfdf414f7a : SUCCESS in 14m 45s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/aeb5e48befcc465d87a54253fd341983 : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5958d3ea493a4f24b238dde61263f0be : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c7dc822df164505a1acfccea59878f7 : SUCCESS in 21m 54s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e2903b8f7c6c418c8df386e3f78cfce4 : SUCCESS in 4m 22s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7600fd5ab68a42aba2328f97b5a21c13 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/20b23ded357b410f8f9d5617ff657a8e : SUCCESS in 6m 36s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4cb3fff8d1d744efbac056720f4d8a59 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78039c4f45a54f89a4d283ee40a4a05c : SUCCESS in 37m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c84838c9d3b74ed895ef89fac81960d5 : SUCCESS in 1h 35m 27s","accounts_in_message":[],"_revision_number":15},{"id":"d2c521502c93e6ac75d634b4381613c31edfc8fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-18 09:12:04.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"2884971ac3053c3bf8f20ed900c58a610b6caa71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 09:12:20.000000000","message":"Patch Set 16: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/896097 is needed.","accounts_in_message":[],"_revision_number":16},{"id":"502dbd1279a072067444fb0edd1726b8d6e8e2ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 09:12:20.000000000","message":"Patch Set 16:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/896097 is needed.","accounts_in_message":[],"_revision_number":16},{"id":"0bedffdfef6f4a96548eba3cda9d270e1c98ff6e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-18 09:12:46.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"12329bc95b120917ab052202046d20ce455cd5ae","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 15:53:04.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"8c58efabfd2c6554bcb4c313ab4c147b4977fd47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 15:53:21.000000000","message":"Patch Set 16:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/896097 is needed.","accounts_in_message":[],"_revision_number":16},{"id":"1bc396c8af0a4d2059d938fad6b1799530fd61ef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 16:00:40.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"844ba020d6ba1b10c854fd333fc01562ddc2a5bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 17:24:39.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d2e041d6b2d42ffa5cbb3f7a8aedb14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2a63f9ee7e0410e9240d9ca15db672d : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6eb16f410f3487085c3dcbb3ccbfd24 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/206e587c45314fe58881bbbdfb5d8a0d : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc965577368b46c48a0aaa88def9acf7 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5295ab3c6fc41948153b87294bd3f77 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fef7ba3016e34b75894edf4fe4c8a105 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9d64fe131c734753b6de5f17ed76131b : SUCCESS in 11m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/808456fb339a44a08bea0f26aff8c90a : SUCCESS in 15m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c3924f00aa7447e594d0a52b2ea0c4ea : SUCCESS in 10m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7112d6e25acd4104bb854875688e7e33 : SUCCESS in 18m 31s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/bc8e34ffd9804268a117150b76d210ae : SUCCESS in 4m 43s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/1da19f36d8bc4f1b92fee22b1c4c520e : SUCCESS in 4m 30s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b5127ce9a4e744ada7957d22062caa4b : SUCCESS in 4m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/0d1f153a16494c518160dc1d4dbd4647 : SUCCESS in 11m 07s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6779e1b1321b43d7b7058f71aa7c9848 : SUCCESS in 6m 01s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ac30da64c68f458a8293b31073cbb45e : SUCCESS in 4m 54s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/fe8d072632c04818aca59918c2fe3049 : SUCCESS in 7m 31s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/1fe670ed103843eba6b05c7872d5aaae : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/812852ba43e0420aabee2e101c5a3eee : SUCCESS in 7m 26s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/69e0a9359293472196d306c2732a8ecd : SUCCESS in 9m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/6d2aec25810e4302b756af4751b5a987 : SUCCESS in 9m 38s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ac8e6b333d814ca8802c201ab84fbfbc : SUCCESS in 4m 15s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/843a243b5d5f48c5afcbd00d789a94f3 : FAILURE in 7m 58s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ba30fde986f74b5c9d41d10350d95e90 : SUCCESS in 4m 36s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/87ee5a80f4fd4f35893a20752927e26c : SUCCESS in 11m 28s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9f963033630349398d074c31ba1f5c7b : SUCCESS in 24m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/92652854874a44ecb694ce3e880d75a9 : SUCCESS in 6m 19s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/97e43eff8a454aa784b1b1bb2f24ef30 : SUCCESS in 12m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/332c81ab6ba44755b6a7036a10a47d2b : SUCCESS in 6m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/806a78f6ea524b4ab09858e490ce40b4 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e5dd24fefc384213987587e8cb62838f : SUCCESS in 22m 08s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8cbf17af32ea42bfa207701f5c436443 : SUCCESS in 4m 02s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c307c4578c3247ea9e32b02916a850e7 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0943a945c4644d3aa0dc6a15295d3fef : SUCCESS in 6m 56s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/51242250223844029466350de8c1ffcf : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cff9d15b272647f58db9da487f0c66dc : SUCCESS in 1h 11m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2d91a121fcb4d76a58fe1e3357ca4b6 : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":16},{"id":"0c0451e54eeaa03ed2fcc2cbcd0fe1392240e2c1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-26 16:54:30.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"d23814ec7bb4e122ca4fe2e7c05257b817516591","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-26 16:54:50.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 879743,16","accounts_in_message":[],"_revision_number":16},{"id":"231a1dad97000530bdcac2e2c3eafb91317fcec0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:55:17.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 879743,16","accounts_in_message":[],"_revision_number":16},{"id":"515ec9855c9dd20a75b3fc6714ec0112cc2508fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:55:31.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 879743,16","accounts_in_message":[],"_revision_number":16},{"id":"d3e09accf1d5a7c5825dc53ad45ce73225995933","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-01 07:17:36.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"2762ce27a66d759cbdf023a164744670f656dabd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-01 07:18:40.000000000","message":"Patch Set 17: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"4fd14a6f14ee060c87ef58159a9b565e47d81028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 08:47:29.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5ca40713c29404780a54a8da2a01a3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5bfcb2edb434de6972a8a9b3b93cd6c : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a3b812f4ce4bfc919d1d57ce36e11b : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a0e06855c7c4bc29984057b90fadf60 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51f4afc5605d45bc91106eb8085aadaf : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c123e6857705490f8f045cebce20ce56 : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac449862e4854322828219459a17ff16 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4efe4e1397ba40ab8103c967d2bd70c2 : SUCCESS in 8m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b2b955e7bada4a9f84a671bee7309217 : SUCCESS in 14m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ade77d10614d4620aac585729df36f43 : SUCCESS in 9m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8fffa6dc0a5f4e01be8e4f0317ef3d8d : SUCCESS in 16m 51s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/3942327868fd40b1bc05bb10a02b4b55 : FAILURE in 4m 44s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/8315b39da117497bb3169fc8026df66f : SUCCESS in 4m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a93ad827711d46418d849a5e9760ab09 : SUCCESS in 4m 56s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/feaf861f09af47bc8e958109d921e2c7 : SUCCESS in 10m 13s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/79487acdc5ec49b584dbfe5d5e2884f4 : SUCCESS in 5m 06s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d419fd9c12bb48b7ad0e24968bb53292 : SUCCESS in 5m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/32b16783df674f528e0dab2ca5e13124 : SUCCESS in 8m 03s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/65656828ebbc4cee9a63a317839fdd35 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fee60a691ed44f11bf9845993c3d9f94 : SUCCESS in 8m 28s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8214c0517bc846768c19da451c51089f : SUCCESS in 7m 02s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/cd98c6d91b9a4907ae1924fcb981248a : SUCCESS in 10m 22s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d03e500b108046afaf9f306cb410ace9 : SUCCESS in 4m 08s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8363eab0890141b9b82670e555540ad8 : FAILURE in 7m 45s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/85c2fd2b78ff4d11a25afaabacd00428 : SUCCESS in 4m 45s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9f923a762ef541d88fcc0f5f2c5fe525 : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f13ea3ade8cd45d186a186df4b40713d : SUCCESS in 23m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e5d6f51fee4d49babe6b652fc9eca75e : SUCCESS in 6m 17s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f12f77273b3046dca3d41c4349458764 : SUCCESS in 10m 45s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/9860528a83a241029a3f3f258d309caf : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a9b6d5c0da64b0ab6fa843c866ab4a8 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e65fca42c2394eacb64bc0a1a9815640 : SUCCESS in 21m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/04167488238c4665a569c3d473bd9f19 : SUCCESS in 4m 44s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/823a492ba987436aa9d095826c094067 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2fd03f664ee941d693f700dd0d20417f : SUCCESS in 6m 27s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/db1210eb11ce49bab0e3e4475907f51c : SUCCESS in 11m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2677569048744ab68998af66b89c5767 : SUCCESS in 55m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4623673d2ac4bffa759acac09b5926f : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":17},{"id":"7b5420875d86555f620734c9fcbf8d6f91a03dd6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-28 18:59:04.000000000","message":"Uploaded patch set 18.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"4731840e2ff0972e10fb4383ca7ca644b80d1304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 20:21:30.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51bfe75fd30f48f486d9a2d049d0e58a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4e6e2a50af47e29965077380a789ad : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f45393a83f4b439b717a469d5652a2 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4557af0db5e041a590ea7598aec8d8f9 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b344a6139e0d465589ce39def5108e6f : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4033fbf1edf64c98874d685e31750d67 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bee3b94d529b45e08dee5ff167aad0e4 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17257b5cc28145acb8f42bf253e7b6ae : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5b7e07088dd64979b39ec241096643e1 : SUCCESS in 8m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/71d501fbcf4e4c3cb2f71069c00784e5 : SUCCESS in 16m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bda8e4e87fd54fc281bdf4f0e428ef79 : SUCCESS in 8m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/30142905c49d429ea0070f02312ca2b4 : SUCCESS in 17m 22s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c1d2cd5e1b7747d3bf817d014a935299 : SUCCESS in 4m 59s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/6109eb26c6004d20a0748a47390304e6 : SUCCESS in 4m 55s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/20f55c642acb41a98e20e6145fc69bef : SUCCESS in 3m 47s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/68bd22750667460195f93b162cb7db41 : SUCCESS in 9m 20s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/431610c4168a41729728c344222826eb : SUCCESS in 5m 03s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/544fb1a1b9d04c058b8b7bee903aca07 : SUCCESS in 4m 19s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/5bcd786bd55e4c7ba9af4f97fedb6f61 : SUCCESS in 7m 13s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4cb35adddd00486cbdacddeee259b031 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c78d92a53e0426c89e35716143fdd83 : SUCCESS in 8m 00s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/cbb05ba1aadd4fab87c442a705e4759d : SUCCESS in 6m 38s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/017ffaaf901f46aa9a426dcc68c2ead1 : SUCCESS in 11m 08s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9f2fa2486fe749c59c94480957cc05b1 : SUCCESS in 4m 11s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/169cd48c72c94173b44fd7a73d90c4f4 : FAILURE in 7m 50s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/b9a7d294c4a34f6bac3e2ac576e13315 : SUCCESS in 4m 53s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0174964c1cdf4b5cb6dca41d7add27fc : SUCCESS in 4m 36s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9a5fe3080d5047a0b6cdad4d6ce2d7c5 : SUCCESS in 11m 42s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a5d34f9bfc92406181284ac360de938d : SUCCESS in 25m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a2e9d7f909a6463d97dbc6682f66cb79 : SUCCESS in 6m 49s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/88f03ab72f2e48369e4b6e47371f81f0 : SUCCESS in 11m 38s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d849775fb59948feab796a1d27c5cde8 : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e45bb54e2974df3b686531ad5727a24 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ffde820ae9d1452abad45188889f0262 : SUCCESS in 20m 50s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ff4cfc24e5554013bb1be70952ac1a24 : SUCCESS in 5m 18s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/077b63009ac047a8894a0af348ff9844 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f3969e5456d42fab53728e5cb55aaa6 : SUCCESS in 6m 13s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b1e17ada70d94306881175717177d5e4 : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fab887ce9a44b4289349a8ec6c99087 : SUCCESS in 45m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9649511103ba45688e56a442a817f2a4 : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":18},{"id":"c6ebbbaad8798fc79c0a70a2a8c877717c45665f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-02 10:54:41.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"20ca9300649a0e96a240120694cb3da48da73114","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-05 10:23:43.000000000","message":"Patch Set 18: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"37e6f483386ed7baa2cf0149b435fbde00d1e766","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 12:11:12.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acc267b262e14a3294c65e52caf5d82d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97b691c20e2f4d1a85501e38dcdf2c31 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a767d2b89a5e4ee696d459a724d0695d : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c291f64e43384c21a54ff59780ac237e : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3c4678e453544668ac44fb824487cec : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b4980c041074c839bb6d8f0ca3ebeec : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04512473f70c48ae82bf90cca874455a : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7aa818c79f64fd78688757853330456 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7ae9bf4ceefc4c56be64c9fb8e605006 : SUCCESS in 12m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1bc9904b63304c7490e7055767f13dad : SUCCESS in 17m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4fa0a4590a8a4bf69ea55fabc562847b : SUCCESS in 9m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/43d1b1b022fa46f39d9637e90c0b62d1 : SUCCESS in 16m 29s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/60b00cb8b0514166bef5c3695d49e375 : SUCCESS in 5m 25s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/c4150cb9b49947729efcce4ab1225e6f : SUCCESS in 5m 19s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/705deaccd3f94b658e9361a38501fd7f : SUCCESS in 4m 28s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/f1cb2ff3759b4618baeef0a1b1186cd7 : SUCCESS in 10m 34s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d88df96f905849a5a558df7afddbab04 : SUCCESS in 4m 13s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f13e8f8c9fba4d238ba256b3b81baad5 : SUCCESS in 5m 13s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c24d92f8eaf64fc190cd1a91e8a9f15f : SUCCESS in 8m 01s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a879c76919f84a0995b6c04129d4277c : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78bf4621f6c84c62ac2cbf4df9e4d29e : SUCCESS in 6m 46s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f67aa32ad6394df3986fef20869cfad3 : SUCCESS in 6m 13s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8fd36e4a68ed476889ba3f5b4eb39753 : SUCCESS in 10m 55s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8de4488186e14093808af16d862a24fd : SUCCESS in 3m 59s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/0df62688e3f549ee845ec1d200610a66 : FAILURE in 7m 47s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/fadb06ecbcad42468b3ef830e8ff6f7e : SUCCESS in 3m 57s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/3109547cfadf4fd2a6179f7490fe6456 : SUCCESS in 3m 49s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/fa20fba2930442d2bd898a5b6da3c068 : SUCCESS in 12m 57s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/3edd87c32376458d8014ee4d75bba96b : SUCCESS in 24m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e232923e6a4c47f2819446abf2169a9c : SUCCESS in 8m 04s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c1f2a846c02c4a18a06ce57a614c5483 : SUCCESS in 11m 57s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e483f66d006f4a238a186a83c89a0949 : SUCCESS in 6m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c9fdc8b004f48139577867623c0b742 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6549c2e074f14ab294b2f8d369b2b2ba : SUCCESS in 24m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3072b6b21f444a1c812918d6ee3c7a91 : SUCCESS in 5m 47s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/646acbda638f445aafe2016ff49c7314 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c67fef17afa341f4ada9eca171d02da9 : SUCCESS in 6m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/33a6dc29c2eb4de39f886dced1c3733a : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b7a15d191bb4c31911403c17771b3ff : SUCCESS in 55m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/033488f68e994e5099943d3cce4de51f : SUCCESS in 1h 28m 02s","accounts_in_message":[],"_revision_number":18},{"id":"1684f06844818485233b936113e804641ce2765f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-05 14:31:40.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"c6f5f243560d16ffbecf32814fe56fcff42a1aa3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-05 14:31:50.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":19},{"id":"4d8db7f28b9c226fb7410cd03d223864e5365e24","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-05 14:32:15.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"0770251517e6a7c367eb9d7158b3d4c60467bc85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 17:05:36.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f893de9fcd1842978086f9634763077c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab5b592094b4aad8e68c3ddc8c7ab08 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62bca92a29b64776bdf215ab1ec953a0 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/860630e9eb684f009a5d6d72364aa253 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed076257eb034e68977f51501380503b : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f118f5af83aa452ca7a881f8cccbb4cf : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd0c9fad80714de9b71c1d035896e307 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a38f0ffa13e4e8893000b006d85b526 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35248e6c57c24c70af30e1398ce3721c : SUCCESS in 13m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dea2aff52fb94d17864cc70280286003 : SUCCESS in 15m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5d81a2fe7f5548a5abb4a0e5ee819ab6 : SUCCESS in 10m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/80af83c890c649a797ea2cc0a9bdd78e : SUCCESS in 16m 55s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e1e2b526dfb14c5da08f8e64d2d7b980 : SUCCESS in 6m 57s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/46c5b71fa2ae46a891d52cb55e2ec6f4 : SUCCESS in 4m 56s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/7099dfdbf35f4f57b737787fbae402b8 : SUCCESS in 5m 50s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ec511bfa78404bb28d88bda6c5dc690b : SUCCESS in 10m 47s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b97265d3953243c3b334cc2980b65887 : SUCCESS in 5m 34s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b592c701f7b04236a2efdff302aa2129 : SUCCESS in 5m 33s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2b14677b2ee548a3bdddc7685a9ebef8 : SUCCESS in 8m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a50f9a07fef94647a9440018b61ba206 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d95c586464d0437089ad3ffba4a46a9d : SUCCESS in 7m 16s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/32cc81c2e4334b91ba26d307ded930a7 : SUCCESS in 9m 02s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/4f4ea907330c4f5ca71362b714734a38 : SUCCESS in 11m 30s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/521582d122474ab8bd29f9fa332a1c42 : SUCCESS in 5m 26s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/503d71fb0f1246e6878feb94074a62fb : FAILURE in 9m 11s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/011d22e6344343e2ab93f6a869602b93 : SUCCESS in 5m 06s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/c4fdacf2b30840a5aa31c6e85623ba35 : SUCCESS in 5m 05s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/97aedc4dae7f401b9696e7bd5de5a5db : SUCCESS in 13m 45s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/de7ce41abfca4c9e9157424cca8b9814 : SUCCESS in 21m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/17d05be714da413aae21aeee860db2ac : SUCCESS in 7m 56s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f7c432d8db9c4d39af8c38c64a66b826 : SUCCESS in 13m 02s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/fb7ef82dfc704edf87d016ea8649e27d : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42570fa863e34ac78a8033e3b9042554 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a704389153fe4fd1996c3bf576fbf2a5 : SUCCESS in 22m 16s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ae6f4438241f4176beb98e433703b39b : SUCCESS in 4m 41s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1cfe02128c70437a86ed0b8d0c8bd1d2 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66e81da4314a409d9db3dfe0e13c8415 : SUCCESS in 6m 14s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/bb1632606f6842d692dcbb53c4bde759 : SUCCESS in 14m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/493cc911e4714e92a66182e22c071e44 : SUCCESS in 55m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/436799017fd7406183914bc4061db856 : FAILURE in 1h 23m 43s","accounts_in_message":[],"_revision_number":19},{"id":"a0b8f50faae2a6a9e62eda5226b782eb8410c248","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-06 08:55:58.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"94676c08efba5dde9b603fb78ca28f97b2858c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-06 10:44:55.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b6db53597d477894e387822e62dabd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08855450ee7a49419e4e5e971f8e8c7c : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19f72ee50d8d4fe79283004e6847f841 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa6c9e5773f44ec59b7157e4b7e4de3c : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea9084bcbabb452180f6516271a76bbc : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e8c95b1828842f19101e3653ed79d2a : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/309104f06a3a493c92adce738f4618ba : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0cdc72bf24349b581f1b0a960da6e53 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/06b0363a41a54f73a07567f4ce91ed04 : SUCCESS in 12m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c2e6f4f00eb9404992dca428e584268b : SUCCESS in 16m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/84672679a9a74d7b90f5c7b40f3e14eb : SUCCESS in 13m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fea7fdf803684c09bf57d54de663ec57 : SUCCESS in 20m 49s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ec3c5570fecd4d8fa1203739fd7aeffd : SUCCESS in 7m 03s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/45e8f844ce374555999ee3ee114d7c6e : SUCCESS in 4m 48s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/180656b579d641028233ec4aa48d8cc2 : SUCCESS in 4m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c88e91619d6744bd8a733dd21efcbc76 : SUCCESS in 12m 17s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/34c9532e02df4ef2aa6ff803de9199b6 : SUCCESS in 5m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/dcd3e567bfdc42b783acdd2415b523ef : SUCCESS in 6m 24s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/a101f455e0ea4d49b485f10f77fdb2bc : SUCCESS in 7m 54s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e858baa5fc9c41ac941e602c7bb28448 : SUCCESS in 12m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1259414519eb4ac099f7a3db3cee35d7 : SUCCESS in 7m 27s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/21a02815041a4de394cdf2d6afcc8110 : SUCCESS in 8m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5bb82094ab594799b34148c59b7e43d5 : SUCCESS in 12m 25s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ce2a39f65f5a495da1750c821fbb09be : SUCCESS in 3m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c33c0c469fd6460ca9b126fbf7bd4745 : SUCCESS in 7m 49s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/b5d671a8a8e84bfb81fb33286ef7adc0 : SUCCESS in 4m 54s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/44967a94973f4b7bb4f08361a5c9e416 : SUCCESS in 5m 09s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/89bf3d12ebbe4e4d991e62d1a434031b : SUCCESS in 11m 41s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0ce8c3e50ade43678b1014ec26e0b2ab : SUCCESS in 24m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4a59925911c54bc2aabc774ec1bec238 : SUCCESS in 7m 22s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/0dfb3ea6e2734c87af044d85f9634123 : SUCCESS in 11m 50s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8b36df9704914a18b1f0c238ba64b513 : SUCCESS in 4m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ee6675c28b243e895a6706d08e9406e : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08167049bd914705a9a388f5e9f79487 : SUCCESS in 21m 49s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8c96777e9b3a4983b0d953764b240bd5 : SUCCESS in 6m 10s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/aed81af114e3424e8a49697cf3611b97 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1aa3a4af36b427d8b2d7781800a434b : SUCCESS in 6m 21s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/994d07bb474b46e98adb3cc7a8c38e21 : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bee53f2d9f8496fbfc6cef21f861215 : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69d1b507ba8044f08aec2fed1dc55f15 : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":19},{"id":"506c3b0dd1fd424e5785fed645847bfa3596a5f0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-07 06:26:05.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"14cd053ed10958b41927d190b8448c8cfa0311fb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-08 22:46:49.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"02f8b97ff15da76f6d7d2505220004d4ac4407ad","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-09 09:19:28.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"64190df4c65c22684ef9f55b42e1534d14b462d5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-04-09 22:58:46.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"edc7fc5bfed6356de30b35d6165b91c1afea04c6","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-04-10 13:36:47.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"e91c84ddd42e301bf45d383f918660b9ffe70e88","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-12 08:25:15.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7c4b1118c3f085155f92bdb9b8789df0e9c3cf99","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-12 10:40:54.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f19b7d2b58f46513c312048c84ebfddf1bb186ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-14 17:00:31.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"9350ad96c91bf07461238c51c141ea6379948983","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-14 17:17:58.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"9d587701738ff4cc0b2da1ee2deed897af2b871a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 07:59:22.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"9fcc97aff7e06c4a413ebafc3a5ebeb436b1d0b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 09:28:53.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caeafc4e5b90468fbeaca60f425327b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf71a765cde646828f65f297d2123b5b : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b08a4a77d724866921f56b0338a40bf : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bddce5a541b44e738bb0934f722f1091 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d63f05c111234d119193077ea57328d4 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2abecd5cae004c11b4c460e0bda396ad : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b40d9c436563491a97e6070db495db5d : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2f0295178c340f0874f90b7049040db : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a54c1972188140499386f8e15f9a162d : SUCCESS in 14m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2fe859fdb2a84dd287d28fcb4766cf1a : SUCCESS in 22m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bd0deb7a81d94453b1bc7c6cfc293edf : SUCCESS in 12m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e89e07070aa54f55a5c1ae3119be4f64 : SUCCESS in 23m 22s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a137f2e6be274d8db2e334f0321fe07c : SUCCESS in 5m 49s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/bce47c0b64354d1e868ee3c6f1ed32dd : SUCCESS in 5m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/c514d7362ced4e728c9afcaeb7a168b9 : SUCCESS in 5m 38s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/31b7c09dc5de41578105084153ef3f73 : SUCCESS in 10m 37s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/77970276affe4b5f82078b2a64d36314 : SUCCESS in 4m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2eac7b65dd604bccaced75d6791962f1 : SUCCESS in 4m 14s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1b17e7b3d7f44d0fa99fdb4337d33786 : SUCCESS in 6m 38s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/bb05e873316a47c9bcc936dbdf634ba5 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6e4ce4d72be40428b9279fc8e86ec34 : SUCCESS in 6m 45s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fea484f295b64dbb9e9e7bc35927831c : SUCCESS in 11m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/bae415f44541438f843d42abe2116086 : SUCCESS in 8m 33s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/36f43495c9e04333adc2cf493d766233 : SUCCESS in 4m 02s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b6d39c92f49140d9983e213cb7ba463e : SUCCESS in 6m 27s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/3d9c1df867294478aa007c1685190dd6 : SUCCESS in 3m 52s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/24248007f42b468e833a7d0b8afed564 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f295bad313ef49f6ac1dfbc701d50c1f : SUCCESS in 10m 34s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/be2fa8b7631048b0b07123ef1caab51b : SUCCESS in 25m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5de33ffeec80497ca3a4ae2f37d43011 : SUCCESS in 6m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/fb5bbb75f35f4895bbe3a5fc31e36a17 : SUCCESS in 11m 51s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/04caf4abe4bb4923a276b2667ddc141b : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/94beb2eababd4a9cae2fdad38345515b : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/66cc622efebb4381977961dfdcee1d50 : SUCCESS in 23m 24s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/63438bbd660f40b2a0424d4c8e78e15a : SUCCESS in 4m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9fb86b0cb24a4657827e50df0f2c070c : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58d83803a6cc4c239a1d09d8f2072202 : SUCCESS in 5m 52s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/506f69f136ec48b58e889ee97cf72364 : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00a6cd03918549588a09a30c6853790a : SUCCESS in 1h 02m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/345de73d78a649e6bfeed0ff56ef74d0 : SUCCESS in 1h 27m 15s","accounts_in_message":[],"_revision_number":20},{"id":"3d619f171ba9124ab10d27002d760892e871c42c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 15:26:52.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"e856cf3ff274661b2c2219b5d546d806dc249f24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-18 18:50:49.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"3a480651d4292370ab735081a63d7ec51d5ad203","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-18 18:51:48.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"fe1038d413b1543778bf2dda28f394993c04ab81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 20:11:52.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7e91f4d64f945d2bf3953a9df9158a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c80312e6f24c3aa573bb2a1473bc21 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700072e0c24542869e0b56eb3b6e22bb : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc1252a6ee9e4fafb84fbd60ba7da053 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c397bdd34a154c06876da80c463ae611 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b502fa330bc5414da96e161770fcdda3 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/656575e74aac4a30bd4d99120ff1c199 : SUCCESS in 6m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e0a8550ad5344d7e8b6bd668d54e5069 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8c316f0d31e450fa6435004286dbdbd : SUCCESS in 9m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8738bd8707224d6b8b4445bc28f1ade9 : SUCCESS in 21m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9d954520d3c7489b986b5e4ab9de97da : SUCCESS in 9m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/58322fbe7d6448b8920f40a3b2081a1f : SUCCESS in 22m 32s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/f4b282f1da9c4073af83046866916973 : SUCCESS in 5m 24s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b229136963b54ef6afed54b09df7feba : SUCCESS in 5m 16s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/596cd00a6af74abf8fc80724d5044bf3 : SUCCESS in 4m 52s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a18df6814e4f4bae8b1d558f16322f8c : SUCCESS in 11m 05s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5851cfaee1ec44c283cdacb4c0d4edcd : SUCCESS in 5m 06s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/10fffefb865e435b8dd76b1ebe5efe29 : SUCCESS in 4m 44s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e3399706e970428b9667e178d5143570 : SUCCESS in 6m 03s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c04e2eae68f949bfa9b7bdbfcf60e2b8 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e78013b60e2646af8e98bcdd58035dd9 : SUCCESS in 8m 20s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/1b4ead7746cb4b15a7a384781dfe71db : SUCCESS in 9m 39s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8bb75b24c33c4202b7ac27c6911ea3a4 : SUCCESS in 12m 08s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a8387feb276f4e5fa4d643002d3f41d5 : SUCCESS in 3m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/800a3f3394554dccb989708a579edd02 : SUCCESS in 7m 53s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ec79ff8e5dbe4ebc91e8f08b9e6d97d3 : SUCCESS in 4m 05s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/925aa0fcc65c48728dc2a6b5da6163b0 : SUCCESS in 4m 12s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a28b3af8cc4e412c89bcfefe583c3682 : SUCCESS in 9m 19s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5a66dde1e8c7489a8dd0d9e67642f73a : SUCCESS in 23m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8054f291da1141808b1a1dc2646d35c3 : SUCCESS in 7m 10s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/37a8f5d1e6f84edf9e1c26ce51ae39de : SUCCESS in 13m 33s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/72cb2b350b2c4af8ba0d248a070e2958 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/685eaaf919db404b97801a3588792459 : SUCCESS in 5m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c635fa718d8d4f95bc4ef7da45bd967c : SUCCESS in 21m 06s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c33937d0bf294d00be6ee203f3fe8e0b : SUCCESS in 4m 34s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/949ca0bf065e4cfea2ee83085b1f5300 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fefbc75f1c7745278846899611f41321 : SUCCESS in 6m 36s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b5e95e9b6cfc43ac92a400cc042fd036 : SUCCESS in 13m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03b184fb63b54c098daf2d6205fa40a7 : SUCCESS in 1h 00m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f7511b3e3604a18b6e9f0f8abcb89f1 : SUCCESS in 1h 08m 20s","accounts_in_message":[],"_revision_number":21},{"id":"4c52395cbffa9053d189f8a29655af1ba1dfdc21","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 01:06:24.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"b4aeda221449997185637f14a0cec1f5bd8d42f6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-19 06:54:50.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"3c0e75812f93f3737e5e0ac262509b94e947489f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 07:21:38.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2626ff1df944cb1d711870826d863929e0583b43","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-04-19 14:54:12.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"829e2226d0f1bd96317fa8e8145346ffe3475cb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-19 22:19:19.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"632cde9cc1252e664b77de16f87fe4645002f7b5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-24 14:45:18.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"0eba604808ac110b1df3f575d5977d990b203d58","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-24 15:30:38.000000000","message":"Patch Set 21: -Code-Review","accounts_in_message":[],"_revision_number":21},{"id":"c6d94ffde320d177f7141983689a2beaf2460db7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-24 15:56:02.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"e7cf5ac5be908b8b7b5151259f66752fdbea63f6","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-04-30 15:12:29.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"defa02866962e9c6276cddbad84ac71ffd08f896","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-02 08:01:44.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"3d08347ed99d74fedaff452ef81084c5e399274e","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-07 14:37:24.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"3b11580e399e117b754eb87a8b53e587bf5b5a8d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-08 04:59:20.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"894a58d65406bb8f7692072ba12de3c9721bc9d1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-08 05:06:05.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"1330154b37d4b686b3b05d764c3768372dbadb2d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-05-08 05:58:18.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d48e6c2fd52957345a53a183ac25017ae93da095","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-08 13:04:35.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2c9a574d4e9b572c8478b5f15fa1265c6726b421","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-05-08 15:51:51.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"8d71d8470397cf6bcaf2bf4bf0407f89b374ec25","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-05-09 12:31:38.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"1fdaf570f85e999a203c0c4b912f466a7b20c620","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-05-09 12:32:31.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"bc40a2db37d24b64997db2a017ee1acfcc28d6d1","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-10 01:19:01.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"272e2a7cd55b053fcab548212b5f0403b8a68646","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-10 15:18:23.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"dbcfe9cf2cf1ac9d5e56622c0928aaee0060e47a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-14 10:08:55.000000000","message":"Patch Set 21: Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"ff837503c7f9fda82735c1a0eaf55f627ad7ab45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 10:09:45.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"dd72d020b959e33f96ea04ce9d35bca034b4a198","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-14 10:23:41.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"c25be3a1bc20fdf463710ac8ad319084791cf0af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:06:08.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20ef2997b048458eb676d19bad166140\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9212a7b3df2349c18aae1d0fe2e62a23 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c092d7b4d9104a3982bd45edd757bd95 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ef54b6abc0b4f999b687aef14103dde : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2934747cc7c8408e9cc182671a6fdbca : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/459c446f143e41eabcb16b83bf32e8ef : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25add93fa0a349c9b7b2cb8454e88118 : SUCCESS in 2m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2d82115bdd7c4e51815ad1c228b52f25 : SUCCESS in 10m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/10746bbcd0a641a08c7bb33e10fba963 : SUCCESS in 21m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1079d6579a7044d7911518647c054cee : SUCCESS in 10m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cc9e7352841644609a1709bd07a887e5 : SUCCESS in 18m 08s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/445629cff6af49e6861fb000ae5afbd0 : SUCCESS in 12m 31s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/caec3ab9e2fa478f8e707292a4a37697 : SUCCESS in 5m 09s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/cf39c53b451244ed908aa1c04628bca2 : SUCCESS in 5m 20s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/a4b8ad405fc8489e9db2fe76fe600a47 : SUCCESS in 7m 03s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/d83fe84c58d74997bcdb51eee0dd5097 : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b4f2b712f524e618738a300c50b1cc6 : SUCCESS in 8m 07s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/17f5c142dc4a42dcbab55b9083cd467f : SUCCESS in 11m 23s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/611ab7e675ac45d0b2b717da1c137f98 : SUCCESS in 11m 44s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2ef38533f66a4b9fbe5f698f2d5aa62e : SUCCESS in 4m 11s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7f8cc235f1b74669aaad468914070612 : SUCCESS in 7m 43s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a0a5d535270c4a02a7253ea39ac1e577 : SUCCESS in 25m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5451b9212c3644e6affdefd7f1bd9c12 : SUCCESS in 5m 58s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8c7761cd2d7348cba35ae2d768cccba5 : SUCCESS in 12m 01s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/98cddb1c3b7642a89dae7320553ff5d7 : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba2ab2a71b224c769da50a944da1e0a7 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d0307f3e81b4bba82874f2d61d459b3 : SUCCESS in 22m 34s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c688202c8f3945e28426500952dc792d : SUCCESS in 4m 49s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0932783624944a9685effd60a40736d3 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c4c5a6ff95f44eb89f4c43f3e2a7a6c7 : SUCCESS in 6m 16s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e02a7d15f3e340df9ca43523940f76ba : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8296bbce7362417798695c6a731ba147 : SUCCESS in 56m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c9ba8e115d04b2583796de351537501 : SUCCESS in 1h 43m 24s","accounts_in_message":[],"_revision_number":21},{"id":"ce801506d3d4ef8e0cfc6d578327bb58f8ebbb87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:06:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"6fe30e3a2f3680650db33bce9d9f113bdbb1d1dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:07:00.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c2c6ca21fe44d9daba878752f2dadf5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/408eaf151aed45cea2e4335f9ce7f412 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":21},{"id":"7830f3748bd0b5ec57c3741469f30fbc655aa3fd","tag":"autogenerated:gerrit:revert","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-15 13:20:35.000000000","message":"Created a revert of this change as I7a3f58df1b00cc4cc5aa2270cf4160764635dde4","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"92c8b4c52b3160df1db1768e8891fc19eb704424","revisions":{"9de7cd155a4229115257ed3fcd7610c7309da211":{"kind":"REWORK","_number":1,"created":"2023-04-06 14:03:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/1"}}},"commit":{"parents":[{"commit":"b44da90728ae3db63b95c4ea6291b072b41ff2a6","subject":"update constraint for python-swiftclient to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b44da90728ae3db63b95c4ea6291b072b41ff2a6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-06 14:01:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-06 14:01:13.000000000","tz":60},"subject":"Update SQLAlchemy to 2.0.9, drop sqlalchemy-migrate","message":"Update SQLAlchemy to 2.0.9, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9de7cd155a4229115257ed3fcd7610c7309da211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9de7cd155a4229115257ed3fcd7610c7309da211"}]},"branch":"refs/heads/master"},"d0cfa5b3e56b650d093ea9db0df280c06ef0bbab":{"kind":"REWORK","_number":2,"created":"2023-04-19 09:21:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/2"}}},"commit":{"parents":[{"commit":"b44da90728ae3db63b95c4ea6291b072b41ff2a6","subject":"update constraint for python-swiftclient to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b44da90728ae3db63b95c4ea6291b072b41ff2a6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-06 14:01:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-19 09:21:49.000000000","tz":0},"subject":"Update SQLAlchemy and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db to 13.0.0 since this is the first version with full\nSQLAlchemy 2.0 support.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0cfa5b3e56b650d093ea9db0df280c06ef0bbab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0cfa5b3e56b650d093ea9db0df280c06ef0bbab"}]},"branch":"refs/heads/master"},"9d8c6e3f7b48c7cc3165e226f1d9dfad1c81a753":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-19 09:35:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/3"}}},"commit":{"parents":[{"commit":"d25ec820b127d383d61d64c4b7d9cbf0e1fae3c6","subject":"Merge \"Remove the Patrole deliverables from req\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d25ec820b127d383d61d64c4b7d9cbf0e1fae3c6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-06 14:01:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-19 09:35:55.000000000","tz":0},"subject":"Update SQLAlchemy and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db to 13.0.0 since this is the first version with full\nSQLAlchemy 2.0 support.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d8c6e3f7b48c7cc3165e226f1d9dfad1c81a753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d8c6e3f7b48c7cc3165e226f1d9dfad1c81a753"}]},"branch":"refs/heads/master","description":"Rebase"},"70267faba7aeebba77b37ccf640d87fe97e4ee79":{"kind":"REWORK","_number":4,"created":"2023-06-13 17:42:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/4"}}},"commit":{"parents":[{"commit":"299b0a97569fbee219054cf631b28041619f4cec","subject":"Merge \"Deprecation of networking-odl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/299b0a97569fbee219054cf631b28041619f4cec"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"subject":"Update SQLAlchemy and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db to 13.0.0 since this is the first version with full\nSQLAlchemy 2.0 support.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70267faba7aeebba77b37ccf640d87fe97e4ee79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70267faba7aeebba77b37ccf640d87fe97e4ee79"}]},"branch":"refs/heads/master"},"d70c9d39d201dd995bbef47a740d3fc772dd11d1":{"kind":"REWORK","_number":5,"created":"2023-07-12 11:56:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/5"}}},"commit":{"parents":[{"commit":"e729367f3547fb5754109157ee6b831db81d9ba2","subject":"Merge \"Raise cap for XStatic-JQuery in ``global-requirements.txt``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e729367f3547fb5754109157ee6b831db81d9ba2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-12 11:55:50.000000000","tz":60},"subject":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db to 13.x since this is the first version with full\nSQLAlchemy 2.0 support and bump alembic since recent versions also have\nbetter compatibility with SQLAlchemy 2.0.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d70c9d39d201dd995bbef47a740d3fc772dd11d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d70c9d39d201dd995bbef47a740d3fc772dd11d1"}]},"branch":"refs/heads/master"},"70f2e0ad019f1ddc230e3b8db521e060dc55d77c":{"kind":"REWORK","_number":6,"created":"2023-07-12 15:39:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/6"}}},"commit":{"parents":[{"commit":"e729367f3547fb5754109157ee6b831db81d9ba2","subject":"Merge \"Raise cap for XStatic-JQuery in ``global-requirements.txt``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e729367f3547fb5754109157ee6b831db81d9ba2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-12 15:39:41.000000000","tz":60},"subject":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db to 13.x since this is the first version with full\nSQLAlchemy 2.0 support and bump alembic since recent versions also have\nbetter compatibility with SQLAlchemy 2.0.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70f2e0ad019f1ddc230e3b8db521e060dc55d77c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70f2e0ad019f1ddc230e3b8db521e060dc55d77c"}]},"branch":"refs/heads/master"},"68346955b43cd4f246d866e2ed7fcd4080171ed2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-08-11 17:22:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/7"}}},"commit":{"parents":[{"commit":"720635b93384f91f11b19fb420ae9412e75a1c71","subject":"Update constraint for tosca-parser to 2.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/720635b93384f91f11b19fb420ae9412e75a1c71"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-11 17:22:17.000000000","tz":0},"subject":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db to 13.x since this is the first version with full\nSQLAlchemy 2.0 support and bump alembic since recent versions also have\nbetter compatibility with SQLAlchemy 2.0.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68346955b43cd4f246d866e2ed7fcd4080171ed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68346955b43cd4f246d866e2ed7fcd4080171ed2"}]},"branch":"refs/heads/master","description":"Rebase"},"21cd139038ff137ff53355c96ee55b914124a544":{"kind":"REWORK","_number":8,"created":"2023-08-28 09:44:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/8"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-08-28 09:43:55.000000000","tz":60},"subject":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db to 13.x since this is the first version with full\nSQLAlchemy 2.0 support and bump alembic since recent versions also have\nbetter compatibility with SQLAlchemy 2.0.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21cd139038ff137ff53355c96ee55b914124a544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21cd139038ff137ff53355c96ee55b914124a544"}]},"branch":"refs/heads/master"},"b4be7451d3afd6955754b50b9c0512398e1c85f3":{"kind":"REWORK","_number":9,"created":"2023-09-13 13:45:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/9"}}},"commit":{"parents":[{"commit":"8cf4e053498c7e17752e482cd164c47306424175","subject":"Merge \"update constraint for python-saharaclient to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cf4e053498c7e17752e482cd164c47306424175"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-13 13:45:06.000000000","tz":60},"subject":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db since 13.0.0 was the first version with full\nSQLAlchemy 2.0 support and bump alembic since recent versions also have\nbetter compatibility with SQLAlchemy 2.0.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4be7451d3afd6955754b50b9c0512398e1c85f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4be7451d3afd6955754b50b9c0512398e1c85f3"}]},"branch":"refs/heads/master"},"d5575b2e6ab525e89e6879ddfb873c0cfb1af0bd":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-09-22 13:27:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/10"}}},"commit":{"parents":[{"commit":"f63b576e9f6cafe93aeae00a8a4627a9083cb797","subject":"update constraint for virtualbmc to new release 3.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f63b576e9f6cafe93aeae00a8a4627a9083cb797"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-22 13:27:04.000000000","tz":0},"subject":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db since 13.0.0 was the first version with full\nSQLAlchemy 2.0 support and bump alembic since recent versions also have\nbetter compatibility with SQLAlchemy 2.0.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5575b2e6ab525e89e6879ddfb873c0cfb1af0bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5575b2e6ab525e89e6879ddfb873c0cfb1af0bd"}]},"branch":"refs/heads/master","description":"Rebase"},"0f9bd4c3b76e1fbc412675b23d6a855540b9e2c2":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2023-09-22 13:30:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/11"}}},"commit":{"parents":[{"commit":"f63b576e9f6cafe93aeae00a8a4627a9083cb797","subject":"update constraint for virtualbmc to new release 3.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f63b576e9f6cafe93aeae00a8a4627a9083cb797"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-22 13:30:47.000000000","tz":0},"subject":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate","message":"Update SQLAlchemy, alembic and oslo.db, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nWe also bump oslo.db since 13.0.0 was the first version with full\nSQLAlchemy 2.0 support and bump alembic since recent versions also have\nbetter compatibility with SQLAlchemy 2.0.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/896097/\nDepends-On: https://review.opendev.org/c/openstack/heat/+/895931/\nDepends-On: https://review.opendev.org/c/openstack/masakari/+/888319/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f9bd4c3b76e1fbc412675b23d6a855540b9e2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f9bd4c3b76e1fbc412675b23d6a855540b9e2c2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c9e3844f2e8d57774e787f97a979c1a51459466e":{"kind":"REWORK","_number":12,"created":"2023-09-27 10:37:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/12"}}},"commit":{"parents":[{"commit":"a3b055de9631c3149f721eebdd74740be06f5b75","subject":"Merge \"update constraint for oslo.db to new release 14.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3b055de9631c3149f721eebdd74740be06f5b75"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-27 10:36:00.000000000","tz":60},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/896097/\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/886152/\nDepends-On: https://review.opendev.org/c/openstack/heat/+/895931/\nDepends-On: https://review.opendev.org/c/openstack/masakari/+/888319/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9e3844f2e8d57774e787f97a979c1a51459466e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9e3844f2e8d57774e787f97a979c1a51459466e"}]},"branch":"refs/heads/master"},"d2fb14dd054aadfbbed5753c4a079819b21b36d6":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2023-09-27 12:37:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/13"}}},"commit":{"parents":[{"commit":"a3b055de9631c3149f721eebdd74740be06f5b75","subject":"Merge \"update constraint for oslo.db to new release 14.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3b055de9631c3149f721eebdd74740be06f5b75"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-27 12:37:31.000000000","tz":0},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/896097/\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/886152/\nDepends-On: https://review.opendev.org/c/openstack/heat/+/895931/\nDepends-On: https://review.opendev.org/c/openstack/masakari/+/888319/\nDepends-On: https://review.opendev.org/c/openstack/placement/+/886229/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2fb14dd054aadfbbed5753c4a079819b21b36d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2fb14dd054aadfbbed5753c4a079819b21b36d6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8023649438fde1f4f5bf3ffbc0bdfc74df50315c":{"kind":"REWORK","_number":14,"created":"2023-11-06 12:40:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/14"}}},"commit":{"parents":[{"commit":"d2f135f2a9a9e57ef5d95989e12e9fd986bba441","subject":"Add python-observabilityclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2f135f2a9a9e57ef5d95989e12e9fd986bba441"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-06 12:40:40.000000000","tz":0},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/896097/\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/886152/\nDepends-On: https://review.opendev.org/c/openstack/heat/+/895931/\nDepends-On: https://review.opendev.org/c/openstack/masakari/+/888319/\nDepends-On: https://review.opendev.org/c/openstack/placement/+/886229/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8023649438fde1f4f5bf3ffbc0bdfc74df50315c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8023649438fde1f4f5bf3ffbc0bdfc74df50315c"}]},"branch":"refs/heads/master"},"c842311607448b0f446cb15199fef20beedbf873":{"kind":"REWORK","_number":15,"created":"2024-01-31 06:26:07.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/43/879743/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/15"}}},"commit":{"parents":[{"commit":"b88debab9d2c455f154ea0fa5d5845be689c7a27","subject":"Force use of 0.35.1 eventlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b88debab9d2c455f154ea0fa5d5845be689c7a27"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-01-31 06:25:37.000000000","tz":60},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/896097/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c842311607448b0f446cb15199fef20beedbf873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c842311607448b0f446cb15199fef20beedbf873"}]},"branch":"refs/heads/master"},"fbabd599e3f1d23f8008edbdf88fbe95796941ae":{"kind":"REWORK","_number":16,"created":"2024-02-18 09:12:04.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/43/879743/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/16"}}},"commit":{"parents":[{"commit":"1d8a9890961c44ac9f1b991db61b87baa653293d","subject":"Remove unused abclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a9890961c44ac9f1b991db61b87baa653293d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-02-18 09:11:53.000000000","tz":60},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/896097/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbabd599e3f1d23f8008edbdf88fbe95796941ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbabd599e3f1d23f8008edbdf88fbe95796941ae"}]},"branch":"refs/heads/master"},"7f6522b8d58e0fece8b6efe1a51aa35a8fcccb12":{"kind":"REWORK","_number":17,"created":"2024-03-01 07:17:36.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/43/879743/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/17"}}},"commit":{"parents":[{"commit":"07688cf6d9e5c0dd70b61e4e0e2a91f89fcacf26","subject":"Merge \"update constraint for python-octaviaclient to new release 3.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07688cf6d9e5c0dd70b61e4e0e2a91f89fcacf26"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-03-01 07:17:00.000000000","tz":60},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f6522b8d58e0fece8b6efe1a51aa35a8fcccb12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f6522b8d58e0fece8b6efe1a51aa35a8fcccb12"}]},"branch":"refs/heads/master"},"a7f6ed0b8afae9d0422715fe432cb13b2868c099":{"kind":"REWORK","_number":18,"created":"2024-03-28 18:59:04.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/43/879743/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/18"}}},"commit":{"parents":[{"commit":"93a3e51432519a14a378a729554b5273675e945b","subject":"Merge \"update constraint for oslo.messaging to new release 14.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93a3e51432519a14a378a729554b5273675e945b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-28 18:58:25.000000000","tz":-240},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7f6ed0b8afae9d0422715fe432cb13b2868c099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7f6ed0b8afae9d0422715fe432cb13b2868c099"}]},"branch":"refs/heads/master"},"1fa7310c079bdfbf7aaf8e9d561adcd36c91620e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":19,"created":"2024-04-05 14:31:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/879743/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/19"}}},"commit":{"parents":[{"commit":"b5b63fb81989d23346145a0b1e33db71619fea94","subject":"Partial Revert \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5b63fb81989d23346145a0b1e33db71619fea94"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-05 14:31:13.000000000","tz":60},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/888318/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fa7310c079bdfbf7aaf8e9d561adcd36c91620e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fa7310c079bdfbf7aaf8e9d561adcd36c91620e"}]},"branch":"refs/heads/master"},"a6a440c65d19ca57248e8d2f2f6c5f9b9e35cd23":{"kind":"REWORK","_number":20,"created":"2024-04-15 07:59:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/879743/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/20"}}},"commit":{"parents":[{"commit":"0e0061bc8bd303892836c0ecfe60e473e405f092","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e0061bc8bd303892836c0ecfe60e473e405f092"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 07:59:03.000000000","tz":540},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/888318/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a6a440c65d19ca57248e8d2f2f6c5f9b9e35cd23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a6a440c65d19ca57248e8d2f2f6c5f9b9e35cd23"}]},"branch":"refs/heads/master"},"92c8b4c52b3160df1db1768e8891fc19eb704424":{"kind":"REWORK","_number":21,"created":"2024-04-18 18:50:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/879743/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/879743/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/879743/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/879743/21"}}},"commit":{"parents":[{"commit":"18098b9abacbd8d7257bebc1b302294f634441ab","subject":"update constraint for taskflow to new release 5.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18098b9abacbd8d7257bebc1b302294f634441ab"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-13 17:42:08.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-18 18:50:26.000000000","tz":540},"subject":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate","message":"Update SQLAlchemy and alembic, drop sqlalchemy-migrate\n\nBump SQLAlchemy to the 2.x series. sqlalchemy-migrate is not compatible\nwith SQLAlchemy 2.x so it is removed: any projects consuming upper\nconstraints will need to remove sqlalchemy-migrate integration.\n\nChange-Id: I23af86739e699bdcb42e1b112861497d35edc42e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/888318/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92c8b4c52b3160df1db1768e8891fc19eb704424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92c8b4c52b3160df1db1768e8891fc19eb704424"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
