)]}'
{"id":"openstack%2Frequirements~879915","triplet_id":"openstack%2Frequirements~master~Idd1a82ee50b698c21a070c9c82a51bb863bc8ed0","project":"openstack/requirements","branch":"master","topic":"constraints/noclobber","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-04-25 01:36:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd1a82ee50b698c21a070c9c82a51bb863bc8ed0","subject":"Updated from generate-constraints","status":"MERGED","created":"2023-04-08 08:21:02.000000000","updated":"2023-04-25 10:47:46.000000000","submitted":"2023-04-25 01:36:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"879915-constraints/noclobber","meta_rev_id":"897a933964881542cdf06c95873c696b7d60effd","_number":879915,"virtual_id_number":879915,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-25 01:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-04-25 00:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-04-25 00:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-08 09:25:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-25 00:21:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-04-25 10:47:46.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"ff01253118c91401435235adf904ba87b7ef1e19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-04-08 08:21:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38b5cc2d2b512a39fe4b5f28716d2754839741a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-08 09:25:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bace775c1434468faf7e8cd236aadcce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/200855273e7640e1b73047abf6ff2355 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d581853933f64d5ea683d4d45057d37c : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fc96542342b484a86c424407a60fa9c : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a59e5c98a3348cfa8106e91b4b4973f : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/591b8ef1fa31441ea1a18c139b929469 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c43144db766f4ac3ab39773c73ed74ee : SUCCESS in 3m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2f4eb448e76649ed82cb50ebd4014c4f : SUCCESS in 12m 02s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3627ea2c4bea4d488eb41e8b02a94b43 : SUCCESS in 9m 31s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/53cf7429e1c94447b9eda832c8ac52aa : SUCCESS in 7m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ec476e5b81ba478ba4006193c8e92045 : SUCCESS in 4m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4176d8426f654e9cad1d76fbfbb05e9c : SUCCESS in 5m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/50f1b0d0928246d4aa3bdef982a34171 : FAILURE in 11m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4d4fd961b14946c3b543401722781479 : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8dd62e8c0efc4c4c9d9aad6aa5cbb96f : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4a5d0ee59dd24b31918ca11b3ffd6fa4 : SUCCESS in 7m 58s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5772721ea9f146b5ab765643b5574424 : SUCCESS in 7m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7fbed8fff1f45e2aecf8273f3f5b1ec : SUCCESS in 7m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a324de976aab44eebfd6cac1764d23b2 : FAILURE in 7m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a4126a0f56334ca2a25cd35d123606a3 : FAILURE in 13m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4d3f31d14f59493abd3f0abff7f8c9d1 : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/74fa2d0afb464d0e90f608e11da21eb8 : SUCCESS in 12m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/03aeb51e9aba451fb4a0b07b9f3fa834 : SUCCESS in 5m 36s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/94f45e7b758e45c38691128639f9ac39 : SUCCESS in 9m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/046d173dee994dd6b0a4105cc84d5547 : SUCCESS in 19m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7e905a50db774cebafc374c7933a714b : SUCCESS in 11m 30s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4ad94d24f5744a17908a1b575862b1f0 : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/74e8c03552254c63b63c3af8095ca1f8 : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6d69c50ad464b9392770374942c4af1 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/428db546524c4b8590cf613f6cadad38 : SUCCESS in 26m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ec02cef0007247dfb79092e67543dc85 : SUCCESS in 4m 47s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f202b19161e742a5be2fe36c4c1023c7 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/97f86184da484318b208d3f147cdfab6 : SUCCESS in 6m 14s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f51328b6df8a476c98a3dd3bd415f659 : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95d71a2bd8c5469b976c34eff2abe913 : SUCCESS in 59m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f8bb283f49045e0a9980a39a89e1992 : FAILURE in 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ce225eaee9d89909776c3ee673bbbc86157e0a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-04-15 08:36:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a9cb0ed9c8444cecc6fcf917f5c9f0591c301587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-15 09:24:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/038385ad3bdf4249bcb46c8821f146b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d577965f9bb34f33ace20c01b2f3afe1 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae33d2d8c4534656a9c72602ddf835de : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44924e6f444f48e1a5f68fa58e1dfd3f : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3096e85a5f7e407c956b9e62f7697aab : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e78a6f18588a45ee988fba7b0ee03aab : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd2c5b5e695e4f85a15e4aec329002ff : SUCCESS in 2m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c00bcfcbf3446199d5845bdb9d84363 : SUCCESS in 11m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/523bdb3b330e4983803956aee321a5bf : SUCCESS in 6m 58s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/8aca2c74f2424c728f481aee50809bc5 : SUCCESS in 5m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/07b6d36387d24d07af21d3a344a5b0c4 : SUCCESS in 3m 57s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6dc40d7a71284802adab335583cb8973 : SUCCESS in 4m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fc6ca41643f7490e8afc71dcb6263b9e : FAILURE in 11m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fbc8010b68ba4996afc950c7da7123bb : SUCCESS in 3m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1887f3ea097f4c468336d868f932a18a : SUCCESS in 4m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7de4b51d462f46519ff9bde2a90f1928 : SUCCESS in 8m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/acf40b6e804b4711b3aa6160d3e41360 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50207aff20f14e6e97a84b444d168d2b : SUCCESS in 8m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5099eadad3d84ceea5a10ad8d1e0bbcb : SUCCESS in 10m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/28640cd8ef3f4388a205d3b8ab179705 : FAILURE in 9m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2b3672ab4368426ea1c12c1f16fa16b0 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/85c88f84cfc94e4cb11acf3c93ae1f85 : SUCCESS in 9m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/06b0d7968e1c47a8bf7f280399101203 : SUCCESS in 4m 33s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ce87ac612b4840f3b7fbd690b2db6afa : SUCCESS in 10m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/17ec75f4733e400ab2c82694a9c479bf : SUCCESS in 25m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/32555f26db12419da5c4565cd4abe1ce : SUCCESS in 8m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/94e4418221244fcb99da98c131167dc8 : SUCCESS in 14m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7eca2ebdb42049a0a067efc79ffd91bb : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a9019276deb404cb9aad0c91c319ae5 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed9d9a0185ba4c3ab99668daf92355aa : SUCCESS in 18m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c8d03961387846e09bae199ba1b27d7f : SUCCESS in 6m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6c413cc066fc4c6ebaf3ee2fb5e71e4c : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/438782637ecf4ce5b34c93bbe4596995 : SUCCESS in 7m 03s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/11c741bb324147fc8abc3900b2ba8067 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8233a1ec53674fb89f7a7882552ef627 : SUCCESS in 39m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b07df4a07e054fe89e138c5bd3a348f4 : FAILURE in 31m 10s","accounts_in_message":[],"_revision_number":2},{"id":"ec4752562e4801d9b138e1f541ea465b84771226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-04-22 08:13:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2aace1869d92c1cffe07264f31fd19ac8c2c0e78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-22 08:44:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a3914b29ace43348052e4cb470ba142\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6f2c37880c48e1b11ff0b4d77cd284 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8faeee95d8e407992e4225d8522042a : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/839ce77a4b824c5385743aac53bd54ed : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d31af89c68d348059eace417e4732cf4 : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/740eb1acda3f4404aaa5241157949fd8 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66a07431fdd34ae3a334142172839472 : SUCCESS in 3m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/14047113a56a4900ad69548bdd969bf9 : SUCCESS in 15m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/591ce97f4ae04fe9af7c146c7c64c8f1 : SUCCESS in 9m 15s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/586302546e294ec58ae05a6fe9f515cb : FAILURE in 4m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5f26e93daa3f4214a445dbcec1c6e090 : FAILURE in 5m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/549749f592724c94840adf3371ab13fb : SUCCESS in 8m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0832834ead5a427ebbcbf3b389262dcb : FAILURE in 5m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/db8ef9795fde43bc8b7a236dea4050da : SUCCESS in 4m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/238b20e307ac419faa913b55feb3eb73 : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/65ba02a4449d4c56b5d6958759775ef1 : FAILURE in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3504bc6420d340eba979c0e4b06f18aa : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d55d764d37e044d2989d0ae2e74416e9 : SUCCESS in 8m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2ca47e1ca68e4aff8b6e41aff319eb70 : FAILURE in 9m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d7c3fe7e63694d0ea2b374c75203c56c : FAILURE in 4m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/50aea92174fe45f3b940b513184cdb19 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/16e87a2702884f59862ce18ae74c3bd5 : FAILURE in 3m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9ac50d1d305645d9a77fc14573d602fb : FAILURE in 4m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7fc65767a84247e0aebd7f79b61f2a02 : SUCCESS in 9m 44s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3e5ca0730846440686626aedde9d918e : FAILURE in 23m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/92a0535401d34770bfb8ca607b91f340 : SUCCESS in 11m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e68fdb3844c1487894a386bdcd0e2825 : FAILURE in 9m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5ad49c3730d04496b3d7958c6c39460e : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9bc1d910018f4cf19bcc22c8afa06782 : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c35a9f200cb4eeeb28aaf4c95a58575 : FAILURE in 6m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/62b44b2ccc2d48f0a9cab82cd1b2bfc8 : FAILURE in 7m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/35cd6eb6a46646dfbba1ba9cf0198106 : FAILURE in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/359fb55829eb4d76b76ddf22f2d4cd7b : SUCCESS in 8m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d5f558310d4c4d5e9c476a54735694e4 : SUCCESS in 9m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/131f7ab090ab4ef994e605035d846219 : POST_FAILURE in 14m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3278931b3ae74fdbb8f4e3d0e03b9bbf : FAILURE in 10m 21s","accounts_in_message":[],"_revision_number":3},{"id":"ac13b4dbc53671d8382dc6a1e29b34f34685af89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-22 12:19:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7e9315bd1bfb1c05f2c86103a702d6828711a751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-22 13:28:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71ff9ce0a85f477199147159ccfcd7af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78aff024626b46f8907ade4e93fcad71 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4e46dc640e45e2ae0e334b5f8ff2a6 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5446496867e94aa68e6aacf573d95344 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b68091801bc4224839bd67fbdba50b0 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2aa0c1276ae43f7b4a628aa101a1314 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdfd8bf5a43340c5b3e772e04b81fd71 : SUCCESS in 3m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/35708d30e6a54145a3b2d4d08c69969a : SUCCESS in 12m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b07540615a62476cbc3cd74d883f35b6 : SUCCESS in 7m 17s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c38ce649576f4f948e96dd98ceef6276 : FAILURE in 5m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0481c2f45c8d44fdb2397fb97182db89 : SUCCESS in 4m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/82b09b6b3dce4e8885072d272a288dec : SUCCESS in 5m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3433f5b040ea41c59e76706c55ee9410 : FAILURE in 4m 49s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fdfcb460e17748eda3c2bcb646a2a589 : SUCCESS in 4m 01s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8ba8f8c5f2214d5f8fb8637062bef18e : SUCCESS in 4m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c68e0c7a01bd49de8b60a92192fe6b3e : FAILURE in 7m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e7fa853ee1934d42adbe66cfe3911fb0 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5bee49e584f74b94a244723f7137c6b2 : SUCCESS in 7m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8c44b942eb1747ab851df5e627cb8507 : SUCCESS in 8m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2fc5d467cdf041f78b6560a8a229fdfa : SUCCESS in 9m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0288de53271d466baae09cf36d803811 : SUCCESS in 3m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/eb916f8597af42568905a2d45499cb8b : FAILURE in 4m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cf87db32cdd04e84995a9d4c27e820ca : FAILURE in 3m 50s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/53dcd96bf8f24950993ed5919278dff9 : SUCCESS in 8m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1f2021dad8a94eb3aef5ba92bee821e4 : SUCCESS in 19m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/af7c476cd0784f31b50d459e44c9a840 : SUCCESS in 9m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4b22d99cfd2f44ee963a7e52bf9368b9 : SUCCESS in 11m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f21ea69ea7424c6095ca2ed996b632df : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0285a2a780984232848f7353a7d09bbf : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9803ae94ad14437fb70dea249ff8ac01 : SUCCESS in 22m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b2fbe15f8c514b4191429401329d5a51 : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c0ff0b96b4f9425fafe638e2a00afabd : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/981badd5e91a43dc8b97e1326a81b6bc : SUCCESS in 6m 25s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5a316df135bf4f90a5ca5ed847d4077f : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8becdafd3ea24b40a1ec1ea6d3ee032f : FAILURE in 1h 01m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8284e96e0b9746e5b45767e4776c614d : FAILURE in 29m 52s","accounts_in_message":[],"_revision_number":4},{"id":"59937013e1a58b25ccd3095b276d9edb848d5257","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-04-22 19:12:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2d6c2f9380aa81ebd0bedc2d89796564af990efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-22 20:21:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5348d6e828f7406aa35a167cf1522576\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26a6f76685134b53b9400990e025e37f : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1353efe8722e467da5af0a1412c2a127 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f21414fc66364fc9b1c792ab4add87c1 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71dec869f87b4529bd6825a4be3967b7 : SUCCESS in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/deabf2860e9c4442ab508f9f6422cba6 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b77aec885a7b4439bd01aa2b63556a4d : SUCCESS in 2m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a15cb718a89348ada47c8de703c4e22f : SUCCESS in 10m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/06db6e49ad2348fc98f76c72f63a08e5 : SUCCESS in 7m 32s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b9a946fc7b644a65b0ba510668ba4d55 : SUCCESS in 6m 32s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/94d2830ad3064be585cb84023aab1afd : SUCCESS in 4m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9672cf4a37f74a0894e064ac1596e66c : SUCCESS in 5m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8365df0cd48e44a88b1636fc6813c457 : SUCCESS in 11m 54s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0acafd3b42574a1aada58e7eefbff285 : SUCCESS in 4m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/68cc7cd36a0a4db998123e7fb296f200 : SUCCESS in 4m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/415a2e3b81a149838a58188de528711a : SUCCESS in 6m 23s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/55e6a6f0d49c4918aa255d107d5f1075 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/859f8c1bb5ad4dc39d76d120d1087038 : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/95e67aaaf00e42948c6a40b567c4a898 : SUCCESS in 6m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7143e919166f48d1a1f3af9384a322f9 : SUCCESS in 11m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f7f350886cb64084a985950dc1160852 : SUCCESS in 3m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cda24f93fcf846fa9df605fba8e5bd5c : SUCCESS in 9m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e3bf3fc99ee6400b8fcfd91dd939eb81 : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/77d60f74e170419e9051ad4eb9356120 : SUCCESS in 11m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f2914c63b82e4e699f006fe4f3255872 : SUCCESS in 18m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e8fb9f54af514b65bca8fdb3419a6632 : SUCCESS in 12m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8497a9d085154a3f9873d035a5f44b4a : SUCCESS in 10m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/76edf0aa8be84a9d92749195b0ad536f : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/630fbd71d1e7479aa7e973c1c983db10 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5af5d2acf9414922843cb2046fe1d9c4 : SUCCESS in 23m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/cddf3c880ec04d92ac3c71f980662a66 : SUCCESS in 4m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d8ce3c2421884e698739269aa36c7beb : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc83a7c7384f4b4f8d9ef92675644a0d : SUCCESS in 5m 21s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7ba260ab72aa49299fb74ab3080b8728 : SUCCESS in 9m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff7474ba0d95461aa70155bbd2b561e4 : SUCCESS in 38m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1413dc684974fb1b36ce65d4f6bcefb : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":5},{"id":"f9c74e64260dfb5fa0b85cd67c5247b1f1a33f4f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-04-25 00:21:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3dcd8cdf8df122730147c1ae52142b1bae99c3fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 00:22:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bb35b4653a81f3162f29389bdf26069418ab27f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 01:36:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7db1060a5dc44f194f305ecdb2a3f96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c08d26160b4ca88a95ef1d2149058f : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c16482288f4da89a2582290fa778ce : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08a466ba3a6f405690733441eeecbe0f : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5d4d508faac43e2865abf621a232b6b : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/435009f893414243aa4f76352d6d58f6 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c51e122f68e340e3a44a67c8031f7015 : SUCCESS in 3m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f99e31018f9643ef8651b4504e3d6090 : SUCCESS in 13m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fee4b124f2e643adb6ddf3303cfb23ec : SUCCESS in 8m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f365cfeda5db48a9b180fcd68db2a3ef : SUCCESS in 13m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/36fe5ef1c93e4beaa939f833cbc0a787 : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0488b1de1aca4e29927b5e283bd0c657 : SUCCESS in 5m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/82c260acf626449bb2c20df685e44d29 : SUCCESS in 7m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3406d4651b564b3fb5f20e0073cc0525 : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80b9f6b4ccb7439eb74ffaff8a4ed111 : SUCCESS in 7m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4926f04b60794790a2bb91e29a10ed77 : SUCCESS in 6m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/272cd4e60d60430dabe7875028c25dee : SUCCESS in 10m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/79e433eb030b4da1919848ba73ccd194 : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f6cc5a58204844348d79da6314af86fd : SUCCESS in 7m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5c9cd2040cce43fe83ebb0809a427fc3 : SUCCESS in 4m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8903be51f66947048179ace994c7a91e : SUCCESS in 22m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e4351613df2a490dab3b666af6de77d2 : SUCCESS in 10m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7ef8847501ae41d783ad8d0808ba5a5f : SUCCESS in 14m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b934d40b085b4b8fa7872a1af4a8fd14 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca8d65242c7a4ced9d71859fd0f3a1ba : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d34b8c66e4a6445b88fd8f7ba6b974c7 : SUCCESS in 22m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/16b7a88cde2a4640a7964f17b8d0da7a : SUCCESS in 6m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7e21b31e187c4631a7c97ab2168201cf : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/556fdc97143740c08a8e1b982844fd3e : SUCCESS in 7m 46s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c21885ddffa847978176be2a034f854d : SUCCESS in 9m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59c81e29d45749ad95930254ee222e34 : SUCCESS in 39m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be33d76f37b947fc98fce5be75945887 : SUCCESS in 1h 13m 02s","accounts_in_message":[],"_revision_number":5},{"id":"1bfa397d146d97ba5a92bb2ddc8056a2df4bc049","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 01:36:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9ed74deccb0ded78f2e85b877b42a0f1e3c54996","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 01:37:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bd64268bd3d44b38a5c1f128341d9a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25cfeebf4ab841a4ba89fc55f46c37c3 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5},{"id":"897a933964881542cdf06c95873c696b7d60effd","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-25 10:47:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bda7d7fb77f93df3938b873b3d4b28e5594367ef","revisions":{"dec1ffc418b7fdbd0bb82e0dcb0380bf740ba9df":{"kind":"REWORK","_number":1,"created":"2023-04-08 08:21:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/879915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/879915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/879915/1"}}},"commit":{"parents":[{"commit":"b44da90728ae3db63b95c4ea6291b072b41ff2a6","subject":"update constraint for python-swiftclient to new release 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b44da90728ae3db63b95c4ea6291b072b41ff2a6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-08 08:21:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-08 08:21:01.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Idd1a82ee50b698c21a070c9c82a51bb863bc8ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dec1ffc418b7fdbd0bb82e0dcb0380bf740ba9df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dec1ffc418b7fdbd0bb82e0dcb0380bf740ba9df"}]},"branch":"refs/heads/master"},"6a82e19ed52be969cca3dc594bdada68cabbbfac":{"kind":"REWORK","_number":2,"created":"2023-04-15 08:36:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/879915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/879915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/879915/2"}}},"commit":{"parents":[{"commit":"d25ec820b127d383d61d64c4b7d9cbf0e1fae3c6","subject":"Merge \"Remove the Patrole deliverables from req\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d25ec820b127d383d61d64c4b7d9cbf0e1fae3c6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-15 08:36:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-15 08:36:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Idd1a82ee50b698c21a070c9c82a51bb863bc8ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a82e19ed52be969cca3dc594bdada68cabbbfac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a82e19ed52be969cca3dc594bdada68cabbbfac"}]},"branch":"refs/heads/master"},"e5ea8787eb2e2c725ca5f5f4dae0c541182e3e8c":{"kind":"REWORK","_number":3,"created":"2023-04-22 08:13:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/879915/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/879915/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/879915/3"}}},"commit":{"parents":[{"commit":"8a8b39afed8dcbb732eafd5c8ad8314620c1ed6c","subject":"Merge \"update constraint for tooz to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a8b39afed8dcbb732eafd5c8ad8314620c1ed6c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-22 08:13:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-22 08:13:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Idd1a82ee50b698c21a070c9c82a51bb863bc8ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5ea8787eb2e2c725ca5f5f4dae0c541182e3e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5ea8787eb2e2c725ca5f5f4dae0c541182e3e8c"}]},"branch":"refs/heads/master"},"ebc4e5c3aa0980ac62ffe886aa7de9765f7cfffd":{"kind":"REWORK","_number":4,"created":"2023-04-22 12:19:05.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/15/879915/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/879915/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/879915/4"}}},"commit":{"parents":[{"commit":"8a8b39afed8dcbb732eafd5c8ad8314620c1ed6c","subject":"Merge \"update constraint for tooz to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a8b39afed8dcbb732eafd5c8ad8314620c1ed6c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-22 08:13:10.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-22 12:18:16.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back:\n\n    SQLA 2\n    alembic\n\nChange-Id: Idd1a82ee50b698c21a070c9c82a51bb863bc8ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebc4e5c3aa0980ac62ffe886aa7de9765f7cfffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebc4e5c3aa0980ac62ffe886aa7de9765f7cfffd"}]},"branch":"refs/heads/master"},"bda7d7fb77f93df3938b873b3d4b28e5594367ef":{"kind":"REWORK","_number":5,"created":"2023-04-22 19:12:42.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/15/879915/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/879915/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/879915/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/879915/5"}}},"commit":{"parents":[{"commit":"8a8b39afed8dcbb732eafd5c8ad8314620c1ed6c","subject":"Merge \"update constraint for tooz to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a8b39afed8dcbb732eafd5c8ad8314620c1ed6c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-22 08:13:10.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-04-22 19:11:44.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back:\n\n    SQLA 2\n    alembic\n    XStatic-Font-Awesome\n    oslo.db\n\nChange-Id: Idd1a82ee50b698c21a070c9c82a51bb863bc8ed0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bda7d7fb77f93df3938b873b3d4b28e5594367ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bda7d7fb77f93df3938b873b3d4b28e5594367ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
