)]}'
{"id":"openstack%2Frequirements~880280","triplet_id":"openstack%2Frequirements~stable%2Fxena~I0342d60f4a102748758e9869f8e93bd669556d0e","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I0342d60f4a102748758e9869f8e93bd669556d0e","subject":"update constraint for python-manilaclient to new release 3.0.3","status":"MERGED","created":"2023-04-13 10:29:21.000000000","updated":"2023-04-13 15:48:36.000000000","submitted":"2023-04-13 15:48:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880280-new-release","meta_rev_id":"0f98942027a211c233b1e9f23e0ed85e716a2945","_number":880280,"virtual_id_number":880280,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 15:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-04-13 14:22:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-04-13 15:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-04-13 14:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 10:40:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-13 11:52:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 14:22:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-04-13 15:17:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"1605a531f7e834b7c527d230fd63c9c768e4280a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-04-13 10:29:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb7ab08f311fbe127c0a51c8051c88941143a3d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 10:40:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3118d88715d14fa3b5d1856f99c513f4\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7d6119ebb66b448f8d76248f3180e206 : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e3dd3fceaab3a023e1a212b5899c7082ffe30cac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 11:52:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec0753468a5d4c398af5045e0a1fd03b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab27406ee49c4a79bd5d168d1c0df8e8 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daf4aaff1fe24a4daf84a28fdc25e8e4 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e91b2eb997634c488a3d9d0a33b704c0 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/900fe25019b549eb843d146fb91e8d5c : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b523e60c357d42db81b716f027190bd2 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c40d00b40a7f4772a621ee228479764a : SUCCESS in 2m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e382f63fdb9d463aa11e3748617eba6f : SUCCESS in 15m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/30f80baead094e0ba3d14b47fa6fb634 : SUCCESS in 8m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/53201d113bbb49899645f8b7aac9ca0d : SUCCESS in 5m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/db0e27f8072e44649a40ca1b069c1145 : SUCCESS in 7m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/438abe076cc94d899dd5c59660920d1b : SUCCESS in 4m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/302162f6c7ce46fcb9c87f3a142a1ca2 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4bdf1092c16242d3a688d830d451b4b7 : SUCCESS in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f3f11c76103f4562a5a49121441a2280 : SUCCESS in 7m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4358335c7c844e77927c5fdd3f94f6f0 : SUCCESS in 13m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4e6d4855652047358c53af6d1ca89945 : SUCCESS in 9m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/29e6d5832d6e4960aebdaa9cf79530ae : SUCCESS in 4m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c64328fc75024cbbb3aa26e1471149b8 : SUCCESS in 24m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/76bec7ca0edd40e0aed074e76a2677de : SUCCESS in 10m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7990850f8fca4101bb1e9012afc5dd20 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0d7f190cac0453584b88fd4f70e17de : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16fc1e1280304ae7b1f6e300e471e478 : SUCCESS in 19m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7eb5526b40634f93be9bca6b5ee38f2e : SUCCESS in 5m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a3c2594c6dc741888242f16bd95f34a5 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40619c1f14994342baa5be5c85994547 : SUCCESS in 6m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3ae6fa03a156474d831ba7a109221cbc : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/791a0154cf054602ba6f9bee5648466f : SUCCESS in 56m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ada0b8fd9d604e7da6eba0e30fca2963 : SUCCESS in 1h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"239f6ff6572179b59e6840c3a90419fe81321c46","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-04-13 14:22:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f259e87fb9c68ba8729c192c1f4dc1437436a218","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 14:23:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ede80cdd3faec7d1b45622d2ceb5efb65a4f7ec3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-13 15:17:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"16fb2bee543d4566661a490785ced2e8b429f95a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 15:48:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/562afd3683bf43a0ae34146e0523b459\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4577726a8b3740298bbd8905b0dc4391 : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e81fc50a9494486bae2b4dc3b3e87ee : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6ae0e4c8cfc4b4bb1042f10fa93cda9 : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4577b43fd9e34a3199a06bc9664286de : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f153d173bfa4088b483778699e4f1a7 : SUCCESS in 3m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de21618fbcf6461d89c9475e3b4d97a5 : SUCCESS in 13m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/371d4f6ce7e94b7f9ae86ab05521f94d : SUCCESS in 9m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a0c19688fd4a4e5a91bb61ba68328d4d : SUCCESS in 10m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c7c54fb0cd704365880b2df85aad2954 : SUCCESS in 5m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3c3f0de8bbab453b9609129f0ec149ed : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d8f2bd5f68f4bceb53cbc8f439bf654 : SUCCESS in 7m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/98333b8406ff43d58064e2e149aa0307 : SUCCESS in 7m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/05f3e8000b1c4d82aa91675df23bc7dd : SUCCESS in 14m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/28daac70d79d4ae899f5ed8bcdfb5a0a : SUCCESS in 8m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/67ff8125105c42078220a58e71ce7fe3 : SUCCESS in 4m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ecb155f5ff47415a8f0f8ce2017ccb4d : SUCCESS in 22m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f8c521be8e0a40b7a776adc36d5352dc : SUCCESS in 13m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/22bec37b33dd46a7afcbb25e55fdcb32 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8475d5fdeada4e4cb8f14081ed1a0b46 : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5bba5cfac2d4481e9b6b3109e8a7708f : SUCCESS in 17m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2ee68d5ba9d24576b8583e3f5039cd1a : SUCCESS in 5m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/39a8e423e27948519b8b1a6b1c63cbf4 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3eb1b8e6c71f4630bfdf3473651d7098 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5503474a41df45c6be9743c0c66c739c : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d32d54c312d94d2f82001c1d3c226d34 : SUCCESS in 46m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/649610415924484ab9e3681f13506994 : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0f98942027a211c233b1e9f23e0ed85e716a2945","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 15:48:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"970b9d648dfa66baf46907be9f1d1757c7fc8ede","revisions":{"970b9d648dfa66baf46907be9f1d1757c7fc8ede":{"kind":"REWORK","_number":1,"created":"2023-04-13 10:29:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/880280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/880280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/880280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/880280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/880280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/880280/1"}}},"commit":{"parents":[{"commit":"45e3947967109cc945fbde7bb149ff701f7bb871","subject":"update constraint for neutron to new release 19.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45e3947967109cc945fbde7bb149ff701f7bb871"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-13 10:29:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-13 10:29:19.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 3.0.3","message":"update constraint for python-manilaclient to new release 3.0.3\n\nmeta: version: 3.0.3\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Commit: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Change-Id: I9f9a6d2261c70ef416ddf02a9ff3767890b9960b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I0342d60f4a102748758e9869f8e93bd669556d0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/970b9d648dfa66baf46907be9f1d1757c7fc8ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/970b9d648dfa66baf46907be9f1d1757c7fc8ede"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
