)]}'
{"id":"openstack%2Frequirements~880283","triplet_id":"openstack%2Frequirements~stable%2Fxena~Iab448b53fa629e92158bc0977dd28a0a2c966330","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"Iab448b53fa629e92158bc0977dd28a0a2c966330","subject":"update constraint for horizon to new release 20.1.4","status":"MERGED","created":"2023-04-13 11:34:38.000000000","updated":"2023-04-13 16:06:54.000000000","submitted":"2023-04-13 16:06:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880283-new-release","meta_rev_id":"e87b675778a539a7faffb0276348df1d927d7b62","_number":880283,"virtual_id_number":880283,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 16:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-04-13 14:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-04-13 15:17:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-04-13 14:22:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 11:43:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-13 12:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 14:22:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-04-13 15:17:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"2e604995f5c548fb48cc89d828bd864c663098d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-04-13 11:34:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23d8bc3ff50df0e9bd4b602145ab228f869d686b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 11:43:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc2d8ba49c2442f9901254335710a5a1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b17e8c3faaa54d2cb8c1d01123944ee2 : FAILURE in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"31e1c276846145080305964eebcf8817bca37fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 12:43:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/517817111bd3413494f08b8d566d6804\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9dd31afab3348d79de69406269a2064 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9f8f4d38e6348b99609f0e5fc3a2026 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2ad728c3c794e11b37778d3ea31f0f0 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2399f554806141d7986a60cfe513939c : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80a8b973db6f4c029e61503fc414c4a9 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d64b4008b9942b3b526fb8b70313219 : SUCCESS in 2m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5fbaedaa073e4748b6883bcc4b6e0f06 : SUCCESS in 14m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ee5bf507ace47fdb9793bfa35e57aed : SUCCESS in 8m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2adc00439a6d453b9e6195ddf44d4254 : SUCCESS in 4m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6528124561ca4304afc13b8688562539 : SUCCESS in 8m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5fd2db5e094043dcb55e2aea7b9abb34 : SUCCESS in 5m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/84be3482a4104edca604e55760a82484 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/425e4a795a8848b48e1500ef97e66527 : SUCCESS in 6m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ff27506e7b1349fcb698832f922bd146 : SUCCESS in 7m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/328612da79574109a646f3e0c26517af : SUCCESS in 14m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/11e21dd92848439982c2112c5b3798a3 : SUCCESS in 9m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d7afca5f29ab48f481aac0542ac141e9 : SUCCESS in 3m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/774b830a21e547168dfba973ba98a708 : SUCCESS in 24m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/038681549c9342769ce931a5328314a7 : SUCCESS in 12m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/42f166b6ae4742719780d9619c0e17ba : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aad4fa4b4d354976a12bd830136d6820 : SUCCESS in 2m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/521fc7c1d1c841de997532d4fd6a9522 : SUCCESS in 20m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e2711989477246458fb7ae475b4505df : SUCCESS in 4m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f38eccd63d2a4e7880db03e50d30aaa8 : SUCCESS in 3m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/463e9a494da14daea31891c7b6b50823 : SUCCESS in 5m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9107d68bd3a34e2aafae09d493966b20 : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee53cd6f24764f1e8a853748e436b680 : SUCCESS in 56m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/993b13694e844e7295f8e1e0f4cf06f7 : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":1},{"id":"de42e8028ca0602b0a0767a88cd51002b52c4fca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-04-13 14:22:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a86e266b5f8e2536f436c66f93333b388fbf6c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 14:23:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0bf936a235f75fb01bc507716b73d02c1cfadaa7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-13 15:17:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c7cbe4493acb0eddf73502e0009a050a779f230f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:06:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f8023b42a2547efaf6d16f1cffc6003\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7084f70662d94e22bfcef33b9a350062 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c18b9602b00a4aefbfff86c339ed2ba0 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/570982de51cb4d768551ae4c512a2c83 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80980a9ca73f481a90b387ab8f0ca5a4 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f7f8de174654ee98c30e03a6b21e8ff : SUCCESS in 2m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/49ee2b0e29e84364853b39a295a53edd : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99a06eb70d3e4f17a459791399e773e4 : SUCCESS in 8m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e64410f4ef314387b21c298d388a2334 : SUCCESS in 8m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a145716c70e24516b84e82e7e1901ef2 : SUCCESS in 5m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/72e549ad0a694b6487cee226ca868716 : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8312538538b4eeda56caa16c7ddbf54 : SUCCESS in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c4a8d50476a7439fbb8fbfc18f3dd667 : SUCCESS in 9m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fb6e0f9b7ece428ab67e0a23147f8cd3 : SUCCESS in 12m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/db6d51902bff4555af2c7d7adbecdc09 : SUCCESS in 10m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b40062a2e245431c86413da277ed5402 : SUCCESS in 4m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/582723c6724a4416a388447373e2f005 : SUCCESS in 23m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/009a115548534727b885e5f97e671699 : SUCCESS in 11m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/018c5b01166b49f2a6811ef83e2b0134 : SUCCESS in 5m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fee25ce3b33f4aa3a6bb956e947c2f16 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93b89eeb9e5a4289add297d52451ac28 : SUCCESS in 20m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/33e2d4110d1349c6be082f35a6d368ae : SUCCESS in 5m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d316c3c981d04a0c9c14762d5c81b143 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/282c84e8624f41549202def19ec12b02 : SUCCESS in 5m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/adbff776320a4d46a6987664e7fd8117 : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75c34ecbeb7c4e5293b1decfa21523d0 : SUCCESS in 52m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff6abf175ae744169078d6dd21ff43a3 : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e87b675778a539a7faffb0276348df1d927d7b62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:06:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67e4a7de3cded81efe96e4d83d058ce101093c53","revisions":{"67e4a7de3cded81efe96e4d83d058ce101093c53":{"kind":"REWORK","_number":1,"created":"2023-04-13 11:34:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/880283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/880283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/880283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/880283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/880283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/880283/1"}}},"commit":{"parents":[{"commit":"45e3947967109cc945fbde7bb149ff701f7bb871","subject":"update constraint for neutron to new release 19.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45e3947967109cc945fbde7bb149ff701f7bb871"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-13 11:34:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-13 11:34:36.000000000","tz":0},"subject":"update constraint for horizon to new release 20.1.4","message":"update constraint for horizon to new release 20.1.4\n\nmeta: version: 20.1.4\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I753c2598ba1630da0ed636e304d7d2b6f1a51535\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Iab448b53fa629e92158bc0977dd28a0a2c966330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67e4a7de3cded81efe96e4d83d058ce101093c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67e4a7de3cded81efe96e4d83d058ce101093c53"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
