)]}'
{"id":"openstack%2Frequirements~881539","triplet_id":"openstack%2Frequirements~master~I90274181c1e027d9b17a556fa1ac4767cbf6b5d0","project":"openstack/requirements","branch":"master","topic":"add-pysnmp-lextudio","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-04-28 13:27:00.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-04-28 18:11:48.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2023-04-28 18:11:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90274181c1e027d9b17a556fa1ac4767cbf6b5d0","subject":"Add pysnmp-lextudio and pyasn1-lextudio","status":"MERGED","created":"2023-04-26 09:08:09.000000000","updated":"2023-04-28 18:13:34.000000000","submitted":"2023-04-28 18:11:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881539-add-pysnmp-lextudio","meta_rev_id":"ed15ec7c3b420e8c4bb03960b6d6ea67460d81ce","_number":881539,"virtual_id_number":881539,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-28 18:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-04-28 13:27:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-04-28 15:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-04-28 15:49:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 09:12:17.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2023-04-26 10:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-26 16:04:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2023-04-28 13:27:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-04-28 15:49:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"130951c20bc31a94ddb763d488188431c87d7148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-26 09:08:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d23b8b6acd3b74137952e7607c86094b524b9b4","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-04-26 09:12:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"80b61a853c6ab5f42a4970aea6549d07ae8f6447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 10:52:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d570ea48c14e49179a278d0fb9befa35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c20e8d951e448c8a7a8871544ac34d : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51504fbc37b941918836a224b30c0c25 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19b4ca3c80b8454d9792959d086e731c : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b3dae7e988a4cb8aa5338745daeb7b3 : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49514224a34a4f1899f21bc7b8e20ae6 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e83c34196a32452badce0b77ef33d0cf : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ec5857506394081bfc10f584258c3f2 : SUCCESS in 7m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c94a3820b50f495cb7ecd706834d4ab9 : SUCCESS in 11m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/11c85a2b156d41aea4d9a6be1c65419b : SUCCESS in 8m 36s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3274b2c1f24a4953ae4ef6927ba7423d : SUCCESS in 7m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ebc947cc73b94c6e90809220f5e677aa : SUCCESS in 4m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/67af7ad6370b42c3a693fef249eb2042 : SUCCESS in 5m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e2d95c50da5b4948ac999b368955bbf5 : SUCCESS in 11m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/393a80e12cbd49c7a83f20de7036326a : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cafcdf2c39ba41f89b0f21a88685244f : SUCCESS in 5m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8249876c84794959b29710994f12aefa : SUCCESS in 7m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f9ead45d509743dda5a4140404c2e5c1 : SUCCESS in 13m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/afbb72268cc245bb807d90a060e1db55 : SUCCESS in 6m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2c42259f64814a2fb99a531d91f7bda9 : SUCCESS in 12m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/de1d3bd791e9473a8cec599ab88be480 : SUCCESS in 10m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cb78c9c225984df08359a6b4768aa945 : SUCCESS in 5m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/da7187bf46874067bfcc6cdecc58ad98 : SUCCESS in 8m 28s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c0dcfb4521a54b4989448c4344167057 : SUCCESS in 4m 53s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f9aa4c10fdf442b2b37702bab03e2e32 : SUCCESS in 12m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4dfe0b4bd9d04ca58ff51c0d57bf165a : SUCCESS in 26m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5fb97aa7ee784c588f19314f4216e2f5 : SUCCESS in 11m 20s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/88918c5749b7489b804e0dfc7a14c88c : SUCCESS in 13m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0a684f69615e456299eeb3d3146babdb : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/849732c667c14182ab16360f4ef93247 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6bf804192a5c4337a29a1a03d999cf97 : SUCCESS in 23m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4ef497aab62f477fbe94be213ab353f1 : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f0f7670b35574aa5b1950583823576ec : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4054b86445f7458e999abbeee6d2f5f7 : SUCCESS in 7m 45s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6a1ba7eb30774f45b689a91857c72f19 : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e04ad1a38754a829c3a37c0896a06a4 : SUCCESS in 48m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cd4f4d1ad7a49e8bb8383325caf8723 : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":1},{"id":"699cbfb66686d521ad371c6a2ef069d8d4d7de71","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-26 14:17:23.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9b249a7e87a5e5f0a43e62341138cbc1b5b919f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-04-26 16:04:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6444466ee074cb2857bcdd9a527413c0e8dfc0c1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-27 10:28:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccb4a3f4c10aeaf0c8b3e87f3ff21e6eda3531b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-27 13:55:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"aa4bbbf6558d8da9398cc44db3210a0bf0f57c8d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-27 13:56:03.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"db6c2138471939a93b46b10b44d792f2667a7cdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-27 14:19:05.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f79b5f1391971305b1f783599d431b8c0474ee8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-27 14:19:36.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f99e9800612fedb878b5f62f6ebad36d00ecc768","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-27 14:28:55.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ec2785a0fa4779ba296b93a6880371e6999e436a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-27 14:32:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"6719fc8c56f583fea4b09ca88142d18186b77243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 16:25:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/890489bfd51f4bf2ad2d3172b0cdfe07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269f3649cce640b9bb99f57e34e62936 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629b7c59377246f5adc054703b1073a0 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1e090d70b15476eb4c23fa948017762 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/601e328a42894881947682d4c4b618a9 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d181937e858b4da99c3af2d6538fb74b : FAILURE in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a91032f0170444f8699579560cc8b69 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e866094f75947ca8a7fc64b782327b8 : SUCCESS in 8m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ae8556adf8bd4e72bd9b2ea5a1aaebd3 : SUCCESS in 15m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e5c1390f82c74dc6b54663eaeec8b242 : SUCCESS in 8m 15s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d711468bd310499893403cdcc761b0c0 : SUCCESS in 8m 35s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9a27897a3c0145a5b3e5fdcca97afef3 : SUCCESS in 5m 29s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f5e890c156264d419c01bf58cdd48bf9 : SUCCESS in 6m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ac62318c5f994c7795bce03414416d12 : SUCCESS in 16m 25s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/83795f7c62794103aea0072d8abdec69 : SUCCESS in 5m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/75e4a69c134245969cec69f68757072f : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0c10b42e70e347718a166c323d869128 : SUCCESS in 8m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/053c382a69c14bc8b2466711b8cccdf1 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/076968fd8e74481499aa8d4bdf891bad : SUCCESS in 15m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/49d3a0387c70402fae602810f3135c16 : SUCCESS in 8m 51s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c30057ed09ea46b2a75c015f4162d873 : SUCCESS in 11m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/72f40ae5285342549048e39bf8e58c2c : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b26b239e49af4218a4f13a2bba787e00 : SUCCESS in 8m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/46024340afbd4dcba5d394e5712b7be5 : SUCCESS in 4m 18s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/435229896d2a4ae6a4eb716b254eda15 : SUCCESS in 12m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f8296fa24b414882869a657352269b18 : SUCCESS in 28m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/553ca84429af461c874e8aef411d9c15 : SUCCESS in 14m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c80596d33e4242e3994b4e0da48d48f7 : SUCCESS in 16m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/578230c2cf1a455796dd0320659711a5 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57c2eeaa27a84037b8ceb686dc68a52a : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97f6c766d42c485c9dd9e9f656ce6c1f : SUCCESS in 21m 38s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3bb0db8dfa0b4f75ab6fbf23a58f548b : SUCCESS in 6m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1409ee4a1d8c4dda82edc9d3a1237566 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6da43285aa2d4ecfa4e88660bd5f01c9 : SUCCESS in 15m 22s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/2deb43bc073d4342bd13d1cb9564d509 : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c3672493f77489cbe18020ebf45281a : SUCCESS in 1h 02m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8aff8b690db14b90bf4fd5d8b7aea6f4 : SUCCESS in 1h 34m 46s","accounts_in_message":[],"_revision_number":5},{"id":"57db01ff3745a0300e6c6823f8424594b157370f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-04-28 07:24:18.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5a8110c1db5f96fcfa6b0f1216a2788dc73783c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 08:59:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b5501634144d54bf44c54367bb7984\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09e0c2ffc2d43be914d616617f93ea1 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae33dec9078b4295bd10172e45c293e3 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4b36051f2e84e7a8da0e0686ded0f4d : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/069d98a5709f4494b55db73e951edbe2 : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c2df12ce1f24da696cdc2a92c6ce875 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e596af0efe0340b79fa5282401993a13 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/46ad226610ba4a18aaaeccee865f36f1 : SUCCESS in 8m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/04c7825bafcb4def82ecadc59c2ce310 : SUCCESS in 12m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7a2b494ebb8c475c90aa4183878d753b : SUCCESS in 9m 05s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/68861119165b4e1f8cc76405e279da4f : SUCCESS in 7m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b937c8a7051843febbc629108b7a2a9d : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8a3447a8649a451f8cd62ee8eb822c61 : SUCCESS in 6m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4d83f20427c841dba3bfa36fc5923f24 : SUCCESS in 11m 24s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e45f7e63f6da4dcd876b78515b0f137d : SUCCESS in 6m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d7f5a36629b34affac4feeda07a1a87a : SUCCESS in 5m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/dddfd692296e40b2ab5cba102983a2e8 : SUCCESS in 7m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f7d4c90c723b4450b8c6c8be36c41366 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5895b9cdc864ed2b1a3671b613f7e40 : SUCCESS in 22m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/717404a1715645a58f240ddcffbbff0f : SUCCESS in 9m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/715afeeb630140ae923972f81cc96196 : SUCCESS in 10m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/eea2668583294a0bb21256e5d49cde76 : SUCCESS in 3m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e40e80019f1c43d08d6ae893f45f186c : SUCCESS in 10m 46s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b6a73dd1ccb5447b93c984efa1e389c2 : SUCCESS in 4m 28s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b0dd105e44d14c59b3b03807e81906f9 : SUCCESS in 10m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/388b20045c8a4707bf03b83ec09adc31 : SUCCESS in 26m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4a48d89b855d42f3b592515f6c6e8d29 : SUCCESS in 10m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/95ef48f8b7ab4de6876b136b699e436f : SUCCESS in 16m 43s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f1d6e4e52b174654a1d34ff100b4207e : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fa98c207915a41149cb3370a012d0b89 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86d78bcb7eda4e53b78723f4d9a475db : SUCCESS in 22m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fb449fb275f44117a690dd89075bac64 : SUCCESS in 6m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b074e001aabb49949c03a9da387473cb : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d540b257d6ec4960b7e59bd0efe4343b : SUCCESS in 5m 53s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/10ba852102334863b1e32f23f985c789 : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be49f5bf56a643969bc97b81e0a83e5b : SUCCESS in 1h 13m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56394442da574018b1fc719b7e2b8d23 : SUCCESS in 1h 25m 27s","accounts_in_message":[],"_revision_number":6},{"id":"941391bd4cd873ecc8bad81ecadab5356a678a4a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-04-28 13:27:00.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4b3c2648549a3d641840424d80ab0d43eef59acc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-04-28 15:49:32.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1476f5e33c3f835840aa659f0a2e2a1d628b57c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 15:50:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"99c5b6ba6b3984575c70d3a81c161c7df15dd166","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 18:11:48.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8279939117e4666804526d568b679b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05bfd1ccff145738e7e8a0530f76f2b : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488d1419bf984a8188ac4c3a6abf6e22 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6e153c5cc6d4f6fbed705dc25f7dfa1 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edcd024a449a47168ae81b6abe3ce66d : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5aa76733278437bab657b4a65c5863c : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dec592fbd23743f199bdd3efec2275a2 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0285d61b5be8434f806550cbdca3e182 : SUCCESS in 8m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6d191ac1a94549f5ad5def940b8be806 : SUCCESS in 12m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4e32320ced7a4183bcac6becc4c29ee8 : SUCCESS in 9m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3fa38bd6c7eb4345933f7f295ef88d2d : SUCCESS in 12m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f9b040a59dde428daa364b3a403fa4f5 : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b76c8dd6316a45cebfc1fb19bada1083 : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6500056153cd4144b7f6b87e86c07f0e : SUCCESS in 7m 23s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a6a6a2fa961f42f9af9d044797e11bf7 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e2d48ce00fde4764b1ecaf4953dc59cd : SUCCESS in 7m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e4a3e29ecfed49edbc1d16a88b8cf0e4 : SUCCESS in 11m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a5b3d6bd687043a8af8ed0c4c7fef7d0 : SUCCESS in 12m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c678a4265bb64c6ea408a2d7d32030ae : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fbdfad0b8f534e539d70ecc4d8d84fb3 : SUCCESS in 10m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/690d391835164307b7c12f299acb98ee : SUCCESS in 5m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6cf23ecfcbe6414688bf88caaae92a08 : SUCCESS in 30m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1787d6ddb8774a238ba592e1a7925ab6 : SUCCESS in 13m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c0acaaa488b846739e33bd9edc47533e : SUCCESS in 15m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/de6ce12f8c274a278f1f844b738d9b35 : SUCCESS in 6m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3267f633ce984fb8a6a27db032961272 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc5e67789b124873a4ff79cd8ef348ce : SUCCESS in 24m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a9efaee0076f49b4bcd96ac17e677bbb : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a33ab635922f407c872f68a1a964be9b : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f9265df4ea7c4d1cb7b970af8fb6c23f : SUCCESS in 8m 00s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e72adea9c6c94bb7938be8c0e2c94231 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52b4cff53f234e968c3265a3ba6a289c : SUCCESS in 1h 00m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f18be2602f2f486591587437fc0b2802 : SUCCESS in 1h 18m 48s","accounts_in_message":[],"_revision_number":6},{"id":"63de220eb56c304f07c120a802eeb4b726e3e043","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 18:11:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"ed15ec7c3b420e8c4bb03960b6d6ea67460d81ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 18:13:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c223a9fe3db34a7dbbaf3ebebf143144\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d48cd11b6e98484b9541cf362fa21b32 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d875d3d4dd744c3b183e80810cf60b70ae8fb5d6","revisions":{"e79d6d6d7a912b278c53551fc9c479e88eeef707":{"kind":"REWORK","_number":1,"created":"2023-04-26 09:08:09.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/39/881539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/881539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/881539/1"}}},"commit":{"parents":[{"commit":"e83fc2217d04a5b32ac067aff2bc15b6adb4e319","subject":"Revert \"Drop check-uc jobs for py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e83fc2217d04a5b32ac067aff2bc15b6adb4e319"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-26 09:07:01.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-26 09:08:07.000000000","tz":120},"subject":"Add pysnmplib","message":"Add pysnmplib\n\nThe pysnmp library is not maintained since 4 years and it\u0027s\nnow incompatible with other related libraries like pyasn1.\nDeprecate it and add its fork pysnmplib.\n\nChange-Id: I90274181c1e027d9b17a556fa1ac4767cbf6b5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e79d6d6d7a912b278c53551fc9c479e88eeef707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e79d6d6d7a912b278c53551fc9c479e88eeef707"}]},"branch":"refs/heads/master"},"68b95f7f0f1016cadf2ce18b8e618729099742c2":{"kind":"REWORK","_number":2,"created":"2023-04-27 13:55:42.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/39/881539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/881539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/881539/2"}}},"commit":{"parents":[{"commit":"3442fd6ffe1517716743205973e2c7b84777acea","subject":"Revert upgrade of pyasn1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3442fd6ffe1517716743205973e2c7b84777acea"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-26 09:07:01.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-27 13:55:31.000000000","tz":120},"subject":"Add pysnmp-lextudio and pyasn1-lextudio","message":"Add pysnmp-lextudio and pyasn1-lextudio\n\nThe pysnmp library is not maintained since 4 years and it\u0027s\nnow incompatible with other related libraries like pyasn1.\nDeprecate it and add its fork pysnmp-lextudio with its\ndependency pyasn1-lextudio\n\nFor more info on the new pysnmp ecosystem please look at\nhttps://github.com/etingof/pysnmp/issues/429\n\nChange-Id: I90274181c1e027d9b17a556fa1ac4767cbf6b5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68b95f7f0f1016cadf2ce18b8e618729099742c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68b95f7f0f1016cadf2ce18b8e618729099742c2"}]},"branch":"refs/heads/master"},"8995e26b5efb04afbd0047b133e93bb9625371f1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-04-27 14:19:05.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/39/881539/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/881539/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/881539/3"}}},"commit":{"parents":[{"commit":"3442fd6ffe1517716743205973e2c7b84777acea","subject":"Revert upgrade of pyasn1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3442fd6ffe1517716743205973e2c7b84777acea"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-26 09:07:01.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-27 14:19:05.000000000","tz":0},"subject":"Add pysnmp-lextudio and pyasn1-lextudio","message":"Add pysnmp-lextudio and pyasn1-lextudio\n\nThe pysnmp library is not maintained since 4 years and it\u0027s\nnow incompatible with other related libraries like pyasn1.\nDeprecate it and add its fork pysnmp-lextudio with its\ndependency pyasn1-lextudio\n\nFor more info on the new pysnmp ecosystem please look at\nhttps://github.com/etingof/pysnmp/issues/429\n\n[Q] Is the library actively maintained?\n[A] Yes.\n\n[Q] Is the library good code?\n[A] Yes.\n\n[Q] Is the library python 3 compatible?\n[A] Yes.\n\n[Q] Is the library license compatible?\n[A] Yes. BSD\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] No.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Yes, but they\u0027re unmaintained or not part of the same ecosystem, which brings\n    incompatibility in the long term (see the issue with pyasn1 0.5.0 update that\n    is incompatible with pysnmp, even if it is one of its dependencies!)\n\nChange-Id: I90274181c1e027d9b17a556fa1ac4767cbf6b5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8995e26b5efb04afbd0047b133e93bb9625371f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8995e26b5efb04afbd0047b133e93bb9625371f1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bebe1af961f763fa854f58c7deb2a2a9b0092295":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-04-27 14:19:36.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/39/881539/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/881539/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/881539/4"}}},"commit":{"parents":[{"commit":"3442fd6ffe1517716743205973e2c7b84777acea","subject":"Revert upgrade of pyasn1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3442fd6ffe1517716743205973e2c7b84777acea"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-26 09:07:01.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-27 14:19:36.000000000","tz":0},"subject":"Add pysnmp-lextudio and pyasn1-lextudio","message":"Add pysnmp-lextudio and pyasn1-lextudio\n\nThe pysnmp library is not maintained since 4 years and it\u0027s\nnow incompatible with other related libraries like pyasn1.\nDeprecate it and add its fork pysnmp-lextudio with its\ndependency pyasn1-lextudio\n\nFor more info on the new pysnmp ecosystem please look at\nhttps://github.com/etingof/pysnmp/issues/429\n\n[Q] Is the library actively maintained?\n[A] Yes.\n\n[Q] Is the library good code?\n[A] Yes.\n\n[Q] Is the library python 3 compatible?\n[A] Yes.\n\n[Q] Is the library license compatible?\n[A] Yes. BSD\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] No.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Yes, but they\u0027re unmaintained or not part of the same ecosystem,\n    which brings incompatibility in the long term (see the issue with\n    pyasn1 0.5.0 update that is incompatible with pysnmp, even if it\n    is one of its dependencies!)\n\nChange-Id: I90274181c1e027d9b17a556fa1ac4767cbf6b5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bebe1af961f763fa854f58c7deb2a2a9b0092295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bebe1af961f763fa854f58c7deb2a2a9b0092295"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9c0990eb83245f7398f1aa4a5b1a61f7c78d5ed0":{"kind":"REWORK","_number":5,"created":"2023-04-27 14:32:19.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/39/881539/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/881539/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/881539/5"}}},"commit":{"parents":[{"commit":"3442fd6ffe1517716743205973e2c7b84777acea","subject":"Revert upgrade of pyasn1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3442fd6ffe1517716743205973e2c7b84777acea"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-26 09:07:01.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-27 14:32:01.000000000","tz":120},"subject":"Add pysnmp-lextudio and pyasn1-lextudio","message":"Add pysnmp-lextudio and pyasn1-lextudio\n\nThe pysnmp library is not maintained since 4 years and it\u0027s\nnow incompatible with other related libraries like pyasn1.\nDeprecate it and add its fork pysnmp-lextudio with its\ndependencies pyasn1-lextudio and pyasn1-modules-lextudio\n\nFor more info on the new pysnmp ecosystem please look at\nhttps://github.com/etingof/pysnmp/issues/429\n\n[Q] Is the library actively maintained?\n[A] Yes.\n\n[Q] Is the library good code?\n[A] Yes.\n\n[Q] Is the library python 3 compatible?\n[A] Yes.\n\n[Q] Is the library license compatible?\n[A] Yes. BSD\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] No.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Yes, but they\u0027re unmaintained or not part of the same ecosystem,\n    which brings incompatibility in the long term (see the issue with\n    pyasn1 0.5.0 update that is incompatible with pysnmp, even if it\n    is one of its dependencies!)\n\nChange-Id: I90274181c1e027d9b17a556fa1ac4767cbf6b5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c0990eb83245f7398f1aa4a5b1a61f7c78d5ed0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c0990eb83245f7398f1aa4a5b1a61f7c78d5ed0"}]},"branch":"refs/heads/master"},"d875d3d4dd744c3b183e80810cf60b70ae8fb5d6":{"kind":"REWORK","_number":6,"created":"2023-04-28 07:24:18.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/39/881539/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/881539/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/881539/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/881539/6"}}},"commit":{"parents":[{"commit":"3442fd6ffe1517716743205973e2c7b84777acea","subject":"Revert upgrade of pyasn1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3442fd6ffe1517716743205973e2c7b84777acea"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-26 09:07:01.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-04-28 07:24:18.000000000","tz":0},"subject":"Add pysnmp-lextudio and pyasn1-lextudio","message":"Add pysnmp-lextudio and pyasn1-lextudio\n\nThe pysnmp library is not maintained since 4 years and it\u0027s\nnow incompatible with other related libraries like pyasn1.\nDeprecate it and add its fork pysnmp-lextudio with its\ndependencies pyasn1-lextudio and pyasn1-modules-lextudio\n\nFor more info on the new pysnmp ecosystem please look at\nhttps://github.com/etingof/pysnmp/issues/429\n\n[Q] Is the library actively maintained?\n[A] Yes.\n\n[Q] Is the library good code?\n[A] Yes.\n\n[Q] Is the library python 3 compatible?\n[A] Yes.\n\n[Q] Is the library license compatible?\n[A] Yes. BSD\n\n[Q] Is the library already packaged in the distros we target\n    (Ubuntu latest / Fedora latest)?\n[A] No.\n\n[Q] Is the function of this library already covered by other\n    libraries in global-requirements.txt?\n[A] Yes, but they\u0027re unmaintained or not part of the same ecosystem,\n    which brings incompatibility in the long term (see the issue with\n    pyasn1 0.5.0 update that is incompatible with pysnmp, even if it\n    is one of its dependencies!)\n\nChange-Id: I90274181c1e027d9b17a556fa1ac4767cbf6b5d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d875d3d4dd744c3b183e80810cf60b70ae8fb5d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d875d3d4dd744c3b183e80810cf60b70ae8fb5d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
