)]}'
{"id":"openstack%2Frequirements~882916","triplet_id":"openstack%2Frequirements~master~Icdeb56c9890132b19bf65934bbee135df56d0341","project":"openstack/requirements","branch":"master","topic":"bug/2007574","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-05-15 15:52:16.000000000","reason":"removed on reply"},"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-05-16 06:34:41.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}}},"hashtags":[],"change_id":"Icdeb56c9890132b19bf65934bbee135df56d0341","subject":"Bump XStatic package version to latest i.e. 1.0.3","status":"MERGED","created":"2023-05-11 06:43:10.000000000","updated":"2023-05-16 09:24:52.000000000","submitted":"2023-05-16 09:23:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882916-bug/2007574","meta_rev_id":"434e3bb0dd05e99d720b60c4d7bc9760d79d96c0","_number":882916,"virtual_id_number":882916,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-16 09:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-05-15 15:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-05-12 18:12:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-05-15 15:52:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-11 08:09:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-11 16:28:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-05-12 18:12:41.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"1f58df072f94591555c29c7a8a2c5bf57714e4f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-05-11 06:43:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bacc7e1580046421106a9b561fb6d8518f2e6e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 08:09:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34eaef0de9f40aeb24144a075f44ca6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2720ca66557c444a8db9ddd86222e65f : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97faae5ba3d9480989c100792633a387 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1863270dbee94444a582fb5c880be852 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efa5040605c54fa58aece1e1dd01b855 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/822488dbf716463ab4e478874d6c9f0c : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1ecea34ccc34610b90bc37d120c8919 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e487d60053241d4948c3fa326388130 : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/62d4ef69ec5644dc9d6c07ba288849f7 : SUCCESS in 8m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/67b76beea367410d94f628fdc640e894 : SUCCESS in 11m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c36425496301430b9d43e9159e9e9b34 : SUCCESS in 8m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9dd864d3d1c34f37baa0115361bf1b75 : SUCCESS in 14m 41s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1655d55157c8442a90be8cf476d8a852 : SUCCESS in 6m 50s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1be91d375f3a4f34b6ed6f7cfdcf6039 : SUCCESS in 5m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d7f4c26db0bf4ca2ad1be08a4ea9be5e : SUCCESS in 6m 22s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/445d6313898e4e368e8fdd7c62f638f9 : SUCCESS in 10m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cbb7b25769a548c7af86f04784668d76 : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/41185755de924f73a02bc038ab4f92f7 : SUCCESS in 5m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/293bac4a649a4a5e989d0f1c76179c7d : SUCCESS in 8m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/70b835078f16421fb82d8693d0aee330 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e292fa55ddf4d9bad1dcdf5baffb0d5 : SUCCESS in 6m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b41699432ced448999df42af0a935122 : SUCCESS in 8m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f0b477d899b8451c9cb4c959368f0d4a : SUCCESS in 16m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6bd868ff48594db3bc72d7d13bd2b725 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9558b9b9e182412daec8cc5b1e4981f5 : SUCCESS in 9m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e2aa0f328c3f42c0b41fc7a67f8ae6fb : SUCCESS in 4m 40s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/05d3ec7aaf3c4fa994ba969f89d704ee : SUCCESS in 9m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/446d859521bc4dfda67f332be207843c : SUCCESS in 25m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a62d789dde1748c08dbbaa8763a53164 : SUCCESS in 10m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6cc81c1a173f4e1a881647d2ccd376c6 : SUCCESS in 14m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6ee2a8cc8b1541de8b9f6c0583fa355d : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c32e26a4b434094ad59c0904eff336c : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74836e92fc904a608ffa6088fdd95263 : SUCCESS in 23m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e9201c1890b248498c5c8cd98348c4e1 : SUCCESS in 6m 14s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cda5a965c5f140598f4587c7e2444cb9 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fed2e9d97d1d4476a1965dddcc00f588 : SUCCESS in 6m 40s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/51d5bdc00f614a42b8c56caa0da4df7d : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9550b9bc80ed4353808d2bae542691a1 : SUCCESS in 54m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08c4498cd8c94746b3c504f65ad8c53d : SUCCESS in 1h 23m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4df99183cbdb96fc17715353329c72216125bf94","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-11 16:28:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"98d7b08bb55c484d9c6347fa88d8131515c79802","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 20:25:35.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 882916,1","accounts_in_message":[],"_revision_number":1},{"id":"1ae50450e3f8be828cf95837dc0918de0c4b701b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-05-12 05:09:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2174be74d30eaeaf385f9d777ddf97a543065651","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-05-12 05:19:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf3577f5fad895ad72bbb5cd985e1554c8de0f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 06:19:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c743ba5702914a658c6a382a5a94f8ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41e0f83f8fa4c4c87939e4fd6876616 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc3a9d086e1469fa7fe41100eceef16 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f712477e25f4ea59231eff616d50bd6 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95a1b37f1e354af18a64348cb843fae2 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03bf156a63274491813868f064f80fec : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/713e7f0800c74984a2e261613d28f607 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7da9edc913ec476498e0c8677c377434 : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f2b5b4915f9d47cfb580ad68db11da39 : SUCCESS in 7m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/75949bbfe29441249fd55b68e0c018f2 : SUCCESS in 13m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ecd6a2e445e642ce9e59bd23f63ca9c9 : SUCCESS in 7m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dd75dc0b2a384a5786e864480de07d16 : SUCCESS in 12m 12s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bf70599615e04670be8d62bdc970f6fb : SUCCESS in 6m 52s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f8e50fe7b8a744b29961271c8a245466 : SUCCESS in 4m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/15c75421e00a4b1789cdf837bc1a897e : SUCCESS in 5m 47s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/164b9741bb784f56ac5e833b863525e0 : SUCCESS in 12m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a8f11f47f45d4a4f9b1fe24f717e847b : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2a5326d54fbd4a10a97bbbec22db7e90 : SUCCESS in 4m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/785d6a4163834a6881a431118a033d93 : SUCCESS in 8m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/da73b0f9305a41f799a9ff3d1d0c20c8 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7205ee8acec54bc3b96f41650c042fa5 : SUCCESS in 7m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4b85e0b7d134456497b126a0983212b5 : SUCCESS in 8m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5f2ae109294548bb9da057aa9fc2f968 : SUCCESS in 10m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0f4f9bd424e045aabbb57463a8516d19 : SUCCESS in 3m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c74ccbef2c6d483e95c7d18a27a56b0e : SUCCESS in 8m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b8cf74314dbd43c992626724da62b88d : SUCCESS in 4m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/463163154f31498fb4e2d060f2ff78c7 : SUCCESS in 8m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/741ebe7952274b41b98086c1eb1cb76b : SUCCESS in 24m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dff78544d913477d889c83fe383390a5 : SUCCESS in 11m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4adac93afe584ad1a52f2d888ad01de5 : SUCCESS in 11m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b445c31f062c47c8957e50dde3b92932 : SUCCESS in 4m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60ae46289b5e49e5849571ba0f4e21ec : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d6c5b55f48014848b9485bd3252d9b4a : SUCCESS in 19m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7d72e2bae79b4373a371560c0987510b : SUCCESS in 5m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/49dbcee0f1874a2f8a58629809340053 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66ae217061ef41a0ac515a70d0d4bfc0 : SUCCESS in 5m 53s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ae982781b844422995dcb0f126af73c9 : SUCCESS in 10m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21e41ac15b0e485da616bccd26a39547 : SUCCESS in 40m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/552933d5ebd8463886a7997358766eb3 : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":2},{"id":"1b130929d8e09d087deccd30d13e2a8285d9b14a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-05-12 18:12:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2c74377667d481884c774179f3a4b584dbac0c02","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-15 15:52:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9cc02a61c39c07a038a337e6819b400f74e49ad4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 15:53:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"57a7bea36c06c512e5aad82323d2641d492ce099","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 17:37:13.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f310d6fb12b4f758c07d711f140a7ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2336f7e525fd4e92b9d07dad24b7f953 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/729f34c375714898be4f5ee85ac58413 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/382bf0e5a4214688bdf72ff612bba87d : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fdfc1363bc74d44a92952f9a099ece4 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fccc2aaacf04450aaf13acd9a265589 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59906d5f1b9945dd9cf560807571b850 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95f64c457ffe4e9d9f19a96f97899eea : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a62cf7eca5024093997af097f2c00966 : SUCCESS in 7m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f22ba7129b1a45a0b8284fac4a2c244a : SUCCESS in 14m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fc358435e4884088a17c880fbcdb046d : SUCCESS in 8m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a22f30ded358416bab3b28b8db7e1ed7 : SUCCESS in 11m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/be20002ab5c04c249e09ccccd097dfba : SUCCESS in 5m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b241a00e34c4496babbcf1c12e4fea03 : SUCCESS in 5m 32s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b18f3e116fd24181931c1b8d674f567a : SUCCESS in 7m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/87cc3a3f7b744f86821075314a0feb7b : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5a5c921acc54ca8a3138c79ad5d3cdb : SUCCESS in 8m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5c435c4897944223b09320292b436238 : FAILURE in 7m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/60fcb14d062c414e88e02e4e3af0d363 : SUCCESS in 12m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8f2a897941bb4794b34d25e2ffab5081 : SUCCESS in 3m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/48ceb5713f3e411290e30b2c0c4090bb : SUCCESS in 10m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/26b27a2e52f24d2ca63db77cc0e497c4 : SUCCESS in 4m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/14aacf1c37664d1f8103cee8d50476dc : SUCCESS in 27m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8f7c44232d7f426a9a9e55e30573492f : SUCCESS in 11m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f2376626565e457491c492bc22c8fb05 : SUCCESS in 14m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/843da57d380c45dbb6e096191443223b : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8e474e8dfbd4e8ab8f49d38fc01fdf8 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b970800506034bda87c60a1036df49b2 : SUCCESS in 23m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/32478d7f352645b1bd07811f76da2e97 : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/25cf97560da046fc93450df088a54465 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b9501012125f4bc4bcece78fc60ed2b2 : SUCCESS in 6m 35s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b0f7c636555e4e6a8feab3e38ca53336 : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb0e5022273e4a128d840664cafd3231 : SUCCESS in 51m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83ad40406bac45a4bafe8d27ad3cddcb : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":2},{"id":"caa9f47fdecd776e055116ba805e94caf9745636","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-16 03:56:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0ae61aca87c84f942db08190be4717f5ca4ef12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 05:16:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7133133d31064d269100cf0979a2c791\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff157a240304070a8f2a5b3f6de87a2 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9366f2bb8f2b4220b9f599322a230960 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b81c74cad414298a67c7fe08a1e0691 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0f8ab30f6ab4568b2d5bfd938861cfc : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6482dc5459184d32a3f5a92426d78110 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5eb827e633d54ce49f71db4538ecd305 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77bdd0920ba84701be5e18882ea3e80c : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2d578e92a33841e184f0362e90466d58 : SUCCESS in 10m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9c3e0dd4c0db43989f9c912c74768cbb : SUCCESS in 18m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1c65730048ae447e94d900a3735f60f7 : SUCCESS in 6m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f9069dcac7a54ae299b0cad1bfb2b375 : SUCCESS in 11m 28s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/dc98b3fbb0f64a4294960dacaa49e1f1 : SUCCESS in 7m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c38cccd6d1ad49b19b5b48a5dd786903 : SUCCESS in 4m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/13bce7011afc44dfb81a8bb62613be58 : SUCCESS in 5m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/658b2f53dc2643819df816fdc022056c : SUCCESS in 10m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f9146e889a8a4a638e2db5daec8583a4 : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/01f2b5c60f3f40aa9b8b85ddfc2182c7 : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b4cac0348b96483e8322c9d462a93fb2 : SUCCESS in 6m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/810ffcead9a74f0f916a9319d774575f : SUCCESS in 7m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8cfcc9a0216240bb9dfb79c380d172a6 : SUCCESS in 6m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ec095b2b255e48f8a024b70a69b20b78 : SUCCESS in 9m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c425021a42944416b14b2bc96ffccd9b : SUCCESS in 11m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4d802a6cdcce44af86e5a01dd8c9ce65 : SUCCESS in 3m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2190f64046c54b2ca471603e0e4b5bed : SUCCESS in 8m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2b1418a8f1464adb8300c7edb2310026 : SUCCESS in 3m 59s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3fc4b189d45b4005a9d957ab4d9b77ee : SUCCESS in 9m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/90f87abd3dff44758b247725b3ed5cfb : SUCCESS in 23m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/95bacd91fea449138e28880b7f27255f : SUCCESS in 9m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bfa42465c17a4f7ba852952f920be036 : SUCCESS in 11m 38s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/12435dad3a9a47d5998b4e9042d602ea : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d586815d34774e758ca0ba53ede76dd0 : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b28ebc0585bf429aa6a7acb8173d6f08 : SUCCESS in 18m 19s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a4dc36e315024905b0873401242d521b : SUCCESS in 4m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/defcc46ae80d47828b083fcde7e3e7b2 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d5fed6b5bf14649a9bddc13be9c4b06 : SUCCESS in 6m 19s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4e14bb7e94b2491c8dd52f70bab439b9 : SUCCESS in 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebfa97fc244d42b798fa753471183bd0 : SUCCESS in 40m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/420570a1731f4c6f92c41b060861e43f : SUCCESS in 1h 18m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e4cec6094af63ed7b1efbd83eb405b7394344aab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 05:17:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"42fb0684cef05c84be69879965b99d7b29407172","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 06:29:57.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec130c6e1c5d46fe996c0fa65a345bb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e94fc76c2374bb48e9d879b2849127c : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d155ceee903747199f1616adce07c6c3 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d009049bc04c4d10aaa51b7c8d2e9ad5 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40b0f6e6d02241f08705a00c93a577c7 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/715b3199e0a94173a26ff22077011a45 : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b407fc9ed89240e39026a99acdbbce5f : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/823c19d082b843f6b8011381b6ffd508 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/be77d444e2f54f66b8b8ff293886969a : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1658db56970a442386cef3110fca784d : SUCCESS in 9m 42s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cf6844a265994b2993a695518579ca66 : SUCCESS in 8m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a3b62e56c2a842a2961a2ffb18b89705 : SUCCESS in 11m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/34636f36883e4c8e9473c5a0e5c864ee : SUCCESS in 4m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/394636337d264d73a6f0e854112b3ea0 : SUCCESS in 4m 56s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c2558031092d4f6ab84d7f458df9bc08 : SUCCESS in 8m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f2b4653ac1ca430ea91cb0edbefc7d82 : SUCCESS in 7m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ba798a68bdf41aab31805de9a2b2578 : SUCCESS in 6m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/72d62b9244af4a4ba79a686072a4f624 : SUCCESS in 8m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e7e43bce7f164b699f85114984fdfc3a : SUCCESS in 11m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b08ace88d2e44f27b67dd92fc74069d7 : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/361623bbe37a4c93909eba48bbc44757 : SUCCESS in 12m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/04a6c4bec50d4df3a186b8d07b4713de : SUCCESS in 4m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6a65b00272914eccaf19c0491496e6d2 : SUCCESS in 21m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4fe57fb841984074a352ba4f9d48d413 : POST_FAILURE in 20m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8fb090db17a6466aa6fe3005ee820770 : SUCCESS in 14m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/27f14858175e479eb2ae6b4eb78b8fbf : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43028a8e14c14734bef611e543e472a9 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf270e7eed0447b6841ee89745811a49 : SUCCESS in 18m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/87e502ecc0794cfdb6a20eacac12a444 : SUCCESS in 5m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1812d25a024a4b5c8459c5728b8b224a : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e5d87972ca2b433481297afa2aeea4b5 : SUCCESS in 5m 39s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/203fb06c3226406e83f09aa87cc722d9 : SUCCESS in 8m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/092fab535ffc411190fbf41c71730de4 : SUCCESS in 42m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4689fb78c767426cbd91d86fa3f18e90 : SUCCESS in 1h 11m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a4a142b751dc190372e6e3f03f2bfee40c7317cb","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-05-16 06:34:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2480c97ac29ad7227e30682abef545400b3ffec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 07:52:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b9461c280ed4cbda5fc68899f88acff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a1daa05f1844ad91a92aca32560080 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2caac38e200c404183cc940f79bdb7bd : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/914112bccabe489a87fdb0136f363e10 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9014dcf8f513428da02507e33a6e3c7c : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b56b31af4b9447ad9aba11c91dcf0ee8 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6038e45873a64441b5520ddf55521118 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4993088d9d2e4c0b99b78488b086b258 : SUCCESS in 3m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/68c6a43f946d4f6aa19d5a9c665f75f0 : SUCCESS in 6m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7cc53416e8b248d493c76b75b59f0317 : SUCCESS in 15m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1ac22a7e46bc4df8b32c48e1517bc7fa : SUCCESS in 8m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bc946b92f3b44174811462ad1a004cb1 : SUCCESS in 9m 22s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f239217d064c46e5ac1bcbd6fc7cc2c5 : SUCCESS in 6m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/842f5549cb2b4a2ebf30af5a99fd5756 : SUCCESS in 4m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7ca98b2affac464991cfbe8295517c1e : SUCCESS in 6m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7703c68faba94ca4b09f587fb0909d00 : SUCCESS in 10m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e0d52859bb2640d1b9de4308de3cda38 : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8d17897aa7124738ae21bdd9ce9fa70c : SUCCESS in 4m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/663b0beab19141f6a1446a08301abada : SUCCESS in 6m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c7d720d1422f4998a4e0f7210ccc2528 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1578e49a454040de9aa5b54895450317 : SUCCESS in 6m 27s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/43efa3cd5ef84efcbc87b5a50d8f6fcf : SUCCESS in 8m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1fca85351eac4b8da6547ea092e74027 : SUCCESS in 10m 30s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/97ec2b2c656747bbaa9524e15ef20814 : SUCCESS in 4m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1f4f3d8e30a94f36bfdde45af8b5522d : SUCCESS in 10m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/79f31b5cdda4490a80fbe4e03f1c866b : SUCCESS in 4m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d3c0484b9e414272b379328400e9c56e : SUCCESS in 10m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/36fcf395f71d418280b54415d279290f : SUCCESS in 27m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bc7c2402eb084023a8a4c2f73faf573e : SUCCESS in 11m 10s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/18de5e338a294870b963c44aab85b57c : SUCCESS in 11m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/899a005fb2ba436aa20a5f2d68319ae3 : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6a094bb45374ef480cddfa22b8db715 : SUCCESS in 2m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/43cef9105bd34d539e01c32bab3235da : SUCCESS in 23m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/62f056f4ad2b4d42b319d57836a87123 : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/610d7c79370e4e1eb5ff76c0f4ffb974 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/987e27bd4670491ab63048579d0d7c5b : SUCCESS in 5m 52s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4172dcdce55941e980ae12631ae884a3 : SUCCESS in 9m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/348a9fb983be45c095a27733b58c55da : SUCCESS in 44m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9bd50ec5013439db39e3b8e1957fd0a : SUCCESS in 1h 09m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e61826c5c39e9efb8c9b869dc6f4b347dd78a948","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 07:53:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de6daef8fe422d51de2d32e61ee13622e017f5e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 09:23:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e65331dcae8944218f2b7e0b43764c28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe72407e9ab47c0b5cfe3a0b3c80011 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52afc6f47626411e85da9eeaa91f2d43 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f54802df8164657a5315ec40ce367b1 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4acc33f226594467940de3a697553003 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afe9efb0d9f04d4bb5f7d9d9e1e75000 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8b30f2ccef14a3582dc2fc966bd751f : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92eef62208ea4002a9e3138d6230f9d9 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c76a1c1a9ba4296bdabfd72f8202df6 : SUCCESS in 8m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b6f97ec1d8f544068c5bd5a24ba4f3b7 : SUCCESS in 12m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b2f3dfd638bc43cfa0e40b4ec737478a : SUCCESS in 7m 42s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8d4031f4496e49bd80dab84967326548 : SUCCESS in 11m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/254d551c5b484ec2aca52e63a15d521a : SUCCESS in 4m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f9afc4043afd4a188e9da530adb2ae21 : SUCCESS in 5m 26s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/30fe9311773a4329a3f6bc2be3fff583 : SUCCESS in 8m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/580ab675e33548fca363a36ca1342ed1 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e34bd12cc4b74198993b6cedf886feee : SUCCESS in 9m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1d22d16b4e074ab7b5360fc8b77a6047 : SUCCESS in 8m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9da9c04979ea4324aa9f2aaae186ca80 : SUCCESS in 15m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0343222e5c6a43269a885f05b5392de1 : SUCCESS in 3m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8c1247bf9a7b47319be3eaddfb93f99e : SUCCESS in 14m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1d0b777e4a934cada6c57fa78c9f2d80 : SUCCESS in 5m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/87111e9d6b5e45519bacad8e9472198d : SUCCESS in 24m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c5540956937c49a89875c7c9361704c4 : SUCCESS in 10m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4f1bc7e022d14d9790c50fae691a8d72 : SUCCESS in 13m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2e56fd8b02f948a5a8d4c2e3ebc6ec2e : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c99312d9579d4943b2b6f7521833156b : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7063fe37ece8484b8d16cd5febedf4b5 : SUCCESS in 17m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/401e060e27334ab2b94d03dcc54722b8 : SUCCESS in 5m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7957132eac964113ae36b916bcb39aa1 : SUCCESS in 4m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/156e173c8d1140a88af8af825c5a7b05 : SUCCESS in 6m 34s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/dc41cabd67784b4d8e828086bd23c727 : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f502a9841d4e46d2b6f5ff67347f1833 : SUCCESS in 50m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/667560ca5d704ed095fc70983ddcfa42 : SUCCESS in 1h 26m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d6a4816cba79095f5e349636e73f3bc6d7a88000","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 09:23:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"434e3bb0dd05e99d720b60c4d7bc9760d79d96c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 09:24:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f47a7abdc1443fa19dd6f68c59f0f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e67e83d964d4b1ea3396f04527f4210 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2732c3a707144058c54db50896bc9ca14a4be79","revisions":{"3dc883fcde7511798b2ec0260dc75dfb30a6062b":{"kind":"REWORK","_number":1,"created":"2023-05-11 06:43:10.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/16/882916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/882916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/882916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/882916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/882916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/882916/1"}}},"commit":{"parents":[{"commit":"29f5dfe34e72e86d6e99aba21c5a7c40b4a7dca8","subject":"Merge \"Add requirements-tox-py311-check-uc job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29f5dfe34e72e86d6e99aba21c5a7c40b4a7dca8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-05-11 06:32:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-05-11 06:43:05.000000000","tz":330},"subject":"Bump XStatic package version to latest i.e. 1.0.3","message":"Bump XStatic package version to latest i.e. 1.0.3\n\nhorizon does not support setuptools 67.3.1 and it broke Kolla\nHorizon container builds. We recently cut a new release of Xstatic\n[1] which may fix this setuptools issue. So this patch bump the\nXstatic version to 1.0.3 to fix setuptools issue.\n\n[1] https://github.com/xstatic-py/xstatic/commit/5cd7a2347046b71056ae29f00ffa4f71218a5c0c\n\nRelated-Bug: #2007574\nChange-Id: Icdeb56c9890132b19bf65934bbee135df56d0341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3dc883fcde7511798b2ec0260dc75dfb30a6062b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3dc883fcde7511798b2ec0260dc75dfb30a6062b"}]},"branch":"refs/heads/master"},"a2732c3a707144058c54db50896bc9ca14a4be79":{"kind":"REWORK","_number":2,"created":"2023-05-12 05:09:38.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/16/882916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/882916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/882916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/882916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/882916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/882916/2"}}},"commit":{"parents":[{"commit":"f4238c0a45d422093636b9682f9085f3e89d9481","subject":"Merge \"update constraint for os-brick to new release 6.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4238c0a45d422093636b9682f9085f3e89d9481"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-05-11 06:32:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-05-12 05:09:14.000000000","tz":330},"subject":"Bump XStatic package version to latest i.e. 1.0.3","message":"Bump XStatic package version to latest i.e. 1.0.3\n\nhorizon does not support setuptools 67.3.1 and it broke Kolla\nHorizon container builds. We recently cut a new release of Xstatic\n[1] which may fix this setuptools issue. So this patch bump the\nXstatic version to 1.0.3 to fix setuptools issue.\n\n[1] https://github.com/xstatic-py/xstatic/commit/5cd7a2347046b71056ae29f00ffa4f71218a5c0c\n\nRelated-Bug: #2007574\nChange-Id: Icdeb56c9890132b19bf65934bbee135df56d0341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2732c3a707144058c54db50896bc9ca14a4be79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2732c3a707144058c54db50896bc9ca14a4be79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
