)]}'
{"id":"openstack%2Frequirements~882933","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Iffbfb7d62ae53a49d3f4ce156cfa602cea73e29e","project":"openstack/requirements","branch":"stable/yoga","topic":"bug/2004555","hashtags":[],"change_id":"Iffbfb7d62ae53a49d3f4ce156cfa602cea73e29e","subject":"update constraint for os-brick to new release 5.2.3","status":"MERGED","created":"2023-05-11 09:28:10.000000000","updated":"2023-05-12 04:44:41.000000000","submitted":"2023-05-12 04:44:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882933-bug/2004555","meta_rev_id":"3e9e9a9117a646857dc17d38eb006b151b6d1397","_number":882933,"virtual_id_number":882933,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-12 04:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-05-12 03:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-05-12 03:13:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-11 09:34:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-11 11:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-11 12:19:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2023-05-12 03:13:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1951cd11faf4ceee46ba3e9731e63c3742bf18ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-05-11 09:28:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bb56b7691865ac030cdc6c03be4f4fb74705b76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 09:34:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9129c7ec3d9647c2958c3f187fce8203\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0e5564c4bc05415ba929dd4ed28b82f9 : FAILURE in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1685fd167b38b9221e6d661dfbb02a236e691d4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 11:00:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f560c0bd37264cdbbe6c00fcde18cb05\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e39d451f7b340e699ed55ace1395e15 : FAILURE in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64295feaa92248dfa4d8ed4790da8aea : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/867756cee61440a89298e1c00b9b9118 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9049851545ce4277bb9bca3b89fa5e01 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e06a0c9ef9b94812849c8ccf3fcd4d6f : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71cdf5640df94e3d97b927d3b365477c : SUCCESS in 12m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4167fbd03423468c838913b831a3e04a : SUCCESS in 8m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3f3cb56891784889ba0bf9a0e41829d2 : SUCCESS in 4m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7a96e861b5f4400cabe5c106a3b5001d : SUCCESS in 12m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a8a6ac79eb9c45b9a016e569813afd66 : SUCCESS in 4m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fd29e49303574a3d8fe7fd65c79a1a47 : SUCCESS in 5m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/859b07ad2bdc4c31b27fc5edcab698bd : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/995e4caaf966408f8415155efdd6618d : SUCCESS in 6m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a8ccbfa46a794a3ba7da468aff459181 : SUCCESS in 8m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1cf21ee671e34612b70dbf0d7ad071bb : SUCCESS in 9m 47s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4504fd1c4dcf4120abbcc1ca8965e016 : SUCCESS in 4m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2938d04dd098407fa63114345e411065 : POST_FAILURE in 8m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cf6058b66dfd4290a0da182663b54025 : SUCCESS in 4m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e331c20614174cefa380734bf4bf3e1c : SUCCESS in 24m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/37a5852e617c46cbb38bd8d0b12561f2 : SUCCESS in 11m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cee6864f5d4d4fd8a4329c10456beafd : SUCCESS in 4m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a7f018ed6684ee89bf085fbe811d1b7 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d950bba9650c478ebdfa830a7024f663 : SUCCESS in 19m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/41e2697ac37346e3b82104ad75a9a08d : SUCCESS in 5m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4f7d6abd9c7d4ca0ba407d4eee1c90a5 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a62353733fd456ab3c5237265caf0dc : SUCCESS in 6m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/820ab9f43615400faaf47ead7dae45d4 : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54a2d58f431348c79c8de61a1a2e725e : SUCCESS in 55m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cc553d70df945ac9384be5fd529747d : SUCCESS in 1h 27m 16s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file update.py has only 1 lines\u0027\n  Comments left for invalid file openstack_requirements/cmds/update.py\n  Comments left for invalid file update.py","accounts_in_message":[],"_revision_number":1},{"id":"5147877c7349f67bb337733a95573efadb1997a1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-11 12:09:38.000000000","message":"Topic new-release removed","accounts_in_message":[],"_revision_number":1},{"id":"213f06e540b967dd3ad304e80f8d9773177d9d3f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-11 12:09:48.000000000","message":"Topic set to bug/2004555","accounts_in_message":[],"_revision_number":1},{"id":"c965bbedffbd75136813f05f04bf3cb1acd081d0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-11 12:19:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a0c943fd1917c4941843198bb6a05c8626c5fec","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-11 12:26:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"078eff8a2e7ed6d30646b5b5b23a7f84dee46388","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-11 12:33:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dd0cc04b3a01655c0dd0662973b77379c61b26b6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-11 12:40:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9915c314eb300a9538512fdd06f647766062431","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 12:46:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c37b72e2260499b80f586771d218188\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af541ea334044288ba6f4637329b1c53 : FAILURE in 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"39b18ee0ab7c1641059a3d5333b234d096118d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 13:44:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff98519d8bd24792b10868424eb6f2e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f548b20efbb4f44be62e2a007dac669 : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d761d97feee4732a0176a06e1fc73b5 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63aec4d44aff4d98919ba22a5439e1db : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/763d54422a3d47df9cfae6c6f07912eb : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56bb38885b15488f97cdd48ad76cd303 : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c2f86134e441414fa022a66633956038 : SUCCESS in 10m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb7de52822664dc5bac00c722cab938a : SUCCESS in 11m 45s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/bfbe5a78d9654a88b1092d3845d7c801 : SUCCESS in 6m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ad3b07a1ce964c058585059a9531e03c : SUCCESS in 11m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ff648356d0114355962494113c142d9d : SUCCESS in 5m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4552ae8878e04251badd265fdad1857e : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1c370e6ea54f4e0da4a975c4adcea096 : SUCCESS in 7m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d4fdbb0e24d4458881278ac47dbf9f4 : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/83a61b4657c946ceb9e9513287686c1f : SUCCESS in 10m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5b6609d44c3d4eeb875f68dab30f5123 : SUCCESS in 10m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/506d0024520e484fb9242589aabcd516 : SUCCESS in 3m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2912a03d991a4cb1873c3914e3d4d1f0 : SUCCESS in 8m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2bfffa624e0d4da4bb6297a01cbe0319 : SUCCESS in 4m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8666ca0401494c95a315d956846a448a : SUCCESS in 25m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8934ba05ed174fa8937e8fe2b6b55228 : SUCCESS in 11m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/81ec4b6996774668ab20ebce1c6e8412 : SUCCESS in 4m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43597dc39f64467e9d2c14bed1c4504a : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aff4bb31bdea419f8ed6ae2fe93150c4 : SUCCESS in 21m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/85efdb0141d84ad68b85d1871663fdf6 : SUCCESS in 5m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c2f07b1b61e442f6b2c562a36cc57011 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a53920386154660b4504bb2a4328225 : SUCCESS in 6m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d4f3f36d22174703acc6209ac45de46f : SUCCESS in 9m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccaeec4d3810437192f98af46a998bc6 : SUCCESS in 1h 07m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51e97bd5aaad439884d5da29904411a1 : SUCCESS in 1h 19m 56s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: file update.py has only 1 lines\u0027\n  Comments left for invalid file openstack_requirements/cmds/update.py\n  Comments left for invalid file update.py","accounts_in_message":[],"_revision_number":1},{"id":"1c739748573e946130b4573584e338a94e58699b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-11 20:29:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb252ec5ce360cc9c38ac8b5ba8b22b85ea005ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 20:38:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ccdccbcce3b4963a36caab0a03a2a90\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/016365cdb2784ee0bffe4be7a57eb74c : FAILURE in 8m 20s","accounts_in_message":[],"_revision_number":1},{"id":"adb42ba50a48940e3c63e6371ff6889e70c331ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 21:49:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08c0bb158c394c2dac40dc8d88caaf01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aadb336314094a3c8b2d6603feee820b : SUCCESS in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ce46b71baac4c0497dfce2f770e8c5b : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7681483532d94bbe883a136b148eb730 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48fe30a4e79d4adda339401fc09cbeb2 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1efd7f1ed82849e596780d9b2d195130 : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/00b0ae2367324bfebf263e6c85cfab75 : SUCCESS in 8m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bfddbee840a44c9b9a470e3a5ebbdc4e : SUCCESS in 9m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a4f700538f404028934c9ed47b23c1fb : SUCCESS in 5m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/398f275e6b3a48069461017e66cdd677 : SUCCESS in 17m 55s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/577990063d72493ab3f42e75e848edbe : SUCCESS in 4m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/58627d97235d4fd8ada901eb9733242c : SUCCESS in 5m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/26f8f017882e4e3cbaaeec9de06bcc9f : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80c47c308bbd403ea68dab55d7fd8719 : SUCCESS in 12m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c3f33bbcf2394999acc9bfeed9f76831 : SUCCESS in 7m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/320bb238391f41a5ba87c09e5757b34b : SUCCESS in 11m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a94fd8438d5242a69a78baab42df3a6a : SUCCESS in 3m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c0bf9670c2cc443380fb3e25ae7e0fe7 : SUCCESS in 8m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/369068ba144e4862af837c6abb2f8a88 : SUCCESS in 4m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f114bbe636954790920c10227575f486 : SUCCESS in 24m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cbb5e3cf78e1464f91399025ddb063f4 : SUCCESS in 20m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8c088eeeebb5439ca7373c0004472390 : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d31a725fb4c5419ea5c2dff541f2bc41 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e41b174e4fe46b2ad5ee2ad6ef0991d : SUCCESS in 22m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/babe2f7a56e046f9b2f4624b81a616fb : SUCCESS in 6m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5f94e40eea38411fabdb98007e493f52 : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f516ba8e3fb44db69d227d28c5923afa : SUCCESS in 6m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7d9bf7caeb564cce852731710723c993 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/222403e9a05343dd9fa67bd9de315814 : SUCCESS in 50m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/442169e6d41a416a9a61b5d7545e1ecf : SUCCESS in 1h 15m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e5cb941c99675813214db983dd3a73978097442f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-12 03:13:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c377f38e652ac49a83793e599e5dcc9260322313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 03:13:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2581a799846683e0e3a66cc3ecffa9ba158194b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 04:44:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72aef9bb74cf457c8f0f8ead5e2aa262\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4bb2d68b90243b187ec1bf4c39ba45f : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0b8d60c850a4056b7bb793804d87a78 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0b3a556050f4b89b6ac3700b0ce47e2 : SUCCESS in 3m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ac3bb700a4148e7b993b836ecb03e40 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/097671697bf14456b62995e99cb5e05a : SUCCESS in 3m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4b34768bb76343a297f6f490bea0fd94 : SUCCESS in 8m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b3b4bcdfc77f432e8fa550452b5d8bd2 : SUCCESS in 8m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/591cf87f42b745519e90381da84e60f2 : SUCCESS in 12m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fb67d247397847e893c75b38aabc9720 : SUCCESS in 4m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/abc9772ea71f41ae93498fe526a703c1 : SUCCESS in 5m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e7935228f79146f682bc4f161e4667e4 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d621bbb979814d60ab0c15c499a73326 : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d3a3b764cba94dbcbc113e86ca5c5593 : SUCCESS in 7m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/06cd99452619455085cc8a09ae3602ff : SUCCESS in 12m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3b4377e80eec4da9a2e398a3088712ad : SUCCESS in 3m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9f4481047481405e8200484fb94bc431 : SUCCESS in 11m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3e45140b76ab4463840d403477234142 : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0bda779163b3473cb038833a447d0155 : SUCCESS in 23m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b3f2258182ac4c46aaa4c8a475deb683 : SUCCESS in 13m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/823f6929bc1b42f38c32457432c10472 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/19784cb332824addb866382823d0859f : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d61dd7482b274d44b7f15d5498ee3c28 : SUCCESS in 21m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2186eae4696e417992fb05b0dffc0a95 : SUCCESS in 5m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4dbe3b45c8a64edf977f59dd1476a0b8 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3fd7278b174d4d16a9c71684fe85efb9 : SUCCESS in 7m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fb283db005084b999ad960b3cc2d086f : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16b716a8b6614916aeafca11f89201c2 : SUCCESS in 59m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0afae9d64fbc429086d2cc1e019f3a6d : SUCCESS in 1h 28m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3e9e9a9117a646857dc17d38eb006b151b6d1397","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 04:44:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d58fd2b0e1258462029db6999fed6a828b99f75e","revisions":{"d58fd2b0e1258462029db6999fed6a828b99f75e":{"kind":"REWORK","_number":1,"created":"2023-05-11 09:28:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/882933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/882933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/882933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/882933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/882933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/882933/1"}}},"commit":{"parents":[{"commit":"70106ea80ce77cc6425ca09790503ba03be14618","subject":"update constraint for python-manilaclient to new release 3.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70106ea80ce77cc6425ca09790503ba03be14618"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-11 09:28:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-11 09:28:08.000000000","tz":0},"subject":"update constraint for os-brick to new release 5.2.3","message":"update constraint for os-brick to new release 5.2.3\n\nmeta: version: 5.2.3\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Change-Id: I6fbe9525f7f0262d25bf9a7979912f5fd8ce05e2\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: Iffbfb7d62ae53a49d3f4ce156cfa602cea73e29e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d58fd2b0e1258462029db6999fed6a828b99f75e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d58fd2b0e1258462029db6999fed6a828b99f75e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
