)]}'
{"id":"openstack%2Frequirements~883036","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I89ba1699b2742066ca5eb8e46e73f6954df869e5","project":"openstack/requirements","branch":"stable/2023.1","topic":"bug/2004555","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-05-12 05:50:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I89ba1699b2742066ca5eb8e46e73f6954df869e5","subject":"update constraint for os-brick to new release 6.2.2","status":"MERGED","created":"2023-05-11 20:39:51.000000000","updated":"2023-05-12 05:50:59.000000000","submitted":"2023-05-12 05:50:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"883036-bug/2004555","meta_rev_id":"de93435cfe44f25e78d980aa078f918aa8615c89","_number":883036,"virtual_id_number":883036,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-12 05:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-05-12 03:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-05-12 03:13:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-11 20:49:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-11 22:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-12 03:13:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8efcec4c0721177aeb62af132155f711980335d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-11 20:39:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36f850e7e6534f8c69e1f9d28b071c50d2082a29","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-11 20:41:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eebcf84665da4c7c193ade92fda48477ffa30c39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 20:49:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feb7148fbf9541f999b75aaa85fc19b5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/271afd6a66004a3289b8282450f124a0 : FAILURE in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fed081495407c12249d00520497fabc66d612c9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 22:19:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ea1336aac55449387962206dd9fd3ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ff0c5a5b1e4b49a0c9fefeb3a867e8 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f5b84bb7e484c1983e29f26315e58b6 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ededdba8e6443839e180a58d807dca6 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15c5aee264bf4c91a1ad2a90f8d8dd5f : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/efac7d68f2e24a83b118c4a381379e63 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5883090a622a4300934338fbddf3d772 : SUCCESS in 7m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a00c839f698c49e9bbe5d629a4497b13 : SUCCESS in 12m 59s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/534b17c2544d4da89788eaf4ab519487 : SUCCESS in 8m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/77564e3c0cc94cecb34468be5b9b3bb9 : SUCCESS in 5m 21s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f041bffe9a1c4c15bc4b389cbaa952d6 : SUCCESS in 6m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7c8e16fde05d47af926727195056a9bf : SUCCESS in 10m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d6d0f533e21f4739864ce2475b6d09fa : SUCCESS in 4m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/20f4e6114b1c4fbfbd257b515e969e26 : SUCCESS in 4m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6824124c36744b5290899c1818972cfc : SUCCESS in 7m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1dc36fabbe0045579f881ac21af6d5fc : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/549259f3e892467389f0be7bd80f72e7 : SUCCESS in 6m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/eaa12d6ea64c4692b8974b4f3715ec66 : SUCCESS in 11m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/51ef4ab9a3164c71af94c7a94a865b54 : SUCCESS in 11m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f5fd539fc2e443ce8ae982e0c6d439e0 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3d4cce47cb7e437ea5cb2656a252ec90 : SUCCESS in 9m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/db2c58ffa59b44039baaaca2e0c63f80 : SUCCESS in 4m 03s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7c89e1a2c5d74c6d87ac0426778d3f65 : SUCCESS in 9m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d67ef3091adf477eaae94a05c8a8a0e3 : SUCCESS in 23m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f4247c291bdd43f1a9f1e886bb6850da : SUCCESS in 11m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2073f258627e486e96af64d5ff3661bd : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d99e331e6f74c35bb3086cefef01e40 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e40aebd697cf462180e1cce6d4c66c17 : SUCCESS in 24m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8b907ece062342898f4e04dd25b7f523 : SUCCESS in 5m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c7826569336e48448a7551e98360f0ff : SUCCESS in 7m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58b4c31d71bd44f598ca72debe512b43 : SUCCESS in 9m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/46548a1ee2af45dabd2ee8c9e411a8ae : SUCCESS in 11m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7668b20df2b44380b2d6db1ba13135fa : SUCCESS in 53m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edddcfe418f54c8ab5090066ffa86807 : SUCCESS in 1h 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4cc96391f58b6f592695fe96d5a1e5a149c72c13","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-12 03:13:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f67b61442c900706f9b20b573be3754b1c5309a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 03:13:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96063eca84882162787e17f4ce3cc880f51f7de8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 05:50:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc1027be791e42d28dfafd0475ba0eef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb54b7eaa994119afdbdcac2cddae01 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25c40431e1ab4f8398757d8f8ab2aeff : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2be5c66fb9954902bf924107b57af45d : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5050a5058df4eb791867a840f13726a : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1249017a290405a962e4228ed4fc815 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/21f5a6f34057443cb5a6d6decb8c5f97 : SUCCESS in 11m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2b83dd1ba28b4bf9b5dd5b6f1ca58027 : SUCCESS in 11m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ba885bc7ab314ac9aa401d38c2cbbdca : SUCCESS in 10m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/721f232d37954f4f930c030ace349e8b : SUCCESS in 4m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1bf608322f634ac1bf0f8bfdc678c81f : SUCCESS in 4m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2e09af2b1cc14187aa72f60435ed887f : SUCCESS in 8m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0cdbf9f76f8344bebe935d5eab0531c2 : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd51980c4e3649769f8fe6ad62de362f : SUCCESS in 8m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e0b6aa4f85184e48a23572700de75d74 : SUCCESS in 8m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2d27602830b44a788e55d1728f3ad856 : SUCCESS in 14m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/30eac3e51d8748e4afd4ea50d7946a11 : SUCCESS in 3m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/70488f961c56464ab4b7fed666372f76 : SUCCESS in 10m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/123bcf83101546b68ac6802c849693be : SUCCESS in 4m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/36b9e888f2b041b8b487d1ab3f68db2b : SUCCESS in 25m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8adca68eec804ebcb714f00b23f756b2 : SUCCESS in 13m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f9c2d914d32449dea0c6fe484bfaf9dd : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/25b46d25d80a4e278b5037615b6b89bb : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0f8b8cc0a0647da96e9b0eca1628ae5 : SUCCESS in 23m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8fa2548e5dc54baeacddcdd74a4c8746 : SUCCESS in 4m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/770ec263a9e544029697e8de40f66d18 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/19cd8d0be88442e3a897cabb98c2d035 : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/871bb84467c04eb08053b944a95f5245 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/525f808eef394102954a2b3dab59f6ca : SUCCESS in 48m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/501c0745c7134b088eb6b53d2417a6ed : SUCCESS in 1h 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"de93435cfe44f25e78d980aa078f918aa8615c89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 05:50:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92e17ec0794879e010ed4ba754b5ee1246bfe917","revisions":{"92e17ec0794879e010ed4ba754b5ee1246bfe917":{"kind":"REWORK","_number":1,"created":"2023-05-11 20:39:51.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/36/883036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/883036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/883036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/883036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/883036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/883036/1"}}},"commit":{"parents":[{"commit":"b1ace5c1dbdb9b0cf49b9ba0ad2687b60827219e","subject":"Fix overlong line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1ace5c1dbdb9b0cf49b9ba0ad2687b60827219e"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-05-11 20:33:28.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-05-11 20:33:28.000000000","tz":-240},"subject":"update constraint for os-brick to new release 6.2.2","message":"update constraint for os-brick to new release 6.2.2\n\nLooks like the bot is not keeping stable/2023.1 up-to-date, so\nhere\u0027s a manual update to allow os-brick containing a fix for\nCVE-2023-2088 to be installed in 2023.1, and to bring the bot issue\nto your attention.\n\n(The meta below is from the bot-generated update to master,\nchange I56e2c5f3de66d133.)\n\nChange-Id: I89ba1699b2742066ca5eb8e46e73f6954df869e5\nmeta: version: 6.2.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Change-Id: I6fbe9525f7f0262d25bf9a7979912f5fd8ce05e2\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92e17ec0794879e010ed4ba754b5ee1246bfe917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92e17ec0794879e010ed4ba754b5ee1246bfe917"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
