)]}'
{"id":"openstack%2Frequirements~883139","triplet_id":"openstack%2Frequirements~master~I0ea9448614da35bb9ad119e1f57395bc8aef7875","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0ea9448614da35bb9ad119e1f57395bc8aef7875","subject":"update constraint for stevedore to new release 5.1.0","status":"MERGED","created":"2023-05-15 08:20:04.000000000","updated":"2023-05-21 16:54:06.000000000","submitted":"2023-05-21 16:53:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883139-new-release","meta_rev_id":"15564fa322dc94ae5462511d0a9adfd657fcdafe","_number":883139,"virtual_id_number":883139,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-21 16:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-05-21 14:59:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-05-15 09:04:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-05-21 14:59:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-15 09:04:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-05-15 10:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-21 08:12:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-05-21 14:59:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b045c5b87f59bb8b2e828eddebde175c9687ff6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-05-15 08:20:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60bd9e06d2aa9cd2dfe1cdc66bae8b92890a6e89","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-15 09:04:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0728456cbf21f481b1066956fc76107b2f6a46fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 10:07:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75d2bc116bed451fbc392af1dcbc2f17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c65352bb5749efba7f85848cb1db83 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ac296d3db94a5aa0d66b8d51121aef : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5799653d714a425aa98cbcfcdb1be6e1 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b910182e792543d5b24b9b9a558ed424 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0134f9381f544449b196fdd932bbcd1b : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/284501c60d984fcaa8b547fd0b8a02a7 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a904bab72c094d85b8ee57fab0f4890a : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d2b2b053a79f44fba1031bede15c3d83 : SUCCESS in 10m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dc2e060895db4b8f9ffe656d037c9e4f : SUCCESS in 11m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/226ef513f652482eb05e6bff9feba870 : SUCCESS in 8m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a745fe8bb5c4461da98bd62b86b9826e : SUCCESS in 10m 18s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a29978113dac408a998468a2a1d77c44 : FAILURE in 3m 23s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1cf1a554a0064146bc29afa6462b9eba : SUCCESS in 5m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c22c86e41d614d72bb16867374a1f2d9 : FAILURE in 2m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bef6282834df40649bf20bef129e83fa : SUCCESS in 9m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a1006c0538724fe4b29d2e67417916a4 : SUCCESS in 4m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fc03ccabcf614644970eb6fb407703dc : SUCCESS in 4m 42s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7fbbb4785bf84d4386221fec03c320a7 : SUCCESS in 7m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/57ed0199c9754d3685cd4a7731d369af : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b70ab1289e5b4e94899de4be970f8c50 : SUCCESS in 7m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d0b710db2a5c4703a6ab258f8457e692 : SUCCESS in 7m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4d7749498381430ab618c349d7220571 : FAILURE in 3m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bafda8f4299f4032a86fad8a86efb2d9 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d82727171555477281fe066ef3de56b3 : FAILURE in 3m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bff2177b38aa4946bb1c6ae19c301817 : FAILURE in 2m 49s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c71447eb3afe410793b845aa29564bdf : FAILURE in 3m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/390ed99dbf304211ab204f21126d48f0 : SUCCESS in 27m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8fd889a5e4ca40908af2802a58c2e791 : FAILURE in 3m 38s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/55f49113bbdf4704806cddc229c1f6f2 : FAILURE in 3m 45s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b6f237c67cc44dbe9c853b28d85d20de : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c08ba0c3139406e91cb8f1a27502d54 : FAILURE in 2m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8bcfd24e524a44f891778e2e84b8fe05 : SUCCESS in 23m 17s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b6518f476b9646799009d03e2fabfc75 : FAILURE in 3m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/746ad6fce0274ac48868e527dc90bd35 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/adca28810cad4b2cb507c2553ce9f655 : FAILURE in 3m 06s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4acf1879a3c7458ba889175e4d66a37a : FAILURE in 2m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edf5a115b72549069e88fa972d95a7be : SUCCESS in 59m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc41e1e17df84585ba90e85d448181e5 : SUCCESS in 1h 39m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0e8165979b2ccf9c1265a51f737ab34500156ef8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-21 08:12:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daa55bc54e447d585216cab0fe40c3ffeb9511a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-21 09:50:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4500cd173a9f41eb850ddf5be39e7268\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e367ebedb74e60b5517bc9f5ac40fe : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd60d3c271f64da68d384c8f45742551 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16d6cbf211534f93bed86678ea0f4d44 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c848cc3021f0437ba09cf2c9c2b8053f : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e83827cdbc0141b88f9ba074a111d767 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8401efb73bb141c2b558e639171d1632 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9300cac5406b4521bd38be03b0b6c745 : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/088b04bd47ba43999f0a48558b818a66 : SUCCESS in 7m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7a174d8357a144e59edb64d6ada9b845 : SUCCESS in 14m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9b116788b0e94a3f9080bd7c07999b9c : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f3ee6df043dc4675b0f4146aeb48a377 : SUCCESS in 12m 40s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e86b0435ebbc43078af3073ff141d2bf : SUCCESS in 9m 27s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/8ba03d977c204419958fa18c7235276e : SUCCESS in 4m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f016cc26a16b46c88c587e0ea99f2a5f : SUCCESS in 4m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ef883e4690684a9cb5430f6cd8ca0522 : SUCCESS in 14m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d77200b713cb452da5c4d3b4405761b4 : SUCCESS in 4m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/919ac60d47424943a3c59adbd5ff38d7 : SUCCESS in 5m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/cb9137ad55fb4854b645f94ebdd90228 : SUCCESS in 6m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/32d9c87c010a403396ef71814b147b21 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c42f8f2b8374ae18090342c95fd0402 : SUCCESS in 6m 54s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a2fab0a27a1d43d795b12698ea678a35 : SUCCESS in 7m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/33bca05d9a484388bad8f7c46395642b : SUCCESS in 15m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ce1e9910e4e94ff2a8dbd129f432c17b : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9b9f26a5fe8e4a659a9063520f23133f : SUCCESS in 11m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bc87d0e6eb784104b058257fd7f8f7ab : SUCCESS in 4m 21s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/528c54f94c124198bb1ece240f629d62 : SUCCESS in 10m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2e0140f805ec4046ad86521c4e902f25 : SUCCESS in 34m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d6fde51070db410aa33a414862e7bb9a : SUCCESS in 10m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bba988ea474143fcaee69db472845e1d : SUCCESS in 13m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c3369d03e80f48c79551166c9cea0702 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8f6998f4316450dbde9396b74c8b912 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0a74881985d4f2f89dff36dc72a263c : SUCCESS in 21m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/91168788710e47a7bad980b7a30b71dd : SUCCESS in 5m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b17f5bf127494cee8149f9c491b1ca3f : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/397ade3bf04d412ea13c6ecf26fd4105 : SUCCESS in 6m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/8712ce3c56b743c5a2b0274abad4f06c : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a2577bfefd741af9d6f70ab8827c1a0 : SUCCESS in 1h 02m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7112476bf5a4cb0963c521eabd35ebf : SUCCESS in 1h 33m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d0da27fd43284dfc899141ff22798266fab0f28d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-21 14:59:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6382e4bd7837401380bd075102998115173bab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-21 15:00:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99dca404425fee638f9a8e80660925af552883a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-21 16:53:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4501c3ecc47c49c083670e66539c41dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969fc040381d40b2bd2bc188daacb1a7 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99b5951772c43d2aa8a29ee10bad5a3 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30f03f8bd8ca464788b7d4304c4b2bb9 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11ec537d4b2b491a9c0839516a4a1c9b : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/443e099c022144808169f2aee08c443b : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c47aa81e4dad4fc4a30c4153dad16df0 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0dd5707be664dc6ae26b21b1425c866 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85c8201c48df49789db43041aae7382c : SUCCESS in 7m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5ad3930f2a184d76bfd642aa73f086a3 : SUCCESS in 10m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/59e2e7fb33564e71b6f1916f71620949 : SUCCESS in 9m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3a29f8b572bd408c858ccca8add2f573 : SUCCESS in 10m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/032deed48a63489ab5589ba4a4eb5366 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0a7ef59e9b544b0aae49f101ec514128 : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/db120e9f80d4461584eb8d26a57a0930 : SUCCESS in 7m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9d0286c41e3641a08f45b99542e276b2 : SUCCESS in 7m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/64b1c8c93b2a49d3a04f7c14d97663ce : SUCCESS in 7m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/18c79d1df30c4c519460f1481e9ae4c4 : SUCCESS in 7m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e50a173f24f944cab32b982a0d9b009f : SUCCESS in 13m 39s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1d1c81cbc6364d67a4cfe72bd0fdf4d8 : SUCCESS in 4m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/50cef898d3ab49c687022853c6466c44 : SUCCESS in 11m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ba039e3d59d348d58c93f9020094b9d7 : SUCCESS in 5m 06s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0bb6b4a71610402ab65d941b44d71a26 : SUCCESS in 25m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fbcf78b02c87473283bafa73a5834cfc : SUCCESS in 10m 30s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a14fd3ed68124f75805680c5b63fccdf : SUCCESS in 14m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/940945e8dc1a4729a53006693268495f : SUCCESS in 6m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c31bb4d8b203419498e5372c49daf379 : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b58c59e8281c40eb9fb8434d48503588 : SUCCESS in 24m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/35150a39a7fa4a5893056e24f7bec895 : SUCCESS in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9b1094068d3c4c4b9cfdb435c3ddf054 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/319ea0522b5342b1962524235ff05ecb : SUCCESS in 6m 17s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b5f5903e76a74477a49862cdb22d1a24 : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b5a264b25e1417b996f2e3672035d02 : SUCCESS in 1h 10m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33ae712caef04fceb1be4283cf1674c0 : SUCCESS in 1h 45m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b0da1518d6bf4bafc6b33a3b3596016c015e163e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-21 16:53:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"15564fa322dc94ae5462511d0a9adfd657fcdafe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-21 16:54:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd09b6fcd7a3420ba9e545a4e5c320c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b24acf353b48ecb5630206caa4e479 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0e48e507bf4c18b05b7bc1b691784e62b8bd51e","revisions":{"b0e48e507bf4c18b05b7bc1b691784e62b8bd51e":{"kind":"REWORK","_number":1,"created":"2023-05-15 08:20:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/883139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/883139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/883139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/883139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/883139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/883139/1"}}},"commit":{"parents":[{"commit":"f4238c0a45d422093636b9682f9085f3e89d9481","subject":"Merge \"update constraint for os-brick to new release 6.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4238c0a45d422093636b9682f9085f3e89d9481"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 08:20:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 08:20:03.000000000","tz":0},"subject":"update constraint for stevedore to new release 5.1.0","message":"update constraint for stevedore to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0a5de85554044658334bd6c93882646173f41589\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0ea9448614da35bb9ad119e1f57395bc8aef7875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0e48e507bf4c18b05b7bc1b691784e62b8bd51e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0e48e507bf4c18b05b7bc1b691784e62b8bd51e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
