)]}'
{"id":"openstack%2Frequirements~883140","triplet_id":"openstack%2Frequirements~master~Ie1acadb1149c8500e3fdf31688a0e0da7f483e8e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie1acadb1149c8500e3fdf31688a0e0da7f483e8e","subject":"update constraint for taskflow to new release 5.2.0","status":"MERGED","created":"2023-05-15 08:39:05.000000000","updated":"2023-05-15 14:21:21.000000000","submitted":"2023-05-15 14:19:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883140-new-release","meta_rev_id":"149008dc6fe1571035f1f3e4f75a93591b787ab9","_number":883140,"virtual_id_number":883140,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-15 14:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-05-15 12:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-05-15 09:04:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-05-15 12:27:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-15 09:04:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-05-15 09:53:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-15 12:27:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"5217323d2967ec178fef604e836b02b76f5d0718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-05-15 08:39:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0210e74274760dde8cd9f0eac561bd6bee7b6a36","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-15 09:04:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81c14943983768c1680ffb292413f38750d7e143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 09:53:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85a1cee2262e4f25ba7b670c4bed798b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b410c6d55df47d6b08bbe1adb27e000 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b78ad2c0af4d27be053d8921c6f162 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a9ad52269ea499db9fddc48a4dc2048 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97270c217f554a84b2e094e94870c76b : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80153f0f781c4f7ba922422a5b4bb565 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bba9096586d44925be02026ad34e4317 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b16db35a31643b6ac851fb34432335c : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aeb1bffef40c46e9947089db3fa4ef4d : SUCCESS in 8m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ae92ebd779954f318516f08b9616a25b : SUCCESS in 11m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a28f3914558e41358153066d820e260b : SUCCESS in 7m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0782549a270f4b1e9d596bcbc7c819dd : SUCCESS in 14m 26s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5febbf54b51545569291b75b1496ee3a : SUCCESS in 7m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/619ba2b07285409ba62050db28017940 : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/14690ebf629e45c7807a66482e07adbe : SUCCESS in 5m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0ac0371c153f42159d944590b05d1097 : SUCCESS in 11m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1f34213ae4b24085ab222345d95384f6 : SUCCESS in 5m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/263b4a2685c045d385702dfcfbf94468 : SUCCESS in 4m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/92c2f2bea8b0494ca995ba827eae420e : SUCCESS in 7m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bcd6d5799d4d4fcaa36bc97385ec98d5 : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f69ac741ae341cb8082d6bf19622ae7 : SUCCESS in 6m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d9641d3a872346c5af4f16082ef7f8b3 : SUCCESS in 8m 30s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5ee77bd38a94407295a0f3c4b91824ae : SUCCESS in 10m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4f6b8882078f407da22f1bf985300f75 : SUCCESS in 3m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/41eeb01c04204e828f0630a316c96179 : SUCCESS in 9m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/463fbe2428ec4ff88913f57d5797647f : SUCCESS in 4m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/04eb37f3b552440fad09614c7034745f : SUCCESS in 10m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a481831a9305401880abf3271cbf02f1 : SUCCESS in 22m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cd9c98a5b124445eb7b3bf14b9691bd0 : SUCCESS in 10m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/66b1012788a24173a9f681a4edf77cb3 : SUCCESS in 11m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a0536987ffcc496da4d63d6eee89a3e8 : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/24f0a729a8044b40a15879e12dd0c0da : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1047627fb6124ab0b5d4dfdf7888f0fb : SUCCESS in 21m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9f6166de31894aa8af6f1b22d875d6fb : SUCCESS in 4m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/66520dc1cd7f49348ff3dbbfca3796bd : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b7cad74bec341e58317a84859e4cf01 : SUCCESS in 5m 17s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/cfd46879c56c4260a04004ece4372aae : SUCCESS in 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a8be4f4868749a9bee6cf9806e05e0f : SUCCESS in 52m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/853ad5b542a542e89f93c7d98cbfb624 : SUCCESS in 1h 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"fea82def8fab7ac81ff2bc5cf6996473a37a41a5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-05-15 12:27:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32a46c1196c3831114c0fa6a0a4ac15717bccffd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 12:28:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"512f20520f42ded9928c603bae010274e3435184","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 14:19:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd06d16ab5254949a1570c0377223bd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d390848e3149088d9a58b049d47754 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/323868396659453eb39468d53f2d4ff6 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2b32116c5ea469ab21ada1731738a5e : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bde1f5239f774c31ab8b2c535be878a2 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1f5f723afdb462482d5e3b1daf8a0e1 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1a5d8291d2f462b87480362d04501d4 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7efdf8e05bad4d9fb6813a4e2d812a14 : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f24271f237e543e198e9e6c3b38873d9 : SUCCESS in 11m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/90ece5b270f34357b23c0395a9a676b7 : SUCCESS in 13m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/78cde050ce934a6a9a54b5dff86be092 : SUCCESS in 9m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/256aa81d8ebd412485f86c2e271a1ace : SUCCESS in 11m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a1e05619e35648aca2cc467c64d00492 : SUCCESS in 4m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0f7f4322b0484ccc91027ad420f8fbae : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5f15c553a6714ed5bebabfc8bd13492f : SUCCESS in 8m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c9d63dd2e3de45e9898979639fc77277 : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b8d1dd0473740ec89e8dda991e2c343 : SUCCESS in 9m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0ff74e6d213843ed9146e0ae4c9476e3 : SUCCESS in 6m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/80049ba88633435fac4d027863cfed12 : SUCCESS in 10m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/230919cf52474adcaa32e7d1af4250d9 : SUCCESS in 3m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8f7e9434f2b1436aabfc6b70878bfe05 : SUCCESS in 9m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/54a6a08a4280454f908744f6de84ca31 : SUCCESS in 5m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8ed5082bbd364a75a25df5076b289a36 : SUCCESS in 27m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/62d246270154417ca3cb4a07f55b75b6 : SUCCESS in 10m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1682b80efb214f3da87cd5392ac73263 : SUCCESS in 12m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/69dd21146e42442ebbc7e979d9168e47 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/19ba01a9f3924a838960ff1edb7a42aa : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfd9c7b93e184231acd8c6b6b4f3094f : SUCCESS in 19m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8c20e33f416945aa8a3ad3006f5105df : SUCCESS in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6103a0589989449ba239ae15b77910d2 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79b0f65bd9d34f8db1fbcc856c370cbd : SUCCESS in 6m 32s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a66e8a33ce5b441ab9b82d93c685ce2b : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcb7196ee9ff48b3a2a6a7fdce08ab9e : SUCCESS in 1h 01m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddea048957d649059ca255fbdee23144 : SUCCESS in 1h 20m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a067039acf2045f89e227e792e1d6fd4ac5c9923","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 14:19:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"149008dc6fe1571035f1f3e4f75a93591b787ab9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 14:21:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc09d40b22af44c9954e756e6865a941\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded85d54a4ba433c8cb33df2534687e3 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31d44f38b22ebf8c426efce108e004c8532261c9","revisions":{"31d44f38b22ebf8c426efce108e004c8532261c9":{"kind":"REWORK","_number":1,"created":"2023-05-15 08:39:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/883140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/883140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/883140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/883140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/883140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/883140/1"}}},"commit":{"parents":[{"commit":"f4238c0a45d422093636b9682f9085f3e89d9481","subject":"Merge \"update constraint for os-brick to new release 6.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4238c0a45d422093636b9682f9085f3e89d9481"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 08:39:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 08:39:03.000000000","tz":0},"subject":"update constraint for taskflow to new release 5.2.0","message":"update constraint for taskflow to new release 5.2.0\n\nmeta: version: 5.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I77967e5e92e304fffc815ccd7f9f7f7fbaa83aed\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ie1acadb1149c8500e3fdf31688a0e0da7f483e8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31d44f38b22ebf8c426efce108e004c8532261c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31d44f38b22ebf8c426efce108e004c8532261c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
