)]}'
{"id":"openstack%2Frequirements~883141","triplet_id":"openstack%2Frequirements~master~Ide706e2cd151a074024e4b1e60465718f59b6c9b","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-23 06:45:53.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-09-23 06:45:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ide706e2cd151a074024e4b1e60465718f59b6c9b","subject":"update constraint for castellan to new release 4.2.0","status":"ABANDONED","created":"2023-05-15 08:47:59.000000000","updated":"2023-09-23 06:45:53.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"213d152cf6626487f07a69ac2d21586f2ccaad5c","_number":883141,"virtual_id_number":883141,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-07 13:28:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-15 09:04:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-05-15 09:26:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-21 08:20:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-02 17:56:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-09-02 17:56:32.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-21 13:34:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"ea4807465ffa4d3e7514bd9002ac2d2b168cd2c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-05-15 08:47:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ce93e3a13bcd2692890bbcb864590c20a265c92","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-15 09:04:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c63977cc1b0256403f661368b66cb92ae48e207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 09:26:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fb21052ddb746e2b33bc5f45e22f32f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98be1bc69ac4c7394875bbbd709cad0 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02e52121c1fc4c1d996e5b59a743171c : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c88fbccee48249c28cc5b3fc0bcc8811 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91d969145edd4d74b687d0161a11586d : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d59fffd8f0b47dfbcfe2aca633ecbd5 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24b24c42b661459887259368c09053fc : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d218f2bebf1e4c61b7b2cef05c11da82 : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/33ce31b110584349a02860b873f90613 : SUCCESS in 10m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fb0cb9b2b1ea4affa6c103560e0c6595 : SUCCESS in 13m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f53d8e17d8c44c68a65b603da48a36f3 : SUCCESS in 9m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/582373b78fd04ff1a746de7da04ff5ed : SUCCESS in 9m 36s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c6964a413586419d8f964a33dce3f917 : SUCCESS in 7m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/aeb1d38ec05e41e2ab76d20f9eb4d39a : FAILURE in 2m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0fafebd0688649e98f04fecf2539ed09 : SUCCESS in 5m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2dfac0cff17d48d2ba6319f39b5ba03f : FAILURE in 11m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e22d73000f0f4e16a2476a2afa29cd4c : SUCCESS in 4m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c146312e49b4422b8f2c37dc11bd0ecc : FAILURE in 3m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fd242de0536548b693a5066ff702cb92 : SUCCESS in 8m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/703c7efaaa1746d7a935284e8b6e603a : SUCCESS in 8m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56320829ddbe42d6a196bad83490bd67 : SUCCESS in 6m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/39099e976f01429798d2425b0572d626 : SUCCESS in 10m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3ea0e48988294265836be92e96369679 : SUCCESS in 11m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/924f93908c3c49a09d1334788b1dfcd4 : SUCCESS in 3m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/33fdf6a2efbc4c22af20e08b134a1b37 : SUCCESS in 17m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1911e55fbc19464ea05d5aad39af9e1c : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/eb680741a55c4dc595ba6fea09b07b51 : SUCCESS in 10m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/79a1a1f6a1694354a766704aa186ded4 : SUCCESS in 30m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2f429868f52448359b194f056a9e341a : SUCCESS in 9m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5af5a3196e6e4ede84bd41ef6c8a524d : FAILURE in 16m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c172c218ff874959a17eb66cffdaac90 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf54ebb5fc8d440d906b13d100b83b66 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a7b45a0301f4979915291ad04cf56b6 : FAILURE in 24m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0679f3368e4c4c22be3ef86afd8cd00d : SUCCESS in 5m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d49c5b253a244793baa8e3f03c6fc7cb : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ebae86bca444186aa5423b1b3dbdde9 : SUCCESS in 6m 30s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a2226e7f4a9d4f20a52397745c2d0a12 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10c0e3df32554a82852dc63833c011f7 : POST_FAILURE in 17m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5525bfc2f6d4aac9201928625cff2b6 : FAILURE in 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"bfb73b4a6b820837f2388af937eab315e68ded0e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-21 08:20:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d7d763006672786a6f38a29ea50c7ae27bc2e26","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d851ae23154cbbe885cd08f35fccf9f519254ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:20.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 883141,1","accounts_in_message":[],"_revision_number":1},{"id":"8e1c186d447fb69ee35fd8319786313efc41337b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:21.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 883141,1","accounts_in_message":[],"_revision_number":1},{"id":"8191237b51c8fa694b53cd55acbb150477fc9f0b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:32.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 883141,1","accounts_in_message":[],"_revision_number":1},{"id":"9f7741a3bc9a36d4a67adfd2d029bb3b1a83426f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-07 12:52:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"35bf7fa28157326040ce1b5618d0dc0a979614fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 13:28:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66638da29e4340f9adfc15db5c8e16af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a569f27cf6641f8866176e8f7e74926 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7b379a555cd43edaf4344767c1b2da5 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ce7894c7a09446f8d3bf8825d502cda : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc289b2506b342abaa9d65fa38948903 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b23e59d8157f44a89f159e723a5c196f : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/403cfa10de1f409ebf5f53d290457bd8 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c6a2f55d7e14e26a4290e93d8d212f4 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e3bb2e92dad4911be27eb1fadaf987b : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/01b03e20123e40e9abef318a1ee26ee1 : SUCCESS in 9m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cc27c4e05abd4b77b6cbaa507bc4cb92 : SUCCESS in 14m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dd1916491b9848618274437cde672570 : SUCCESS in 9m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6b708bf21cec401185e740a61c2281d9 : SUCCESS in 14m 24s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b0d20afb3ad3419c859bd16dc7622bbd : SUCCESS in 7m 28s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4402c93a7dcd4c25be9e57a93713d8b8 : SUCCESS in 6m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1a5a2188b5f34574bcf505f0546f8dc0 : SUCCESS in 5m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/68cc84aa68ee4cdb8ccd4a5e865d8998 : FAILURE in 11m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/83a0928b0e2e4edf96fd402bfd53968f : SUCCESS in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c1d944f6839b4222be78efec278e1b2d : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/96ac29de04e54f52acec3d810d9feaa8 : SUCCESS in 8m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9fa82de229624a65943426cff215e51e : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96c667c638f74c5b8c1c4dc3abef2e60 : SUCCESS in 6m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/877b821d32e54845bbf00c1634df8665 : SUCCESS in 8m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/29c3cae301c148c7b190b10c4942a8f0 : SUCCESS in 10m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/30dfadf763db4a758665543673841f29 : SUCCESS in 3m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9918ffd2159f4003b57427e3193e8d94 : SUCCESS in 8m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f5a40686a39e4afebe3ed22dec20954e : SUCCESS in 4m 33s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d5df0f92df6e43d09ee3314bf732f84c : SUCCESS in 13m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3e5cdcfe0bdd4547beb36575bb09b552 : SUCCESS in 23m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ce98b31d0df045a886b256159d03a0db : SUCCESS in 12m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f951b473faca4724a0c55085a21eb455 : FAILURE in 10m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e5c6afcfa0ac4c1b861a9ec6f0d162b4 : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a58c0f3162174a0f9b0e74cef4c78317 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aab2a6f8d82d4d20ba3333e81609f54e : FAILURE in 23m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f4099d7399b147a796b6b4db4520005b : SUCCESS in 5m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/af4b4613ec024f0aba5814ce08975247 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad2e37c8941b49e6adf36c202b9b8397 : SUCCESS in 7m 03s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/26ffb4fef35c4caaa6d320bc0b0aabb3 : SUCCESS in 13m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17f3bacb4f6d4f2c8a20d09120d2856e : POST_FAILURE in 23m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f8a9f6aed3d464792a0c9419ffadac4 : FAILURE in 30m 33s","accounts_in_message":[],"_revision_number":2},{"id":"347f6c9d071b7dc3b6c47530c72ea8892676bd80","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-13 13:32:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"845fb73b0eed97fd04b632c59c296b248e5d9644","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-21 13:34:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"213d152cf6626487f07a69ac2d21586f2ccaad5c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-23 06:45:53.000000000","message":"Abandoned\n\nsuperceded by 4.3.0 https://review.opendev.org/c/openstack/requirements/+/896254, castellan should release 5.0.0 from main early in C","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a026396bc759a746db54bd96b0368b387ac0986","revisions":{"43c28ec53314076d6d53582234d175870748593a":{"kind":"REWORK","_number":1,"created":"2023-05-15 08:47:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/883141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/883141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/883141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/883141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/883141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/883141/1"}}},"commit":{"parents":[{"commit":"f4238c0a45d422093636b9682f9085f3e89d9481","subject":"Merge \"update constraint for os-brick to new release 6.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4238c0a45d422093636b9682f9085f3e89d9481"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 08:47:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 08:47:57.000000000","tz":0},"subject":"update constraint for castellan to new release 4.2.0","message":"update constraint for castellan to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib833fe8a73158723cc30975cbab98248295d5bc8\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ide706e2cd151a074024e4b1e60465718f59b6c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43c28ec53314076d6d53582234d175870748593a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43c28ec53314076d6d53582234d175870748593a"}]},"branch":"refs/heads/master"},"0a026396bc759a746db54bd96b0368b387ac0986":{"kind":"REWORK","_number":2,"created":"2023-09-07 12:52:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/41/883141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/883141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/883141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/883141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/883141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/883141/2"}}},"commit":{"parents":[{"commit":"237d49d5ba63b817eece6376a284e990654f4ea5","subject":"Revert \"Add Spinx RTD theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 08:47:57.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-07 12:51:53.000000000","tz":120},"subject":"update constraint for castellan to new release 4.2.0","message":"update constraint for castellan to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib833fe8a73158723cc30975cbab98248295d5bc8\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ide706e2cd151a074024e4b1e60465718f59b6c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a026396bc759a746db54bd96b0368b387ac0986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a026396bc759a746db54bd96b0368b387ac0986"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
