)]}'
{"id":"openstack%2Frequirements~883143","triplet_id":"openstack%2Frequirements~master~I58b9356999e81ab2a5116f3ff573d205a05cacfa","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I58b9356999e81ab2a5116f3ff573d205a05cacfa","subject":"update constraint for oslo.policy to new release 4.2.0","status":"MERGED","created":"2023-05-15 08:57:39.000000000","updated":"2023-05-15 17:39:49.000000000","submitted":"2023-05-15 17:38:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883143-new-release","meta_rev_id":"8943c753a3bbd91bf213ebab3bda46358e3dcde7","_number":883143,"virtual_id_number":883143,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-15 17:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-05-15 12:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-05-15 15:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-05-15 09:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-05-15 15:52:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-15 09:04:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-05-15 10:26:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-15 12:26:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-05-15 15:52:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"cf46e00c0ea2c74b09be86ed80202d607dff176d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-05-15 08:57:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b679cb5f82adee8552577bbf7333fd0b8965aa6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-15 09:04:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77cb955e2998ba2f43d4394c987d9d1c161b46d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 10:26:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37cbec78a07e4f46a4f74645507842ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a75e56f60034d7a92b2e174faaa8614 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7adbac4efa44eee971f630caa58bce2 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c77379359b93446fbc41f4f1f99a7690 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e39401cbeff944ed905109bbcbd071b2 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/821919ac887548dfa62c6cf744185b14 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a51fdbeaac741489a9d5149007c4e1c : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a27b42ba0977475db9fb3e283a75f59c : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/74acc7e9970349e190b76926c1558ae7 : SUCCESS in 6m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d917357dcd864d8999efd561807fea63 : SUCCESS in 11m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d51fbd989c99495697ff1d5c89c317b1 : SUCCESS in 7m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b704cfe986974152817652097b66b2cb : FAILURE in 3m 53s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d4fca428fb394ea9a2de4eff3ea6ea1b : SUCCESS in 8m 28s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b12e425bb41147449070dac233dd9901 : SUCCESS in 5m 11s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d82aeb3af90f41c8b474d7e6ff651fc8 : FAILURE in 3m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0e813840b0f54e258e92f7ce3a5d3216 : SUCCESS in 11m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/99627a19f9914016969498350ae5f960 : SUCCESS in 5m 35s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2385c5c75554439994f2cd30a9420124 : SUCCESS in 6m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/95388dbf86f94d1fb71de8196d68d8a8 : SUCCESS in 8m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fede8ea3ff874b048ab90369d72ffb1e : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b4ba45a27d04593b4be57328162d13b : SUCCESS in 7m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/90b27744993b46c59b24346d86228e4f : SUCCESS in 9m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/506fe9a8658040629c41f5e12e22191f : SUCCESS in 11m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ac3c76e6923d4cdaae414c4339f243e1 : FAILURE in 2m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4c43170ebcde4e779c8db68cc4682f88 : SUCCESS in 17m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5e369113629340cea5016645dabd2ecc : FAILURE in 2m 37s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/57573e53e5614fd98a16a4cd193f3d8c : SUCCESS in 10m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c4f451799dd6486081c9f8083831a117 : SUCCESS in 34m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a97a3af4073a4a0ab8b49704fba7252f : SUCCESS in 11m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a076f8cbcc3543dab0d9d8b94b552e76 : SUCCESS in 12m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/266bb5a6d0884c81bef4c0eea215309e : FAILURE in 3m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e51a61c186e04eec8d87d78490c580d1 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0da5c38562724763b27e0833c9c25baf : SUCCESS in 23m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3a7352f6f5fa42da8001a68ae12beb32 : SUCCESS in 4m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b57e8c537a2e45e9af16e84378f32ab7 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9f94dde6a004626a43118313a0516dd : SUCCESS in 6m 42s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/2bec701ef7124a7ba10c95512c4361fd : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd4c627c5b1a4abba3c4e4e932c73a3c : SUCCESS in 44m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d00c893ccd974ff391352a251522b632 : SUCCESS in 1h 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"899844845748a2532b77c6eba8b74743bb377bee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-05-15 12:26:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"865886d951b3e1f36a2030f78d2d8e366581b51b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 14:16:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed0fb7dfb8b453795d9b5fd816e368b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af0e3075ccda4ea3b35727e07178da32 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/056ec7429d584221bd9d5415887b64d5 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccacbdebceef44caaccb105161772a83 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07e015b3f68847af81779b0b1ba93c68 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c22ceb66fcd840f9a78fcfee273d8fb6 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/030300845d2b4c03985612e0ef139f8e : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/554b74d8a94e46c39c26791b7cd26497 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4062e62fd22e4faf9c41a8dcaf7c5636 : SUCCESS in 7m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0017367c241b4fc4b140184eaa8bce77 : SUCCESS in 11m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/97320ed45a2f4824996576308244ca88 : SUCCESS in 5m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d62e258631a447d0ad71e3c6e2767ece : SUCCESS in 11m 22s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b10197539ba643d0a20881c035c4b887 : SUCCESS in 6m 03s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d0980bee4e454b13936d253c2e0b99b3 : SUCCESS in 4m 42s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/05861dc4e9834b2dbfe09b7b3c7560bc : SUCCESS in 6m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e7643a6084b94cc8aee85ad0633a5529 : SUCCESS in 11m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1cdb949ccf9d4295a2631928cc553b7d : SUCCESS in 4m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ea109f9e2345413d939b285e372f5d0e : SUCCESS in 4m 34s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/997ea983656344ceb0bcab7735dc0d11 : SUCCESS in 7m 57s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1a5964f8d11b4ee29f30eb7aa3d42223 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a8ab1e16c924f7baa98476eb3293d6a : SUCCESS in 9m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ce9c888c631a4ed0add499b9647f9b57 : SUCCESS in 8m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/419ab1f30b2e4f0ab3154a7801268785 : SUCCESS in 11m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3ca38632111b4026bf9c3c067312f822 : SUCCESS in 3m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5c6d10a2a89942a5b27cfdb72819537a : SUCCESS in 8m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/17b0a06c22074a9f9ccd25c21f029f43 : SUCCESS in 4m 03s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7f1426c104f74e5b8ed881d41a2f69a1 : SUCCESS in 11m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/25e553db87524789b472e21c704a59f5 : SUCCESS in 32m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9270847d4fa94f5ea7cd16cf37a036f3 : SUCCESS in 11m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/af3c365a7f65439cbe75e0eaf8f2ac61 : SUCCESS in 12m 32s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fbd813c0f0a441c0b588b272492ee786 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a3e508fcfe3465295dc8e07bf5089fd : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97e9c288fedb48288bbff17e2bd55b4e : SUCCESS in 23m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ccf150cd131b471794f7991fe225e806 : SUCCESS in 5m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2789f01b2cc04fd4bb858ccbe2acaf14 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10c3c1018b874701a5bd133751f1550d : SUCCESS in 6m 07s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b4e5e871fcec4e9da325f8a8eef349a8 : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efe94469fb704ae785dff74b5d2f7aae : SUCCESS in 47m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be6c9496c75c4df2a18f1f0d612f8458 : SUCCESS in 1h 44m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1f349b298360579c5954986cb0c38fb2654ec5f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-15 15:52:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0ce353e8f62c55d46e94f96c549db6942ad6351","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 15:53:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21ee99bf560aec74e604a80268fad93063877658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 17:38:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779f01c9b1f348d7b989d472a8b0d7d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392547f6d73345ae9d39d2f2a266f71f : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b76b7ae3ba44aecb698f5555297ccff : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8431449922a48faa9d7848f08af87e1 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfd8a581ee644aacb1ec4596cfc93549 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f01b652ee89c4aaebe7b3f4b97c6b8d8 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80052a7f4d6c420f8059c4b17387f9ea : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3caae05aa25e4d41a9469260ff99a305 : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5b2b111e16294f9f8ff59c68488652a8 : SUCCESS in 9m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ad716cb0372745f8ad38c9c8e644301b : SUCCESS in 12m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6d8ca6c21d08421eac849e21e4bf0730 : SUCCESS in 9m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f29d4f1777514954abe5b4d66257e348 : SUCCESS in 12m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6737c87949764d9992621c6bb19a23d6 : SUCCESS in 4m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3daa101956ef4331a7181bdb09b98019 : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7297f3b4a3b4433c8eeeb2c278e249cc : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a6b341c6953e457c98caa27a58377560 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff1e537d855c469fb0922fe9b95afcd0 : SUCCESS in 7m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/746d8dfb320341889553e7d7d3293076 : SUCCESS in 9m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f77b4ad053ce48dbaf6700ce5a0d85eb : SUCCESS in 10m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3f6b43d521944b4a8a9530551248c3d0 : SUCCESS in 3m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7bccff1b23af4d82b74ff7a766fe9ac6 : SUCCESS in 8m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2acd0126f80a46269bedaf6af9fb9fe9 : SUCCESS in 4m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cad38849c4c6443bbd3af8555c7d78bd : SUCCESS in 24m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e95f028157f847c5b11125d56316bf05 : SUCCESS in 9m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8fc556f66acf4b80ba522bd13d65fc72 : SUCCESS in 11m 22s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/824e0940671042848bf3c69046b6fcba : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7674978d3d7a4ee49d99388291795994 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0802580adb4a47a3bfeb734ac259339e : SUCCESS in 21m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e71994aa6e274e0fa06d665ff97f3605 : SUCCESS in 6m 09s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/332e110bfabb46afba09ee20f1980b70 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a86a900d6464755ae71376a3b2df870 : SUCCESS in 6m 28s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7ad706b25aa3473aa087404bec336a1e : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c16b7c169fc4e0c816c09b7a666c8b7 : SUCCESS in 59m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dda0eb7f312c43f5be6cfd318c1959a7 : SUCCESS in 1h 18m 07s","accounts_in_message":[],"_revision_number":1},{"id":"261c27d51d9df4596efcc5a8f89a45403b16a87f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 17:38:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8943c753a3bbd91bf213ebab3bda46358e3dcde7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 17:39:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d43d9e332f4c4dafbfc182b6d4d5635d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b663e69e9d940c1be9a0e0c9a77bbbb : SUCCESS in 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aef0fc0d7723467ff094af412a4f274b6698811d","revisions":{"aef0fc0d7723467ff094af412a4f274b6698811d":{"kind":"REWORK","_number":1,"created":"2023-05-15 08:57:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/883143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/883143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/883143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/883143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/883143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/883143/1"}}},"commit":{"parents":[{"commit":"f4238c0a45d422093636b9682f9085f3e89d9481","subject":"Merge \"update constraint for os-brick to new release 6.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4238c0a45d422093636b9682f9085f3e89d9481"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 08:57:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 08:57:38.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.2.0","message":"update constraint for oslo.policy to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I2280d24ddfaf7b970c135993573fd97113bf324b\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I58b9356999e81ab2a5116f3ff573d205a05cacfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aef0fc0d7723467ff094af412a4f274b6698811d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aef0fc0d7723467ff094af412a4f274b6698811d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
