)]}'
{"id":"openstack%2Frequirements~883144","triplet_id":"openstack%2Frequirements~master~I95f88a04dcec43ac1c93bba8260073317bccb435","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I95f88a04dcec43ac1c93bba8260073317bccb435","subject":"update constraint for oslo.messaging to new release 14.3.0","status":"MERGED","created":"2023-05-15 09:01:52.000000000","updated":"2023-05-15 13:37:20.000000000","submitted":"2023-05-15 13:36:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883144-new-release","meta_rev_id":"819533b063b0d5917d6517ff601414ab9772e716","_number":883144,"virtual_id_number":883144,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-15 13:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-05-15 12:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-05-15 09:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-05-15 12:23:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-15 09:05:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-05-15 10:52:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-15 12:23:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"2faa166df353c102886f4e0f74b40c73aa75ccdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-05-15 09:01:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1a48c749e0f703521553388f3f633fe257728b9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-15 09:05:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5c86b7a47ad2b0985fee032383b3afa2cba1d3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 10:52:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8583e554ea5c41329345d531feb98610\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ddf68b7f58480ab53dc4383d1bcdc2 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b19ab87e31124656a1562c7333abfefa : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9df93238847949e48dd7592d8c4ec45c : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6c3fa8cbca143d281301d3e9c3b7469 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c5e1d6dbed5429b9094dadab9c92404 : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ee1834a62804505b2fd91c91070b9e2 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b54db7f369524352a9a5d49acf46956f : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/49a5ffb1c93244ad828c2b38710c9fee : SUCCESS in 8m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dd2cf4260c4b4128a0cb660388d065e7 : SUCCESS in 13m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/639c924fd83645e694404f6dc5da8509 : SUCCESS in 8m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1a682b83b9e64a30a41bd65b1995e797 : SUCCESS in 12m 38s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a4d16488750f4bd69b67c2d25dddcf9b : SUCCESS in 7m 06s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/44c5016310754083ae7e1b6e931ea57a : SUCCESS in 4m 38s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c259b886b4d9410e97cad6669df74af0 : SUCCESS in 5m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/68d15ef10dfa4833adb9edc61cfc39f5 : SUCCESS in 10m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/18108741a61b4e3fba60f778028c9002 : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5db413532bf44a75a9ebac036c4f932a : SUCCESS in 5m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/af5ff65b31d14d37b327cbd8a1c1ac02 : SUCCESS in 7m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a7bc8bb9a05a4790951f76f59137100e : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65aadad826aa4fd380e7ac80e7ecf063 : SUCCESS in 7m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/36acc9783e504b06bff69d32a5ef8b9a : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0afdc1024d464ad29abc0e3d2272c92e : SUCCESS in 10m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d13f6acc1853472694c0ce24719dbf5d : SUCCESS in 4m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/913c53499be44ccc9dd640fe350b7ad9 : SUCCESS in 9m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/60a02a10c8c6415d81c509d0f65e459d : SUCCESS in 4m 42s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6724e9f871f248c18ca17220c8ba0f76 : SUCCESS in 10m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3a036a9d62f7415faa907c59513e0418 : SUCCESS in 25m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/04c54c0083ad429b862e8be360c57942 : SUCCESS in 12m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/39373fb7267048f1b952a77bbdb1c352 : SUCCESS in 12m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/803a160c7d294574b10f71b96f8eb533 : SUCCESS in 6m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1953ea9f27714d92a72be7de9c1cfec2 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d78b8a1c4e7d409cb83ad937eb463fad : SUCCESS in 23m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4c6bdfce2e694fac831979953afe87b9 : SUCCESS in 4m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c9babbeada394b6a93af765df5085c2a : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5fefd081e8924688ba3a325ec02da4b6 : SUCCESS in 7m 29s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7cb1c0ac9ed44759850eefaa0b376962 : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5d41cf71a594bc3a31d7b200ad56380 : SUCCESS in 59m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0d9de0aff6244c897a13aa6c454e246 : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0dcd13e081c1c549b546f5d83c625fe79b9a6d21","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-05-15 12:23:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"362762c4d31b43260c95bd9b1042426a98ba7c48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 12:24:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a8178dc867cd4ba523b161cce26a61cd5434310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 13:36:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47ebb8c3d73a4ccb8d8862066f2b0e90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a51d2cfad044a78705b9d2020f93b8 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/670e8366697149199e7b04b1e50463ad : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eafc25384af94111b72875432e016225 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6e2de0f28804587bef10fde9dd048b2 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edcb646519954135aa35ccfb4035e3e2 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd8c576c6d134cc2b8e3794891fac95a : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c2e194ef2154f279c40d822d82f012c : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1fa45a99819b468a83f41742881dcc86 : SUCCESS in 8m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fc3753c7e3c047169ba734ae7aead6a5 : SUCCESS in 14m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0c6ad2d701bc4e56a5410c180e29527a : SUCCESS in 8m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c32de9ce27db4b808bc1d8c979b27710 : SUCCESS in 11m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/be11fcf137464f86be7ace9cb7dc163b : SUCCESS in 5m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/91b29c0c31214e2e9e3eea70fe9b12b9 : SUCCESS in 4m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/87b6b0b952b943169b23d2e01ce411f5 : SUCCESS in 8m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d99636dcc55948deaca70aa58226b7cf : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c3d87c4d75f94549b71945f0bd1f33db : SUCCESS in 6m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4edea7775d9d466ca457965fd50c3356 : SUCCESS in 8m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/79213aca6aaa43939cedd6bf4ebe6f7a : SUCCESS in 12m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5a7bcaa297d342b7b1abb41051ddb2f7 : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/463f8249228647fe9fd6f902147e581b : SUCCESS in 10m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c813a8243af34970938da272e905c1b5 : SUCCESS in 4m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1068d559204549339ad98c3073c0eb25 : SUCCESS in 25m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a856bddf9d104067b5c3571f4b1a7367 : SUCCESS in 10m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cc728d60a71b440097d377c58baa3642 : SUCCESS in 12m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/82c4da8d021944f2a56fee502be99076 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a49b8e6b06d84383b99b8ab01e868b5b : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/827d96b49f8d4de6a6e7c3ac2f84bc03 : SUCCESS in 19m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fdfa23dd20c849239822cae217820bfd : SUCCESS in 6m 32s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/aaad136b12bd40f19d87094f1b67b9f5 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa566e51f298484d802b76a0a3e04189 : SUCCESS in 5m 57s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/647ba7b59f83465a960cf9a476ae0e62 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/037ec57b35f84b69b7d230f762db3960 : SUCCESS in 43m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d41b9cca3f7f4cc596a6dc0243c13d9f : SUCCESS in 1h 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9325c888c432acf10aae84c4d94a11087158b3cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 13:36:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"819533b063b0d5917d6517ff601414ab9772e716","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 13:37:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f2d1e37103741778c4b28c6dc4f077c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e87664e3e9ce46a0a93e35a479822033 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9348331c134669c949812b8cc090258000156b8c","revisions":{"9348331c134669c949812b8cc090258000156b8c":{"kind":"REWORK","_number":1,"created":"2023-05-15 09:01:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/883144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/883144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/883144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/883144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/883144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/883144/1"}}},"commit":{"parents":[{"commit":"f4238c0a45d422093636b9682f9085f3e89d9481","subject":"Merge \"update constraint for os-brick to new release 6.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4238c0a45d422093636b9682f9085f3e89d9481"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 09:01:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-15 09:01:49.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.3.0","message":"update constraint for oslo.messaging to new release 14.3.0\n\nmeta: version: 14.3.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic67679a03512ccf4599d2f5aa615b0877e4795c6\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I95f88a04dcec43ac1c93bba8260073317bccb435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9348331c134669c949812b8cc090258000156b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9348331c134669c949812b8cc090258000156b8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
