)]}'
{"id":"openstack%2Frequirements~883402","triplet_id":"openstack%2Frequirements~master~Ieba70cb9e9300ab027c4b2ac7a72c4df7a21e660","project":"openstack/requirements","branch":"master","topic":"block-jquery-migrate-3","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-05-19 22:51:18.000000000","reason":"Change was submitted"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-05-19 21:39:06.000000000","reason":"removed on reply"},"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-05-19 22:51:18.000000000","reason":"Change was submitted"},"1736":{"account":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"last_update":"2023-05-19 22:51:18.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2023-05-19 22:51:18.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-05-19 22:51:18.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-05-19 22:51:18.000000000","reason":"Change was submitted"},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2023-05-19 22:51:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieba70cb9e9300ab027c4b2ac7a72c4df7a21e660","subject":"Revert \"Revert update constraint for XStatic-JQuery-Migrate to new release 3.3.2.1\"","status":"MERGED","created":"2023-05-18 10:58:30.000000000","updated":"2023-05-19 22:52:13.000000000","submitted":"2023-05-19 22:51:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":773823,"submission_id":"883402-block-jquery-migrate-3","meta_rev_id":"24c6ee2737165be6ad5c9d1eade17c5ba57cc5ef","_number":883402,"virtual_id_number":883402,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-19 22:51:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-05-19 21:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-05-19 21:39:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-18 10:58:30.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-05-18 10:58:30.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-05-18 10:58:30.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-05-18 10:58:30.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2023-05-18 10:58:30.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2023-05-18 10:58:30.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2023-05-18 10:58:30.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2023-05-18 10:58:30.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-18 10:59:01.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"3b6d288eb0a5384372f4d4838ddd4254cab2c732","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-05-18 10:58:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5224e04166623f86115233795f7797203e54464","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:58:47.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 883402,1","accounts_in_message":[],"_revision_number":1},{"id":"df858a04b86fe1ab91573548aaf238c45e4b354b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:58:47.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 883402,1","accounts_in_message":[],"_revision_number":1},{"id":"dea45b0625dcb5003291d5c605aa8c5c1f939c4a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-18 10:59:01.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 883402,1","accounts_in_message":[],"_revision_number":1},{"id":"9bb4e7779090c06835b9370b87ab7c668b357021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-05-19 03:34:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e7aada92c05945c891d789bf7f2ab87e6a52b753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 05:03:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f90838fd11d495aaa3722f0b0d60e54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ccafd04d2154779a549a0237f722323 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e566baf284464c8392315dbd04bfdfeb : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92e35a9e43f34e79b17ecb540ffe32ef : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bc37bb423394c5c955abadc9c1845c8 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd49911ebc57494e8c3c2abc6d2086c7 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bf4cb750ea64590ab5cc264a91b5a04 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d81544819481487bbf7e7076534bbb99 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4c708007a14246d189f54c549f71f1e8 : SUCCESS in 12m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/55df94dd101d48f4b1c2795f78780e30 : SUCCESS in 13m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c4db197e4da74a6e88074dd5abe9c88a : SUCCESS in 8m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/54159c0902f3439cad0d34364411499d : SUCCESS in 13m 51s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f37f89ac9d0541aea1d5bbfd092f759b : SUCCESS in 6m 48s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a02023825b304d63b56e429dbb0a3556 : SUCCESS in 4m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/61034398bc9c4b23a9f569b1296cb8a8 : SUCCESS in 5m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/15b958f283f24fcb8225c64d139e3879 : SUCCESS in 11m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d667ea8cb237434dbae29f3adb17780c : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/baf85ab1c64243879c27c359304cbb81 : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a2c503c049264ae693b9c42092726402 : SUCCESS in 7m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/199e506c13ab401e9abc579a6e730342 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a65f7cdd90a421eb2d6622819a11604 : SUCCESS in 8m 13s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/14c924d05e554347b4979607278b779d : SUCCESS in 9m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/62f38ab21e134540b05c468c9dd0b720 : SUCCESS in 11m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/090b1ef9798146fd9003f8115c487ff6 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/65b72d025e82468297848f1b710e3a48 : SUCCESS in 8m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/95d0313df4a5435f8642af39ccf0ead4 : SUCCESS in 5m 52s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/23ddef27c5b94fb5be0f461df84c2d5f : SUCCESS in 10m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/625d80ade7554a86a035520b3dbe6804 : SUCCESS in 27m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/94a8e3931eb04e25a2e11db4ce25d91c : SUCCESS in 15m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6cca59252def4deb981810d28ed9a315 : SUCCESS in 15m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/52d69d3c46c84e7aa38b5bf610c82a61 : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d0c91ae8cb24fddbefd24cc466a790b : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53f9b9c3a18d4cc193e3f90f7f159151 : SUCCESS in 23m 25s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/95a4777cfdf048aea1d583bbeedd343e : SUCCESS in 5m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e194e88cb2f148febbbe4e230fbafc40 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e6734c91b5be4dd78834b03e216d187a : SUCCESS in 7m 51s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/fd4ea5e1d04842e4bef3f92b5b04effd : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d32d086770be464fa785fefe66bc9368 : SUCCESS in 1h 01m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec7e439b7d3d4d0e943d0d78354324ab : SUCCESS in 1h 18m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6ff1f390fad997165c3b15c1c5e40d463e6282e8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-19 21:39:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0342d01d1a82faf97c7692982fc93ebc7f2b92ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 21:39:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90a7f43e02a579625d04b8aaf92351516fc4f015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 22:51:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/659ad61c4f484e45892f615b98762b01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31051fc4fecc4c5ebd6ad74dc5ed0610 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31868fa4ed924976a208849563b9e8c0 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cfad6c96bda443c96a4178244851bc2 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24442b987a254613a307d1147c4740b7 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d25c882bcd7441548ad9cabb85aac538 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8bb6366c346c4738aba5ac46872551ca : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1521fd709ba847d88f2850e616087976 : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a9321b8a765f401cb9ec1a2d54aa4a41 : SUCCESS in 10m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/84f04835ddf3448b907bc3bdaf4ee886 : SUCCESS in 12m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/75cc9e35925d490da9491e6db7f6fc2c : SUCCESS in 11m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bd2ac2d3811f4560a25ce821ad6d8286 : SUCCESS in 11m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/686a8208ffd84cc586c13586c385ce55 : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1fdd58bdc28b4dfa8f85a73552c813ca : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/96147f4f897d43c3b7201e7479fdc22c : SUCCESS in 7m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/45d6ed222ef64e6d8c188a103b50b7b8 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f270702b6163447c836513e247f00866 : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0ae034db1ffc4697b3e3255b5e8206d3 : SUCCESS in 8m 25s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/01d3eead36604997a866077d60d1d440 : SUCCESS in 11m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/91736d410b514a2394f28b9bf27c9f4b : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b992570c017d4ba88335b3d42910f0df : SUCCESS in 10m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9f30bc2a9c32416296417c3824f2cf6e : SUCCESS in 3m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f37ff86abf51475e8223ba326be6cb00 : SUCCESS in 28m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c486154b53d744ba90d1a6552d40aea3 : SUCCESS in 9m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ad7c47335121406b9d499038f290c341 : SUCCESS in 14m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f395017952734391b6c813f36376f571 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/22e7b40e2ee14eafaafb3992da449210 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3da89e4d2e0a438f986413e0cc572f17 : SUCCESS in 24m 39s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/93ef09a7732340f98a2ddad231153d59 : SUCCESS in 5m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/edbb9c6146f24a2abd8cd9651aad1269 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87b9d0f051a44e67991d3886d2cadfbf : SUCCESS in 6m 58s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/49b149ad53ec471586bc356593c19501 : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dede17327352403b916c53a40a7bfb75 : SUCCESS in 54m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14d2bb6787b94775b18ec869f21db0f9 : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":2},{"id":"0a1a94370081b1e43f9b3ad425a075845e53931f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 22:51:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"24c6ee2737165be6ad5c9d1eade17c5ba57cc5ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 22:52:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8727e2876047d399e6ee3de8a4f190\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02275082db564129ae1998c79db0e98c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1bb72ac575d019558ddbf301004f798db88304ae","revisions":{"b9dd296a07cfea9f0a1683924d7ae60530e69303":{"kind":"REWORK","_number":1,"created":"2023-05-18 10:58:30.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/02/883402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/883402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/883402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/883402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/883402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/883402/1"}}},"commit":{"parents":[{"commit":"dff4253a1f87d734117835aa0b17a2026f5ad9f2","subject":"Revert update constraint for XStatic-JQuery-Migrate to new release 3.3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dff4253a1f87d734117835aa0b17a2026f5ad9f2"}]}],"author":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2023-05-18 10:58:30.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2023-05-18 10:58:30.000000000","tz":0},"subject":"Revert \"Revert update constraint for XStatic-JQuery-Migrate to new release 3.3.2.1\"","message":"Revert \"Revert update constraint for XStatic-JQuery-Migrate to new release 3.3.2.1\"\n\nThis reverts commit dff4253a1f87d734117835aa0b17a2026f5ad9f2.\n\nReason for revert: Most of the deprecation warning horizon get after upgrading to jQuery-migrate version 3.3.2.1 is fixed now.\n\nChange-Id: Ieba70cb9e9300ab027c4b2ac7a72c4df7a21e660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9dd296a07cfea9f0a1683924d7ae60530e69303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9dd296a07cfea9f0a1683924d7ae60530e69303"}]},"branch":"refs/heads/master"},"1bb72ac575d019558ddbf301004f798db88304ae":{"kind":"REWORK","_number":2,"created":"2023-05-19 03:34:15.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/02/883402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/883402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/883402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/883402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/883402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/883402/2"}}},"commit":{"parents":[{"commit":"79ef195f2b39c301c29653aa1b0873e6fcc61f67","subject":"Merge \"update constraint for keystoneauth1 to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79ef195f2b39c301c29653aa1b0873e6fcc61f67"}]}],"author":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2023-05-18 10:58:30.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-05-19 03:33:51.000000000","tz":330},"subject":"Revert \"Revert update constraint for XStatic-JQuery-Migrate to new release 3.3.2.1\"","message":"Revert \"Revert update constraint for XStatic-JQuery-Migrate to new release 3.3.2.1\"\n\nThis reverts commit dff4253a1f87d734117835aa0b17a2026f5ad9f2.\n\nReason for revert: Most of the deprecation warning horizon get after upgrading to jQuery-migrate version 3.3.2.1 is fixed now.\n\nChange-Id: Ieba70cb9e9300ab027c4b2ac7a72c4df7a21e660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bb72ac575d019558ddbf301004f798db88304ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bb72ac575d019558ddbf301004f798db88304ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
