)]}'
{"id":"openstack%2Frequirements~883794","triplet_id":"openstack%2Frequirements~master~I91d596f8f6ed08dab0c8d8904bb62ec7e33faf39","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-05-22 15:26:24.000000000","reason":"Change was submitted"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-05-22 14:07:46.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I91d596f8f6ed08dab0c8d8904bb62ec7e33faf39","subject":"update constraint for neutron-lib to new release 3.6.1","status":"MERGED","created":"2023-05-22 08:53:33.000000000","updated":"2023-05-22 15:27:24.000000000","submitted":"2023-05-22 15:26:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883794-new-release","meta_rev_id":"7a6c8a0587cf56cb15d9398656ee4679a852da78","_number":883794,"virtual_id_number":883794,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-22 15:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-05-22 14:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-05-22 13:05:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-05-22 09:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-05-22 14:07:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-22 09:38:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-05-22 10:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-22 13:05:10.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-05-22 13:05:10.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-05-22 13:05:10.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"4855147f608246325390c82531defed1b135df73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-05-22 08:53:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ffd993871e5c0175e5e0d589f8660a02d8343d0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-22 09:38:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8ff672808033b539c9f2d7f6fa56cd87ea046ee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 10:25:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97aa76d4fbb5427b8cc13eaf1635a190\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4f7dec025545eb95c2a1bf1ba4f445 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f30c20a9c084ab2957a7d3c56eefb54 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/165f0b5cb104458fbef010507b0e3535 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2de7189758f44ddebfe2438a525a16f7 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9e8d16655ac46fd91ad75c6fbebaca7 : SUCCESS in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00112e56a101422e8cc0bbf574e6e7cc : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/602d627f6890478296d951bc7c49cb9b : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f27cbb4233364bbdb87d3a602382f8de : SUCCESS in 7m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e4bb27b4a28147e1b8b0eb816401a6eb : SUCCESS in 13m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/023895ec3d2f4646946a4c6d9891d902 : SUCCESS in 8m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cb7b4f2a561d45f8b6761ae8ad6e937d : SUCCESS in 13m 07s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/21bf0592762749ea9b0657c5f3bed014 : SUCCESS in 7m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cae097913d154d01b560af091f747b81 : SUCCESS in 4m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f0b9227854f549c5a3a2f25ea2996b65 : SUCCESS in 6m 20s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/eeef819732f5420190151ea43f01815e : SUCCESS in 12m 09s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/736c2b533ffb4280a0e7a0be62023405 : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/560d3848895e465ca1d246611900e40a : SUCCESS in 4m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/39bd20d80eb04276b4e1c1c3bc66af89 : SUCCESS in 7m 40s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ab32b69ea02b49a1adb87506712d40aa : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0190dcc150084084a9585a2483b86e30 : SUCCESS in 7m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7fd01523073741f4a651a2423a110634 : SUCCESS in 8m 50s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1904547eb08c43518a28dd6e8dc2808c : SUCCESS in 17m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c3d392ad90ab4272bd3d4d002d25bed9 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9f2925a1982643809088d070f6101782 : SUCCESS in 11m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d1d1d805ae424deeb752a17b13c0b7f9 : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7bd06047c9144682854823d30c386210 : SUCCESS in 10m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d417dee862b94776a54cb3de2ff4779f : SUCCESS in 36m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/37543d8d7b5845b295bf9e00c0a4fc47 : SUCCESS in 11m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/56d23c8d06f54824b3fb621a21f56e5a : SUCCESS in 12m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/80bcc17612894eb0a5ff4139969ebef1 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6287bf395ec428a922ed991f0f04d15 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef2b565718294b71ac1bce06fa8ff498 : SUCCESS in 28m 52s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/624f603da8d44acebd67a05fdecb61c1 : SUCCESS in 4m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a1cde405ca8a4dd387d222b49e15a87f : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b78dbbdaef194b7289838112ec931d82 : SUCCESS in 7m 54s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/761d2f0ccdac499f8f65f543548f4d3b : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ed5c09013bd4a898a80cd4ba1347b1f : SUCCESS in 44m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/28cfef73d0114c62920eeb4cbf94bdea : SUCCESS in 1h 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"cbf82b07c0498f5191df7d30642c2e0fe23e4fce","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-22 13:05:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a3338cc3ce993826b2ef1420339e6a93201a0a25","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-05-22 14:07:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fde9f6922e22c39e76d3bb1a8a79fdc56007ad9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 14:08:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e0146d76deaf4f1dd063f6d777efbb40a6fef51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 15:26:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/165144d1e040485f83dfcef7ef0394ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679dd0d7a61f46c388ef4fe3edf41002 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc648d670434cddab78c48d78e6ef83 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8952446baab442b85763acc67fb016d : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/403ef02fc18e42858d914130e0a35535 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/603b694e358a417886f4ca30534c6306 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/082206da153e4aa58b2b2f05c9f8a300 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aca39f670fa34b10ba208b7148e70e33 : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0956e33eaa7c41be88b8a4cf650d853c : SUCCESS in 7m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0c7719b33d134f7d92b72286e03bfdce : SUCCESS in 16m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b86485f9e9e84a7cb74d09dae4589e89 : SUCCESS in 9m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/77121132f38c4a02a151f084fde3397d : SUCCESS in 12m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/95cd2f9d70f3440691e5696dcdc6650c : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/392e379f4bfa4bfbb278077c26df4d65 : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bf18b3427b8d47a28508ee9b1adef1b3 : SUCCESS in 8m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a2ac949b91944be9a41ced016e0f0ec1 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b58f6f0897834c7ba4dce264e5a731f6 : SUCCESS in 7m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/76519cb14fd84e2f9c26f3aba1b46b87 : SUCCESS in 9m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/47450ae19a3e4552bb34f619313aafb0 : SUCCESS in 13m 57s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/46f7a964e806482ba62551b91fcf5e8c : SUCCESS in 3m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5e8c8e5117ea4877adaccbb0e4d6da0d : SUCCESS in 9m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/15f96d629029423886dd93f427ae0b6b : SUCCESS in 4m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3b143e084d094383b94fec2792a39396 : SUCCESS in 27m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/43401a5afa814db8a44acb321cbc7274 : SUCCESS in 13m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/95c89e7177a1444fad4502acb53224eb : SUCCESS in 13m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3e4d3c37f5d64bab8febb3b6d5eeaa4a : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac89109c2a594e36a4f2285c8d76d963 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09b8e560b37c4b1f96bd0a04c169b40a : SUCCESS in 22m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/de45f87c33874a48a277ca75bf0676f7 : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6f4eb53648fa456fbce6ec346bc1bd1f : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a722f4cbbb4541f3b380a1ce2bbd5737 : SUCCESS in 6m 24s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/83b0e2c369a6436599611613bab23daf : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48204c0cbaf24e3588d7aad4c9b19902 : SUCCESS in 58m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f79154fce0e4531aa474e6bef1df141 : SUCCESS in 1h 16m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7bee7001cad6adce1a4bc1da576e5a1c623d02a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 15:26:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a6c8a0587cf56cb15d9398656ee4679a852da78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 15:27:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6380377e4b594f02a2a8492ea1631c2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f226c17190294c85a1af7e1a77e8758c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb4c7b38305841578dd89f3168136808fc351ee6","revisions":{"eb4c7b38305841578dd89f3168136808fc351ee6":{"kind":"REWORK","_number":1,"created":"2023-05-22 08:53:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/883794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/883794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/883794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/883794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/883794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/883794/1"}}},"commit":{"parents":[{"commit":"b34becd25f65b7dd9e1750fffebd3f0e25fe4da9","subject":"Merge \"update constraint for cliff to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b34becd25f65b7dd9e1750fffebd3f0e25fe4da9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-22 08:53:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-05-22 08:53:32.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.6.1","message":"update constraint for neutron-lib to new release 3.6.1\n\nmeta: version: 3.6.1\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Id9836472648a6abaa35c7a8b45f2500af770588e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nChange-Id: I91d596f8f6ed08dab0c8d8904bb62ec7e33faf39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb4c7b38305841578dd89f3168136808fc351ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb4c7b38305841578dd89f3168136808fc351ee6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
