)]}'
{"id":"openstack%2Frequirements~885067","triplet_id":"openstack%2Frequirements~stable%2Fzed~I0d7a517598871ff1d6848840b9888f95d8b27f48","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","hashtags":[],"change_id":"I0d7a517598871ff1d6848840b9888f95d8b27f48","subject":"update constraint for oslo.db to new release 12.1.1","status":"MERGED","created":"2023-06-01 15:31:27.000000000","updated":"2023-06-01 20:58:21.000000000","submitted":"2023-06-01 20:58:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885067-new-release","meta_rev_id":"a5c8d9166ec617ae3175de44b1cc5f757ebc32ec","_number":885067,"virtual_id_number":885067,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-01 20:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-06-01 18:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-06-01 18:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-01 15:45:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-01 17:50:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 18:55:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0247078edcdf276feff5dedaed77f36ce7575fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-01 15:31:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65d972e63148457c9198225583b3934b7f0b1ba7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 15:45:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dba065ec1c64181a7aad47959b23492\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f0cc4b9dd2cd43d1a27d08d01652694d : FAILURE in 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"98926d66a36b365c08506a779ff92347f7988d9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 17:50:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30d26167088141d3828522defc7be1d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c0df20ebae4f71af055e183bc675f5 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63b31095beab4607831bd959f37f17d0 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96d2ffef122f47459e2f8aecc02f828b : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0e0fe33544c47dea2b1890b79f9cff4 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43df2d7c4a3a4645951b77dcfaf75ec3 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4126293782d40b1accb341765a85fe8 : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/732319edf63148128177a50214cb205b : SUCCESS in 8m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/483407513d1846f9b677f1284275bf73 : SUCCESS in 13m 32s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e767581220114aa4b3e50c1f639bd629 : SUCCESS in 11m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f2686f9b152e4b8582c9a3a15e482eda : SUCCESS in 5m 12s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/67896e8311404ca3b2309b9a5d45b220 : SUCCESS in 5m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f654f5dae38148ea8aa9129041c029b1 : SUCCESS in 12m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/99fc98f73b5f4caea77af64deea2384f : SUCCESS in 4m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c98e3eb164784552aaed32bbe01fd797 : SUCCESS in 5m 38s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/561a8e8ee0834dea94460f13d247b396 : SUCCESS in 7m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/99a5144c764e4ebcba60cee0479978d9 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/995c1f87ee1241ceae4800e265fb2786 : SUCCESS in 7m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/70876bfbda244462abd1c258fa05f1c0 : SUCCESS in 11m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b5db615cf02f4fada78d3efa7c965614 : SUCCESS in 10m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/59c18e9773b74f98b9e1eca7fc819e1d : SUCCESS in 3m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/adde24e6521140d8845b87ff0342456c : SUCCESS in 9m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0ec7517b7b484e3a87ddf8f0dddd5856 : SUCCESS in 5m 01s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/236b1307478e4871ab57793d43a34872 : SUCCESS in 11m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4f7e89f47e4d45098eb7b437ba8c1177 : SUCCESS in 23m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1027d9937410475488bb2328c6c20902 : SUCCESS in 12m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/29cbb77fd466455c951840550b808d81 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c41a83905b04f0e8a21fde1c9946dd8 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8c0c6605f604788bd19c52b07bc3382 : SUCCESS in 24m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2a3e24f4b08a48cc84451c03fd0b938c : SUCCESS in 6m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8b2b0b32914c44849686805d65c9fe43 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c10bc7c472d145a58cc5c009633f2801 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9a1625d5d4414c6c889e15d6a536457a : SUCCESS in 11m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/324eaa3d98fa440796077a3d6e42ab53 : SUCCESS in 1h 02m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7974da584a544b0a50c277bc415f81e : SUCCESS in 1h 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6c8776c6bf965bbba460c5436fb02323fa307903","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-06-01 18:55:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a56ff80351bf6ab8caed1e50640dad63bd40e61f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 18:56:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c51d6cc3cfd763840aa6b2a6ed642eff697d28ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 20:58:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b47dd17fc184bad96b7f32e630b72d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c198c7e7af28488d87971b8c268ef2d0 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a22627777d854066bb4865b7023a7c35 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a659d8d68bb4cac9e0ca4a0d3a3922d : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25e75adde1d247318c6b2369dd1fa5c8 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3549ccb3679e4a8aa76c46454898a730 : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2283f847837f4b878ef71049a0a8c7f9 : SUCCESS in 9m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5b7b708f13d847e296530159d91d403b : SUCCESS in 13m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2abc7d28db54423e87454f8a342629dd : SUCCESS in 11m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a63b6ee53e904e46b9213ccc2e60dc50 : SUCCESS in 4m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/51a4cfbf0c1a4b98971716b3408a5a28 : SUCCESS in 5m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/af63f4cd7061478abdbeb4e5cff16233 : SUCCESS in 8m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f1cbc9ee2dd24e32b082bc70e6017759 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27193da75d8348dfbd38b66c303b8cf8 : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7bd00246b53b45f59e9bd050d2897cbf : SUCCESS in 9m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8df919fd88b441d4b5179f161dbb5d32 : SUCCESS in 10m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5a8ff90f48424fb69df88e951d9b3205 : SUCCESS in 3m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/34dd6a1363774f248984fc339c12c94c : SUCCESS in 10m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4ac10ccc57c7443b80c4ba1f523af872 : SUCCESS in 4m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/05c11447183d429db8eea36d4fb49800 : SUCCESS in 26m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/300db017789142d39f148a1445ccec1c : SUCCESS in 13m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e2bd033e84f64987b86f12e9ecaf5a07 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/19a220be62e34be29499362b44ab49ad : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75cf571d572b4aa18cc986b51553ce0b : SUCCESS in 23m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c55a24e304664758a764aff121fdfd1c : SUCCESS in 6m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9bc096a65664458d99fdd8457b10d36e : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b3461240c1e34bd79e32659cc4b60393 : SUCCESS in 6m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/53cc52f6635648f48358dd9a8465976d : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2b9d4f0fa054f6b9d86849e32f8231c : SUCCESS in 1h 06m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d3cef08ea104c1d80154161a01d9ca9 : SUCCESS in 1h 24m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a5c8d9166ec617ae3175de44b1cc5f757ebc32ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 20:58:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b684e74405dfb416c49b9a0cb0f7fbc38ae6021","revisions":{"9b684e74405dfb416c49b9a0cb0f7fbc38ae6021":{"kind":"REWORK","_number":1,"created":"2023-06-01 15:31:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/885067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/885067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/885067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/885067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/885067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/885067/1"}}},"commit":{"parents":[{"commit":"8a9f69d09be03926242bfc2a04168a166f23209f","subject":"update constraint for glance_store to new release 4.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a9f69d09be03926242bfc2a04168a166f23209f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-01 15:31:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-01 15:31:24.000000000","tz":0},"subject":"update constraint for oslo.db to new release 12.1.1","message":"update constraint for oslo.db to new release 12.1.1\n\nmeta: version: 12.1.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic20b8ca7832ad514ceb87819a7f94b5a344aa5bc\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0d7a517598871ff1d6848840b9888f95d8b27f48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b684e74405dfb416c49b9a0cb0f7fbc38ae6021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b684e74405dfb416c49b9a0cb0f7fbc38ae6021"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
