)]}'
{"id":"openstack%2Frequirements~885883","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I808e3d677d63095a96cd12f82851939807b8e002","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"I808e3d677d63095a96cd12f82851939807b8e002","subject":"update constraint for neutron-lib to new release 2.20.2","status":"MERGED","created":"2023-06-12 14:00:21.000000000","updated":"2023-06-12 20:07:52.000000000","submitted":"2023-06-12 20:07:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885883-new-release","meta_rev_id":"1efb3d8b60ce7f65401be94aa295320bf507de88","_number":885883,"virtual_id_number":885883,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-12 20:07:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-06-12 14:41:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-06-12 14:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-06-12 14:41:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-12 14:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-12 14:22:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-06-12 14:41:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-06-12 15:14:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-12 17:01:30.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"}],"messages":[{"id":"1a2bfac5d25be8c803121b37c429e7ad1d67df69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-12 14:00:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a811741ce5c277e87ae9284f095c0b8221d8ce3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 14:17:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/950df29b59ef42c0a3d0c815257351c5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ba86322a038141dab470944a1c4c4039 : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c5f03cfc4c128f8fbe141e246fa95b07faa76bb4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-06-12 14:22:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6d3e6dafbe337d8de1e58e565507c290fc37a99","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-12 14:41:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5326fcbf9dfa48e5cb1786cbebe45c0d689dadc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 15:14:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82f753d613054e4faa6cf80e7db70b07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cfc16d35e56428897b40ac80b6e65b1 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/005a4bfaec6c4033b875bf8803f85ddc : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a30cb8db40804e13a7952303fb9fdec0 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7bc6c4995c0b4743b38362f631d5fe59 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b87bbf3ed0834f09bbe6c128e218b4c3 : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0acb0fbd6ed94acfa6c012243aac88f7 : SUCCESS in 12m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e11d361bb5f4adebfab3844df895f15 : SUCCESS in 10m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b4d3eb703f1b40e4adf31a7c99bf063b : SUCCESS in 5m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e46f447686054ce1bcaa39dd016ad4f1 : SUCCESS in 10m 30s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a2523f4360c54ec0a5038b525f3d8740 : SUCCESS in 4m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/59dbbc5bb91547a08b31b6fee38087f7 : SUCCESS in 6m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f2a6d58069bd448fadae328874b59e7c : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58ea3dfb0b6c4ce49faf360e1dc4276d : SUCCESS in 7m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/22a6de7212864e6bb0cb2d7d88442ca4 : SUCCESS in 10m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/91929bdbbed24a8a8c90c1dd4a785f38 : SUCCESS in 12m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d20f9a4d9107482db758eff834f8ac33 : SUCCESS in 3m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/adeb214a53864559a3be0d0356d83970 : SUCCESS in 9m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/95eebab42b08493c9cc713fb495f3789 : SUCCESS in 4m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3e02607636fb4311837561912276821e : SUCCESS in 27m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7ed66f076d8a42ca81ccc84d5c7fc7fc : SUCCESS in 11m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4cfbfee56f924bada98451f029a3f406 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/001b5216c4f947648b9f5f99dd7e93b6 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac8d73c582404b3480c4aaf872ac466f : FAILURE in 24m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6248083d4ac345dfae3037afca7bf4c4 : SUCCESS in 5m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f58e43a609b144458c4e96e55f23ddab : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3346c44d2e6448f8aa381a6f7a119bb8 : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ea5183cc687c4e39849c806442a5e2a7 : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5804d3d7782d48d98a44c9c510fbc568 : SUCCESS in 50m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a91ffce0104f431d8e09f56be6a04a2c : SUCCESS in 1h 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2bef10659c125c643e65769e61a446081ee4789b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-06-12 17:01:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f34a5e787a5993c93928d5ed165af241e5a7b47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 17:19:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38facf7d1b9e404e8adad6695634b8e4\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b250c2f7635c4390a4f1310c8baa6991 : FAILURE in 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c62c54ba70c9065a1279e23007905a7e01d2af67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 18:08:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1a7108998854e4cb9ebc3c2cdffa0f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25fe302ab66f43a480b4b3c2b5d0a882 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a94bdb29a85c4eb990ebeda869d5fd94 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b895c7d4f2024f8b87719635f904cd29 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f44fc271a65041c5afcae033a3542de4 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34be28707ad449aa8c951b98c77c3d0e : SUCCESS in 3m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8c127249e6ab4962abaa8ac1b5da38f8 : SUCCESS in 8m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/733e0ad9b26c426fb2dc401c0e36f14a : SUCCESS in 8m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/84f109ea64044a9ba718f2342d2a9289 : SUCCESS in 4m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c3120e8569ed425f95c498523f07b788 : SUCCESS in 10m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ce9853eccd14471c93483bd84248797d : SUCCESS in 4m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4b2c98753f4948e599d2c89137793a52 : SUCCESS in 5m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d5a46aa0faa84a7182a9fab4b2e93682 : SUCCESS in 8m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e190e5c304240a2ae49dc9d78cf07a7 : SUCCESS in 8m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/965cd239df914e61a48d3a30f34371d8 : SUCCESS in 8m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2c01003f091d43b5a952bfa9596b015f : SUCCESS in 13m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9c6f1574643d435d9091014f8f23d50f : SUCCESS in 4m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/71246f54b5d349a9b1f21b6e94e4eff2 : SUCCESS in 12m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/910fd9ca5fad41a686ecfbe59d3410e1 : SUCCESS in 4m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7b4402c140bb4469bf503d39dc4e9720 : SUCCESS in 21m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2992e949efa442df9290b0af4cb46312 : SUCCESS in 10m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fd397266e0b74d56adbc4f6e8ba1d155 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/335c36a9901a4220a275b6d7569d551b : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/976424cd4fe14e43a9280e80c635344c : SUCCESS in 19m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7022c3ec47b44358b0a947481f7009f3 : SUCCESS in 6m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7968dba2cf5045d9a997ba705bf105f9 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d41d8172f4b64e0c9afcc9d986beb247 : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/88f4dc47a6e243d999fca020526e3b4a : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ce400946b114a8085f42a3cc8e9185d : SUCCESS in 48m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b811b36e3954345b2e64af9ebea8ed1 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6d76256723d5fd3f55698905b2aa92d9d3ce20fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 18:09:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b3bffbe5906e990d231226a24a1a63234174f0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 20:07:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a819133fc344f60908f0797b3e8cde2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82deb06b0e40458fae1202906aebeffd : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbaf8efaae604db3bab23e7edf102307 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37c55d9cd3584f4795654857dcf123ec : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03710cfaebb341d39cfca82cfb6bb431 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e62947dbbf754b87a296a04f4926217d : SUCCESS in 3m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/976d42d1148e4d3d94fbfee7e8802010 : SUCCESS in 11m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da2f1b7e7e364862bf53ea0305d17152 : SUCCESS in 8m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8136ce3ae86544d29fc0e504cb94bae2 : SUCCESS in 10m 24s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b8141c16957040eaa978cc6143b77c97 : SUCCESS in 4m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d8f02c7cf2e04679a5e07ce5007be29f : SUCCESS in 5m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/aa759e11f9be42e4a95f2e47c4b6b39d : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb6752b76ae64cc28468b18611e7ef95 : SUCCESS in 8m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d24bdd185ebd49c6987e7283b848585c : SUCCESS in 7m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3dfa2f2822bf44c7ba0f6ce618a33286 : SUCCESS in 16m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/174b0810c90a40429abf10899fb5bf23 : SUCCESS in 3m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2edf0c79f31b4994a460f4cb0212ec55 : SUCCESS in 14m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ad95d9fc24bb4823b4a84a2a94dbbee0 : SUCCESS in 4m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dfb8a85086f5458ca35eb93d6168f3e4 : SUCCESS in 22m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a006ede8f9fb4794b06165cae67bd23f : SUCCESS in 11m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9982c669abe7459fbf9425a96c429ec1 : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa76ab030f1d457ea3794b87dc9e98c5 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f569054304142ddabdc61bc31bb20d1 : SUCCESS in 20m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/66f164b255794dd3a2c1de2da7b456c9 : SUCCESS in 5m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fb6003e79f454458918076f615823d91 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/085db7307fcd41b7a1e7593e704a97a6 : SUCCESS in 6m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bbdec1b6a2ef449482df3063ec307569 : SUCCESS in 8m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4ab5cdbffbb49868bcf9c43f6dfda15 : SUCCESS in 59m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b9a32293ade4b05b9d8cd6907b6067e : SUCCESS in 1h 39m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1efb3d8b60ce7f65401be94aa295320bf507de88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 20:07:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2faf59db501f5684fbc7d663b31f699c6aa6599","revisions":{"a2faf59db501f5684fbc7d663b31f699c6aa6599":{"kind":"REWORK","_number":1,"created":"2023-06-12 14:00:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/885883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/885883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/885883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/885883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/885883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/885883/1"}}},"commit":{"parents":[{"commit":"b30d68b4306f8e4d411794279505f2593f9e3be3","subject":"update constraint for ceilometer to new release 18.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b30d68b4306f8e4d411794279505f2593f9e3be3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-12 14:00:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-12 14:00:19.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.20.2","message":"update constraint for neutron-lib to new release 2.20.2\n\nmeta: version: 2.20.2\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I42946e7a9ce2cbd466ce59702de70eb5c9019a02\nmeta: release:Code-Review+1: Miguel Lavalle \u003cmiguel@mlavalle.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I808e3d677d63095a96cd12f82851939807b8e002\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2faf59db501f5684fbc7d663b31f699c6aa6599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2faf59db501f5684fbc7d663b31f699c6aa6599"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
