)]}'
{"id":"openstack%2Frequirements~886599","triplet_id":"openstack%2Frequirements~master~Id1feedcc496959cea1183b14028f767fca674aa5","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-09-22 23:37:38.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-22 23:37:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1feedcc496959cea1183b14028f767fca674aa5","subject":"update constraint for os-api-ref to new release 3.0.0","status":"MERGED","created":"2023-06-21 11:14:15.000000000","updated":"2023-09-22 23:38:31.000000000","submitted":"2023-09-22 23:37:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886599-new-release","meta_rev_id":"1d61224709692b08a2507d2206272b83029235ac","_number":886599,"virtual_id_number":886599,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 23:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-22 22:34:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-22 22:34:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 13:50:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-06-21 17:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-02 17:56:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-09-02 17:56:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-14 19:15:33.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"CC"},{"updated":"2023-09-22 22:34:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"79301946bcda54d0181c09415715419ab9f8c3da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-21 11:14:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3d8c19d8bc86d8918eee1bcd11e37deb227005f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-21 13:50:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"25c5f01897d602c1139fd4e5e9313ff00433b118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:34:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d54ba290523411194cb96d83abd42dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21c5e1de2ee4c0bb7076770c19eb2e1 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dbd7ae12ec94dc8b12d2d5b32232c27 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c41e3edfff494498b3581f1ad7a84d22 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fd3071543a844f6847b4f421bb6fb68 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe5f09024c854f569aa848fbcc8ad430 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef7094de8613471ba714752885c0fa3b : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b8138e6ea1441ebbd9469ac797983b8 : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/feb4b736063b47c993d340c1accea41d : SUCCESS in 11m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/beb5ee16750a424a926a29ecbf97ac85 : SUCCESS in 15m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/485395beae1b4e4fb8fd4134a92548a8 : SUCCESS in 11m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/32164fcc92124b468526d2ec85fec464 : TIMED_OUT in 31m 33s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/dfd5fbb21d13428ea2f16e072803492a : FAILURE in 5m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/733946875d2a4ad9b8d8eae3de76297d : SUCCESS in 5m 18s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2c14953413fe4c3ab12d5b7dd5e5c969 : SUCCESS in 5m 57s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e5f58b4158e14b9a8c4a6874cabf368a : SUCCESS in 12m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a1077703f21548efac98b72f8eabf0ad : SUCCESS in 4m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6668d5d78f924c4c820cf441a13eacce : SUCCESS in 6m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9e42a8dd2482441ea8ccfc6e1d6a6c5b : SUCCESS in 7m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e17e8c8b6d4d412aa47eca9d4ea0439f : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f05557dfff04b9db3753bf04299f335 : SUCCESS in 8m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cd5dbca817054cc795c9f9b38c34d5fc : SUCCESS in 7m 49s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7f675d28547c4c629eb318622fd20f81 : SUCCESS in 29m 42s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9e29a5db7c394e619a98e0b97a006833 : SUCCESS in 5m 24s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c4f85d090a70487a9a1ff32727ef0c3d : SUCCESS in 11m 20s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/47d582eac17c485f9cb1293a4c7dc6d8 : SUCCESS in 5m 34s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/124522925e9d40a899e3e834b2cad630 : SUCCESS in 11m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cfeca4ddb8a0453a8e64c1d1345c17b8 : SUCCESS in 25m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e666efbbd78843fbb84655804b21fde5 : SUCCESS in 11m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8bc4326eecbb4ee382fa959df3780512 : SUCCESS in 14m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b06776c446b24b1c955fd1faa46abfa8 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e19ce222b94419892ffcb97bb93f301 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/adb68f0f13c34b76adad1f543ff22241 : SUCCESS in 24m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b7d599be05d94c289d50422eb59341a9 : SUCCESS in 6m 51s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/48e45fd7aa4947a482e63435f8d71305 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/567d57ba59dd4897acdbb043809d9a81 : SUCCESS in 6m 45s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/27974a591db8420aa31517f4d53b7809 : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d77f2a2fbc64d2893a2b2e5e8dfb369 : SUCCESS in 1h 10m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/759261dd2a574a6facc8d31eb149efc4 : SUCCESS in 1h 38m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5a8d3b181a91f720c769b030d5e5b222ef8595df","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f96642282d2b9943f689f3ab32133c5f89baedc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:21.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 886599,1","accounts_in_message":[],"_revision_number":1},{"id":"91e28d11b1588d4499a647bc3eaff85f9a284dd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:26.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 886599,1","accounts_in_message":[],"_revision_number":1},{"id":"3ec22e8b068468d006937399871e56bc72d13898","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 886599,1","accounts_in_message":[],"_revision_number":1},{"id":"830255db31a3536329668d9772c2d5973ff0afbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-07 12:52:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f8404b25b0f26278164f48991cd07d9c2fab4ae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 14:22:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df2635e335bd4969ac4256385aef0edc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25fcce7441a44c61b50c74b7461abba6 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312aa70911264a9b8bd50f962098fb78 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/063b6343082e468698bf33736f04061f : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/521847f99887431c8889cf5e30aadfb1 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e05f0e247984905afdef5e0592dc459 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d3fc47510ea4d2391585be25f1daccf : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ee8b41124d9405e8d2590dc8b1a4bd2 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc6ce8dc1ef0466394e0b1145d772020 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d1df3cdac5a04403b47e7e04640882e5 : SUCCESS in 8m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6e133002a8e148d7a1950213f854f624 : SUCCESS in 13m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b72f182edba54085be267454267c23bf : SUCCESS in 11m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1fb7acccbaab4b7f9c554315d1bcf4a8 : SUCCESS in 11m 53s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/99f9df7b38094547895b5b97fcc1f282 : SUCCESS in 7m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a319077de1f349fa8d9e929f3921b8a8 : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4990cf11f40b44159ff711a707a8d87f : SUCCESS in 4m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/94dc78b0713e4d63ba36211bb9e884b1 : SUCCESS in 11m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6c6b41c097b248c59fb2f9b4a0adaf2b : SUCCESS in 5m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/951c35d95d8c4a7ba422e25c358a6bc3 : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2e01dc1d18a94faaa2eec0ba8fc82f2e : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a37d420215cd4c07ad18be63f6efc6bc : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3cf5ff8bdc04e108dfa72eb32590770 : SUCCESS in 6m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/15953349b8db48d4a180b831284032a7 : SUCCESS in 8m 11s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/66c5b5d1d4ff4feb9bde1ee9293c1d39 : SUCCESS in 12m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8747cb72710a46ab8495dfac0a25d3d2 : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4d89dcb4288b45bd8c388dcd4d180620 : SUCCESS in 9m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/619b25de38d341069f6d7256b7cb3c55 : SUCCESS in 4m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1cfb5e76f3b145dc87da60b9506a6abf : SUCCESS in 9m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a52676ba008a4bd2ad71591403b22d3d : SUCCESS in 21m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ea0fd99a421f44ab9a55a257acdea39d : SUCCESS in 11m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/02c1611922b7431d8f4033219d3e3d3b : SUCCESS in 14m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/515ff4d02aa94c099ff1209418c7be47 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2339ff330ad046fdb6d22f144df8d37d : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e492e5d04074277ade087698ae39f0b : SUCCESS in 24m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7599204607b64a8280b6585c5937ee96 : SUCCESS in 5m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6071ad43354b40ffb2c7cd499b55c13f : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbcb32d8af3a40eaac6acb9f2fe35c3d : SUCCESS in 6m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/48a2de3db2dd4fb4851365b7af12253c : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f199bd74cf284190a678f7204333e3ba : SUCCESS in 1h 01m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c88b1f5b50954631ae5386f79a9ec621 : SUCCESS in 1h 26m 48s","accounts_in_message":[],"_revision_number":2},{"id":"307d7d112963f7395123070ef98d70296fa5788a","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-09-14 19:15:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb9fea1d6fc49d6441c2fd821d96db80e4e35f55","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 22:34:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4deab814508bd8d04bc2672e87052f92f4421011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 22:35:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"01e7277f40936bf1c4c9c6bf59feed9bea8d358f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:37:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/094c9d57e1f043849387e33c248126e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39dbbc13b5dc480f93f2a1404033d13e : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52e3215a3e4d45c88734012dbc05ea90 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/997b037cf1284b7299a4ede7d0e4b029 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62519bad423947e886b0f88f146cc0b1 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ed07e1a67ac44eb8a2c3cb3c55f99d6 : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3e7ff333c06426eaf08ceca4debcfc9 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ce55bb7b0ff47fb823a87ca940eaf91 : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f7fc5bcfac5b40e2ad32e26683041a83 : SUCCESS in 6m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e5646d3991e2432a9536bcf06f680ca3 : SUCCESS in 11m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/de8f31e8494c4eb7b8ab51904e14284d : SUCCESS in 9m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c55c54eccbb343cea5b2f68accb38adc : SUCCESS in 10m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3c70980cfbba4eb5af17a5ad6b95d7e8 : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1e21c95ad5bf4219b95bcc5f859b0473 : SUCCESS in 5m 33s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ab8122d3dc2045e3b5f566f1d137a0ce : SUCCESS in 7m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b8660a80d90048d3b1e49cb8abc092a0 : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e987bdb38bb949c882b04fb0efa72a6b : SUCCESS in 8m 09s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8c219d231c7b4e049d1ce5bf911d449c : SUCCESS in 8m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2237a009e6e8435ba3b62c062d15a023 : SUCCESS in 13m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6f826419ede44d29888874251c8d2485 : SUCCESS in 4m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ff0a18a8dbfb47dbba8e0b371561def1 : SUCCESS in 11m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9e55d7d6b5744567ac4c5c2459d3e9b4 : SUCCESS in 4m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c6841730d9734d05b0af3e6e68df17b1 : SUCCESS in 25m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c6ca10c619834784a34a6a643cc33c94 : SUCCESS in 13m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f97db468d97742039196c266f8a8f70a : SUCCESS in 14m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3a2412e6924a4c0494e0a8b5f75a3ad9 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c7c550ae9ea4441b440bf27eff64721 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d4c0770dbdf84d4399eb122b0008c353 : SUCCESS in 20m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c48122d44eaa43ff8f248cbf32b68b8b : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/edf5f7a7f7164f8aa28a26eafeb8deb0 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee6a0827da6d4f83a490b98251fec4a2 : SUCCESS in 6m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/15d5d0fe12c94ab18abf40f914d76680 : SUCCESS in 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6ed8287831a4915a753f1c0991f1607 : SUCCESS in 50m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d70bbb130704c989c6662193aef3d53 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e3890b6afb0e4573fbe192b124aaccc66faea655","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:37:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1d61224709692b08a2507d2206272b83029235ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:38:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5856988d3a0482d9dc87cafec0274a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a451c11f9a48e9969c17b36551b0d7 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"35c58c69fa1440ce9db3f6183efa85b534de0a3b","revisions":{"2a59a5da71b66f3171e8bca15ea4cacdb031ca8f":{"kind":"REWORK","_number":1,"created":"2023-06-21 11:14:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/886599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/886599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/886599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/886599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/886599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/886599/1"}}},"commit":{"parents":[{"commit":"265cad8956fd4d018f4aba4702247785d9b379f0","subject":"Pin XStatic-Font-Awesome to \u003c4.8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/265cad8956fd4d018f4aba4702247785d9b379f0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-21 11:14:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-21 11:14:14.000000000","tz":0},"subject":"update constraint for os-api-ref to new release 3.0.0","message":"update constraint for os-api-ref to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ifd37cec84e021ae93986e8cdf09ce64e02c5e006\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id1feedcc496959cea1183b14028f767fca674aa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a59a5da71b66f3171e8bca15ea4cacdb031ca8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a59a5da71b66f3171e8bca15ea4cacdb031ca8f"}]},"branch":"refs/heads/master"},"35c58c69fa1440ce9db3f6183efa85b534de0a3b":{"kind":"REWORK","_number":2,"created":"2023-09-07 12:52:51.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/99/886599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/886599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/886599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/886599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/886599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/886599/2"}}},"commit":{"parents":[{"commit":"237d49d5ba63b817eece6376a284e990654f4ea5","subject":"Revert \"Add Spinx RTD theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-21 11:14:14.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-07 12:52:41.000000000","tz":120},"subject":"update constraint for os-api-ref to new release 3.0.0","message":"update constraint for os-api-ref to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ifd37cec84e021ae93986e8cdf09ce64e02c5e006\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id1feedcc496959cea1183b14028f767fca674aa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35c58c69fa1440ce9db3f6183efa85b534de0a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35c58c69fa1440ce9db3f6183efa85b534de0a3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
