)]}'
{"id":"openstack%2Frequirements~886743","triplet_id":"openstack%2Frequirements~stable%2F2023.1~Ie052bb5578619001dc24f39b98955565eaafa925","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-03-01 01:22:23.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-03-01 01:22:23.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-03-01 01:22:23.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-01 01:22:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie052bb5578619001dc24f39b98955565eaafa925","subject":"update constraint for oslo.db to new release 12.3.2","status":"MERGED","created":"2023-06-22 12:06:25.000000000","updated":"2024-03-01 01:22:23.000000000","submitted":"2024-03-01 01:22:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"886743-new-release","meta_rev_id":"20bab70b7866cd6fe2855225892a762894053403","_number":886743,"virtual_id_number":886743,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-01 01:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-02-29 22:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-02-29 22:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-02-29 22:56:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-22 12:06:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-22 12:13:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-22 15:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 18:16:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-11-09 14:30:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-11-09 17:20:32.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2024-01-24 21:33:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-02-29 22:56:09.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"1e22ea9914ab9fd24abcd5f674d0eb061c77636f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-22 12:06:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"040679a5b68437571a5b0d05e484a2a3517296ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 12:13:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b969c61d5ae04c76a18bb47255bd7108\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aae14dab0b1841e1a2710f8669cbc7ec : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4c7194d4c7fa7cb2450856b21dd6b8e5721a1985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:50:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7bbf1618c824b00adb5a6931fc03314\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35f2ce4b65554f548ab36cfa941fdcfc : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/398a3b70905349b7b9d5857d71d8e9f4 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5650759efcd42069c6e462b9c37e4e2 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1397d88fa3084c0d868abd95055cff62 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/188cd384bb89423f8c18637f7e3f8593 : SUCCESS in 4m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5253a062d13f4ad3b7ead9ab3b5aff1f : FAILURE in 6m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f9b30b0596784ec4bd2073ee35eeecaa : FAILURE in 6m 16s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0d22d035befd4d36862ac23b5efb9d2b : SUCCESS in 8m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/68fdfa13c4234b7ead09e9480c39d227 : SUCCESS in 5m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/024cc10d00194bb980c21b055869e364 : SUCCESS in 5m 54s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/92496962ef8e40e98529a381d43fc3b1 : SUCCESS in 10m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a6f6db80567b49e186c3283082134654 : SUCCESS in 4m 34s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/63f76610f75a41779b8d34f6bf72bb5e : SUCCESS in 6m 15s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/284a98640cc34fd1a266eeacbee9616e : FAILURE in 4m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f26db8d5e3fb4bdcaadffae1bb39d409 : SUCCESS in 13m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a4a4731a90348d88ba159de877f7e86 : SUCCESS in 7m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4535951eac304793930f031e761e4b74 : SUCCESS in 9m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8d2af741630647adafe567bb75d081bf : SUCCESS in 19m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/935237b959d6462dafbe39b8536d3db8 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6d7b64059ed84fbab9ff8542e5fe4f3c : TIMED_OUT in 41m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/158399310a74438ca3b2a5ed0796f4e2 : SUCCESS in 4m 35s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c59c7578653744bb819411fd9b7f6801 : SUCCESS in 10m 54s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0c149e11e13c4cbb8e24beb1b75c7a29 : SUCCESS in 31m 44s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cf01ab86d3444d6aafaac00caf1fc696 : SUCCESS in 14m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b0a66c9211494296a193a827a1680bf6 : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b5b2784e2353453ebc21cd10a88f8996 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63cfe8c2ccc748a0b28eff9f8a067296 : SUCCESS in 21m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3a047d3022e64fff9ad35d4d9d601443 : SUCCESS in 6m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c6ea20238f4941428d544e239e72b65a : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7384b921becf46e48972f008661d32e0 : FAILURE in 3m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2094458d1d3144f99451e7b0fe5683c0 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec147567c63c46e1afefa68b2354fd62 : POST_FAILURE in 12m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcc3e61c75de44b9a1cc0db8206a26b5 : SUCCESS in 1h 28m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b55c670bcd9302584800fac74cabe75821839390","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-22 18:16:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2062f42495dd919b6c1918fca1de632f51320ae8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:23:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc6d2f929524e07ad393bd8c19f2be0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7a18aa2861544bb592ae2406d16dd808 : FAILURE in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"94a3160f40704ca317e45948ee48bea73be1261e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:08:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b1648ff49f64ea494110f9b69ee4128\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737857ddb8b24569b59d9225c80d45d6 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/592fd17636fd46b890e639e5e1eebad3 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a019964f8fc847e1a12c8e1745f26064 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aede7bcc41674497b4a03230206744ae : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b6771e671ba43c680b2a40390db7a86 : SUCCESS in 3m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4699eac9bfe9450390f3efbbda018e6a : FAILURE in 6m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/982629dba904493babe6ea8e2b1351bd : FAILURE in 5m 45s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7764aa1a984d48e29cb8ab5eb5ed1efb : SUCCESS in 7m 22s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/84e43fbb91d84fdeb865c853e96ed09f : SUCCESS in 5m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4e6b7adc8a0c42e29ab0997afcbd7c26 : SUCCESS in 7m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1af7d7cc0b5d4319b27c8b2317acee9b : SUCCESS in 11m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8003346c5dc34cd0bfe0371dbd1521e8 : SUCCESS in 4m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/694fdf00a5fa4a668622c75dfe931598 : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5834da9bfacc41a793c9b173c7669a56 : FAILURE in 3m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f7aa71e5b5d24392a9643f6f9fc9fab3 : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2866da527ba4d92bb557b3dc6cabd03 : SUCCESS in 8m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f516067a5602480a863669f9ec423f40 : SUCCESS in 10m 39s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d8cb616dcd6f4e9fbc55c3c9f1881f7c : SUCCESS in 13m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/71e5add96ee447a08f21b5cc5f228cd8 : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5cf0de9c1c9d4790807b931a0e6bc61a : SUCCESS in 10m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9207335362c947e3bed713e5c37889f0 : SUCCESS in 4m 41s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4bd45e64c9cb4067bbeb1cd6a9e3bcdc : SUCCESS in 10m 53s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3dfb11e1fe6849e7947e737b6f3eef52 : SUCCESS in 27m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/df6f5b82fef44bfab95b3a0b93b448c2 : SUCCESS in 14m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bc25d85cb59d48e28a4a9f03e21c05d0 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c890fefbeb8840afae23e962d27a807a : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c0e904da1da4749bbc40020f68d8346 : SUCCESS in 23m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d7bf85039d1444e4b1e88c5f2fff6f5a : SUCCESS in 6m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6cb81f8fd05849e99a6ecde87afc14ac : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d98676b5f6f4de8811dedaeffa8c6b5 : FAILURE in 3m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/54702a12ee2b4b788b030b16420ff4d8 : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddc37ff223da4f5d89a40c862fc0d8ec : POST_FAILURE in 13m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/decc986281774faeb57a1940fe470786 : SUCCESS in 1h 44m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1c2ba8f6490a699d14526718ebbd9a64bd7a86fc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-24 05:01:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"904744fb4b080e8b404d3214940214e9be510fc2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-11 06:53:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"704723f0af7a5ef6e0877ac2928d6ae19f27abef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 08:13:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f327d2266b943168572a3e27691daae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21d376c6eecb4da4bbaa7f9af84c01c9 : FAILURE in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f58dc10e30e9490c9a1958135df641e9 : FAILURE in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60d807fad0114e1ca0ea2f985abd72dd : FAILURE in 3m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5dda4b577274916afa7b8d49a61a4ff : FAILURE in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b362329fb6dd4b64b008e8a58bdb9ef1 : FAILURE in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c91ef24846314622a3a9b104214544d7 : FAILURE in 12m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ee5cea35d532468785ef3ba50eb56e1a : FAILURE in 6m 37s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5c1cba1090d743a094dc638d00f1085f : SUCCESS in 7m 24s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6ba147895a3a45ff8b60f0e06b71f2f2 : SUCCESS in 5m 51s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1f76c2a09cba4110a6bd87b5e4bebb18 : SUCCESS in 6m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/62c5f739d4104df090d39071ef127272 : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0072d6f07551432fa6984fef27997e9f : SUCCESS in 5m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cad6c014b36b41448e6672644995e5cb : SUCCESS in 6m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d7948aea91bb488394bcc5fa9d7bc428 : FAILURE in 3m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d576420cafef48caa4078fc5ddf161e9 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f321cbd47354d39939074d1b3a0b37f : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6c76760efb8b442581ae31b0646c3318 : SUCCESS in 9m 07s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5fae6eb8fda248df8c8022e719f39fc7 : SUCCESS in 13m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1c080ca5966c42b6a27858a8e60e8745 : SUCCESS in 3m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/97551b332d82413e8f39784bf8d305e2 : SUCCESS in 8m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d8b1583262d841d2a4e75cc8160ed74a : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/997ee4192c6d487c85f2c25fe106193f : SUCCESS in 11m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/efbb0aca17b7497ca4ade516e0e3df00 : SUCCESS in 18m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8eef668871b44489a6f1d77153584efc : SUCCESS in 12m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5e0611975cee48498873add61d265101 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88d8685a86a8487e9606eb4414921b41 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/446cee2971054c2dbbc02033a08c8f90 : SUCCESS in 20m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/dc9274322f0442e480ce2b7a9d160788 : SUCCESS in 8m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8a1694e52d754349a279fd8c1b2f391d : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e0ddfaf8e6c74fbbb237da9817050f62 : FAILURE in 3m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ada98c6882dc42238064bf7d9e3510ca : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16d51bccedb143c3823a4529f06de57a : POST_FAILURE in 15m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/713b06d310c54509b7d64cb3374a6075 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e78f0ed4b6a6d88bef75db803d3cdd12de6a2b3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 09:36:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c3e0a84ed42414980c77046fb6688c9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/32ad6b5100634a5e9f5f4c56aa7136ff : FAILURE in 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"835cfd483b8dbdca06649430e0c38b6f522b610a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-11 13:48:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cb2cb244fda244ddc6585e1a7915585f86a5974e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 15:12:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef3ec5d5a0304e8997774e165c2c782c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f33f8d0d7af14f13adca1babd3becb85 : FAILURE in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48d2ef36127e427e96be6e79f4711c72 : FAILURE in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c1a0c6b9be1469099a17f14a7ce038f : FAILURE in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2333274a1a940799243b7bf5647acaa : FAILURE in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a3100fa9dd94bb8be0f9060e89a951e : FAILURE in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73668645799a449ba0da3c5707c76478 : FAILURE in 5m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3a93f6bc288d4bf392277a16ea082368 : FAILURE in 7m 26s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6a0d9a2c8ec748b1a035847528a9a9a7 : SUCCESS in 7m 24s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/813535bdf5bb44a692463d0f82bc4223 : SUCCESS in 5m 01s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/2dfc24b6ebdf459c98ea604f041cda06 : SUCCESS in 6m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/530f792048034ff2b52ed68abc4779a8 : SUCCESS in 14m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b29e63fdf53a4e4482f2bed48bf25df1 : SUCCESS in 4m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3d9dfa689cd64cf3a47c140094fd6a3b : SUCCESS in 5m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b90d83fd4a9d4bdaac7bb4bdfab9ce4f : FAILURE in 3m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3e74fd301eed4f5bb94c414118d3c7d0 : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66b3c859a7f94c11bf4bf02681e092b5 : SUCCESS in 7m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7dda1906837e4d1a95c941332a1c036d : SUCCESS in 8m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/42af821eeafa4d4594f9be04e5998c4e : SUCCESS in 12m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bae7cefee33b4542b509b4d34423b4a2 : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9c2a55dfe3184155a1f2649d40a30195 : SUCCESS in 9m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8aff7fb334644d379910ee55c82c135b : SUCCESS in 4m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/388a79e6987c48119d155225f374ca1b : SUCCESS in 10m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2ddbe69a516041729a0d57a9d67f241e : SUCCESS in 22m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a613131e54b84efbab2a276f3c3d87b0 : SUCCESS in 13m 22s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/63ead240dba14a369ab681db52df2d51 : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/83da4e9503bd41e2b1b898df63a082fd : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88219adb19b5434e911192e4f2a989da : SUCCESS in 23m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f9f46db9539f4eef9e8a010c66af3f04 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/79187ae4681f4d6197df58e333bc8f61 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a82663ed560e4688b39aec47cd51963e : FAILURE in 3m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/af94ac6082324ea1a5d2bea3ffbeef22 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b4b323ac48b4794bfde1ca6eb35b4d8 : POST_FAILURE in 10m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79d0a545a5b2437591e3ddcb661d70d2 : SUCCESS in 1h 21m 13s","accounts_in_message":[],"_revision_number":2},{"id":"fd89cf64fa7c023c1179ec22025197fae3003bd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 15:28:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3677264f684f4fd2894be2410828d7ea\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b7cea604a78f43fa8b15227c53e2abdd : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":2},{"id":"13a9493ca03d9de003c8703d06f9147e419d7a5b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-09 14:30:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7804804d39bb6327c0a664873562982c71c40cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 14:35:31.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d6f2838098742ea9c5412b29d7d1ce2\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/04911c81ee354e058a79239320dc0b2c : FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"a29d46d49e42263928b7a84eec130ec7e2ff3f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 15:53:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73c751eebfb94f6a81cdccb0b89f6287\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5429031eec41589ee568073a1a5090 : FAILURE in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/505db99d09a54009a78bc5b93bfa1715 : FAILURE in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f90ca6b41c2042019b2345e55785fdea : FAILURE in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d351bcedfc3f4c66b911882af2f43439 : FAILURE in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78e9fde6d30b4a9594afd9151cba22c1 : FAILURE in 8m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8f4ac17eb8c9495daef164abfe7da859 : FAILURE in 4m 54s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/468f320204814af0b325309b98685741 : SUCCESS in 7m 53s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9fb7e187131b45019b0909b04b68d16b : SUCCESS in 5m 43s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7b415c24f245493e9993e891cd8168ec : SUCCESS in 5m 09s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/37a50fb50f9b472b976f4ae4835b95ec : SUCCESS in 13m 17s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/77a6f1cc3a004b4aa7def9db8b774e01 : SUCCESS in 4m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cfaaffe2faa14aeb9f401ae3d210daa1 : SUCCESS in 4m 36s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/34d7801f51cf446bb096b40632204277 : FAILURE in 3m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9d3466b2814e42ecb1fd2b0100164793 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c147c1bf7f554dee8295c6a26e92056e : SUCCESS in 7m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0b1d6f861d234271a963481ee877a109 : SUCCESS in 8m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/1641a838a89445af8492a5e3563d97f7 : SUCCESS in 11m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8663260d70464c4c9aa49a4f4d4f1877 : SUCCESS in 3m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/41e1d20280694bbb87d8d96c975fdf07 : SUCCESS in 9m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c61e78ac732340c6a75b8abd7920df01 : SUCCESS in 5m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9f38ca83c2324d3d8828d5a7fe0e116f : SUCCESS in 10m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2a437c2b26fe4918a18e9f1a80f26d96 : SUCCESS in 25m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9f9d9c7b801c4f4b80f7b8a68ebfe5d2 : SUCCESS in 13m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2b949c4af2a74b2b8ac2cb01ca98282f : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3defa944e34945078cf590345afa645c : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6860aa15296341cea8690b84de66f19d : SUCCESS in 24m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e79ea51abb0b4394ba99500fcd857679 : SUCCESS in 6m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/46e61561311244ccb50557f3e8452784 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b687e5c24b9248babf3ed7f1d9380aac : FAILURE in 4m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c2c017d38d17496da6de1d96549a9b21 : SUCCESS in 9m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9646c31f3f3454eaa06c8bae33f4533 : POST_FAILURE in 25m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb73f7c64eeb4e3b81b3fff43eaee135 : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":2},{"id":"489ceab5acbbc5a289d76bf07c3b72e759a67319","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 17:20:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57317449756a35f22be0f0627cb4acca534dfdc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:25:43.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df5307545274ae885bedd9c3bd0fd9c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/646d311002cc430e9d5111be8739309e : FAILURE in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"eb6c940cddc7c397eef04c399a1dd995b1191319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 18:51:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/602fbd70e9fd4fe1a3e4e658ff1fe6d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457caea3a5234275bbe6738a8124bb71 : FAILURE in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd76069373cb4bab931a3944c34c1a78 : FAILURE in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff71d31d91c04e6daddd1f180dca2a04 : FAILURE in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c1d534e864d94523989e5a9bde5275af : FAILURE in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/61ccbb7aa76c4f5091503c9d20ad29ac : FAILURE in 4m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5ec718929046441cbd3024d75f80fcfe : FAILURE in 5m 25s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4a338e1677c6426f8eb75311f7dbca07 : SUCCESS in 9m 19s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/93b880e86bfe44d981466e898f1de763 : SUCCESS in 4m 48s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b1f21230868640f397bcd5ed380c8613 : SUCCESS in 5m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f35822ceb7464b14926176f8f68a22ea : SUCCESS in 11m 49s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/46a18dd7b66848baa1a0123b7eb41461 : SUCCESS in 5m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/662a4c69fb8046b88212aaac8e964380 : SUCCESS in 5m 15s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2521f65614fc4ea580f4b2df04d8cb2d : FAILURE in 3m 45s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/55261bdf6fb44be9a4105b847fc008f2 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/77580fb688b743fc8cffe3955eb68be2 : SUCCESS in 6m 42s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/995abc9626fa4392b0f58a371bbf84a1 : SUCCESS in 9m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3ece714da44143af80a86cd29ac7102d : SUCCESS in 13m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/19d2a7556ad5438084bd54d750c454a7 : SUCCESS in 3m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ad247e10382542ef925703b5fa145927 : SUCCESS in 10m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/34354194332341028cf3837eea7c8744 : SUCCESS in 4m 58s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/aea0ea5b4c654952a5a556811f31e86b : SUCCESS in 11m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8c0b0e67cff845368d8a061900711a72 : SUCCESS in 24m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/31abcf8ce61e468aa033d719772fa04b : SUCCESS in 13m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8deb7b0ab50d465fa3722838e2cf847a : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/02583a771ae349189645b01d7eb94252 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00226fd5cca0425f9d261fc204759ee7 : SUCCESS in 25m 01s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d7015f4ca85147acacfc334a3d36ce1e : SUCCESS in 5m 39s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2a9c95ba146843e1b544ac36cec3e949 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cda885f4bc124540b2b4ab903494ec5d : FAILURE in 3m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/08cd9452cc5a4fbfa3e29535c66ac5b5 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1e81cd265e0463cb8496c11dd639964 : POST_FAILURE in 23m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d24f5347ebc4b258cc322fb5fbe2ef0 : SUCCESS in 1h 25m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4d9269c2bb14dcf788dbd551f3716e8e619b4d71","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-15 13:44:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"139a534eebb21380597d3b62d2db806bbbb9fff2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-15 13:51:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6171fd9cce7c02c18b9b0dcb1e333172e02a624","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 21:33:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99d410f9aee51ac6244708442611bda7f985cdfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:39:56.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3dd31dc1c843788136210d799db166\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aaa546e42ccb4c6e8afd4315092c7a29 : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"218cf49e09353274758b7912162269d07d35b6da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 22:42:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7f79648a88e4b559f29816b7b6c8309\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547792b4bd754be586721a701223ecd8 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5737cf4d6b9040baac749c67c9ce9937 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/494d52853940402ab3fc8c81a6f344c7 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/817b61b7e10e4312890c67cbeb241c3c : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/be073223cc2e4b5eb00eb31c764b96d4 : FAILURE in 4m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0e2c2f899ccb4b85aaee3ea37e65275e : FAILURE in 5m 31s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e3cf8318f532494cb5d902985faafc47 : SUCCESS in 8m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b6267b58a3ca4d7394c61e4163d1b80b : SUCCESS in 5m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0f9251e117f149ad9136f6cc938022c7 : SUCCESS in 5m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/eddd50d39d224a32a8f521b0ecb2789a : SUCCESS in 12m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1644aa2148c7493da2d8055fe7a9304b : SUCCESS in 4m 56s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7679aad7b2cf481aa909df2ae23b2687 : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f899363b3e224d04b38b457bf47a69a0 : FAILURE in 3m 23s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a1154c309a744000b39a9af9f3771bb9 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c03329867244ce59931909f4325f5c0 : SUCCESS in 6m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/915ade5e52a247c68888cc3430fe0911 : SUCCESS in 9m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0ffe72f1ec514cc1a7b0699047a0e9ec : SUCCESS in 13m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/bb7f84e990d949d6bf54f183aa6cbcf6 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f93af062a2274b7792eabb0291a5c0ba : SUCCESS in 10m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/64d8fd7af9a844b4bad566705cd571e9 : SUCCESS in 4m 24s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8d43355ce3764a1286980a3d335073a2 : SUCCESS in 11m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/254b815476a3448fb9ece30da74054b7 : SUCCESS in 26m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/58a22e17fcd5467cae1c9b797b517741 : SUCCESS in 11m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f51ecae09b8744e7ab8a6b498c1c1b05 : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/751be5045f374e6ea754939de481d338 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e4f8b4d64fa4a63b6c93c715ea6d869 : SUCCESS in 22m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a6f373fc681f4ac8a3c928102a7d6b60 : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/afe410e3d5fb4a998c875595e527c461 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7545f99341224d53a9d8dff7d6ad9f04 : FAILURE in 3m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c2342f65471246deb679ca2be945dfd2 : SUCCESS in 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6af1e0030bb74560937f88788fc2d9a5 : POST_FAILURE in 28m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2bcc7e993fe4a7fadd0adaa2ff78048 : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":2},{"id":"7672c1179c47fe2bba448c0cf4b74f5dc9a84e35","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-30 08:52:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1410f66dd19e99ddc39a58a9392ee90d0a9103e6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-31 13:02:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7e3bae548cd8107c3c3fa63bb8ee09f7d364e95","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-22 14:40:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4aa716a3bc2aa4f48c900ad49e16cc09e40b4d03","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-23 02:22:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c47819476faa17d097f79b6972e54e79187e1220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"real_author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-29 22:54:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_11904\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"_revision_number":3},{"id":"15f86e8798a55d031b43633737cd6b4580093ac7","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-29 22:56:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a99454679fbc2fb8425dbbb79e977b0ba8719bc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 23:00:26.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58f2995410f44e4ba43fc5b8a9a6c7c7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dc269972a8ef46c694e022198d3559eb : FAILURE in 4m 53s","accounts_in_message":[],"_revision_number":3},{"id":"c55b5cbb478cad6eacdb5188d254c2267cebf5a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 00:10:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cfaceb504a444c788d5770c9f997b0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2263fae9fbbb428692e758bf31dc3f4d : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c6afb402147402b90756bbfb2411ac5 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b9a541d9d0b4fad91344568582718f9 : SUCCESS in 5m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74effcff173b4323b2df4e21fa2aaf07 : SUCCESS in 4m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f796a4e748524c9b8f59f18920353976 : SUCCESS in 13m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e30b73f4ed20440992ea076fab1e94ee : SUCCESS in 13m 37s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/20d95de995714df28853b5710c969ebf : SUCCESS in 7m 22s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/727a77860e25482d985b59f7d4c67286 : SUCCESS in 4m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7189bc20782b494d817ff7050742dd49 : SUCCESS in 5m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b8f15e32c7f14e8bb21cd3c34d07d8f8 : SUCCESS in 10m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fdbfe40ee590431589619b0b7453943d : SUCCESS in 4m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/63318fcf987649398230afa11d4edb3c : SUCCESS in 4m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6a6f8f33576b4d38ab509dfb9dfe0360 : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0ec3ae968b5341c5a501676403f5521e : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb92879e347e48a29b6d136acc62174f : SUCCESS in 6m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/85e93fea92ff4487bacec1af3462be75 : SUCCESS in 8m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3c7ddfbb2c8049acadb4820e9de80c32 : SUCCESS in 10m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/74ce2a2ccaf64f5ca2efd87e91ba10bd : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d13da2edccad4495b1839edd26adc570 : SUCCESS in 16m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/97bee39ce24b4b1795b396e2f9353c1a : SUCCESS in 4m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d88806bc42214c389cf7e295200d0b45 : SUCCESS in 10m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/afe10a24a6a2438a94fca90d5ab253f9 : SUCCESS in 21m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f7ec94f9041643fa9fe5bcb8211376dc : SUCCESS in 12m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/717c81cbf43046fb9589088ff0c03318 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4fc8818e19a41bd8e1d55f4a4c83285 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f24838e9fb8143bf98a0bc95823d6821 : SUCCESS in 22m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8e8ef987c62d4a398e2ac44bbfb54e18 : SUCCESS in 5m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8020df37485846168c09ce5edb22fd52 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9b17997f9654fc5a4fb6e8cccb0c56d : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aea3177949ce44adab2d74dc894ffee3 : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e51b2c18ecc47dba720b0efd278a5bd : SUCCESS in 52m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4acce5a3afb4e92ab8ae5cf81ddaa4d : SUCCESS in 1h 13m 47s","accounts_in_message":[],"_revision_number":3},{"id":"cfe1d66655d16c18271282d3f26ef2ec094dc5c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 00:11:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"404cf946072990b747807dc91fc533fbf2281883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 01:22:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c07149fc43d4347b6a775158889b7df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86fd868b920643bdaca31909707ca3e4 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/434953e6b416431088527df4ce34bdc1 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d5d79f47f424d9093dc05c73deaeb08 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7afdad97758c412d93c184d9523234cf : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a63d22220e4a40f6849c5ca9e5262a1d : SUCCESS in 7m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6c0b717e8ae74f1f9eaed2e5e54e2c90 : SUCCESS in 10m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/03b247196fef42e2929363ba8de327ba : SUCCESS in 9m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/76a05220e09f494e8d43bac7560cceff : SUCCESS in 4m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f03db9af6f134fdebaa4723e89d9d587 : SUCCESS in 5m 15s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/41ada545c382450882b0d9f5d7688cbc : SUCCESS in 7m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/be7a5ea5993d44d9bd8b94ac40fa5134 : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4120e9ca795d476ca8058ca207d7abc5 : SUCCESS in 7m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/9da9ddde85814541b1fe70a421f2ca44 : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4cd0e1d6707c4fabb9d9d80aa5ad6134 : SUCCESS in 11m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/eeb2609d5b0945cb9614a71f04558c6f : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9ed217cecc704652a97b55bca3456390 : SUCCESS in 10m 16s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/97f0e47e1a954c36ab984d8177a4a059 : SUCCESS in 5m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/38e5340ecd5a4a7fb1d16a089093efae : SUCCESS in 17m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d3c17197d4984d0c9362a9a7ec6e5258 : SUCCESS in 11m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/360a3d2e763c42869e04bb22bbee3931 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f883b2b97354a2a839205c5d7504397 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44210ac1685340be87226f098553c31f : SUCCESS in 26m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2a5310ee8b1d4e9e908ae99fcd11411d : SUCCESS in 4m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4cf4c4f0c7544090a69b6fb078ea272f : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22d5be594fee4291bc0fc538b65e475c : SUCCESS in 5m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8360c993f3414b5b8548f7a63a8023c2 : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/565a6addcc6d45318b1b5ce75785145d : SUCCESS in 1h 04m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3e07f5deb634ed497fa2aebf2c4c51c : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":3},{"id":"20bab70b7866cd6fe2855225892a762894053403","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 01:22:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4f8e5815b58f3a1d694d116de7eb0d31a466d6fe","revisions":{"505b5b233601e8e458092a601ecf26b25a950dad":{"kind":"REWORK","_number":1,"created":"2023-06-22 12:06:25.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/43/886743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/886743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/886743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/886743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/886743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/886743/1"}}},"commit":{"parents":[{"commit":"42f9e16674b7c873778803b454e3ccd562db8995","subject":"update constraint for neutron-lib to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42f9e16674b7c873778803b454e3ccd562db8995"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-21 11:40:13.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-06-22 12:06:05.000000000","tz":120},"subject":"update constraint for oslo.db to new release 12.3.2","message":"update constraint for oslo.db to new release 12.3.2\n\nmeta: version: 12.3.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I9f5ca543ca85411edbe9431720ce05f5d5897139\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie052bb5578619001dc24f39b98955565eaafa925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/505b5b233601e8e458092a601ecf26b25a950dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/505b5b233601e8e458092a601ecf26b25a950dad"}]},"branch":"refs/heads/stable/2023.1"},"1d8a672a1c5730d928fad380186c036866a429d3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-11 13:48:41.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/43/886743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/886743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/886743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/886743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/886743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/886743/2"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-21 11:40:13.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2023-09-11 13:48:41.000000000","tz":0},"subject":"update constraint for oslo.db to new release 12.3.2","message":"update constraint for oslo.db to new release 12.3.2\n\nmeta: version: 12.3.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I9f5ca543ca85411edbe9431720ce05f5d5897139\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie052bb5578619001dc24f39b98955565eaafa925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a672a1c5730d928fad380186c036866a429d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d8a672a1c5730d928fad380186c036866a429d3"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"},"4f8e5815b58f3a1d694d116de7eb0d31a466d6fe":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-29 22:54:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"real_uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/43/886743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/886743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/886743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/886743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/886743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/886743/3"}}},"commit":{"parents":[{"commit":"c3ea54bd5476bb1a19c230d7993b820271758f6d","subject":"update constraint for aodhclient to new release 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3ea54bd5476bb1a19c230d7993b820271758f6d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-21 11:40:13.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2024-02-29 22:54:38.000000000","tz":0},"subject":"update constraint for oslo.db to new release 12.3.2","message":"update constraint for oslo.db to new release 12.3.2\n\nmeta: version: 12.3.2\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I9f5ca543ca85411edbe9431720ce05f5d5897139\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie052bb5578619001dc24f39b98955565eaafa925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f8e5815b58f3a1d694d116de7eb0d31a466d6fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f8e5815b58f3a1d694d116de7eb0d31a466d6fe"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
