)]}'
{"id":"openstack%2Frequirements~887260","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I53ee8da295d761b6b1d7c5ee68741d544656fd5f","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"I53ee8da295d761b6b1d7c5ee68741d544656fd5f","subject":"update constraint for oslo.messaging to new release 14.2.1","status":"MERGED","created":"2023-06-29 09:34:52.000000000","updated":"2023-07-01 03:40:31.000000000","submitted":"2023-07-01 03:40:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887260-new-release","meta_rev_id":"513d1f5a2f2e99bc5f443e431849d2fae6f2c960","_number":887260,"virtual_id_number":887260,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-01 03:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-06-29 13:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-06-29 15:02:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-06-29 15:02:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-29 09:52:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-29 12:44:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-29 13:53:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-06-29 15:02:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"429639a4c6f54506a410e9479f72b4b5eb898b6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-29 09:34:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"831afd960fba7e8bbc7c6e3894ab4e475216e27d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:52:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f97f1abb2c944bc8acd36e198d12328\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/297a5be606e746e491c7075d03b57d0c : FAILURE in 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a4fd631c2d69cd508c7b278d818afab354e21a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 12:44:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8177fff673554702b895bfb1c6ed9e62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cff9c2e250f4c89b59ee6c2dd05dc79 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bf1c27d66f24e9c8fe3a4334c02e633 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f4c4ac19136467cb1213405660bc2f1 : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/142b7e2795c14870b5be534794550cd7 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f6c6e8b768e416f8da59079ea4cf2b1 : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3d58b2cba63d406f9e6555c9d08638db : SUCCESS in 15m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f452040596dc4fe7867c3c5733f28754 : SUCCESS in 17m 59s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/003b336d43d245faac6d6cf301b2cf75 : SUCCESS in 9m 05s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2998a47ec47f42ab9914a55858c7e985 : SUCCESS in 5m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/36231b59720d4d1e949d080b9bdadafd : SUCCESS in 6m 45s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ac61bde87a2648899157b4d4f5a4d724 : SUCCESS in 11m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d0a506ea8fdd41c3b13fdb13cd3fe4f0 : SUCCESS in 5m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7756aa7931ab4741b022dc5a5dd0b738 : SUCCESS in 5m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f944528a325346fca063b0b492eda7aa : SUCCESS in 8m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0a456dbb8eec40d0ad10259ca7fe2917 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9d36bc4a6c04e3083c01eb7962565d4 : SUCCESS in 7m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fc8e0c1549e94d2bad605d281c6d380d : SUCCESS in 6m 37s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4fc5478766b74291b7a1638564130f9f : SUCCESS in 13m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/477074e7036745f89c15e1c4b3791c85 : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c05c8eaedd2f42c69ae1f78d2343929a : SUCCESS in 9m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f43368a51b1b46afa5fd6c95b0d8617f : SUCCESS in 4m 25s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/bcc987e53678450eb90a2b807117742d : SUCCESS in 10m 59s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/532618820797432bb9d871bf82ebf7e9 : SUCCESS in 22m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4bb0498e739b44c5928149a4a37129df : SUCCESS in 11m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9735a51a3ee043d28324cf6766308f99 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/495d2b869a204ea7a452b90d0c4e6f6c : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a450bfe922d14bc9a6f3b49df79d26d3 : SUCCESS in 22m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a7988d08099d41649961e08993aed6f5 : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b189c199564347a082a3ef84ff3d6e07 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e18b3ddc27f42399692df211ccc1d58 : SUCCESS in 8m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5eba5a8e10824ae48f46449ad1bc4d4e : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8acc764947d44dfbdc026662c0806fb : SUCCESS in 49m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5ee0f640fb64a00a2d9602cc6b6643d : SUCCESS in 1h 34m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d5aad74a970a4ba971141d0aed576ef5914e5212","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-29 13:53:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa24f88c811bdcdc3cb55210d7a81bef8ee73b43","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-06-29 15:02:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f580556a289afd2ce3710487ebf2cfdd73d47c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 15:03:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b82ee8ad6c3dd8fe3e684650732a9a8c396359d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 18:58:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7677527d53844dfab8f0539b0d4acd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6dbb3e3c09c4f9eaf8abd97b7799486 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1d3fd89d2d3448387cdcb133b3ba70a : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1992164ae67c4372a841ec97279d71e2 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/094e48dcab3443bd83fc2d87643c44f3 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83e5c7715f63437584617a30f038131c : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f6f5361f21d48a0843e3dd4cbae268e : SUCCESS in 15m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8e8b650212774fec897b7dd2af267347 : SUCCESS in 12m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/de3ada33bcc74135bb107131b18b21b3 : TIMED_OUT in 42m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d02e27657db640b392bb9e42c4c3fe0e : SUCCESS in 5m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a22501881bda40c8a4afecdef45a8f71 : SUCCESS in 5m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7564936e3dfb427fae558343f59c1692 : SUCCESS in 8m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d6b9af874ca340f88a9aafd84b0b185d : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd799d50398145889e03d1f58e9b7820 : SUCCESS in 11m 40s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/446f590c7e34475486f8c5a6a5f32956 : SUCCESS in 7m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ac5d29e22a0b4e8a8d66e6103189aa01 : SUCCESS in 50m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fbcef4d1f8d044dabe4e1c56aaafebd7 : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5cc1b4f0fd184b85bc6b23106f3512bc : SUCCESS in 9m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5f28cbc622c2463e9ce62a5a8306fcb2 : SUCCESS in 4m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/8c4729997e54470bbdafba96cc8c4c6d : SUCCESS in 25m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6cad7a25dc64468d8f28377ae3ae76ea : SUCCESS in 15m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/6464d438838e42e6a272bdbd0c121619 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99b3cf5298ee4756b2254386c8591ebd : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98b123429d3247dbb1d91e36291b8ddd : SUCCESS in 23m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7d2f57b0ddd845cd9edb4db5333107c5 : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/52975fbfcb584f04807bf8b6fd8735dd : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c507a5a6d4840f7b070a4b8aa394fc8 : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ef05baf784e14810b2ee6525cb95ddcc : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efd47b5b9325413fbac33230b28cddbd : SUCCESS in 1h 06m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e774777a4eb4951af09b4f2f9bdfacb : FAILURE in 1h 22m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e7f962f041d0f9b0aed930c9799af7f0e9c23b39","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-06-30 02:08:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d19556ad0c4578cce9ccfdf2248956fe88453180","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 02:21:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a89ebb7af6c42289ed98bfc0cef8044\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6b952ca872a54ff692fe32ccb348b98b : FAILURE in 7m 41s","accounts_in_message":[],"_revision_number":1},{"id":"932366485250c4bf548402b553066baa0ba7046f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 05:02:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35d893ce39824f0c87afb95bca7d18c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e73b778cce48e4a3577313525e98c1 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4802855e8e5e49cc97d10ecb0bf61dd8 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e19a3841505c46e491253dd4f8934301 : SUCCESS in 10m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cafa653e0c1496985067c7b5e21cac9 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/826dd8e17d7e44a09ac4955c7a7f5399 : SUCCESS in 3m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0edcf8d320e44e94a4be989bce406280 : SUCCESS in 9m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/680099356d3847d692435c26210bbf44 : SUCCESS in 17m 21s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/395cf16bf9074d78be2cbacd1f713131 : SUCCESS in 7m 28s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e4215b593b824bd19802d152d30bcbdc : SUCCESS in 5m 25s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6aacfa9f8a3c467eba82e7bc4212cd81 : SUCCESS in 6m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a1403509fd82470e92a0e4751564e5bc : SUCCESS in 10m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a3496773e1c54ccfbfa3829dd2e9ec9d : SUCCESS in 4m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fd44c9b800534de0b7c29c2a4b9d27ba : SUCCESS in 4m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4984ad4f23e640919c69dc7613946051 : SUCCESS in 8m 30s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d450e95eaf8a4504bd8c6aa974bb77dc : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d1c37cb31a54056add058a8bfedb904 : SUCCESS in 8m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/86f424b08a074ca8b7686730bdb99907 : SUCCESS in 9m 04s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3c0b662293164d6ea5f20fb2fac43eaa : SUCCESS in 12m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7e670053d89643c99a7686b29998e1f5 : SUCCESS in 3m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/86a31519459a4fc3935c366bfefbb087 : SUCCESS in 11m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/016aa1facfb94144b15ba10710d8e2a8 : SUCCESS in 5m 05s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b2149063b44d4dfaa12c8d846b0b2f90 : SUCCESS in 10m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7f128fad11994d199bee3bb4662368d1 : SUCCESS in 25m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/01c7ff7b396a4a27b349a91b2966ab06 : SUCCESS in 16m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/57adb50c763047658c3331d6f658c1ab : SUCCESS in 6m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fdcde01549154a83a6a9ba1ff0b68dc8 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8eb552f70ec34f599948b172999a8298 : SUCCESS in 24m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3466c7a5802a4f899dfbbf30efd234ca : SUCCESS in 5m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4d3d8066cae145879f340c13cf899fd5 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6785d782e764478aa24994fbf8e7c6ea : SUCCESS in 7m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5c33a1c79a98428baa3a0ac5e0713636 : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cacd524438e4c2393e2b079be529db9 : SUCCESS in 54m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc76fc56db1b4f02b9392590433ff032 : TIMED_OUT in 2h 03m 04s","accounts_in_message":[],"_revision_number":1},{"id":"77eea26d667ff5179d76fb4f1e07ef48abf26c1a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-06-30 17:01:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"df0c28b709ea7012c62097bfe65d1eaea1edb1f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 17:07:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/436e7e615e2f4eca872cb88824984e3e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/194a59a8d776480885a03e0b8fd8676e : FAILURE in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d1c6d96562de45d4eca524faf176573d377c3d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 19:03:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdcc86489d1a4fcfb0f8cb822258c01c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58b8e6fe0b95414293fb4286581ea831 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2554515bed9348a5849f82b42ad9736e : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36be5af8528e4cf2b8561fe48bdf631b : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a83610b433440a4bd01a347380e5c66 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21019ad6b19c4fe4a16e5aab6a88d4f9 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/96c552aa23474d5eb5583d9023206c01 : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7b10340257424f25b8914752fab10b58 : SUCCESS in 11m 38s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/609fa8cdef34453b821f521dfd7a10c3 : SUCCESS in 6m 13s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/936aa83d9fd944898c5be8e9d3a43678 : SUCCESS in 7m 15s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/70f44a0922f042598aa5594c2194afff : SUCCESS in 6m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c8bcfe4b60ac45728e84e24de8f4e7f0 : SUCCESS in 10m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/25b74daf8f8042dc8b73a925040f884f : SUCCESS in 4m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5d7cda85a480442cb731a0d7242c0fc0 : SUCCESS in 5m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/09fa23b57f604b4da97487a1804c45b5 : SUCCESS in 8m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/811e3bb5f24d4cc2ba77d79a7ea47d74 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/385de30147b24358a21ca9bfb7d49935 : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1ea48071843d4300ba8d30e8f1ad71d5 : SUCCESS in 9m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7121a550d8aa4b2b84641c3918c1ebaf : SUCCESS in 11m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8a52b267404340d585b5326b5c7bdeb6 : SUCCESS in 4m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e26a57ac63fc461b918ecfb919bf6b52 : SUCCESS in 13m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cc8f5f0d04e24b18afe71e6176f0e890 : SUCCESS in 4m 52s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/34fbf6ac096541a6b9a23b06e973012f : SUCCESS in 11m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/004e0b6962da42d0aa0544f61ed6e591 : SUCCESS in 24m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2851b9e5681a460f8ef67d52eb30948f : SUCCESS in 13m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d3feaaf200a34c46af3bc3bde46b6cc1 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f32785a1de2e4fa8970eec50ac76515d : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76c70c2009064538b1f721acc108c235 : SUCCESS in 22m 33s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7f5bccf84c454643b82c11b23781136c : SUCCESS in 5m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d54fcd987e8f495db21ea3e841551f2d : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e5c680c2b2a24de7a2cdcf03bc6a41e6 : SUCCESS in 5m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/69c67d41609949eaaea45a385637dae3 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/526637a1f7344d1296316cc1acf06f54 : SUCCESS in 52m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/584c3a8979894300bde668d1949dbcdb : SUCCESS in 1h 38m 11s","accounts_in_message":[],"_revision_number":1},{"id":"479a2fe7ddeb61989178220cbfa6e91ebe0ede88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 19:04:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"288bbfcc2cc8d63d13774d2b4012fd14b80e32a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-01 03:40:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d0f70e51fb417c886dd9cfc8eddefd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89aa13f125c5495e91dea83fbb7e58a6 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5be04f2a3a0c4fd681b56467430a9f1b : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29d0b5a1beec4aedbe97ca425289851e : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c5bd7f6fd304865b29272b9b59eb6ef : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a466e3e6ba9496db13d9feebe35a60d : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0064bb8afc614b65b8c83942a1a1450e : SUCCESS in 7m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c4ba775852f44ccdb2a3a489afefc2d2 : SUCCESS in 11m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/069540f292964722bafbfb0ee9d4437a : SUCCESS in 10m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1741fc43e6f540b59ea3135e2f80c7a8 : SUCCESS in 4m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3d5b54d49c7b4cbb8b957eba4b80002f : SUCCESS in 5m 03s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d89c3f08493b41108fa92d82c78b97cc : SUCCESS in 7m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/90eda4123fa44c8ea9d39e79d371c29b : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9541b12ae0134396ac41c6b455aa68b2 : SUCCESS in 8m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4f5db722d6ae4d2b8f169f2d5f664e33 : SUCCESS in 9m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7e734bb1930f4fc6ab5e7c232fd30023 : SUCCESS in 16m 49s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1f1bb865f1864e23b0c9b185931778fc : SUCCESS in 3m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/42229698025441ffbe783fd6c9e7026b : SUCCESS in 16m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/335154fd6b0a434abbcdd04d558dcdad : SUCCESS in 5m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dc65135db5254ac9b3cbefe9e94bb23d : SUCCESS in 23m 02s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b3ab5dcca69743778c75854c17d6958f : SUCCESS in 12m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7ba0e5ba667348aa96a904591ae03b49 : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eda7d5c94ef54142baef354e4c7597a5 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de9a81ed18874a6eba2e7df0057ec627 : SUCCESS in 20m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/259d063c5ce64a498d72350edbeda26e : SUCCESS in 6m 19s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ce9caf215ba14b0fb554e7c06aa641f8 : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35eebf92af53469ca86113395a824311 : SUCCESS in 9m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1c3c277e73c54716befa6c8ec9e4436c : SUCCESS in 18m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b4c9ed4f0bc417fb0685fe8be73fd58 : SUCCESS in 57m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c548389918c42ce87fcb411e3144d76 : SUCCESS in 1h 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"513d1f5a2f2e99bc5f443e431849d2fae6f2c960","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-01 03:40:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d450d25d953cc53e902f83c6a0123efa2536faeb","revisions":{"d450d25d953cc53e902f83c6a0123efa2536faeb":{"kind":"REWORK","_number":1,"created":"2023-06-29 09:34:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/887260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/887260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/887260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/887260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/887260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/887260/1"}}},"commit":{"parents":[{"commit":"42f9e16674b7c873778803b454e3ccd562db8995","subject":"update constraint for neutron-lib to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42f9e16674b7c873778803b454e3ccd562db8995"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-29 09:34:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-29 09:34:51.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.2.1","message":"update constraint for oslo.messaging to new release 14.2.1\n\nmeta: version: 14.2.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nmeta: release:Commit: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nmeta: release:Change-Id: Ie32a4b008e8e64b423a2e2799488559eb4bd1ef7\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I53ee8da295d761b6b1d7c5ee68741d544656fd5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d450d25d953cc53e902f83c6a0123efa2536faeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d450d25d953cc53e902f83c6a0123efa2536faeb"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
