)]}'
{"id":"openstack%2Frequirements~887261","triplet_id":"openstack%2Frequirements~master~I6c561cbf2b3d22574e8d85d7884d927df202cbc2","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-09-20 05:32:09.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-20 05:32:09.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-20 05:32:09.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-09-14 11:13:21.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I6c561cbf2b3d22574e8d85d7884d927df202cbc2","subject":"update constraint for oslo.db to new release 13.1.0","status":"ABANDONED","created":"2023-06-29 09:38:04.000000000","updated":"2023-09-20 05:32:09.000000000","total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a7964c7f344b62ba89fd06f6d175236ca8568110","_number":887261,"virtual_id_number":887261,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-07 14:39:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-29 12:32:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-29 13:52:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-09-02 17:56:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-09-02 17:56:43.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-12 09:42:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-09-12 11:17:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-09-13 07:40:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-09-13 14:16:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"77e02ae1866acd62cfecfeff593300d4423b8d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-29 09:38:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05266c97ac99b4c73342818e47e53086d4ba4c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 12:32:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/def8d582d00f40aca764f38cd95d02c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c72cf3ea27644c59577f7873ab16554 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/555d425519f343878164c5df561ec6fa : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d251ac5c0ebd4dd49854858a0f2b1b76 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19d0d74b07b5465abdc1a6ed5301d0d9 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62a53f0b5ded40a9875fac48d84a3077 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/17b8ea03137d44b5948f5218c3aa2938 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/42abb256004c434f88190151bb48d508 : SUCCESS in 2m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1ab20b8d280b482fa210039ae47bbec5 : SUCCESS in 9m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e42228a67d4a45aa97ebe30aa79ad01e : SUCCESS in 14m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d94a25da2ac441e3b9017a67d6686702 : SUCCESS in 8m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5382da7a66bb4e93aadcf1b8ce469f11 : SUCCESS in 10m 56s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/18e1c0a34bc045b5b061ade2f4cd08fe : FAILURE in 33m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c97a30ae248a4dd7aac9c1059d544c6a : SUCCESS in 5m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/35cf165f281f480eb87b9df6f0814604 : SUCCESS in 5m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d7e7b18b32b340b298c2cdec140a1eec : SUCCESS in 11m 53s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/035795890e344db6abce412c5ece4e93 : SUCCESS in 4m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/103d4f7a866745be813f56bc39add9a6 : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/7d47fc945c484283abec6fca72da5597 : FAILURE in 8m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ac540a1b83e44b20a1b09a540f578f39 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c06fed3b3bd64353b6d4add549306cf1 : SUCCESS in 7m 34s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d617929f14ba4ac18d2d2c08db9f9d0e : SUCCESS in 11m 05s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/abad86fbd43b4e92a74028a915a3604f : SUCCESS in 10m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2524a8476f604c6ca36931fd11317141 : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5f559c2643434f42b22f19609fc7e783 : FAILURE in 4m 19s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/680bcf1cb2944629a85e5c29e7428660 : FAILURE in 3m 44s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4e634f6849344570884d2903de722b48 : SUCCESS in 9m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3e5459a6b1a643a39c54358c12218451 : SUCCESS in 27m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/33cf61968c0f4d3891fd87bf3cfc6f54 : SUCCESS in 9m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b9e0e0d743d246c8be88e2dedf55c5d0 : SUCCESS in 11m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a69744d6b8114a4eab8aea9cf2e48176 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e239bcba6352431d8887fe8f392f2d46 : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06962eae8e1445158d34d2d6e69b1b23 : FAILURE in 27m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/83bf6f60481b4c44b5645d429520c4a6 : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/664e564d359f4336a42397224bc2cc45 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/11280711416b48f28798ba7cd62bcf82 : SUCCESS in 6m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a4c381c35b7542089fcddc1f0fa99429 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c42f3abdebb4470b65e4d6dab8fc730 : FAILURE in 48m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ecbe36c9c0534f4a9be76f37ceceb6e0 : SUCCESS in 1h 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ccbb9f5e0bacf33cbfa345a407c11fd7f6d4cea2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-29 13:52:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2578cc8e8c20e4d7811eaa3cb5bffbc72fa7320c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 17:24:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27bf6b3ab8b34b2b84a4cc077f8638b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/726a8556911b445aa977a5581c573642 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf18f05a46d49e2bb7b019e77a555ca : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b78f2aacff4f458b99552518ef77e01b : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7405ef28899443eb229f31c8ce84758 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa7b2b97c37545059036ae5ad6e4f73d : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/058b37fa242d487a94cfc3266b359922 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45f11025f10147bb9a20ee366c884324 : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/127a5e06ba3244329d3d4d8e6b024558 : SUCCESS in 9m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9ddab5afe52442a9b357f93e7d9c2acb : SUCCESS in 12m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e4d0e4743c0441a896bc764188006dee : SUCCESS in 8m 12s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/95605a2bf2984c2198ea7023ac0b638e : SUCCESS in 10m 32s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7df7fd802a814ff9858ac9b7e971943e : FAILURE in 5m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/aa4a0f77c4724fc8ac455f4aaa9fcd62 : SUCCESS in 5m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ae7e69b43e03447eaea9e97b9b01fd70 : SUCCESS in 6m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4188022f72a64709a321889867103666 : SUCCESS in 12m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/789813bd735e46798abdf984a662a223 : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a3f7eb7944084deca69dd998c942ef32 : SUCCESS in 6m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/121dba927c2c42dd9dfbb0867421c1d6 : FAILURE in 9m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9a53e4a0eba34b23a571e8cc286c4775 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ecac6191a4414e8aba438c098a6737a0 : SUCCESS in 10m 38s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/196574112c3c40cfb6add0e2ff67777f : TIMED_OUT in 40m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9dc90257459e424587572af3242c77a1 : SUCCESS in 18m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a035ac850c8b4c21b10c58fd45eb1e1f : SUCCESS in 4m 08s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/241370b5b2314fe183f3f0ea78d8f9f9 : FAILURE in 5m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1c16fab47ce1437d89e637d76ee4959a : FAILURE in 6m 40s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/11608bfbe6e948d5aad58a3736c562d8 : SUCCESS in 11m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/535f6493a86a411c95af52e4f815085a : SUCCESS in 30m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/80f96627ec3d4fc4928b2c8e0c32ed8a : SUCCESS in 10m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8bb614c7ff2e438cada7651a6bdabd99 : SUCCESS in 14m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d684ba8a0b614e518fbc0f92c8530ec9 : SUCCESS in 7m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/295b2d9d72c64a59840ac6b1799759ab : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/11feee8a5a304ba080f6c66c2d2bb262 : SUCCESS in 28m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9c6564b93aae417b89b84d3729c98912 : SUCCESS in 7m 48s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/963825eadb894465811a83cae04801b6 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/504ed97df56249e99365fe5482904bfe : SUCCESS in 6m 46s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6ba27be0cf6a4eb1bc5bb0db1490eca1 : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/797ac5bef301428995898d76891190f2 : FAILURE in 47m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/630f9b4286284ad292d3d189f51b91e1 : SUCCESS in 1h 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ba38b152f75f9904cedeff0cac6d8a2aa9697b6f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0f8aad0b50a6e94f4606e66351c8cb0ec3e4e273","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:22.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 887261,1","accounts_in_message":[],"_revision_number":1},{"id":"bf678406f00c8770a736767bc737e364502df49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:26.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 887261,1","accounts_in_message":[],"_revision_number":1},{"id":"2c7e1a93775ba86856781a9002599718f214fceb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-02 17:56:43.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/requirements for 887261,1","accounts_in_message":[],"_revision_number":1},{"id":"f3ce9a364b0d1f9150831d3945c6d618f002f60f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-07 12:56:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a1d1843dfcc18cccb769389b68bd65aba4175f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 14:39:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1631d74912ae41d8831652453b10c95d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8485e91136464c7fab848534c11a295b : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a6cfa71744245e4aaff0b45f9f3f1e1 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/237b23ada19f4363897757530ee3c6e3 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a63d0971391a49a88b86660fd21ec83f : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5eb8c5510d0b465682efd5a06a319014 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82ecb1ab7ade48d29f5ef4e0bbc1169d : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3995c33ae44442a85cc59e2f6d57a99 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61696b51ac5a464facd2462cf99d63a5 : SUCCESS in 3m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38ca4f617fd746939f46a3f8f206a3d2 : SUCCESS in 9m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3d47a2f4c22d472c96ebf63a0fda7547 : SUCCESS in 13m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6316850ca8234ac7bb7bbdce85cc6126 : SUCCESS in 7m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/98be716b776f4ee19626551376cbc105 : SUCCESS in 12m 27s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b9d171ee1691452a8e6bd7b9372a2f57 : FAILURE in 4m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9baf9f6d8c2a4c32bedd62b72118fa5d : SUCCESS in 4m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/327b084bbcf64f8092ec9782584cb2b5 : SUCCESS in 6m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/346e175c14734979903043e266fda607 : SUCCESS in 11m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f0223c24ba1c42d6925f0504001d4c0b : SUCCESS in 6m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e9b482bb3a2544e4ac4255db4752d1d9 : SUCCESS in 5m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2857911347e844b89f9557a9722e3f3f : FAILURE in 8m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a8a0e749f9fb4848a3f4408e761c0372 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f950eed91e54344a06f1e63d06d6fca : SUCCESS in 8m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/239ec970f5294366881cbbcefb0fd6e0 : SUCCESS in 7m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4527dceecb5345d5adfeff90fbc67246 : SUCCESS in 9m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6178cfa10e394e62a39f3f4b251e60f5 : SUCCESS in 3m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d2753f4435c94ebaafe083b3aa0eede5 : FAILURE in 5m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d1db1e2e7aee4222bafec60571cfbd53 : FAILURE in 4m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/15ed390c308b433ca855180e91bf977f : SUCCESS in 8m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2286714f860947d9aa67072ac4a0967f : SUCCESS in 26m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ae22c20a47e543beb7480fb80792dc7c : SUCCESS in 9m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9f71daa10c62433da6a323cb0e1ac4e7 : SUCCESS in 13m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/216b4bd5f9834318ad4480e4b214aa71 : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/742e7692141a483b8595c5bbf04798dc : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/029a6066c9494986bc8627c4584f116b : SUCCESS in 23m 53s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/83f50c6e237940069a89db55ab949524 : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/122713e8021c4e668f713a8206dd2805 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17e602d163c14db7951b70610462ef95 : SUCCESS in 7m 04s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c2470ee685db4e47aeeba8e2078153fe : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9d3c13969ae402bb9e9ce18e34f64f2 : FAILURE in 1h 01m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/330f7b9058b84a89aa68410d8cfe44dc : SUCCESS in 1h 36m 33s","accounts_in_message":[],"_revision_number":2},{"id":"527adb76b3c1d8399c6972e417b40911f0406377","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-12 09:42:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63f5d438dfa350cad95fb9049b845907f121223b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-12 11:17:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76e9dc04b6034d6253bae7e30d80ea20e7e71613","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-13 07:40:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff4b735e5dee7a98332ac8928f799c74b95a5a4d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-13 07:49:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7905e0d445c87d7f15ca2e97b918c7d07e9bde3f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-13 14:16:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"831eca572b9269c91029aab72ad5532f93fded1a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-13 18:38:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8498fcc23c42c70ffd199d9184e39ed02feb2014","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-14 08:45:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a9c3eee666375018c74849dd0e2208709d67696","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-14 11:13:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7964c7f344b62ba89fd06f6d175236ca8568110","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-20 05:32:09.000000000","message":"Abandoned\n\nwill be superceded by a new release for bobcat","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2f62f347b8a01fa0700ad44f830daa0d75f81c55","revisions":{"7185bad1dff7b170610cb059628e31ca8429d1b5":{"kind":"REWORK","_number":1,"created":"2023-06-29 09:38:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/887261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/887261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/887261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/887261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/887261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/887261/1"}}},"commit":{"parents":[{"commit":"8dd61e9bb7da448a6fa05ca34dd14e03d22971df","subject":"Merge \"update constraint for glance_store to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8dd61e9bb7da448a6fa05ca34dd14e03d22971df"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-29 09:38:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-29 09:38:01.000000000","tz":0},"subject":"update constraint for oslo.db to new release 13.1.0","message":"update constraint for oslo.db to new release 13.1.0\n\nmeta: version: 13.1.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Change-Id: I1d747c8a2459cbc93e6fbde0e547b63577005910\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6c561cbf2b3d22574e8d85d7884d927df202cbc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7185bad1dff7b170610cb059628e31ca8429d1b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7185bad1dff7b170610cb059628e31ca8429d1b5"}]},"branch":"refs/heads/master"},"2f62f347b8a01fa0700ad44f830daa0d75f81c55":{"kind":"REWORK","_number":2,"created":"2023-09-07 12:56:38.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/61/887261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/887261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/887261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/887261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/887261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/887261/2"}}},"commit":{"parents":[{"commit":"237d49d5ba63b817eece6376a284e990654f4ea5","subject":"Revert \"Add Spinx RTD theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237d49d5ba63b817eece6376a284e990654f4ea5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-29 09:38:01.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-07 12:56:13.000000000","tz":120},"subject":"update constraint for oslo.db to new release 13.1.0","message":"update constraint for oslo.db to new release 13.1.0\n\nmeta: version: 13.1.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Change-Id: I1d747c8a2459cbc93e6fbde0e547b63577005910\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6c561cbf2b3d22574e8d85d7884d927df202cbc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f62f347b8a01fa0700ad44f830daa0d75f81c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f62f347b8a01fa0700ad44f830daa0d75f81c55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
