)]}'
{"id":"openstack%2Frequirements~887770","triplet_id":"openstack%2Frequirements~master~I9ed31162f7db71ac6135e069fe035cb3abe86b7d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9ed31162f7db71ac6135e069fe035cb3abe86b7d","subject":"update constraint for octavia-lib to new release 3.3.0","status":"MERGED","created":"2023-07-06 08:41:52.000000000","updated":"2023-07-06 22:32:24.000000000","submitted":"2023-07-06 22:32:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887770-new-release","meta_rev_id":"3d58c4325ebde7a4fd02cde5b11a5717389e4a4b","_number":887770,"virtual_id_number":887770,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-06 22:32:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-06 19:47:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-07-06 11:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-07-06 19:47:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-06 10:32:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-06 11:51:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-07-06 19:47:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c25f749140a78268b17aebc3e279b934a7af48aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-06 08:41:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2e53fa4fdbeb37a83bc2bd1a33e3e0c51dfb3c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 10:32:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26a4720b34ab48e3a098bf9f70dff3e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f94117f2bc84c4ea19603a4fc468910 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c04af70120742d6af1f393c97567609 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff7b8d2020bd4c239c5cb7447361a3e4 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87761e6de4b84ab6892251d1e1c5c6d1 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/565e9e13d3654926affe52b28e6dba92 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/baed295845e24ca39f8d1492f7fc6498 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66b712f919b94022bec00dcbbd08a0d5 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0bcf896d54c14954a569d4609c73b9b8 : SUCCESS in 9m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1441def2a2ef4e2bbf2d4724193c2aff : SUCCESS in 11m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/10bfc1256ee34476943397fe874bade4 : SUCCESS in 10m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/daa1446c5bc649518ea1161ee9b1f11d : SUCCESS in 13m 28s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c796e2a7e834434ca5d5d772860e7c43 : SUCCESS in 7m 36s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a03954b5a97142a193534f4f54f662ef : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9bdfed05dcce4c87b638632c9ff8485c : SUCCESS in 5m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/13a88e94baef41f98b23c7acbde69edc : SUCCESS in 11m 35s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ddf0ed570abd4c1cb2da5dd9c9544275 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ba3f17a59fea4d7e857eb01193078dfe : SUCCESS in 5m 11s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/59e86a15a4e0479f987f9dd3d84ddc14 : SUCCESS in 8m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/16827fa6fcf04765953e98f6cb65642e : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8150e7a987f74ac99166755e5424819f : SUCCESS in 8m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dc1bb30c93c04e3187051a1a625d1dbf : SUCCESS in 9m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/04477eb10c6e4cf096f44957a99f4d15 : SUCCESS in 12m 28s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1396a448c7584d7092a872e1c4b0c667 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8c6255780cf04d4a920f723caab6af8e : SUCCESS in 9m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bac29c9ac9b149a98c8bc3a959eaf766 : SUCCESS in 4m 24s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/33421c1971334a98a4c059b6365f05d8 : SUCCESS in 11m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/206c1c63548e42c2b191fe4eb9e5f298 : SUCCESS in 28m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c7a5c27447b34ba19df9fcffaea83bfe : SUCCESS in 10m 55s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e2fe170a4a9b4485bd44a13f88deb4fe : SUCCESS in 14m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7073c4f11c4f4ee49953865f4c867e1f : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c38f116319f47fc8598937fc312a9b3 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6793a192ed224fe49ebbefa08654f794 : SUCCESS in 26m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1f37eb6c64034b40a1f633bcec7494d8 : SUCCESS in 5m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8924944cbef5481eaf47c63133ea5246 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0689b3e7e929404b8f6132f929adb921 : SUCCESS in 7m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6df06ec9999c4ecdad4616cfc87d7d3e : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd48a7d06c034913ae5e74920086d2ab : SUCCESS in 58m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3259c411b57c46adbec9a42a8a0a5861 : SUCCESS in 1h 43m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b591a37757a1e08e07a8dd66cf174c935cbe1da6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-06 11:51:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"89c50597a5ec5f12ad10def78b86becd767b99fe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-06 19:47:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbca50fe283ec5735b0b70ae456abe28e0736ce7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 19:48:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4c6b3c353d1fe6a7a4472a5d8595f0d241cae24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 22:32:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7c2c94e46ca4aa4ac35a44dcf2577aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ead698ad0844a2ba4c196ca3ba56782 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff984dedb214d54ab6a039cb7aeb988 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6f1f2fed5b94bbd8f25615de1cdadb1 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee1830b9e5d04f72af124c7964c8e400 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e050e66e5ec64d80809e0ab0d9609fbd : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34f9a2c37f1f4ba7a1c412eb02a9c3ca : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3f57cab2b1047c19a894359dc656436 : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cff15623f9db43a2910aec6543679aff : SUCCESS in 7m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8c0e28a59ba442b4aef34d54352323e8 : SUCCESS in 12m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f74792def8834ddc8db16b5c866449df : SUCCESS in 8m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/dc79ad3ae0424907996933a300f61628 : SUCCESS in 11m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/918061e7df7e4c22bde825cd1ee0fcc1 : SUCCESS in 5m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/21b963714cf14c1daad417d19277f3e4 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/602093eb28354a2b8181f6c2c9bb2be3 : SUCCESS in 7m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a02d0a3954104853b5d003534a1ae24c : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6679cae61e274f5ca14e12d8ca708db4 : SUCCESS in 8m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4d92be9f30934a5cb64eca5d4da48cc8 : SUCCESS in 8m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/882db5e866aa499fa879ae0a81f06b2f : SUCCESS in 15m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ba811344881f4bb4b5e1808bdd72a2b2 : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e5b2d3e8980c40c4a3a828562d814f86 : SUCCESS in 8m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d2e1ca4ca704422181cd4514728abb9e : SUCCESS in 4m 33s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2caddcc6c63c4716b89e8f48056702ff : SUCCESS in 33m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4fd523decd514ffba30d62c30edda81a : SUCCESS in 13m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/464447ffb0b44d679932f547ca482c8d : SUCCESS in 15m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/5153f4b529304ef58a92deb8ed734453 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e566544ac8ee481383a7907c618ffea2 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bee7d340f1544ce783b5b16f9aed8c56 : SUCCESS in 26m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/37a1b2b69b874c4cbc59d8b7f579a0ed : SUCCESS in 5m 50s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/248fa75811664489a40028d8408fc68e : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8422a0ebe7f04d219bca955362570f66 : SUCCESS in 6m 19s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/13e7fc0a327942f0ad671b58f8df6d74 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08c3722d54e847d289020b786db92198 : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db753b95a40f4d5aa713790b41e5c491 : SUCCESS in 1h 43m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3d58c4325ebde7a4fd02cde5b11a5717389e4a4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 22:32:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6b4fead39a2b1f1c71ceebf3c1c2996c1d131f8","revisions":{"d6b4fead39a2b1f1c71ceebf3c1c2996c1d131f8":{"kind":"REWORK","_number":1,"created":"2023-07-06 08:41:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/887770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/887770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/887770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/887770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/887770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/887770/1"}}},"commit":{"parents":[{"commit":"ac06a7a25582da6072f54e09e06a655bc5e95db0","subject":"Merge \"update constraint for neutron-lib to new release 3.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac06a7a25582da6072f54e09e06a655bc5e95db0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-06 08:41:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-06 08:41:49.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 3.3.0","message":"update constraint for octavia-lib to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I82779805f32939fb4fdf404875a41361501d2a78\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I9ed31162f7db71ac6135e069fe035cb3abe86b7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6b4fead39a2b1f1c71ceebf3c1c2996c1d131f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6b4fead39a2b1f1c71ceebf3c1c2996c1d131f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
