)]}'
{"id":"openstack%2Frequirements~887774","triplet_id":"openstack%2Frequirements~master~I21282e6916f64d15d87d08d242a57d6bafc3c500","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I21282e6916f64d15d87d08d242a57d6bafc3c500","subject":"update constraint for oslo.utils to new release 6.2.0","status":"MERGED","created":"2023-07-06 08:59:37.000000000","updated":"2023-07-14 23:01:10.000000000","submitted":"2023-07-14 23:00:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887774-new-release","meta_rev_id":"ee77ab1d4bc31bec83f32ad241c861135cdad031","_number":887774,"virtual_id_number":887774,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-14 23:00:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-14 20:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-07-12 00:13:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-07-14 20:17:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-06 11:09:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-07 18:55:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-07-12 00:13:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-07-14 20:17:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"436bf8e3b207d13f96fdc8eecc802a333f7ca213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-06 08:59:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f6cafb33232caba1c1f342197e9463f945a452d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 11:09:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b3b43d3321e45339dd0f02a3481d2e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eba2208c5594248871a844d0e9e2acf : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20c7f657a5bb417083bd59d7c00a1d4f : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45c43847a04546b39e406e8c410f655e : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46acd8bb419d4bb6b61e7b53652ea68b : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/774e68f932b94b76877650e3316d89f7 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cbeedf4f476a4612b142bdc481b2a74c : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48e76da847c541ad8e8bb898f63cc087 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3d44d7444a06475986eeec7e2e7fd6fc : FAILURE in 4m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0f18e718a32847e695de4e0deefb4af4 : SUCCESS in 14m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f538ce0a8b1841cda4575812951b874b : SUCCESS in 12m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ceea5a636baf426f8d254c3e1b545eeb : SUCCESS in 15m 44s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b9710a462045461bbf08935b15a44fee : FAILURE in 3m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/899e6b9141e3427ca61fe319b49a9090 : FAILURE in 2m 36s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/a214f0c4a89f48a69d4fda4230bef9a1 : FAILURE in 2m 30s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e305d5d7e12f453c9e0c21f1e41221ef : FAILURE in 3m 45s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f68b4d3420ca4c90b6b04e42cf506a0a : FAILURE in 2m 09s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1643a306fc05466f86125ad88a1f8efa : FAILURE in 2m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b7f06a238679489abb0a2f409e6b9b4a : SUCCESS in 10m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e25e6d5c019a43dd8ccbb18f94d9177a : FAILURE in 2m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3ef5b15caef407c9ecc3f25bd16a650 : FAILURE in 5m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b9e57b790e2f420f9d9ac295d783202d : POST_FAILURE in 8m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/badd1888301b4c3191cb25f9bedd7fc6 : FAILURE in 3m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/77322156d4ae4e9a8209d046fe1714b9 : FAILURE in 2m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/243a2b2a78c7464d948e57313769595d : SUCCESS in 17m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/81d79dec0f274be18eba0c6bc13bc7bd : SUCCESS in 5m 38s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/38c82f1c45ed4125aa86b3daf7e12807 : FAILURE in 3m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ec4495cb49ad4e70832a76376b5dae0d : SUCCESS in 26m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc7bfd7ae911425b84b3818fcbd09762 : SUCCESS in 12m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/45448bdc701449a2b2dfc6fc32200637 : SUCCESS in 13m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/398ecef3dfb94f42b0651004730f35a9 : FAILURE in 3m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/53ddfd0e2f1d4676ba011be7115861c0 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9d8f451ddfe40509c22068c38df120b : SUCCESS in 26m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ce9710cda3294adba051a694f591fd81 : FAILURE in 5m 36s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/58a7982d96594f929c75e4a3d586fe92 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e052546340d4add8ac75c76a082d5cb : SUCCESS in 6m 24s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3fe9bf7328644dcd8850c55072b082e9 : FAILURE in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fe6537f78e44bfeab43492655c6a1da : SUCCESS in 1h 10m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6c270bd38be414b90e7d8d14db90a7e : SUCCESS in 2h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"584ab2126d0de40a2e83f78e9b250823ae864d5a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-11 21:02:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13878d25734dab8d174d0464544411cd117fae2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 22:43:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4db5c571046f4035a62106eff638534f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6679ab6b3a0a46739bf8308ae004438d : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c1ec57201c84ac3b0856c4462e8ba14 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d687ca159a7c4a31a17308c524748131 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff6120c0eac941ea835886d7498b16d0 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2beeac1e42404a57a7f406e3ba43cf19 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2df8c3cc6ca449f09616e03931650ea4 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66f35b7673354504afee1c8f16000fa3 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ac72c1e62f9f45b39635722790665afb : SUCCESS in 10m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/051d29f790494fe386c684a8ba5a0e0f : SUCCESS in 12m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a9bc5a885dac4b1a8467b4673e4624d9 : SUCCESS in 9m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ea82ff550c154bb4895ff45bd05d4d42 : SUCCESS in 13m 30s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/32e787729bab4c28a2b11c986537589e : SUCCESS in 5m 50s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0d700199ef00494cbdbe76ae3099b2d3 : SUCCESS in 4m 59s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5e5baded7924467abaf0376d9385c8c9 : SUCCESS in 5m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f3c7f83625f9418cacb475c5e252991a : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ae01e4e98e70471ca67f7e46392266e0 : SUCCESS in 4m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e3481b86feae410782b9b1f4cfe3a4b1 : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3947cdff7fe846ce9cc1d8de84067791 : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5afde5d5026647568189ea7078c2d320 : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17e74938e7b744d49e170a365935e969 : SUCCESS in 8m 36s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/34d4961fd5a14f4b968234fa67e62caa : SUCCESS in 8m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3b8e270c5a2847b5b902884f955e1a94 : SUCCESS in 12m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/452ce19e38c9448f82021aef61247869 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ee9471d6de084fb680248341421401fb : SUCCESS in 11m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1e96ce0103594a68b30afb2f92aff2da : SUCCESS in 4m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ed816bdbdab94b1487276494bc1a7187 : SUCCESS in 11m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5a09261e3667464786a384d64baddc8d : SUCCESS in 23m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ca93125781e34a90bf2fc730bf929c58 : SUCCESS in 11m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d08250298ffb4a0abb290329a2233aa0 : SUCCESS in 11m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/42552bd6fd1d4458b94d8af208db7972 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/65073f4d4c554cc69d39f0e083c24230 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79ae4f1090b24040af629f730b870bd4 : SUCCESS in 22m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d945e7dbdd1741fea7b7ed56c00e9d9f : SUCCESS in 4m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2e69107b22104f71877ca0ab5a65ffa4 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e80c5baea29d434bb4f8fe921f9ffb3f : SUCCESS in 6m 51s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/70498e86cf3e4ef0b0bc4f5861e4af5e : SUCCESS in 9m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b60146931765421ca1f6f20cef616213 : SUCCESS in 40m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22a6614c6db84e4b86f1c0b6377b1171 : SUCCESS in 1h 35m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8fa5c266a0f8c59f0135a690cc9bf1de4640caae","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-12 00:13:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a0054e982b79396fb152c3ccb6b725822f549e34","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-14 20:17:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d81cb5a9980f17257f8870ac08f1e767b17bdcc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 20:18:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"760cc171bc965937ca0f5b43e91fcf8ae43c648b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 23:00:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/513e1c5398f24f3db8f29d75bdffa633\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f107429343744005ae758b08783f21d6 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29322232b24a4ed0bd4ad1a956f0abb0 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d555bd49e6e4d5092a5c5e6fcc2a613 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23dd08a827c549538571f45748de9726 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b2f9f74beda481f85318fcab6e4e559 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a8c604e0b8c4ad3bfc61cc638bdcbab : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4492ed9e79454becb1a901855913c758 : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f20c034b96e34596bbfec3d16c23dc40 : SUCCESS in 8m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4de5cfb3aba1480cab2ace77536b4934 : SUCCESS in 13m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4b5b86c002964185ad5050347aee573f : SUCCESS in 8m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b29854d0c185435fae282b71349f5d35 : SUCCESS in 13m 19s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7b50fa5715364316ad83e12904fb1276 : SUCCESS in 5m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b5e90cbdf6044f72ac085d9d51d0e356 : SUCCESS in 5m 32s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c2051cf0737344b89023ea03ea21181c : SUCCESS in 10m 24s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/36317ad5d2764ce7b619be571a493763 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/230b1558fc144379b8a83c8bdf3a17b8 : SUCCESS in 8m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bc3b3eca7b324cd591dcb1c74065534e : SUCCESS in 9m 05s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d4335b4514fd4fa686e01ff8be9b1083 : SUCCESS in 16m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c0721c03a14f48bf8c7f1e98078473c6 : SUCCESS in 4m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ca2e963dd60b4b099d45dfb7b3610140 : SUCCESS in 10m 17s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d0669c45f8364334a9fda96ab7c20a51 : SUCCESS in 6m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4676de5a0a5f4d6d8fd420617c652487 : SUCCESS in 28m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4fd20288b00b472c9a9b24584906dc68 : SUCCESS in 9m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6985e5d96c6e40859617b6c48f09fdeb : SUCCESS in 14m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/93774c751d51441ca99a3bea33f95906 : SUCCESS in 6m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5eb852d470924ae6912317eb3411ebfc : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de0599e323cf4976aa3499e1fb2e59bb : SUCCESS in 20m 21s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9341615e7896429db677edc323b8f4f1 : SUCCESS in 6m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bad5c500c9094a7d9917d6b7648ebb7a : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18c31df6096148fda78e58cc7750182e : SUCCESS in 7m 20s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d574255cbfcf499db13fcb1b5473759f : SUCCESS in 14m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1c1e49916254787ab8caf6ab2137c29 : SUCCESS in 54m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2866b5f0dcf0438e9edbb93167dbde04 : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c02e9e32554c9ccc076e384088f6efee6be48d42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 23:00:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ee77ab1d4bc31bec83f32ad241c861135cdad031","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 23:01:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d94581ddee348cdb2d1a9a4204480a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7095a33600f4b3583a1c22ef4879752 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4156d37e8bf9c3fbbbac8cbf2174da9f117ba711","revisions":{"4156d37e8bf9c3fbbbac8cbf2174da9f117ba711":{"kind":"REWORK","_number":1,"created":"2023-07-06 08:59:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/887774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/887774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/887774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/887774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/887774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/887774/1"}}},"commit":{"parents":[{"commit":"ac06a7a25582da6072f54e09e06a655bc5e95db0","subject":"Merge \"update constraint for neutron-lib to new release 3.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac06a7a25582da6072f54e09e06a655bc5e95db0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-06 08:59:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-06 08:59:36.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 6.2.0","message":"update constraint for oslo.utils to new release 6.2.0\n\nmeta: version: 6.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I62874663f891d0536042af7411048e0205ae635b\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I21282e6916f64d15d87d08d242a57d6bafc3c500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4156d37e8bf9c3fbbbac8cbf2174da9f117ba711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4156d37e8bf9c3fbbbac8cbf2174da9f117ba711"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
