)]}'
{"id":"openstack%2Frequirements~887777","triplet_id":"openstack%2Frequirements~master~Ia1946f42a187c69872744211fef719bc963b81dc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia1946f42a187c69872744211fef719bc963b81dc","subject":"update constraint for python-ironicclient to new release 5.3.0","status":"MERGED","created":"2023-07-06 09:04:18.000000000","updated":"2023-07-06 22:43:45.000000000","submitted":"2023-07-06 22:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887777-new-release","meta_rev_id":"478224a86721e5ad7d785f2dee46e92f0769b70b","_number":887777,"virtual_id_number":887777,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-06 22:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-06 19:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-07-06 11:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-07-06 19:47:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-06 10:30:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-06 11:50:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-07-06 19:47:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"50f1ca63045395c51d1556d6ec418293fe34b4fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-06 09:04:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc42ba6022a8fbcd60bbf3765b9f8c6f03323bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 10:30:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e2e4c6831dd42048b798c7edaf8888e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/908a4997ce5d4a50a64218f178bb5071 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a29784a452142649923dc66232cf5a4 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fc49d519faf4e17988dff023e7294a7 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92233a5f625b44e79ad58b34a6bd514d : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e91380bf286845238bbc5267e42cf79d : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6648a1bcca6441a852f61e596209167 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f33aab87513402e94e5bce347802eed : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ebc7175dec740378dbeb319a5e96641 : SUCCESS in 9m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/99f20cad5f3545e6969da33e57eafcdc : SUCCESS in 10m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cc1f5e050ade489d8db70cace0b36ac2 : SUCCESS in 8m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/aa35657266524cdab4795be58979628c : SUCCESS in 10m 59s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/08d3c97a79c1455190b82882c4729ae9 : SUCCESS in 7m 42s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ea41fd46d44443379f9d5998e3269dc8 : SUCCESS in 5m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6e51aa2c682c44349c3ae6fe00b89e44 : SUCCESS in 5m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e764517dcd0a47d5bba902b429b778f0 : SUCCESS in 11m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7e396149df044cfeb3085b160c876d28 : SUCCESS in 4m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bf052161fe2c4842b2956fbc935b7132 : SUCCESS in 6m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8c0c9666ff9f4d5c8ffe54177c9d71ab : SUCCESS in 8m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/dd0e62f3864e42df844d8a4d373d6d02 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/546b0aceb1044426bb0697e965b4e5f7 : SUCCESS in 7m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7a4eefd25d024b3184f8e3f586aed69e : SUCCESS in 10m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b5947b33942a43d18eb1a4447dcc77da : SUCCESS in 12m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6b75d46171b84cf4a44ed3f3ea53e2ad : SUCCESS in 3m 59s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/169dc286f42943e2bfed195901f865a4 : SUCCESS in 7m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d9c79a51a6184ba1a76b8223b487f9f1 : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/079b5ae684d74bd3bf2758f794b9e6a7 : SUCCESS in 10m 08s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0ae252f3d8ad43a7b31b8789d75f9eb0 : SUCCESS in 24m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cdac3ed11ca045c5ab92771542159753 : SUCCESS in 12m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/21bbec71251c41ffbb4f967ab53c6265 : SUCCESS in 11m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/837ffef4bcab47049e01d428e99a5c49 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/368294ee3596421099def50cd7edfbe5 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/92f6309e27154e3ba76c708f3f5ac84d : SUCCESS in 28m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/28fe51bc2bef41baa2f6386fc4df33e8 : SUCCESS in 6m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3d851d349c614408b8bf45168deafd0e : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc9cb7a263a540678871df9dae92a1c0 : SUCCESS in 6m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/509379ac546b45bc892f8e6cfdd19204 : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/723328e13c5b49758287a31938f6a39e : SUCCESS in 1h 05m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae5db02f49f24c0b92d10db0dfffae33 : SUCCESS in 1h 23m 35s","accounts_in_message":[],"_revision_number":1},{"id":"85c3b2dbe8d7b044fb6be19f1908a64b77998557","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-07-06 11:50:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6733d3445fbbe3fd33784c5e6012d1b83c813c60","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-06 19:47:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4b9d981fb393c12f1226464cbd5c073a2a63f7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 19:48:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48a09830dc101287bfa758fde3da2e93889b46cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 22:42:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/536234a34c664a0483b0eb9c9d6b02b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544a27a8b76643dfb035258457a1db0d : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8556f77b70d40b2b2d37b770f6d90cf : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86670cf1a3be4bcabfa186d9182e8072 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dde0dda4b1b446b89ac021e834ae24b9 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b762926e4fa4e9eab07d8143adcd61a : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d74788264d5c4d5ab8c2f82526505246 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0f37c7a029341a5b900510345b89e42 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b49cf401e90d42a288d90b6aedbd5502 : SUCCESS in 8m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cfee0c2aeb7c4bf492f2484bbefd59b8 : SUCCESS in 13m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e85aa577af854c58ac65535819cc2be0 : SUCCESS in 7m 55s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c2c58d6b5f044a0193a28ae474c986d5 : SUCCESS in 12m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6edbad788f964d0a9f25e363b6ae647c : SUCCESS in 5m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/296976f65b0b406c8230421a03717b10 : SUCCESS in 6m 15s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3b086c2967e14d0794b896cea03c2f2e : SUCCESS in 10m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7c84a24ae77049d884ab2319881b4529 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/57ac40bcfb514885811c4dfa11444ed2 : SUCCESS in 8m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c1bc682a9c1441e3ada3435f8be4c279 : SUCCESS in 7m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9a4f1ebea2e84b709d02a3e82025f883 : SUCCESS in 12m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7d9f5407f99f4db099f76b5646378947 : SUCCESS in 3m 51s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8aa706ee57754ddfb5243d93856d43e1 : SUCCESS in 10m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/71467ec2e5344fdf99008c5ac475f832 : SUCCESS in 4m 24s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fa196250549440b38874fe55f3b25961 : SUCCESS in 28m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f98b1c5efe91421ba92c2edbc723dbc5 : SUCCESS in 10m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a3bab625f3f64a5390d6c9b5eb3c23d8 : SUCCESS in 18m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/63f4563bd8024948991cf2d9e37add57 : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c04fd25fc994460dbd052e2eb2938813 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9dc7fd785f9540d190a9dac21b2b3256 : SUCCESS in 27m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ed630f618c41452f9a83d5ac7fd02150 : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ca45338ef1a64daf9167b0a667d1037b : SUCCESS in 6m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/096b50b9dd59479eb505017ee54007b3 : SUCCESS in 7m 38s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f15b485499344f45bdaa0d31a0d4288d : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfc484314ad544d8a0a0886c71824b4d : SUCCESS in 1h 00m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb1f995fb3c14760988d674e1afdb754 : SUCCESS in 1h 56m 39s","accounts_in_message":[],"_revision_number":1},{"id":"92824c4b56799d4da8af853c3fd2c6bf2c072745","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 22:42:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"478224a86721e5ad7d785f2dee46e92f0769b70b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 22:43:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39460229d3ce490dac4a5f03bc8e846d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1eda31eaca49378aa507245d3ec01a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84bce05233e74771e9c18436cbcb91c0bc01a35b","revisions":{"84bce05233e74771e9c18436cbcb91c0bc01a35b":{"kind":"REWORK","_number":1,"created":"2023-07-06 09:04:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/887777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/887777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/887777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/887777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/887777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/887777/1"}}},"commit":{"parents":[{"commit":"ac06a7a25582da6072f54e09e06a655bc5e95db0","subject":"Merge \"update constraint for neutron-lib to new release 3.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac06a7a25582da6072f54e09e06a655bc5e95db0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-06 09:04:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-06 09:04:15.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 5.3.0","message":"update constraint for python-ironicclient to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Change-Id: I49a0d84f09177dfba153bd1ec62c698c77bae4ea\nmeta: release:Code-Review+1: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia1946f42a187c69872744211fef719bc963b81dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84bce05233e74771e9c18436cbcb91c0bc01a35b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84bce05233e74771e9c18436cbcb91c0bc01a35b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
