)]}'
{"id":"openstack%2Frequirements~887933","triplet_id":"openstack%2Frequirements~master~I82e064cd424639ae5fdf5cb07ab867fac9350147","project":"openstack/requirements","branch":"master","topic":"887933","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-07-20 02:34:30.000000000","reason":"Change was submitted"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-07-20 00:41:37.000000000","reason":"removed on reply"},"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-07-20 02:34:30.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-07-20 02:34:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I82e064cd424639ae5fdf5cb07ab867fac9350147","subject":"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt","status":"MERGED","created":"2023-07-07 08:41:52.000000000","updated":"2023-07-20 02:35:53.000000000","submitted":"2023-07-20 02:34:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887933-887933","meta_rev_id":"d4c253d4e88e61a8157de95ef08c5a86faea8df4","_number":887933,"virtual_id_number":887933,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-20 02:34:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-20 00:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-07-19 18:26:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-07-20 00:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 09:39:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-19 16:31:49.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"CC"},{"updated":"2023-07-19 18:19:56.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-07-19 18:19:56.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-07-19 18:19:56.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-07-19 18:26:10.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"}],"messages":[{"id":"fc7a1ba7ffaff00cafcae0ab78be346826d485f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-07 08:41:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b40709bb843886bfce8463068073c2f4cd05aa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 09:39:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fb3def74c6f4e5c93b57191a8049b3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39508e8c0de64d9a8665999b27054e84 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24ff6ee60d01459f91547f3b04271a27 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05960f96c3c94141a88a60e2c87238e0 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1aca552843e4ffdac2ff704ea7b4eda : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8d7f368bcb84bbb8e264d2398238996 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c44724954d2644269e8b6e39b3cad41d : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc12629da7034b5a83378027f2961f97 : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3297da16c175418fb983d4bb3a359a49 : FAILURE in 5m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/efe27de20e9d4725b029d86976d8a138 : FAILURE in 8m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/25bf286b906349d585ef10a0a740c979 : FAILURE in 5m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7be84852b1cf4ffbb97aa0aeb06a4d1b : FAILURE in 11m 50s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/bb0bfa200c0843ac9c2d92a746ca8e90 : SUCCESS in 7m 16s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/ff24efb90ea742df8f4dd0666a7013b3 : SUCCESS in 4m 58s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f59bd78e45e84b8c8a5984a7d70e5bf4 : SUCCESS in 5m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/fbf9563c54dc49adb0c37f8d144fa847 : SUCCESS in 11m 19s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ebff8e9a92444f79b9225df27a1f3df7 : SUCCESS in 4m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/87ef362291164ae99c21715bf209caad : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/791d7ac7b5e1441cb0a004d72c324c4e : SUCCESS in 8m 45s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b3bb54157b5546098ec00e25d5a17cec : FAILURE in 2m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c977b388b5994097a1589661394a8a3b : FAILURE in 5m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/78214fcd3ad14800a4e6598a409a01c8 : FAILURE in 9m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5660cb41ba4743c5b7ab3f1c39699ff9 : SUCCESS in 16m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/b0ff72faec9a4d11a573f0fcada23558 : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/54d1ca6987a24ef7925e9ce1a89e9454 : SUCCESS in 13m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2128820b005d4cb4a3b6609b8d9e6e26 : SUCCESS in 6m 13s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/36978e785f3f4b2d88776246e63ba51f : SUCCESS in 10m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cf4f2b7aa18f43b68b7cf0965a6d0b9a : SUCCESS in 29m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5b51d328ff4940c3bc7c62b769966fbe : SUCCESS in 11m 41s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/34a6b2e4fe034e60859e0b8ad7707534 : SUCCESS in 14m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/64f6cbef673c426aadf79df0a6878fc6 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4483d66d96c64438a5d4643fda6e7591 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ba9d1f0c3514cb7b0647a3c38a4c232 : SUCCESS in 27m 36s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/79597c996c5d4923af04307669023d34 : SUCCESS in 5m 26s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ab28456f0df44c928f2c2d89bff79cb3 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a43a73b446fc4f56824d1b716ef630d0 : SUCCESS in 7m 55s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/0407b9aae9cc43daa6e8f75a9c0205f9 : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ae9ed18f886476ea82c4e86f4fe3295 : SUCCESS in 55m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bdcf83b1cca4568b3b09ff0289de426 : FAILURE in 36m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2a8f756946ba036ed4694a726c9f4cc0bdd41242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-07 11:00:40.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"abf2fd2e1fe37a0cb1bd42dab726a72c04cdde61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 12:50:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e588844d79654fa7abfdd13af46b0448\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71425ef2b7f942e19322b1758d40c151 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6b689cc2024806b996d26941fa4910 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aec5beed0d64e6b93b90054082ceace : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abd5686c00e74edab258f834f3591882 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93108656d52b4459a22baa4f9956d338 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/157e1ffb5c634a1d9963cf860055747a : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9944388b6cb476cb660cb614d8a48d8 : SUCCESS in 4m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e5528d9d2d0b41fcad0506091e651a62 : FAILURE in 5m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/26a967367f224d54b12109c469a39a97 : FAILURE in 7m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9023e4eaf19c41fbafb17de4d57085c0 : FAILURE in 10m 28s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e8e9629e0c204ddd8376efdfeb5f5ccf : FAILURE in 6m 57s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fea860f73fb34c33a661c6d0440951d7 : SUCCESS in 6m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/779f9aad52ec40f6907e33ed8b586dae : SUCCESS in 6m 08s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b88f85015bb44a148af621274798ac6b : SUCCESS in 5m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e7342f19b0c04808b048ce1f46862056 : SUCCESS in 12m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/931a96f440b24c5e9883a4e387ca21bb : SUCCESS in 4m 43s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6bfcaacf12d4467b95a70558e8b63bea : SUCCESS in 5m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c6916520e6eb4914aa89fca4e8cc1e85 : SUCCESS in 9m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e095a007f6294b6e9242591d33fdf4cc : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52749a3396d54661bd23344be27664e1 : SUCCESS in 8m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c78c98abff4849a28b9a30717663d681 : SUCCESS in 8m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/042c1b684bcb4ace8c65d09cda10d89d : SUCCESS in 13m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a19d4cd9967b41bdb783ba2062a7c8a3 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3d2779e43b91478f86441e63de8d07e6 : SUCCESS in 10m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/9530d074aaa346f583f550042b44f56f : SUCCESS in 4m 04s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/aea99e04815149939ce42dcbc0479c9e : SUCCESS in 12m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cfc3e39ab2434586a52082b1f1b2c787 : SUCCESS in 24m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9cb2501e1f3149c4bc0a5d21852508b9 : SUCCESS in 9m 35s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/664803e3c12e48b38f738a03742a58a7 : SUCCESS in 10m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/aad97ad526a64042b21c139062d9ffe9 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf34f217540343d1b0b33ab3b119a5ee : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/585ae427668f4b8b93f780681d8a6401 : SUCCESS in 29m 54s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7d5fce6dd20744d896003ce8f29296db : SUCCESS in 5m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9f6c117d125240b58b30f49ab786dda4 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a1fcdfbd24b24b09828dff17487e27ef : SUCCESS in 5m 34s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3e256fcbec204879a3a48f249abbfd22 : SUCCESS in 15m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec18d64836894e34a9a85faece8866f9 : SUCCESS in 49m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/287f9a9a36e94225a04bdbfd13b4b96d : SUCCESS in 1h 41m 21s","accounts_in_message":[],"_revision_number":2},{"id":"aca09cdd80bcc936035be0060f5c12527ccd8f29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-10 16:00:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8c9ff7a3746fac652753031cfb2dafb0238d655e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 17:10:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/280e378e7f7f438fbfa8be8d30a6715e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20cd3474de604630bc06dabd1568a3ff : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a8b18d2a3c4f8ab2647f0176cd9628 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1efb48fc4c3404db223b0315a7ea571 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbbc208a9cad4c24a1a74f7d61c40736 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09a8b8938b4c468ca886e3bf7fcc8350 : SUCCESS in 3m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc9d56d398424ca791467ea5c8c6ed35 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bac7dfbe478a413fa2766d0336fae6b8 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a488413faf84c8f9fecba683cc7b0eb : FAILURE in 4m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3e42655bf7924debbcb652deeb5e4694 : FAILURE in 7m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/efa30ccb53b7437abbd4d83a8642f2e9 : FAILURE in 4m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/64adedb9d5be4bfdbfbf0320433ee625 : FAILURE in 5m 02s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/f644d1b156174baabe938574c8eefabc : SUCCESS in 6m 26s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f4f6e6c4bcef43b48fe210842ea9ff0e : SUCCESS in 4m 13s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4a06516659b6449c9a046245addc77af : SUCCESS in 5m 02s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/458e9af7bd6241039d59d58015225dc2 : SUCCESS in 13m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/be3a04bc239541fea1cab124c8f8326c : SUCCESS in 4m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/319635fc39c54cde9c094c00c87b1774 : SUCCESS in 4m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a937c46dbce14c78b9223f62e9125292 : SUCCESS in 6m 53s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d08ce89bd0b240d29c05496eafaa45f6 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e87ac76d5464aeaa879a9b93e97b42e : FAILURE in 6m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dcbf90b298314ef98c8b82f4696e4ae6 : SUCCESS in 8m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/15bb2531bd97421985c7d698e5dbb015 : SUCCESS in 11m 26s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/335657bac86f4384b0499cd57d59df01 : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d9870f42eb9e489eadd25e9740b78a47 : SUCCESS in 9m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ca3b737e06f643fb8ce95e683c66fe1c : SUCCESS in 4m 47s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/024e460c9c054e69991483a9a8fc31d2 : SUCCESS in 10m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4301da1569144207b79194c7ee8c80ef : SUCCESS in 26m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/01cd92c3f76f46ee8bdc3a7d95fdad8d : SUCCESS in 11m 47s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b9849f890ee544c282f8c90bd83a8a4c : SUCCESS in 13m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2babae5588ba4828a344b166c1b32d68 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8edfde3ad8334492937106d938155112 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ad329e268624201a1138642e14137ae : SUCCESS in 21m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/42c9d95e7ba848a2a8cf647997bdfc6a : SUCCESS in 5m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/17534071c6ec4e33bc9a22c97c06eac7 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d21193190e3d4c23bf835d99bcd7f0a0 : SUCCESS in 5m 43s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c3c81ae0fc884159b7974a7e9bf8bfde : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7fdc8ca19024785b5eea685bf53b4ee : SUCCESS in 47m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ec8fb4fa33d49cf9f232170234e69db : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":3},{"id":"a0ac036bb6d657ce179796c6ac4640dc09ecd911","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-11 06:03:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"138a0251190168acb20d263181df24167b892081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 07:37:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46e58471f7594887a8548329feeb7aa3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58dd54321a0744fda4941e6e6f5161c9 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e08b7b44ec3438292609ef740b20a3b : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3285ffd6ab9845b1971af36db6f65cf0 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2db6f886e26419eb50e2129dec00726 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec3c3d2b59844b71a37b33c7b0436f22 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cac5110658824e55863a4d4f025c2ee2 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c2470e7443d4d8989e44a6ec99db496 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d708d64e52148fe86818ce7735f0cb3 : FAILURE in 4m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4be668082c534ad9b33a64d588755484 : FAILURE in 8m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/49d194d37ee745f6bb81fa73fc16c4b8 : FAILURE in 4m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dac356eabe7b4679a72c38d28b5cca12 : FAILURE in 6m 26s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b165751a33724fc9ab34251336b4ff1f : SUCCESS in 8m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b08e91d9f1bf41e8bfb7b89aa2a33217 : SUCCESS in 4m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fdff322c28b143e3a6583b5611064267 : SUCCESS in 4m 53s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/afbd4267c53e4f8699d264a29fe03561 : SUCCESS in 9m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e18fb6217d1a4c269b31af8f9025445e : SUCCESS in 4m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7eceed9d6cbf482b97f2ba70dac28442 : SUCCESS in 5m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/aec71b2e441c4895ac81c48d5282febe : SUCCESS in 7m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4d6c23a2cc294d4f9eed9c42b0aa74f3 : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d16c01047164f449591ee5d71eeff98 : FAILURE in 6m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8ecb15796f3148dc859338167f866794 : SUCCESS in 7m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8f2e3a28a3e64030be68ff6fe57cd48e : SUCCESS in 9m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0ec58b90335947c086cb62c160669f6d : SUCCESS in 3m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/22da9216e2394cb19c422b490e517be0 : SUCCESS in 7m 53s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3811b38ec4cd48e795300f492e0dd054 : SUCCESS in 3m 43s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e2bfbeab90e04004a3f1cf564ea00ecb : SUCCESS in 8m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0c995e63d0c64ed29c77fe252c618157 : SUCCESS in 22m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8865eada51b247248d7b95a3d319ed73 : SUCCESS in 9m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b60ef980428944b38f18442ea6ac3f26 : SUCCESS in 11m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e312444e82de4ef695c0ed4d7c58f781 : SUCCESS in 4m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/803f1a5db652428cb2a6ee3d3b29f48f : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7434a32e492c4b249a92c0e6fea8582d : SUCCESS in 23m 38s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f6ba150047324d9eb788ca57e2681613 : SUCCESS in 5m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7fe9d75a0c004f6d96f466a24a68f51a : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c0970ce6b14456083173cc58c34ab31 : SUCCESS in 6m 56s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/573983af25064d06b124389eb3c5d4fb : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ff07c2b12dc44f4b58621de55c796df : SUCCESS in 50m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ecc362731334ecdb533e326f2c76a5d : FAILURE in 1h 24m 58s","accounts_in_message":[],"_revision_number":3},{"id":"42ff8b60c4f6b4130a3d27a9b227cd68fb1447f5","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-11 17:10:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"68d013dbfac337cd22a44bf306e34014b2d70877","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-11 18:06:22.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0eaeb4bfdaceab6a62c782df45d22fa39c6324b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 19:23:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4af8b14eab2e401dbe6e925b18284837\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/defc62e77b6949b58a1bf5d9c0715059 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f1a23f11249475d9119ffa3f809d162 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57050a68ea63408e8b37e7d9f6ff94f6 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/617a640ea72647758b305c240bb738ce : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/137958d1cc244f38b2776e07642e194e : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/324cd8a4fdf34338a6c4972872034055 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/525216fe5fee41f789ba88fe5afdcddc : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9b7eafe4a814971967da5c75dd0420c : FAILURE in 4m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1442ea710a014564ae619454eacb0744 : FAILURE in 4m 59s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aea5faf3ab554c33876ae4a183bb11e0 : FAILURE in 4m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0bc77e2a21154247a9ccc3cd39c2c2ea : FAILURE in 4m 40s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0f496e7948a644d0ab95051223ed4a30 : SUCCESS in 7m 01s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/781cd84bbd174b1e82d2ae0c7eb35551 : SUCCESS in 4m 28s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7ee873fbfbd3454e9880325c266503c8 : SUCCESS in 6m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/eed0eb9b9880427084a60915c5739838 : SUCCESS in 11m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e0c1158f3afc47b489b1c357f4218b3e : SUCCESS in 4m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1613b3c247854785afd1c098f015bd30 : SUCCESS in 4m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e7053fb112ce4faeb11d0b3a78cfb285 : SUCCESS in 6m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0b3c6ea8f1ad42528275fa8fe8310b4d : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6c0d5a75c8644719191a93a68046087 : FAILURE in 7m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e3855876e7804b9695339a3399bfed99 : SUCCESS in 7m 45s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ce93d679436341deb06fb4e1d09410b7 : SUCCESS in 13m 32s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8998959334374750ab70be70fdf81c34 : SUCCESS in 3m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5aeacd797b7d47ddb31dd00150d04684 : SUCCESS in 7m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4ad02c1e8867480695f793823e17c1fe : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6eae228d3fb844b3ac02ef6f79062485 : SUCCESS in 8m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/67b80f628c564e77b24373b7146a519e : SUCCESS in 26m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0b4787f65ae045d1ad9fde30f90fd30a : SUCCESS in 8m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/67a4ebaafd5a490d900aac9593fd3f51 : SUCCESS in 10m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8260d8a5f78545b48cf21398c35fc29e : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b399b147ccc34aadb0bd4f422a50d62b : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7e96ee51dc44f04b15e39b9a84ce76b : SUCCESS in 30m 09s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/aa6ceac9d8644e5e8e174225402034f5 : SUCCESS in 5m 17s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/26723d52d4a84d5ba834d5ae109ddf15 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da555567c2ee4383827dd39148011973 : SUCCESS in 5m 40s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/42b0a8881f28473dbb19382c07b1061c : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4f32e984ff34d97a61fc85377e3b4a1 : SUCCESS in 1h 09m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7321645292134658b568bff60b435d32 : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":4},{"id":"687f8012c8d9c4a7405b965ab8eb8c5030a78e85","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-13 03:18:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"29ab419aa628b7d4c869ad2ddca25675bf7e6bbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 04:51:47.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59ada468569d46f5a041560813860a7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3880cd58a4f942558ffc0bc6c7ea5d03 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d604da22eb05489ebf032a2794dce899 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d945fb7faed43a581dc0670ebc388a4 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45281b0664a8466c990c05b05534dfd2 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63b40fbb3c2746a2968c0f7db34e2689 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77517171339a4d9997f88c6e4aedb215 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c1f005a6dc1448c99952d823ea085e1 : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/63a655db657f4d58a27efc77862f4686 : SUCCESS in 12m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6ffed75e818947d0b85e236682b0ad64 : TIMED_OUT in 40m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/157dbefbad014d589727a4be059b9372 : SUCCESS in 7m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/131e43adde8e425dad6ac0bfba779275 : TIMED_OUT in 30m 42s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e85170689c794fdcaa9dd46b2181a843 : SUCCESS in 9m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c9e12e18815a4d3f89d82046e71e822a : SUCCESS in 4m 29s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/25a92826e77c4a6a889357a1e93a1683 : SUCCESS in 6m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/38d7bf15a17d467388d9a5572408f595 : SUCCESS in 13m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/13b903a71e6e49b0b2a8e386957bd5e0 : SUCCESS in 5m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/96f1147925a64f2184699a4a3c4c93d4 : SUCCESS in 6m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/722910aca8324b568d7453c28dbd7dd8 : SUCCESS in 8m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/86094ba45a244bc98162ab8fa904579d : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b91213177640488f86f9d1764589722b : FAILURE in 8m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/f43350843b5e45f2b4cbffc7bfa4c44a : SUCCESS in 8m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ed3fbf574e7f422b8c2463f8da13231c : SUCCESS in 15m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ef53244e160d45bc83d386274bd613cb : SUCCESS in 5m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/22a9f86c81ea4322a90a57a9ad0d159d : SUCCESS in 11m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/eb5bc7416f2b40bc81f4188afeba6726 : SUCCESS in 4m 55s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e586bad3d5b34bc4999139e5ae989cf8 : SUCCESS in 12m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/047b2bbb8cd847a8bc05c7c67511f180 : SUCCESS in 39m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f8c271707a254b9e85d4da05edaca33a : SUCCESS in 35m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3544a28dadfa468e894838d527b3c0a2 : SUCCESS in 19m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9a90c4d2b29d4720bfeda108b7419437 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5265b97f777c4d9b9f643d8f6853ca5c : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d712674cabc748abaf0f38d94d248ebc : SUCCESS in 26m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c4147c6217144e179f6578c9fd57e3ed : SUCCESS in 6m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2f0d79c599c6452fa73071a3cbde26a6 : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22bcf8f6d344480db1ddce1eee24259e : SUCCESS in 8m 15s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/4690d93478134caab6bef2676c4a40f4 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb01f9fc9a464a4aa1206e50f1edc336 : SUCCESS in 49m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6bdb840abde45d5aa0997ee42f95445 : SUCCESS in 1h 22m 02s","accounts_in_message":[],"_revision_number":4},{"id":"a6eca57e53bb17378db56fa08f5dec27be003f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-13 06:11:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"70979c1472769f06b81f1b7aee2b63fc27430279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 07:26:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/473a7368d3be4d58861e80971252e758\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2eeca9646946fd85cdba2d0e9e19a8 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293f2750a7f541d3b4ea2375036b38ca : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b0e6fcaf5124ebfb123f8e7002bf78c : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/363cc183614a4f5e92a2c79866fd45f0 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a4b8b540648404d8afb1c6d8598b519 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2568cf88bac4c1c8174a0801907772b : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4acf6a7caf3406ebd52a33801889451 : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1fc904c8badc4b039669e65329e0713f : SUCCESS in 11m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e8d2d50d36364627aa8f3e99cbcd1144 : SUCCESS in 13m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9dec722f408946a8aa24e2c0a36589cb : SUCCESS in 11m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/29eae6a0cf9b421c9e699ed9529fddcc : SUCCESS in 13m 53s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e078133a9d884c91b9f144544b927346 : SUCCESS in 7m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/87be3a9120ee4b1c9dd76c478b897cdb : SUCCESS in 6m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d5493804858c4fce86b792281742e183 : SUCCESS in 5m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7d79c7e31d864931b6fe4d9c81be6b2a : SUCCESS in 12m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/59d28f339e2946b7bdf569e952087585 : SUCCESS in 3m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b263724576cf42f1a4a2adcfdc2a9c68 : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/62ef904b22494322af9fa34f82997625 : SUCCESS in 8m 42s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/88d3203843ef42dc95dc4394c646c657 : SUCCESS in 8m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d430cc8b14cd4c549e7421fbbfa8f9b9 : FAILURE in 4m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e4ce5c1f644f4a298c3d2840916f4322 : SUCCESS in 7m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bc17337a93bb4a568988229cacd6cd61 : SUCCESS in 11m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/07a4857cd3b342ddb59fd0cc36670552 : SUCCESS in 3m 55s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/996836e3eae54969ac46fd7c8df4236d : SUCCESS in 9m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7465c688608b464ba813c47d6288a464 : SUCCESS in 4m 59s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/332d7905311e4ac698081c4af5fe6871 : SUCCESS in 9m 57s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f562854209304b45b8e1b534d3d2f04a : SUCCESS in 19m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b66fd4ca6d2a40fa9889ec22e924498f : SUCCESS in 12m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/31c28c6dbd194ed499389fbb4830d048 : SUCCESS in 12m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1d486622f836491cb4f69346404cffd6 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c84ad4104fc5464c9bab1a8b16fd10cd : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0e19f1865cf4d85bd1965247140c3f9 : SUCCESS in 24m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/607ebfd7e65e4bdca64e8692bf362226 : SUCCESS in 6m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/2a2306ed1c774f70b8a55608ef2e3c6a : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7cbc5d0de344bc9a180d9fb14f13677 : SUCCESS in 5m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ce2635cc66fa48619cc32dfe8f81a2b6 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c026430a02774e67bc0cd61dbce87eb2 : SUCCESS in 46m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2967eef81f5c4f64b01d2fcb50e6dcf8 : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":5},{"id":"1b1a890c3d62a0e44b2884dfb4ffdf895e1e7d4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-19 14:07:59.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"32a3ee786a4777a82f612170deeb325b219bf2af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 15:53:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec449fa80d5c4e6b836a5fa618c4d908\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0557a6a42f4c3d895c35227652465b : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fea698707d74c4d8d1a25c8be1d40b0 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27bd3baa9d59404b85eef5a0fa4e1d13 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e43df39fecf41719136c4c72887ae30 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c98ad18394e494a9d5823869685712a : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16e3301e4df5455382f3b17150c7d715 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e799bc2cec04ab3b4e17bf2cd4799db : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/caf38d06cea442b880bd76fb4106aab9 : SUCCESS in 9m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a6d488e195324f63b34d956d5925d97f : SUCCESS in 14m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/70ed15a8d893402c975b7800aafd94e4 : SUCCESS in 11m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/10e0daeca18c4bad8b95b74a760dedac : SUCCESS in 13m 31s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/a8b6f30f81ad4ba581f38ae39407b707 : SUCCESS in 7m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/98a6d32e2ae440af95afef92c7ead988 : SUCCESS in 4m 47s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d05a0637eafa4f36aefd8025589ebb0b : SUCCESS in 5m 02s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/49069e09f37a4a8eb41059210e0dacb8 : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2854436e3b7d435885f42d71729daa37 : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2582a62d14d14defa945abc8e20ae62f : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/07626f0ed092494c930142e9cce62308 : SUCCESS in 9m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/efb19c195ec64b3faf1254447564e990 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b49bf447a88c41eaa94084c35471bcac : FAILURE in 7m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b0d3da5317fa4a78ae2aad3564499718 : SUCCESS in 7m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ae7c43a50c4f464496bc50139f282d32 : SUCCESS in 10m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e7932888f5f940be8ce451d5cf9b55dd : SUCCESS in 3m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f457625d511e4a69ac43bf20055c3c8a : SUCCESS in 9m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a839a70d13a1402ca31a5018ed2de9cc : SUCCESS in 4m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ca6dd2d2aeeb4db2a024dbd9f3d59c96 : SUCCESS in 10m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ed40d9f588004befa51220c93ee19f0a : SUCCESS in 25m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/27f831bd22d2424784e13d914086b721 : SUCCESS in 12m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ab440c44f3574b4c95ec37332f406473 : SUCCESS in 12m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c08e71dc65b74feab29812eef9961f54 : SUCCESS in 4m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d7fa1818615c48a8825cd581b97eb758 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cae68a5dbd844a0b9ebbe463c3506597 : SUCCESS in 24m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a2d6e17487c94438ab3ba7e006f93650 : SUCCESS in 5m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/860c26f12131481bbfb46b745d881c2a : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca7717555f614bfbadca88940cd15fe1 : SUCCESS in 7m 25s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/79a0b53b91d64e748e865d14617e89b0 : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/227ee1d9c94a40879b14da9516a8f05c : SUCCESS in 59m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d76adc41417148a3ac7dc1b8fa083a7d : SUCCESS in 1h 27m 40s","accounts_in_message":[],"_revision_number":6},{"id":"146919e93462eb286d78a02857992dd537f63fd5","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2023-07-19 16:31:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bb1e6603e9e4c6d431508f829656da0d43d35efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 18:11:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edfda7c0a72e42ef96603a54a1fa9b4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e851c292a13e4a518e337d5340216e7a : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2fb99db8da43b7a31790cf1cd841b6 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ebee90541a34f5c9c9c66d1c8bca044 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98ae19dbb8694d7588715b3ba4ef5ef4 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fbedf22cae24458bb76f6a18f2876ff : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95b283b86b6a4a528eca5fa0bd0ae00b : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84aef6d396ff4d3baa793afb96c81047 : SUCCESS in 3m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f97acc21bbb94fa586b0dcda6e8ac88a : SUCCESS in 13m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8800959a98884772aee07d882d2ec227 : SUCCESS in 16m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/99c89bc83a8544b797e1794a3c24d32c : SUCCESS in 9m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/88e07c4b69b640b8866ffbcd481b650d : SUCCESS in 15m 32s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e5ccf271b1994c15a7e73d8daddbf069 : SUCCESS in 12m 06s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/09097095a95a4db2b37b0e914d55d368 : SUCCESS in 6m 11s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3844ca1fab694fe99abbcdb2170fceed : SUCCESS in 5m 49s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bf4d79d0e7bc4de3a3d7965d25aa44ef : SUCCESS in 13m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d7b570e10d4d4154abf542a2eb471e46 : SUCCESS in 5m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4f913a7cda73458d9eec8e36492e1d8b : SUCCESS in 5m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1dad3336c3c04ee49e458a1a2492cf49 : SUCCESS in 8m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/bb231ea2971c4fa1909983c7568a3bc5 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b0a9a092a61413f96df9baacf3fe184 : SUCCESS in 8m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d8e5e0dd27c04182905c4b4f67bd7e4f : SUCCESS in 8m 18s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9bd85a42edc049b9af2b74542f2f3583 : SUCCESS in 20m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a2f4ffed66d44d9e808632f99c5f0bf3 : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/46a76cc2f4134c638205123ba55f4dd8 : SUCCESS in 19m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1d0af19bad9644c1987bae67c38db588 : SUCCESS in 5m 15s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/eeb813556ddd4553a6725d021b7db6a9 : SUCCESS in 11m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c4ef6fd92c4a441c8dcd1639c1b0df71 : SUCCESS in 30m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cff15c3619c044a7bebd86eef92a92aa : SUCCESS in 13m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/014182f7364144879a64451191fb7ab8 : SUCCESS in 15m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c7140b3d73c84ab59f351707d51f4246 : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/610c5ddd74794efda317b75466ad44e1 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c870ad389bf641338c1dcd7f42ce3afd : SUCCESS in 26m 40s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9e64de714ed14fedbbbb9ce3187e2448 : SUCCESS in 6m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8315fdc70c974d5b99b189a7122de92c : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a113c3237c7149cb9f105c9eab672840 : SUCCESS in 10m 05s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/86151cf0efa5424aabdfa40ba44b1ac8 : SUCCESS in 26m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01b9881400fa45fcba844d16e1aded24 : SUCCESS in 51m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa96c948fb004fb38b727aa1094e418b : SUCCESS in 1h 18m 20s","accounts_in_message":[],"_revision_number":6},{"id":"3bb0f34660811408776683be7abe7bc6eb503d06","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2023-07-19 18:26:10.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"dd4b1afa7e61bc523f90a72aabba58effb2eda57","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-20 00:41:37.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"93b408955ef868ec3ee7eaae1f1a3793689f0c7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 00:42:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"868af15d57356135d3df0ccd2764b7b396d6be43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 02:34:30.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/285559857c0846a99cb304978f3a8ca5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/543735b1a9b7414a937e3f7f82fa593e : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c841598e0e3544d7b1d5bdb2d3c16d25 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22b67139705c4173871ab93e6fe5b419 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aabed025e3e74da0983f9d615b237170 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58d456e6eaa343518c361b5dbef5ed88 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88d1f04a15474a7f98139a0b3ae40cfd : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a067e7da957479d8119641f883ec9d0 : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7a1d3389330c42499b8408926d304a44 : SUCCESS in 9m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a8ae70259b5f4c3c85525b193be8e421 : SUCCESS in 13m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7bdbfc29cdd64410826a7a9cca03c2c7 : SUCCESS in 9m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1efded6c883940a899e031e508d6a643 : SUCCESS in 10m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cbe4ddef9b02402284ba7c78fd3d6dfb : SUCCESS in 6m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/dc7d490ae11a46e88795b843907f6c8e : SUCCESS in 4m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d1b6795b7fd24bd492e55d38b6aa8ef0 : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/eb7e883887f94f978a9d3c9bbd43a772 : SUCCESS in 7m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/93330952b2394cb1808ef363eaae1629 : SUCCESS in 7m 01s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/57cd12aa6e884287af5c9f946dab9ba4 : SUCCESS in 6m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/517d09d5f263413a80db74ac2f46f739 : SUCCESS in 10m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d6079dc259704982b37988edce4394b6 : SUCCESS in 4m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2dc5c1dc122044fe9e344fc4a16c48d7 : SUCCESS in 9m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4f5951ecdad349a98a5ba427701919ac : SUCCESS in 4m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dcfb1fa5b55642c89d83f148e051461f : SUCCESS in 20m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6f8d2165cf414ef792c6a6d63af7d27b : SUCCESS in 9m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c8d6903a72cb457db093234fac11e014 : SUCCESS in 14m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f57564b5966043e1a2021c18040c9995 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52d3dd97537747bf85ef8602ddc2172c : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8715f3cb6e04d0692dc16a3d435c477 : SUCCESS in 20m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f9ef813c409d4fd6a3768dda3c82bd48 : SUCCESS in 5m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/892ec6314f5b41168f96848c5dedf3dd : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45564b314e694baabc7892667042d0ee : SUCCESS in 5m 41s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a38ff42c50574068b7a3e25d6117edfd : SUCCESS in 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7dd9346b7fcf4515b926e0292ec078cc : SUCCESS in 46m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/877ecbc95d274b21aef8e67340870ab3 : SUCCESS in 1h 30m 12s","accounts_in_message":[],"_revision_number":6},{"id":"d69bef6e848a9d7b756f12c4550eb6ed21a9b751","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 02:34:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d4c253d4e88e61a8157de95ef08c5a86faea8df4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 02:35:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53066ad6af5d4fb7bc884c97ad674f0a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f071148d9584ec2b6e67eefc4442de0 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8448ca292f98ac1b14f873bcb6453b014f396e1d","revisions":{"f90aa35fb1944605f51c393fbb319032985ec282":{"kind":"REWORK","_number":1,"created":"2023-07-07 08:41:52.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/887933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/887933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/887933/1"}}},"commit":{"parents":[{"commit":"4d5b93767983d63a6849e6db677b1ea703bcb297","subject":"Merge \"Bump XStatic-jQuery upper bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d5b93767983d63a6849e6db677b1ea703bcb297"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-07 08:41:16.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-07 08:41:16.000000000","tz":330},"subject":"Bump XStatic-jQuery version to 2.2.4.1 in upper-constraints.txt","message":"Bump XStatic-jQuery version to 2.2.4.1 in upper-constraints.txt\n\nThis patch bump XStatic-jQuery version to 2.2.4.1 in\n``upper-constraints.txt`` file.\n\nSince we upgraded XStatic-JQuery-Migrate to 3.3.2.1 in patch\nhttps://review.opendev.org/c/openstack/requirements/+/883402 we now\nneed newer jquery, because that version of jquery-migrate won\u0027t\nwork with jquery 1.2. So let\u0027s make this the default version.\n\nChange-Id: I82e064cd424639ae5fdf5cb07ab867fac9350147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f90aa35fb1944605f51c393fbb319032985ec282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f90aa35fb1944605f51c393fbb319032985ec282"}]},"branch":"refs/heads/master"},"b5cd6a1e577c18abc0f5bbf82a32c09c59751461":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-07-07 11:00:40.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/887933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/887933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/887933/2"}}},"commit":{"parents":[{"commit":"4d5b93767983d63a6849e6db677b1ea703bcb297","subject":"Merge \"Bump XStatic-jQuery upper bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d5b93767983d63a6849e6db677b1ea703bcb297"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-07 08:41:16.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-07 11:00:05.000000000","tz":330},"subject":"Bump XStatic-jQuery version to 2.2.4.1 in upper-constraints.txt","message":"Bump XStatic-jQuery version to 2.2.4.1 in upper-constraints.txt\n\nThis patch bump XStatic-jQuery version to 2.2.4.1 in\n``upper-constraints.txt`` file.\n\nSince we upgraded XStatic-JQuery-Migrate to 3.3.2.1 in patch\nhttps://review.opendev.org/c/openstack/requirements/+/883402 we now\nneed newer jquery, because that version of jquery-migrate won\u0027t\nwork with jquery 1.2. So let\u0027s make this the default version.\n\nDepends-on: https://review.opendev.org/c/openstack/horizon/+/887548\n\nChange-Id: I82e064cd424639ae5fdf5cb07ab867fac9350147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5cd6a1e577c18abc0f5bbf82a32c09c59751461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5cd6a1e577c18abc0f5bbf82a32c09c59751461"}]},"branch":"refs/heads/master"},"f5d80c213dadcb6abf38856059344b8491fba3dc":{"kind":"REWORK","_number":3,"created":"2023-07-10 16:00:02.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/887933/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/887933/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/887933/3"}}},"commit":{"parents":[{"commit":"4d5b93767983d63a6849e6db677b1ea703bcb297","subject":"Merge \"Bump XStatic-jQuery upper bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d5b93767983d63a6849e6db677b1ea703bcb297"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-07 08:41:16.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-10 15:58:44.000000000","tz":330},"subject":"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt","message":"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt\n\nThis patch bump XStatic-jQuery version to 3.3.1.1 in\n``upper-constraints.txt`` file.\n\nSince we upgraded XStatic-JQuery-Migrate to 3.3.2.1 in patch\nhttps://review.opendev.org/c/openstack/requirements/+/883402 we now\nneed newer jquery, because that version of jquery-migrate won\u0027t\nwork with jquery 1.2. So let\u0027s make this the default version.\n\nDepends-on: https://review.opendev.org/c/openstack/horizon/+/887548\n\nChange-Id: I82e064cd424639ae5fdf5cb07ab867fac9350147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5d80c213dadcb6abf38856059344b8491fba3dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5d80c213dadcb6abf38856059344b8491fba3dc"}]},"branch":"refs/heads/master"},"9b73ce3c41ce08ab6a9cda54ba2c70507fff010e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-07-11 18:06:22.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/887933/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/887933/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/887933/4"}}},"commit":{"parents":[{"commit":"e729367f3547fb5754109157ee6b831db81d9ba2","subject":"Merge \"Raise cap for XStatic-JQuery in ``global-requirements.txt``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e729367f3547fb5754109157ee6b831db81d9ba2"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-07 08:41:16.000000000","tz":330},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2023-07-11 18:06:22.000000000","tz":0},"subject":"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt","message":"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt\n\nThis patch bump XStatic-jQuery version to 3.3.1.1 in\n``upper-constraints.txt`` file.\n\nSince we upgraded XStatic-JQuery-Migrate to 3.3.2.1 in patch\nhttps://review.opendev.org/c/openstack/requirements/+/883402 we now\nneed newer jquery, because that version of jquery-migrate won\u0027t\nwork with jquery 1.2. So let\u0027s make this the default version.\n\nDepends-on: https://review.opendev.org/c/openstack/horizon/+/887548\n\nChange-Id: I82e064cd424639ae5fdf5cb07ab867fac9350147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b73ce3c41ce08ab6a9cda54ba2c70507fff010e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b73ce3c41ce08ab6a9cda54ba2c70507fff010e"}]},"branch":"refs/heads/master","description":"Rebase"},"542a5a8a9afd1fc165dc5ae8f9c42eeebf4cecfe":{"kind":"REWORK","_number":5,"created":"2023-07-13 06:11:11.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/887933/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/887933/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/887933/5"}}},"commit":{"parents":[{"commit":"e729367f3547fb5754109157ee6b831db81d9ba2","subject":"Merge \"Raise cap for XStatic-JQuery in ``global-requirements.txt``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e729367f3547fb5754109157ee6b831db81d9ba2"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-07 08:41:16.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-13 06:10:12.000000000","tz":330},"subject":"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt","message":"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt\n\nThis patch bump XStatic-jQuery version to 3.3.1.1 in\n``upper-constraints.txt`` file.\n\nSince we upgraded XStatic-JQuery-Migrate to 3.3.2.1 in patch\nhttps://review.opendev.org/c/openstack/requirements/+/883402 we now\nneed newer jquery, because that version of jquery-migrate won\u0027t\nwork with jquery 1.2. So let\u0027s make this the default version.\n\nDepends-on: https://review.opendev.org/c/openstack/horizon/+/887548\n\nChange-Id: I82e064cd424639ae5fdf5cb07ab867fac9350147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/542a5a8a9afd1fc165dc5ae8f9c42eeebf4cecfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/542a5a8a9afd1fc165dc5ae8f9c42eeebf4cecfe"}]},"branch":"refs/heads/master"},"8448ca292f98ac1b14f873bcb6453b014f396e1d":{"kind":"REWORK","_number":6,"created":"2023-07-19 14:07:59.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/33/887933/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/887933/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/887933/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/887933/6"}}},"commit":{"parents":[{"commit":"e729367f3547fb5754109157ee6b831db81d9ba2","subject":"Merge \"Raise cap for XStatic-JQuery in ``global-requirements.txt``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e729367f3547fb5754109157ee6b831db81d9ba2"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-07 08:41:16.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-19 14:07:32.000000000","tz":330},"subject":"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt","message":"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt\n\nThis patch bump XStatic-jQuery version to 3.3.1.1 in\n``upper-constraints.txt`` file.\n\nSince we upgraded XStatic-JQuery-Migrate to 3.3.2.1 in patch\nhttps://review.opendev.org/c/openstack/requirements/+/883402 we now\nneed newer jquery, because that version of jquery-migrate won\u0027t\nwork with jquery 1.2. So let\u0027s make this the default version.\n\nDepends-on: https://review.opendev.org/c/openstack/horizon/+/887548\n\nChange-Id: I82e064cd424639ae5fdf5cb07ab867fac9350147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8448ca292f98ac1b14f873bcb6453b014f396e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8448ca292f98ac1b14f873bcb6453b014f396e1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
