)]}'
{"id":"openstack%2Frequirements~887961","triplet_id":"openstack%2Frequirements~master~I6c67cd897adb0cf8a84748aea2c92a998964f990","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6c67cd897adb0cf8a84748aea2c92a998964f990","subject":"update constraint for python-glanceclient to new release 4.4.0","status":"MERGED","created":"2023-07-07 14:35:43.000000000","updated":"2023-07-10 18:16:38.000000000","submitted":"2023-07-10 18:15:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887961-new-release","meta_rev_id":"7e646faf87566f07629cdca303ec0420b13804f1","_number":887961,"virtual_id_number":887961,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-10 18:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-07-10 13:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-07-10 16:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-07-10 16:37:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 16:17:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-10 13:26:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-07-10 16:37:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a68215297466110cd51b47d537a6afaf58e76f23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-07 14:35:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a154b9c23dd4e0f809f628471df6631ec981b9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 16:17:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/640718b10de74a77a1cbb26f1cb34fbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d32f30ef1da4d3ea3364c5e61933ab4 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e723ece21d874fada38f08fe4868f00a : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b56360ff2a304fbea471a21f39783219 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9bfd6406a0e40788a4818993e4c1b52 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab825649dbed43a6b3037e93b45b7b25 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a39762ccd75245bf94e94cfe387774c5 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c7066be735446949e4bda49005bf5f2 : SUCCESS in 3m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/01c91d829505478786fb26f93a38b911 : SUCCESS in 10m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4c7a16878e0444daad84b63c8327014b : SUCCESS in 15m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4487d50dba514d88ac7a19ba7b383dea : SUCCESS in 8m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ba7f3db1def04cc2bd73225afb626fdd : SUCCESS in 14m 07s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c5d997d683c242fdb79cdcfa1ea52dae : SUCCESS in 8m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f1f8fd1a99a54a1dbdb44b0a101c15d7 : SUCCESS in 5m 07s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/90faca3584e147c28979e2cb8b57684c : SUCCESS in 5m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/96aa23e1bdf244c58307db988dc3dd37 : SUCCESS in 13m 49s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2c37f3aa8750487bbbad12b9fc35c314 : SUCCESS in 4m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8736f6c2276e4812a91bd09bc88685b5 : SUCCESS in 4m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/480f25725ab64698b0f6381ba8f6034d : FAILURE in 4m 00s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b7112487d63b4ddcbfdef8a1bfdfe97f : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85950cbee9b246199212935a1615466b : FAILURE in 6m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/aeeafeab191e4e4ea927c8c42937437c : SUCCESS in 9m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7cbc6080d78644eaa613b20afa90f839 : SUCCESS in 13m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6ce4539a63ab40f1b83e474307bfec06 : SUCCESS in 4m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/29dd80b376fd41a1a7080d17a788b376 : FAILURE in 3m 21s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3b59497c06094f0ea97079b8cc9091e5 : SUCCESS in 7m 10s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/829cc108100f4195ac85a176c9443a85 : SUCCESS in 10m 39s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/644ea41a10fb42a29704334aaaae051d : SUCCESS in 27m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/98c1245d7e21459ba7db852b2454de3d : SUCCESS in 12m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f3a854b5e3354b089e1c2d59b944d502 : FAILURE in 4m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fc0b74f1751445a78dfef45986306a61 : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5bbb86e9f5c5449e93b9de103ee5e6fc : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2eae18cf3c14a539eeda653cae3a845 : SUCCESS in 26m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f3dc9b4ef2d64ad5b43e52c4ce5f5362 : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d568e63df257464293b1578011f68208 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50260f7236884b80af785b5a2f31f3e0 : SUCCESS in 7m 16s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/74968da2dfe44a2cb0e09d620eb2a494 : SUCCESS in 14m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c619c024dab45d7ab0914ab26faa1a1 : SUCCESS in 1h 17m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13ab765f37334bf6a9a5192fbccc6cbe : SUCCESS in 1h 36m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ef21130975fcdbc88480bbd8da9a648bb1fbe3ce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-07-10 13:26:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1203c8ffcb0c10cef4cb948f2ab62a30d1c67bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 16:04:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c88f889c6add4351ab23ee8f65d0374f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b6a797cf4a411086bc4847d510e25a : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5525f5d59144eb1ac188b822f921b1e : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38c21466e77437181c3fba19c703f54 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c642c9dd942423d8fda86d49e6aeefc : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/837f20e1786644c4933371730b18789f : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dca12e40ddae4d0fa91e32ec37890c19 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f49d7ef125d439b91f1582793e6e399 : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/218abc50126649a78a4a483cf294fa2c : SUCCESS in 7m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/50bcc742a7a540e49d084121d3153902 : SUCCESS in 11m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/180be2e7e4a443b7bbf5d8a0ed957539 : SUCCESS in 8m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4f11a56868dc47c7aacf8ba9e6356faa : SUCCESS in 12m 02s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/17253a6e4a5c4c758910f183401bde75 : SUCCESS in 7m 10s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/fa599f19c60f4faeb6c2eb42bfe3c8c7 : SUCCESS in 5m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fd6673110c434e568805a7b2cd18a5ae : SUCCESS in 5m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/15b577c71f9d4774a2a007d8339d668d : SUCCESS in 13m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6df5755b3332450f87b6a72a9445848c : SUCCESS in 4m 37s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d7f2c9f209b64b6fa5d105845ebb0315 : SUCCESS in 5m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/65fb37c66f0f4ee58e2324e9e9272e0d : SUCCESS in 7m 55s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/11d46bedb0da433b841a67820f0a2a57 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/604f9b7dbe4d43b49d0da37dba78f678 : SUCCESS in 8m 15s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c2efd737f7a341f0979320413e81f84b : SUCCESS in 8m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/de6a0100c57a4f0f93ce514414858ad6 : SUCCESS in 14m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d35f1c3a1e9c45b6a858a8c159d5305a : SUCCESS in 3m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a26d0e34ff594d80b75e4acd8f273445 : SUCCESS in 9m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4ec27836b0e7482c8ffd2f530d15fe60 : SUCCESS in 5m 06s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2629ec78e7624b1384af3f3444a4818f : SUCCESS in 10m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e6bb4b366d634710b456cff76e9bdd0f : SUCCESS in 28m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/27fc4c79ea6e4920928a203bab63a7e5 : SUCCESS in 11m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/36fa6797b7574e73a820f1a37f378612 : SUCCESS in 13m 04s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bf25985b919840ed9187e913d46c50b7 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/470ee067446b4814af39f1ff07110a3d : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d724582c4b84f0cb51118caed761730 : SUCCESS in 24m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d07946df3b554429b50a25304e979894 : SUCCESS in 5m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5dd6f0808fb24e868e287181de6e39cb : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a57e9a974f7d4207859f0c96198d0ca6 : SUCCESS in 5m 51s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/789dbe019845428cb130ecfd3363cd0f : SUCCESS in 14m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65409e9aedbb4a4faa3f5b2bfda8cd71 : SUCCESS in 1h 07m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/646457ce7abd41258abbaa069e6caf3f : SUCCESS in 1h 58m 16s","accounts_in_message":[],"_revision_number":1},{"id":"efd6473534db7008bf310f9ce67370f0dd37d33f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-10 16:37:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7534bc3e0991b8ab968f65ac3cb33132e15ed799","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 16:37:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be0976b8a4479269b905a3809da47b569a0ac10f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 18:15:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a8796edb8e45e79eb4108b93f8f057\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0222f355944e49ed8233e223257e031b : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e039a157370c4daea0b5d5c68783cd0f : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ba4ba512d454cf5b7d1de64c450bd4a : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/709904171e734f428b93c11ed7a517ac : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2343d7b85e4d43be96f3c4b5d49f4030 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/249b76ef04de429fad1c8d56d6d5fcc2 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/996355e53cdd407e83495a16bf41a1d3 : SUCCESS in 3m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5dc000c72e0c4f7fbdd17b5af991eac7 : SUCCESS in 8m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c8e713d1d647496a865be82202f7a6a3 : SUCCESS in 15m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fdb26f3a10974adbac81120105716c17 : SUCCESS in 8m 14s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/531f0e7c73a7437c8c67982c388bdef5 : SUCCESS in 10m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d624c6e3ed9144a7bd4bb8ab2aa07332 : SUCCESS in 5m 02s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5834d9f38ec24cc58e4a62644a11a7d0 : SUCCESS in 5m 42s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/45b27edd4a2d49169337e9a07bb6ae97 : SUCCESS in 7m 44s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/05c2dd22f6b74e9283f74e97e7bdeece : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bdc0bfb1ac4c4dc7912c59c7f4a68c8a : SUCCESS in 8m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ca6a8db2077145b8ab3611684593ca75 : SUCCESS in 8m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/10010c489e974836b3a200bb359c7776 : SUCCESS in 18m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/8f7b21d88f704eab95267337b018779a : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cc502cf00728419294c1bbd78efb0c2b : SUCCESS in 16m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/99149e761a564d47bba4cd0b717a5d35 : SUCCESS in 5m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/1aa8be9880de490dba61c1ed8a3e9939 : SUCCESS in 25m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0ecb46b6d0ef47bbbed869261e7acf59 : SUCCESS in 11m 22s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4000c97794144a2393c014df53b105c7 : SUCCESS in 14m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b31ad4e383154efb866a89e6607fd94a : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a9822ba75974e8180d85dbde47cd583 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78098271d09448eea5522a90f3e32033 : SUCCESS in 25m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/243b15b024fc4f28bc6f239f1dbeb802 : SUCCESS in 6m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6d680b1a51104753ad6c66e8cf159dd5 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b6f109867a44f4c870eb23198a2f44f : SUCCESS in 6m 02s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/20ed71ddc67c43eb9113e36c6498a5f2 : SUCCESS in 18m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12071b685e024c278544fbfe4df4af47 : SUCCESS in 59m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fdabe011505432f9837b7213df44ad5 : SUCCESS in 1h 37m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5e5e8f9f69c96ea91b46a5d6b2f65debd2b1b484","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 18:15:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e646faf87566f07629cdca303ec0420b13804f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 18:16:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d84b3ce0933848e3b75763265387dc66\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89eba5c0b9414bc7bb6dd2ba0c4a646a : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e835b6370d575025c5fdc559b61ff61eac06230e","revisions":{"e835b6370d575025c5fdc559b61ff61eac06230e":{"kind":"REWORK","_number":1,"created":"2023-07-07 14:35:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/887961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/887961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/887961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/887961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/887961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/887961/1"}}},"commit":{"parents":[{"commit":"4d5b93767983d63a6849e6db677b1ea703bcb297","subject":"Merge \"Bump XStatic-jQuery upper bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d5b93767983d63a6849e6db677b1ea703bcb297"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-07 14:35:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-07 14:35:40.000000000","tz":0},"subject":"update constraint for python-glanceclient to new release 4.4.0","message":"update constraint for python-glanceclient to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Pranali Deore \u003cpdeore@redhat.com\u003e\nmeta: release:Change-Id: I327a1884a1c93ece4fe5d7531d9fe4a6db3bd742\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I6c67cd897adb0cf8a84748aea2c92a998964f990\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e835b6370d575025c5fdc559b61ff61eac06230e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e835b6370d575025c5fdc559b61ff61eac06230e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
