)]}'
{"id":"openstack%2Frequirements~888055","triplet_id":"openstack%2Frequirements~master~Iccd7166f0a8091bdf4613186f9c8048864603f29","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-07-10 20:21:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iccd7166f0a8091bdf4613186f9c8048864603f29","subject":"Raise cap for XStatic-JQuery in ``global-requirements.txt``","status":"MERGED","created":"2023-07-10 15:56:25.000000000","updated":"2023-07-10 20:22:07.000000000","submitted":"2023-07-10 20:21:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888055","meta_rev_id":"76970b81d34ec06938c839e60c69ebb9c527ad9f","_number":888055,"virtual_id_number":888055,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-10 20:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-10 19:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-07-10 19:06:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-10 17:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-10 19:06:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1f496e662fd2d0f319977e13a64c9ec89ca6bafa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-10 15:56:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc3d5d1aa2183e59d8940d540af19dfe83e9503c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 17:52:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb3352c1a954fc693dee13a3391d437\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c842d642eb40f884e80f462a0f893b : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58cd43402dbd4f008d804400e44ff1e3 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee28168ee95749ee84eecddd39a37a6c : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e1aa380812846b7bbf7de4cb30e53f6 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bdf7ddf261d451f80e0bebadd5c7672 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/afb9db84a1424c719cdc5637f5672136 : SUCCESS in 3m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41b60dca9889460d8840580f5e0d12db : SUCCESS in 47m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a5c91575a044c5883efe9010b5673d9 : SUCCESS in 1h 50m 08s","accounts_in_message":[],"_revision_number":1},{"id":"266c9fb24b7fdaf447d452cbddfcae7f042c6b1e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-10 19:06:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e5c80df91c7036631c5a3cbef1badf3b2c855be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 19:06:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6af29057f689da130b38a73448a2cd5d62dba26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 20:21:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c3199e60be483aa834d5010beafedb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0160a8ef2f7e4ecc9cd1dd945256a329 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1282b7830114f3790373a0ef46e473b : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e52a52bb3b64825b81150db10f40608 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43e5864c82294cab88be2a531a1a89cd : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69236f23d7874078aee6ccca38e83277 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbade0fa666e487bb0d07fcaf5aa0df5 : SUCCESS in 4m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a51774f85ff40f6a962edb4a679a0c9 : SUCCESS in 57m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7ea0f68bde94f4dba3b511118a5d238 : SUCCESS in 1h 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"53afbc0545e887ab9ccdb945beeb68924654ac89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 20:21:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"76970b81d34ec06938c839e60c69ebb9c527ad9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 20:22:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f614292ab64fedae374e047e85a44e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74def1d0d031431a9d58f6081b68f10e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df5efbd42bd4bf89f38bee2527b61415927c5823","revisions":{"df5efbd42bd4bf89f38bee2527b61415927c5823":{"kind":"REWORK","_number":1,"created":"2023-07-10 15:56:25.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/55/888055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/888055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/888055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/888055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/888055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/888055/1"}}},"commit":{"parents":[{"commit":"287e20544e82e0ebaec3af059a439cd82618764f","subject":"Merge \"update constraint for keystoneauth1 to new release 5.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/287e20544e82e0ebaec3af059a439cd82618764f"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-10 15:48:18.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-10 15:48:18.000000000","tz":330},"subject":"Raise cap for XStatic-JQuery in ``global-requirements.txt``","message":"Raise cap for XStatic-JQuery in ``global-requirements.txt``\n\nThis patch raise the version of XStatic-JQuery \u003c3.6\nin ``global-requirements.txt`` file because horizon\nis already using XStatic-JQuery-Migrate version 3.3.2.1 [1]\nbut XStatic-jQuery old version is not compatiable with\nlatest version of XStatic-JQuery-Migrate version.\nSo horizon is updating XStatic-JQuery to \u003c3.6 and for that\nthis patch needs to be merge first.\n\nNote: ``upper-constraints.txt`` is updated in a seprate patch\n[3].\n\n[1] https://review.opendev.org/c/openstack/requirements/+/883402\n[2] https://review.opendev.org/c/openstack/horizon/+/887548\n[3] https://review.opendev.org/c/openstack/requirements/+/887933\n\nChange-Id: Iccd7166f0a8091bdf4613186f9c8048864603f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df5efbd42bd4bf89f38bee2527b61415927c5823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df5efbd42bd4bf89f38bee2527b61415927c5823"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
