)]}'
{"id":"openstack%2Frequirements~888145","triplet_id":"openstack%2Frequirements~master~I2a52c853b2f90d92d954bd99b914eccbbd0ae91c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2a52c853b2f90d92d954bd99b914eccbbd0ae91c","subject":"update constraint for horizon to new release 23.2.0","status":"MERGED","created":"2023-07-11 17:06:56.000000000","updated":"2023-07-12 22:38:08.000000000","submitted":"2023-07-12 22:37:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888145-new-release","meta_rev_id":"6c442073142533089cf0c379dd79448882b37833","_number":888145,"virtual_id_number":888145,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-12 22:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-11 18:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-07-11 20:19:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-07-11 18:34:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-07-11 18:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-11 18:33:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-11 18:34:14.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2023-07-11 18:52:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-07-11 20:19:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"9911505c9564c68a1826ec1991f8c463f3993d29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-11 17:06:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5401a382df81996c73e24ffc37c5bdd3f89f5ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 18:33:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ffe6ccd8cf411886e3ecef41ca1473\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eecef1010fa41c4a986961b4f8e180c : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c4c415e55a414889edbc43ef70073e : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36fbb14c891b4abf813ee93af42434a9 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/422fbc787a19499282712d054b47d182 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2243cc15a4c847d7979b99cd5d082eed : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21697c39d7ae4257a9ca4a8bdd4a0802 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/151ff1b0362b40438d45987ee9778967 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/609cf3c06b4148469d9aff5014194c63 : SUCCESS in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/62db06d4c90a4a6a9c07bd03241eb825 : SUCCESS in 14m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2b24b2ae70094380aae153d077ce3768 : SUCCESS in 10m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5deb3bf7238249e89f688e01529b62af : SUCCESS in 15m 04s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/cc2db4a89e574b9ab49d1ccf3074dbd9 : SUCCESS in 6m 27s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/732c2269f6c4407ea9b5db6b043d7510 : SUCCESS in 5m 51s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5aae18b7f34a408289c409d12115f0e5 : SUCCESS in 4m 45s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/01dda31dde0a4e80b1561f2698db3972 : SUCCESS in 12m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0ab37e8e48234a0d9a5d546382630739 : SUCCESS in 4m 41s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/826e6af4538a4f2d9ce8f4b5a3d4e50d : SUCCESS in 4m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/df7f764f00454a9bb1f91622f62e7db4 : SUCCESS in 9m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ebb0eafa0313450db028a83ffe34a8e9 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b235a34439f43f39b8c9c03065a434b : SUCCESS in 7m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6dba506bf84440da90bb108b8f80f8de : SUCCESS in 9m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/025a4b8f523f49ffb001fbf9096c2ba3 : SUCCESS in 16m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/556e0f76af164a28a0c282ce204de327 : SUCCESS in 4m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/15da4ac3c0e54d26aac9534e79c92500 : SUCCESS in 9m 56s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/442724afe8494ae1b801b4c59cc2e4b0 : SUCCESS in 5m 13s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/844799f9e7284e758cf7b94fcf457015 : SUCCESS in 10m 50s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0856da812463477a96aee9194d4e3b46 : SUCCESS in 27m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3a22b45724ba4f7787db03f8d0e4ab41 : SUCCESS in 11m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f1607e39aba840088934e58bb57e7d9f : SUCCESS in 13m 11s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/2e8b7a20d6bb41bb8f79cb037780cabd : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8416a6d15fa64bd5bdfe3a7e2e6ded3d : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24093d6336384379993f9f95ffc72efe : SUCCESS in 25m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a91681e8a45e4f8b8fc859b03e438128 : SUCCESS in 5m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/776a4bf66a994ace8f8b0d808d073e73 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a8447618dd144292adf32e6f3cdaea81 : SUCCESS in 6m 30s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5736ef93e7f94bd0a9ebef9d253d3b64 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4ba28173f8f4addbdf624f41bdb3a63 : SUCCESS in 54m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a67a6afa48b4637aa341a1750f37b83 : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1ef06830e1209ce27a8102f101027deac8f2df0d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-11 18:34:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e482361369976174aab1d2ab5334c413fc94c650","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-11 18:52:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d629921e1ce0af1b40c7c4a310266131afb9dc17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 18:53:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f08c944eb489d46ca9e7ea2b6b33e994faada6dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 20:05:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1128a11025d54215bd44bdc5f9dfde54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be127bb3e3d04e0785b91e3c5408aac6 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68426e84981a4cd8a3fafe995e10be52 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c544a82c453a4717a821cac14a16c7d9 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f9bf3b420124020bbcc2344439f707b : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5353c7c4efe540c2bb6a0f9fbf448719 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13aa3730a7124cd2a8ff57274f672b68 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e730aef9c69f40fda61ef8e1ac67c244 : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/196a8f7dcefb4c53807f598b10d9fa61 : SUCCESS in 7m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/633e8a3bf7d24896ae53c2442c0d1fc2 : SUCCESS in 10m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ef1fa8456443410a8e648107a8c4fe65 : SUCCESS in 8m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9339b3e2a1644376a05d0d1264bc9e50 : SUCCESS in 10m 58s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8390261add0043d892753896bbb3829c : SUCCESS in 4m 25s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ad9a86bd0fcb489e802ed703b310904d : SUCCESS in 4m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/66ec0749e9f8410f90ee2e425a9be2e9 : SUCCESS in 7m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e8d468fa0caf4d8084f6ac3bab3f34fb : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83247b1958654544a9cb3179d6273b54 : SUCCESS in 7m 42s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dda9bd95dfde45309f7490d42ab1b581 : SUCCESS in 7m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/138204d484b24aa198c285f4a20c8158 : SUCCESS in 12m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/49ba5783b9404137864450cf8b0ac045 : SUCCESS in 3m 55s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7b5499eb5a26429ea213c45bafc0d4a0 : SUCCESS in 9m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bbf03b1704764a8e8cae3311807dcd14 : SUCCESS in 3m 47s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/30993a27157348ea875012ab24eb2a7e : SUCCESS in 21m 44s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/edbf409219fe421898d664f867f62f5f : SUCCESS in 10m 01s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/77749083832f441a848927dfa41e4ba0 : SUCCESS in 11m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b383256675e74f8db185cb3e61d55d01 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c2f87fa69264bb3833414402f09def7 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca803ade738a4a6585a87f2d1a93f516 : SUCCESS in 22m 44s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/526f85ef1a594fb6bd7cacbdedeb33c1 : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4eab3ee59ce444978a8e5008fb188319 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe8983b8712d4ffd8d54b03b562040d1 : SUCCESS in 6m 03s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/98f482e37914448e975407c93c5ee38b : SUCCESS in 13m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e911412422ab47c49d741212a1b55c07 : SUCCESS in 1h 05m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/430fbcbd56704f2995a73dbc1d2d5a7f : FAILURE in 1h 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7d2495c2745907e3c5dad65a6e0752a0d7fbf0c0","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-11 20:08:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4719eb8858b7af673cc531d748a1a06bdfff8978","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-07-11 20:19:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4faa12560f5ff720b0df01485cd95916e844b416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 21:18:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96f7a977925248eaaa83835b28b2035c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07689cee6e084a2c8ced568529f25e5c : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3f6e9d90c104deab7b61d8c9ea71405 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4af885a4c774f27b1af5e1905981ad7 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93d6bf5facee427bbf0de4a63de20185 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df317dfab3d249608477a00d89e20eee : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6f811ba674c486e94ca4652e743f806 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6fcebcb99d8b4a5ba4108e39c88753c4 : SUCCESS in 3m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0d5b576547e044e89b18b160998d2f04 : SUCCESS in 9m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f1b3c205072d4da48720e0568bab5fbc : SUCCESS in 10m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/290e14e2d2b8448bb3f9b0774e05a72b : SUCCESS in 9m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dc650af563c645e08f55490033a9a9ba : SUCCESS in 9m 19s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3f2a51798daa414b833513f4a81d033b : SUCCESS in 7m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bf6fb102df654aeba1b4d59febc42bb5 : SUCCESS in 4m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ada9e4c2b06d4a3caf49aab23efd6483 : SUCCESS in 6m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d7b2c605c25d4c529d05df161e9ed2ed : SUCCESS in 9m 16s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/954fc7955f4c4374aa8305cf5dfc3ff8 : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3c115b3d0798400c97a0dd824ea2ef57 : SUCCESS in 4m 33s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5744034fbe4d4637992a2b63339523eb : SUCCESS in 6m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/52dc3cef5c1040d8bac062326d6334c0 : SUCCESS in 8m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1bd7494f05024795b4683e5954a07a1a : SUCCESS in 7m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ee9d27a15f8d4919bf7ee520c724986b : SUCCESS in 8m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7b616ee09c0a416697b848b7c0311a25 : SUCCESS in 10m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e3b9cf1bf9de4357b43f0a2c29fdcc89 : SUCCESS in 4m 28s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1615719e87d14104958816933f0b19ad : SUCCESS in 12m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/df9186a00e0e474aabdc57c2239e4ede : SUCCESS in 4m 15s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/61060c0f0f4f49fa935df167ab253959 : SUCCESS in 9m 37s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9b4ea2af43744b8ba00a82364de77b2a : SUCCESS in 23m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9a33f61fd10346d783bcab4c29ad38c1 : SUCCESS in 9m 46s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/620b91d2b9c144f9ab61d933bbf12294 : SUCCESS in 12m 42s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/08c6f551819648e7b91d2eed67b9c16e : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/16b0cb25f0e04c3d906d0e0ddede8d1a : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24a60bd05ecb4812a72efdff9a07933c : SUCCESS in 22m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/705f977b024b4347b3b6359f93533ec5 : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6c9a7d6d6cc748378c85da464560dbc8 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2b2869b338249c3a9a126d16a20df93 : SUCCESS in 6m 57s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/054f0e6e00724a82a665877acce5a207 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee325682da7f491198cfc776f8fde49e : SUCCESS in 43m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/531b7ae777c046bda6d56b31ebccfff5 : FAILURE in 1h 07m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4c1d712fefb3121a955486c8f583d3bc03721853","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-12 02:57:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fba8e8f0328e6404fbffd83e41660333730c98e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 04:58:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92e4b67b262f4bd3a89f6f6b4a989acc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10cd0aecf35242918c1798614fa8766f : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a80307c96e3846b7badf53b751a48493 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ba5d62d8c37403883050062baae078b : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41265e63a51d4e2e8d6a72512af00822 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa4e92b4365444748e7d3f0c4a32e5f1 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ab3c09f843749f295ddf8379207235f : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56f72795e83a40f09725bcad027b4c6b : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/011977e90d71417aac269dd99579f97c : SUCCESS in 8m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a4cc28590aae429fb0c87a2fe79d0f1e : SUCCESS in 12m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/374fdeb9a6a44c6abff47f7f1d639abd : SUCCESS in 9m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4aa0a702a958424684e94a036faddf48 : SUCCESS in 11m 19s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/aa333893e7b34a70b102cf3f32e40a4f : SUCCESS in 8m 04s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a397ec80902f44e5937ca6377a154a1e : SUCCESS in 4m 41s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/43037c3ee5d740589559dac1cd98385d : SUCCESS in 5m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d89e5ca2e4b842afaef15745660ede75 : SUCCESS in 14m 36s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/3a0c4d580b714e94bd6e15a663f9ee8c : SUCCESS in 5m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/03dede7fc5c540b09b137dfdaaac4e15 : SUCCESS in 4m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5cee034c76e7412d96fa5a7b516a3d8c : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/eed85f7b0b294724aaa502255c20d94b : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49df23444212407d905a4d7c1af81478 : SUCCESS in 8m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a442d47aaa6746edbc12c4769fd8c112 : SUCCESS in 8m 47s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8f0fcb8593134169b379765850bd1300 : SUCCESS in 14m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fb6124eaf8754abea19e6f1e8bd034f1 : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/68f0308e25fe4ddc9d979689b8232da3 : SUCCESS in 8m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ab82408a80c84901ac53a14b1cca454b : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/55126505bdab413e9295e3dfd07a10f1 : FAILURE in 36m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/83100d2b706f4ebbab3bd0a76349e4b1 : SUCCESS in 31m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c734d2a8a3564f95a7cfbbb9018b0c5e : SUCCESS in 13m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b8aac07a33f24e299619d353f125e458 : SUCCESS in 16m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9b39edf73b794f3db729fd5bac53dde4 : SUCCESS in 6m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c79b4d6b70a54cd59d2de4ff61481e07 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3a6abeeef5441239c6d86d2b090488b : SUCCESS in 26m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/43899b3a011642389e0254fd2a27529f : SUCCESS in 8m 51s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/91d73bb25a08440a90122c870e6f2cec : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ac2cea1aaec47168242b272e2885c1f : SUCCESS in 6m 18s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/d6d6c980d0624663b82433dbeec25745 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb54d9c6df8d4a6d95d4c905beeffe39 : SUCCESS in 51m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae816c7170f445e484a94fd0a37844f9 : FAILURE in 1h 21m 12s","accounts_in_message":[],"_revision_number":1},{"id":"fdfd1fed2f8c4f658bd7577badc2acecac174ddb","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-12 09:31:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dca7e1e22e8974bdc9695442f7387042b4fb428","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-12 13:16:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54fc18657965f9b16d81c4c0f84454a7f436250b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 16:31:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9bec0118dd641f3860add528c6849eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe59d2a312c4737a2139fa35d2adce0 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d8d2c9f8cf34b0096765d07fd0e49c5 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/310a9e0801354d18a8413d743d5ecf50 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de00f573fb334818979e20ed6b841d5e : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aff8696d3c3d4c26b06bcf450d5379b3 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d82a40d9f049469f884672df8d0d4da1 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4bd7420c29fb4d04b24be7cf98e3731f : SUCCESS in 4m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b39a050b85e0468c86b0a7c90e8e7243 : SUCCESS in 8m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/be1e4b1d43194a34af3e5ff68c783a6f : SUCCESS in 16m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6989c2b1f55b47f38dec6d3a45e017c1 : SUCCESS in 11m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9025a082db2840f2bd37472d4653edd7 : SUCCESS in 13m 17s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3b61f80f1c7047c1a872fb43ce0f187a : SUCCESS in 7m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6053b165bc834aa59309b47f17afc4af : SUCCESS in 4m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4e20209b378b411eb2e6fa586191dacf : SUCCESS in 5m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/3c586f959951457f99d58b3cde0e0859 : SUCCESS in 12m 09s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0656248323a34f0db24e4ff9f0db7fac : SUCCESS in 4m 42s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f439ff8bb3ac4547b57ff69ccf3f562e : SUCCESS in 5m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/22f0321b19cd4e9098df1f60ae20532d : SUCCESS in 11m 35s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e8ce71446fcc4a8a8b79564ac9bc3266 : SUCCESS in 19m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a5894ab60484efcb12509100c25e642 : SUCCESS in 7m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ef002158b6b046e49b51c656fef88ae7 : SUCCESS in 7m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ad96815af720481597d9c00dce2555d6 : SUCCESS in 12m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e424587bebac4029a83033accb931bfb : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/78150ad1ec1244af84bc37020ecc4d59 : SUCCESS in 10m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f7c9ed7966b04c27a05fe6daa18eace1 : SUCCESS in 4m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e329c0a331e04361b09304c74b1abcfc : SUCCESS in 10m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/027419c985f549418a25803bc76615d7 : SUCCESS in 22m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5af44fefbf204d718cc73e1013e20bad : SUCCESS in 10m 42s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f4f14dd19550436bb5d89c21451a1f75 : SUCCESS in 12m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7ddb1d867a8a493dbe47ccfab50750d8 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e77d7d4584d44aa0a61300434909c603 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba934b12afe94e30a7d3dcfe13361461 : SUCCESS in 23m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/31e4c07c20e14f8ab377cc6f698909b5 : SUCCESS in 5m 40s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3943c7d2a35b40b8b030cdadacf60c3d : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56f13d909b154c3596bf6ced249a73df : SUCCESS in 7m 14s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/212dc21cc6e44b839ea986cbe18053c9 : SUCCESS in 15m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e85d4894e7c4f8c8ce944b5e059d7a6 : SUCCESS in 1h 10m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9b26afb1d534420921882558d2b56de : TIMED_OUT in 2h 04m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ff023d50a70f97da7d6caebec6f8273a19223c2d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-12 19:33:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d672de66c032ce9bd6e8769381529a77dfca9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 21:16:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef54167e905a4010b3ca0d8d3d8a4936\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9185d76ffc24a2bbbc914eb1d6f5f20 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d8debf2c654ea0b133ed89275f90cb : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1862793d91494f33bfe93ea07f82073b : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/859e191b70ec4411aac53b01edb1d6b9 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcf91b91d84d4b508326e9adbf40a42f : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9a875969c254601846cbbac06b598b5 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbf6633cbf6341ac9de5221cde191234 : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/739d3429ddaf4bec89dd62dc02c7971f : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/42c57e80c51a438ea12a458988b7e76a : SUCCESS in 12m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/122b460590ac40cb97028dfb3d38b6a7 : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/14c933835e9e42f2a713075877de3414 : SUCCESS in 12m 26s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/cac84190b88a4f7a896844ef92347bd7 : SUCCESS in 6m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9719d11169ed4eab8a771c2b7da29182 : SUCCESS in 5m 19s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3c0fa479de4e4788a5c4d323c0699bee : SUCCESS in 4m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/86637ab3258842c6bb9ec4356df9152f : SUCCESS in 11m 22s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f5dada921e2d41a29e5ca20b512edc4b : SUCCESS in 4m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e054f1a9ad5e4c82b570fbf5b76464d5 : SUCCESS in 5m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f924821a0c5544d8920c645e19c57e02 : SUCCESS in 6m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0d90f2102c944b1e9847d0ab54dee7eb : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6f466973cb7436aa3b9ef11c8a8017c : SUCCESS in 7m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d9908a28f0374327a160d15b71bdd7d1 : SUCCESS in 7m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/86e199810bed4828b5f41bdc009ca753 : SUCCESS in 10m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e4b66dcc6df349f682d40ac3bf333fa5 : SUCCESS in 4m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/98f0cb0042e3427e85e6351988ff076d : SUCCESS in 7m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c83ea73631924c0d907ba05702b18f16 : SUCCESS in 5m 34s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0922c1068344438da2fb2baef1ae013b : SUCCESS in 10m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9b6aa2c69ad74d6a9493dbd40e53ba3c : SUCCESS in 21m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/00250ab2c692452797bfed428e2d2f2b : SUCCESS in 11m 35s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c2658a6d57744544a6c36064f829e084 : SUCCESS in 12m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fdada7d7ac51465c95fcd435cf743945 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08540e86b6b8425a9ee5a40db072b0af : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3449b57621cd4b1d9c3a4b482619c8a0 : SUCCESS in 24m 10s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e1e522a70c644d7997ff0e23a604ffae : SUCCESS in 5m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bc60f42a9f254ea88e23655c2fd8f4d0 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64d81a28ec614a79aa0883581812aab9 : SUCCESS in 6m 45s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e136c5956c0f4ac29bec84d970076813 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7b6ce0b41184952bdcabbfd344370cb : SUCCESS in 51m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfbdaba927de485b972a062e2c5c71e1 : SUCCESS in 1h 36m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2e68f6360a3aed216e58c650f99471aa447ab17a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 21:16:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6e2ce7e4093100a7cd45aef00c6f3efd5502d1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 22:37:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb05e77969e43a88b6f3a9acdf65dbb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f5727cf58a4ca3ab354f53c6c9b150 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c838ef190484184b33bd6637bd19ad1 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3e533777b5d4c2cb3351111d8921fa0 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2542b57948094448b79f7361f1a90ee7 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05e69134a3f54d939ff5b1a9a9bb7dfd : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d88f9afa4c8447e3b4da7a26cc06ef7d : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4279f3a2251f46c6aaba8ff96af38f20 : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/724bb40f3bd242679ab8fab76bb9d25d : SUCCESS in 8m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6a7ba097668f45c1bd3068381278c50f : SUCCESS in 14m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b6fbd6edc8b343a19dceb9814666bd39 : SUCCESS in 8m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/79c484e79e4b4cc3ac1685fb4a5cd340 : SUCCESS in 11m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/84ef04cd34b34b43a0a8c1b36c182588 : SUCCESS in 4m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cbe074f9bef54538ab41457506180903 : SUCCESS in 5m 30s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8e96448c3d5c42a581ca25f076e4d10d : SUCCESS in 8m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ad025e6054304f26b8cbe7ab1cc0fb5a : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46d81d8d55144b4db324c625571cf4d4 : SUCCESS in 10m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c069f5bb79e646e3865fc1fd90930c48 : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d7e812a8d8b64880a155455bddfd9f51 : SUCCESS in 8m 47s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/58b54fb55e0348d5b3037c84e23a4dba : SUCCESS in 3m 39s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e2e4b4e70f39407280014ca6338a0a78 : SUCCESS in 10m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/65af09d8180d41828e0976f5fbb93759 : SUCCESS in 4m 42s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2b6ef273fb8c48efb34e852a20beba7e : SUCCESS in 26m 53s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/70c7705ab0284c65ad0fc1d2897bd41e : SUCCESS in 13m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2d8c2c31954f4734be58cf646b140bc4 : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8c647207f7be47c991c8d2da21002cb6 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5f28e85ca3f4875bde18d51a5bca65e : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3af941dc11d740ceac434d59f18aa519 : SUCCESS in 22m 38s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/157cf76b41b3453bbc92493406cf01ca : SUCCESS in 4m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d189baaabdf34f83be91c2435f920dec : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2650fae22fef4fbbbdfb76ac8b32cf15 : SUCCESS in 7m 39s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/94339ce943664e39ac66ce8a95c58e6f : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f465e29c456740e1b08b7353cbb3a227 : SUCCESS in 49m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd7dfa483c3649aabea9d5a144ae30c1 : SUCCESS in 1h 18m 45s","accounts_in_message":[],"_revision_number":1},{"id":"854d400677a485716fb4f29c07379111238f5c66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 22:37:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c442073142533089cf0c379dd79448882b37833","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 22:38:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4130a76724fc42398b1a0a8e7305f4d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7684e978281e4d42b345c2a9e011e04e : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"665481bbadd6b586a593d93e9bc223b369427923","revisions":{"665481bbadd6b586a593d93e9bc223b369427923":{"kind":"REWORK","_number":1,"created":"2023-07-11 17:06:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/888145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/888145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/888145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/888145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/888145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/888145/1"}}},"commit":{"parents":[{"commit":"e729367f3547fb5754109157ee6b831db81d9ba2","subject":"Merge \"Raise cap for XStatic-JQuery in ``global-requirements.txt``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e729367f3547fb5754109157ee6b831db81d9ba2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-11 17:06:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-11 17:06:55.000000000","tz":0},"subject":"update constraint for horizon to new release 23.2.0","message":"update constraint for horizon to new release 23.2.0\n\nmeta: version: 23.2.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I5ce388ff43e4ee951e56a2461cc1a959dde64426\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2a52c853b2f90d92d954bd99b914eccbbd0ae91c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/665481bbadd6b586a593d93e9bc223b369427923"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/665481bbadd6b586a593d93e9bc223b369427923"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
