)]}'
{"id":"openstack%2Frequirements~888482","triplet_id":"openstack%2Frequirements~master~Ifdd4cfc1d18f86e9d59e0e1271256818b97bc8c8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifdd4cfc1d18f86e9d59e0e1271256818b97bc8c8","subject":"update constraint for python-manilaclient to new release 4.5.0","status":"MERGED","created":"2023-07-13 15:45:43.000000000","updated":"2023-07-13 20:05:14.000000000","submitted":"2023-07-13 20:04:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888482-new-release","meta_rev_id":"a325b83723242e9a96ff4593a9a7468d4a62d205","_number":888482,"virtual_id_number":888482,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-13 20:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-13 17:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-07-13 17:39:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-13 17:25:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-13 17:39:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4e24c665df8861de6c314e5713f90d696b4d9008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-13 15:45:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c458b742c26868f7aaad37c43062b9dfc9b0d035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 17:25:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28292a45b4b943f582234dc338818423\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62817157f631436eb65253a77a57bf20 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eded310a8a94348b11fd686da647c0a : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2805c20944d947c28a5f8323052ae115 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/223bc261dffb4e099b2d63ed71edbba0 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87781ca958b548b09bdb424d433f72a2 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a130ec437f843a4904b94074624bcbe : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8f7e109b5204f019ae2901ea45abe7a : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/688878bdcaf547d98088c7db15ed98bf : SUCCESS in 9m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0dfc6dc44e8a43c8ad5f4b2bda66a8b4 : SUCCESS in 12m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c331a9dba7934801a1aae80ebc17d2c3 : SUCCESS in 9m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/67d3e47df0d5467b87feb08f8e8e6bc5 : SUCCESS in 13m 50s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/85e9e2a53a62438487ce13068e17ad5e : SUCCESS in 8m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e83545ded9de49ceb93604a77402a34f : SUCCESS in 4m 49s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/aa122d6ef0eb4681a140e04f710eb088 : SUCCESS in 4m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1b04a0d3269947d9b278cae2c7b87d63 : SUCCESS in 11m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c0efd35bee9c4f13971924a13e0cc94d : SUCCESS in 5m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b6344a6a1ce34114a31af008ecbd570d : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1197f5e3da8a47cf81a423352fb5f7d7 : SUCCESS in 8m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/238631443c6b46818422fea61aa902ed : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58e09162127e4897b926b28d3da1f768 : SUCCESS in 7m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/00a84566f9d746e58efa5aa5ecb695d0 : SUCCESS in 7m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/87f780708d3b4f7ea6704b3e12691f76 : SUCCESS in 12m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/abd4f2b66a794680b9e73bac60727497 : SUCCESS in 4m 04s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8909f411bb334c8391c5e2a89b31eba7 : SUCCESS in 7m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/05f3ecf01adb4c91845b5d0efd411d94 : SUCCESS in 5m 27s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b2af365e39404f828a2fc9ffcaccfd1b : SUCCESS in 11m 30s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b37aa88a817c4302a65104f5e70ea910 : SUCCESS in 28m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7e2ef41e8a654f1096b82eb8da3b5a14 : SUCCESS in 11m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1896a5384b93464c8195dc79bc437ea2 : SUCCESS in 15m 07s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/be81806e8d694bb6a72eb30d8cf9c382 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3699eaea4cf94606a9b122eebb41b572 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/afa5f7fb91134440afc1a8f6267ddfc6 : SUCCESS in 26m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2d1b216a1d584a55a536214830432726 : SUCCESS in 6m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/51d072aee60047eeb32e82b842ad6ab3 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d6e2650fca1b4c11a3a8634663cf7501 : SUCCESS in 6m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/764a2ca81b3a4eb7ac0500f3a9f7c89b : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0cfafe8052a487b927ad2663cd62be6 : SUCCESS in 1h 04m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41566ef2a83b403a858410e243e6d643 : SUCCESS in 1h 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7a52ec0f4a1c49db5f4c566a4b2ace8df13faba4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-13 17:39:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19210da541261ec6d271822ef53522837785304c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 17:39:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"926a8aba3f310d54cd4899cbafd58060aa26d371","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 20:04:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3689a9f006f84596a183c6651796f8d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e580c9fd004c84a34e07dbf511b40d : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61941de5683047799ff1a870e39cf2f8 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3553f0a92e774b23858dc472615abef5 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ae3a4b6868b447ba852cee4e0021708 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e030ee9020b4a99bbf149a9ad63b6e6 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74ffe53a273846baa7662e3309e03054 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7081aa4f89784660be7fb7818bf8d9a0 : SUCCESS in 4m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fae5ec4aa33449168fb672072e7842a6 : SUCCESS in 9m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3c3960ddcbad468db75df97a07ee5448 : SUCCESS in 14m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/31b4ffb6706d49dbb1e435fa5b7f0aa4 : SUCCESS in 10m 21s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/de5b592ea89b4e6d8f8144f681bbf820 : SUCCESS in 10m 13s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a6d2d2abd67046d4a86abd147254f65e : SUCCESS in 4m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7dad521ce521421a9286bc22461980c1 : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/406a967fa5804b78be04295d955049f4 : SUCCESS in 9m 07s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/416904ca6cee43b2b53330e235455c17 : SUCCESS in 8m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af974b8498fe4352b4b088099ba172b2 : SUCCESS in 7m 37s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a025d11569dc4fdf9d356bb21d10c04e : SUCCESS in 6m 47s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b1e254ac1e024f578448d7a1ad12e2d5 : SUCCESS in 9m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4411873f424946fdbc06732f05678c07 : SUCCESS in 3m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9551ae6dfa21455880a03caf00484f05 : SUCCESS in 8m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/afda66ec2ff542218525ac36145811b6 : SUCCESS in 5m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6c306aa5b10544688e0776ed29e598bf : SUCCESS in 23m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6aef7a2dfe624f69b990fe1f130163e5 : SUCCESS in 10m 06s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/fbfa625e4de14e418ef03aeafd014d28 : SUCCESS in 11m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a109a78105c14febb8687c5124904dba : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8bd5bf8cfb5844a98fbc43caece9b08c : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a273a1b1a9e2451c9f486ec5d4e1b848 : SUCCESS in 23m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/48eb31bb32cc4bd5b4cf5763c89d22fb : SUCCESS in 5m 44s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1590e26b1e55498d9fbee6141bc4c56a : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6900cbc626b44afa994af2cde6dd6cd5 : SUCCESS in 6m 34s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6d7816200c524be2aeaf59233a926a6f : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b37a1e9bc49c4ec680790a50b5aeb8a8 : SUCCESS in 1h 09m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50109b8e59374e6781e591d0730065f3 : SUCCESS in 1h 27m 20s","accounts_in_message":[],"_revision_number":1},{"id":"12be0d39924c6c84fb5a57e87859cd566da75da1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 20:04:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a325b83723242e9a96ff4593a9a7468d4a62d205","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 20:05:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae21caa84dc424a808c39eb1ab8a685\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7a47438d4241b68db3369e1499a13f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"174533432ab6625df112ceb1b04cbc671c2dba83","revisions":{"174533432ab6625df112ceb1b04cbc671c2dba83":{"kind":"REWORK","_number":1,"created":"2023-07-13 15:45:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/888482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/888482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/888482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/888482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/888482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/888482/1"}}},"commit":{"parents":[{"commit":"665481bbadd6b586a593d93e9bc223b369427923","subject":"update constraint for horizon to new release 23.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/665481bbadd6b586a593d93e9bc223b369427923"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-13 15:45:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-13 15:45:39.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 4.5.0","message":"update constraint for python-manilaclient to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: Id7026ec2d2756d7edb830091eb332ea0ffafcb5e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ifdd4cfc1d18f86e9d59e0e1271256818b97bc8c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/174533432ab6625df112ceb1b04cbc671c2dba83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/174533432ab6625df112ceb1b04cbc671c2dba83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
