)]}'
{"id":"openstack%2Frequirements~888484","triplet_id":"openstack%2Frequirements~master~I4a8e2390c0e784893cae5900eb30c7b3ee31f5f2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4a8e2390c0e784893cae5900eb30c7b3ee31f5f2","subject":"update constraint for osprofiler to new release 4.1.0","status":"MERGED","created":"2023-07-13 15:53:15.000000000","updated":"2023-07-13 20:06:03.000000000","submitted":"2023-07-13 20:04:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888484-new-release","meta_rev_id":"ef36316584fc326f6b3c4a48cd9dfc8095303bb2","_number":888484,"virtual_id_number":888484,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-13 20:04:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-13 17:41:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-07-13 17:41:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-13 17:38:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-13 17:41:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ba7772146a88d19daa69d2883ff3fb30d979e82d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-13 15:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36bd539285ffc5e20814e2b2093a16dd806621b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 17:38:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cbc33d023224cf7b88dfa8ba9786a28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adeaf737c3844fb88077c14e0f95e31e : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/140c96a465204d6d875d386f23240244 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1341db2fb01d48d5800d963fae4354b4 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12d84b066053409f9e4e019063d4a0e6 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4591e8d2c12143a29ae58f3228fe0dca : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f9ee67808154166a67011227b4884ce : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9707796255fb45e196a353fa8c8e7e5e : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/54c2f3e8e3c14996acf4f00bcb244571 : SUCCESS in 9m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/666456927913411c95267e8b51baec04 : SUCCESS in 15m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/29266e8591ad4d20b373191c4ecda24c : SUCCESS in 9m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7de6f8fb264e4d16a0d8ec9ed32f7633 : SUCCESS in 12m 06s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/82e86d5d8add4153b5218de79f48d307 : SUCCESS in 7m 53s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/57c6b4817b914a188f6b509c4d90ea28 : SUCCESS in 5m 41s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/717c7251eb4142e7aa3c6a9f4f5f33e0 : SUCCESS in 5m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a38ab806793649a1b8ce62af6c68f7b1 : SUCCESS in 10m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7921a6b7357941b8b2596ebcfe53061e : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e381a2eaf0a34b1590f99779c7709e2f : SUCCESS in 4m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/87bff1d7ef0f423ea16b9cd2d631adcd : SUCCESS in 7m 59s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/50a42f9951d043c495abd63a859ec533 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/638bc225ef574d2793ec49d34a1f8ce7 : SUCCESS in 8m 21s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/00c851225fbf4f48ba30e8814dca10e8 : SUCCESS in 8m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8d24febcde3e426e93533dc2354c2427 : SUCCESS in 13m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c12b675645eb487b8d9a03405bd663c2 : SUCCESS in 5m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/94851b8da9a746f6bae773c56499daf8 : SUCCESS in 8m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c7c5f18a9cd64c2fa14c42f3ae02164f : SUCCESS in 5m 09s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/77bf6cbce8d145f39c084d6d4a8aedfd : SUCCESS in 11m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/85b2c61c68764c1c8fb3703e1564bb2a : SUCCESS in 28m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bb5bc6f2ae6545dd85d86f59c4d21bf2 : SUCCESS in 13m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d6a52803795e4f4f9b904f5811c6ed60 : SUCCESS in 13m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/36d700f76364438d8beabb0dcd8464e3 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e30b639659954e47801118fe57d9593d : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7dce84886174c089b416431843302b4 : SUCCESS in 28m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fa90d960294e4dba99010dfd2232e53d : SUCCESS in 5m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6480d76431214c0e8efe0199c8f4bc5a : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a5bad273538450a8314f9a4686a9a3c : SUCCESS in 6m 07s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a032cf5377c94e40b6de974ba09e3772 : SUCCESS in 14m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/838cc399a23c4ceea442f21f67d3d97f : SUCCESS in 55m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba8ec9e1c0eb4b0dbfd54dc800af833a : SUCCESS in 1h 30m 11s","accounts_in_message":[],"_revision_number":1},{"id":"41d2eb294b228253cfcf59d608c06d56cead199e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-13 17:41:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7cfaac524dbc482cd69f47f724fd0f287842a5b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 17:42:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb0b16b91e1e60ef130d9fa185acf5195e253cc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 20:04:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbb91a4b9bf44ea68de909ef2b337388\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972552bda16f4c548bd9c5d7bf1fad17 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ab6b80a3314510b3a83846dcff210c : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/243a18015b704f5b95ed4475d223846f : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27845bca111c4848b41c195628e86eac : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4801f9916c8641ab9f1ec60f731eb8c3 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8848278aebb47f58af5c20758711a2e : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1ffeaaeb111443badda714bf05dfd4b : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eed3c4e13b9a4d7ea6ede90ae9ef13d3 : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aa1ff95fd7784ada9847f39ea8113d4f : SUCCESS in 12m 35s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4039178d64ac4a2681d5c39e87f4f800 : SUCCESS in 7m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4ea39414fd6d4c878c18f06ee2cb5546 : SUCCESS in 9m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1595d90d6ba7425ea6c5f8e5c97479dc : SUCCESS in 3m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/961a9cdc3cf740c1a59941485153acfd : SUCCESS in 4m 57s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5141cfc1a74c48798142cb7b899b0a04 : SUCCESS in 8m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c582a50beb574dc3a3182e7d196c846c : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1812c5f651bc4380bb1bfede69a3d5be : SUCCESS in 7m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dd88e3444cca4ef29fcd3e3de0a8927b : SUCCESS in 8m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e54dea43dba1487da126ba1c4d25dbec : SUCCESS in 10m 46s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fc738f130dcf4687922fa43832f2ecc6 : SUCCESS in 3m 19s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5cd9ea4efd5145a496cbf53583564993 : SUCCESS in 9m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b16cc222c420422d859dfde63124afe9 : SUCCESS in 4m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b3f43b4e8c88455da3c9c29978f6e725 : SUCCESS in 21m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d5d61b61fcef422195b3f4e2c14d6abe : SUCCESS in 11m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0700f58cea0442328983c0831831e33a : SUCCESS in 12m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/14464f86486841c6a162e45faad4f76e : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ec55284477c4b74bec677d048583ba4 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4606f000daa2469b9139d5ffe44d8111 : SUCCESS in 22m 28s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e09696955cf04b25a10e81dbe48df6ae : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fd374abba6954d07ab649d4ae4e0533e : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7dc516421c70453bb3fe54235866208a : SUCCESS in 6m 15s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/bb3920d359af4c9a8079611d5dd58fca : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/973cb4828e9f4a838cbc2ea6d8eee1ff : SUCCESS in 52m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcc9ba3fae2d4d9892f52187acca6b13 : SUCCESS in 1h 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cd335bfe0691b0f202d9f4cfb4f8dd3b38b97694","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 20:04:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ef36316584fc326f6b3c4a48cd9dfc8095303bb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 20:06:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4c6332cb5e43069737ad3163b869dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df8aa7510ff41be89b30831e468cc6e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d641e5128446d8d283e239e7cc33fd78122ef43b","revisions":{"d641e5128446d8d283e239e7cc33fd78122ef43b":{"kind":"REWORK","_number":1,"created":"2023-07-13 15:53:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/888484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/888484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/888484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/888484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/888484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/888484/1"}}},"commit":{"parents":[{"commit":"665481bbadd6b586a593d93e9bc223b369427923","subject":"update constraint for horizon to new release 23.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/665481bbadd6b586a593d93e9bc223b369427923"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-13 15:53:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-13 15:53:12.000000000","tz":0},"subject":"update constraint for osprofiler to new release 4.1.0","message":"update constraint for osprofiler to new release 4.1.0\n\nmeta: version: 4.1.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I767dfa1c867c3bde3c1356fae2dfc797979bd50a\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I4a8e2390c0e784893cae5900eb30c7b3ee31f5f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d641e5128446d8d283e239e7cc33fd78122ef43b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d641e5128446d8d283e239e7cc33fd78122ef43b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
