)]}'
{"id":"openstack%2Frequirements~889702","triplet_id":"openstack%2Frequirements~master~I5596f70aa052a0f3718defbadf4d7b5b0997b4ac","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5596f70aa052a0f3718defbadf4d7b5b0997b4ac","subject":"update constraint for tooz to new release 4.2.0","status":"MERGED","created":"2023-07-26 12:07:21.000000000","updated":"2023-09-23 01:48:07.000000000","submitted":"2023-09-23 01:46:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889702-new-release","meta_rev_id":"dac5c215821eebc95241c2ea3cb4ebc44c45e206","_number":889702,"virtual_id_number":889702,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-23 01:46:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-09-22 17:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-09-22 17:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-26 13:31:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-28 23:01:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-09-08 07:55:48.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"CC"},{"updated":"2023-09-22 17:36:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"32777d96ade0fb81a1ef62099334ff707045fa21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-07-26 12:07:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4b9a0f331f09a43c796899e98b7fda29a79f1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 13:31:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e56ee2ffcfd47929e9214109a8a1b36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a6e02d79fc459d9e84937a6a092895 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086b43e7f6d241d4a98d7e45ea2ffb92 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/001e7773b91945e58db272d11a710e97 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa83f2662ee644969f9adbf2008613cd : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c8c8064bec34ee19b1c3851cecdc974 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/542e07c78ecf44a892a85ca962fe91db : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c88b85ba2230411da48989fa56db09ff : SUCCESS in 4m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee719f731d4845a3a9b0f6500a97bfd2 : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9c2ae56631474af4bd4c93c569261335 : SUCCESS in 10m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/20bb8dc42ccd41b7bc6ba9e1378fc791 : SUCCESS in 7m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/072d9beb25ff45c5bca6fecb71d69bf3 : SUCCESS in 10m 50s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2e3bb8434e6643a5b70d174aeddd4ecb : SUCCESS in 7m 38s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a96b7d2dcd21465bbd4c5089ccbb8c56 : SUCCESS in 5m 20s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/56d1917deba746f5b61bfaaeb01c8f13 : FAILURE in 2m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/db01630b77134ef9a3e6039c69834ca7 : SUCCESS in 12m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0fa4d6be0bfe42b5be81b627ef38348d : FAILURE in 2m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a8a701709dce409db953d2a8c4eba072 : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fb4b6c6bcc1a46ba88c77d94ab14490c : SUCCESS in 7m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b8d170b586984c39bbf1d97bc04380f4 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d9821424408b401798abd442142ea058 : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/44e87074ced84307b3be7b0ae73f8dd4 : FAILURE in 4m 07s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/0d0e2c18cee34092b38e3a0cbc879dd4 : SUCCESS in 12m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/77c9fc23567942aa9c2074b2918848a7 : SUCCESS in 4m 06s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e4884ddc22644ff08167d894f0951aea : SUCCESS in 10m 37s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d75bb7881288489ebbba1ba3ef56663b : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2ab59a90958046c8acd4830577e29ba6 : SUCCESS in 9m 38s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0656eedbf78f46e18a5575cc0055615a : FAILURE in 4m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a221c611b63342a197077b27a718d37f : SUCCESS in 9m 37s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5e9310eda8284f66b6df1f551f9b859e : SUCCESS in 13m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/86b2f01c390e471b973c367ac3370b1e : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bff472e7f0e647229e7e0664864f9aff : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b043fabf1684461b3afd2dbb5bf69e3 : SUCCESS in 21m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/541e159aa964434c89e7e2f7bedec219 : SUCCESS in 5m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c92dc32913fd4e68a47ad5ba036d8d56 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ce52d3160894d10a26f57e2ec7d1fa6 : SUCCESS in 7m 08s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c68519ca1dac471c84c0c5dd6b8a0648 : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5df8146ecdc249eab45129d2e4aac3f1 : SUCCESS in 46m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54d89880bb674c4a8f16ac1b8249d8ee : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5b11287c2a1f52cbc96b9562cf6ffce60387bbbc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-28 23:01:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"54906e52bac79446b3d16a81ad6a18943d9b4b73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-29 00:07:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad93e36460914f528f82b6152f3073c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da18bbb3028491ea96bf514f2384d91 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54849693f1cc4afe93dba7a6c7b510d3 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1e2445df930448f87c2742f8888c253 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54510a753a4c4c129b5edf8c335221cd : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b4e838f284e4ca8ab6e546a1fbc462a : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ad1baadc15a4a6294e7e7a7e72bba25 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a313f5cca15c46a59d624e2b6ebec283 : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83312d784ff443de8f3b49500c2d94fe : SUCCESS in 7m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b74e350b3e9143819306c68aad13c4ea : SUCCESS in 10m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1e66b90545134d508bd96fbddd4c5aed : SUCCESS in 7m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5deaaaa19a284a278e188d797ab60e4f : SUCCESS in 12m 09s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9fe264445fc44e498be05a493b237981 : SUCCESS in 6m 35s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2a80de7d0ae14e0bbbfabbbc5c589925 : SUCCESS in 4m 29s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6cefec3adfaf4237a99fd6731bccb524 : SUCCESS in 4m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7c64a3bee7ca4120b9def3540cc7854e : SUCCESS in 9m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1628019987d64ce78b5e9b6990ad5e82 : SUCCESS in 4m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/29197cb353584ac0a01da0997b7a701f : SUCCESS in 4m 37s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ba452e0e07ae4f12b98f573ca6c9d552 : SUCCESS in 6m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b66dc30845a740039c45033c563601ef : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a70497c6f42d4d3b9ce2fc93e4c16929 : SUCCESS in 7m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c72d70198a2f4ed1937aab967e40ec60 : SUCCESS in 8m 33s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ecabddce2828405c8fdf488a5fc57438 : SUCCESS in 11m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4e4411e7b39a4eb0baae4784a215beda : SUCCESS in 3m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/686784f3a7be40efa6b14e00b2ae4139 : SUCCESS in 12m 39s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a8aaae63925f4ebca16667a6e8d9ff1e : SUCCESS in 5m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1983025c20ff4faaa840e2611502d3c5 : SUCCESS in 8m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d87fbba6accc44e494a2a3c50a20c7f1 : SUCCESS in 28m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6266f335011040139918051c59fb35d0 : SUCCESS in 11m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/575a04fcd96a492f9487ab60407b33ee : SUCCESS in 10m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bac2525772ca479282a73aeb7fa12671 : SUCCESS in 4m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/87d747b9bd66469eb0ef8e50434b31d6 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e689567cb86448928c2736a3383c4bfc : SUCCESS in 24m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c66628f5831340e2ac95fdd31785d5f5 : SUCCESS in 4m 57s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f6809aba4c074d7095ea5145fcfe1f61 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa9b7d11058c4e039dd46c894efb42c3 : SUCCESS in 5m 52s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f7197f2b5fc84a7080f28ee49311d8ff : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc53a46a836d4e548e60efb8c2b30ea8 : SUCCESS in 1h 00m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/263a201f96f748a8a40ef134d33e468e : FAILURE in 50m 15s","accounts_in_message":[],"_revision_number":1},{"id":"47e522e6eba8134ef43fcb026eb3d57d32402f53","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-09-08 07:55:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af4b5fdd4f763d509ebb192373c3e9164fa32ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:11:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ebb3a4ff9644adbb3a5abfd690e2da0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d91a2bb94a0547208bdda648a320dc1a : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ebf89cfadb4cbc9dc3b9eb1a287752 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaebe849094f491c84d9df71d891c0b9 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eae27d599c5f49d3b34f26b93f463a30 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8411ad4c9a974daaabf83b16568af18b : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0d8e883e3ff404296d234458b32d17c : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87328d75cc2b4747907f7a16ea88738e : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/89861a147abb434ba2546250b82e9280 : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aa964afed1044ddf9f848b28c997d406 : SUCCESS in 7m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/77b50223fa344ea9a709072f973c8b43 : SUCCESS in 15m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/507c35675ec246269f274dc49264d3f9 : SUCCESS in 7m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4a9dcd86e0d14f6ba9991f626bf8b301 : SUCCESS in 13m 14s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/73fdf3e3a08a4b0bbc0d24326950a5eb : SUCCESS in 7m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/3103770078154cab949b5c516d91f952 : SUCCESS in 4m 43s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4904c7a1310146fa93aea3af36f92b69 : SUCCESS in 5m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f23d852aa47d44b492413f90c00f570f : SUCCESS in 11m 43s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ee1675b38ba641fb81e9ea4e242d70de : SUCCESS in 4m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bfe36a5273dc4713b2d881170c0db28e : SUCCESS in 6m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/24c466864ec74f278b902c018c57b2d2 : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/440e65955c1643ed896d6df8e7a9f4d9 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/361f9182e5e240f28a34ad51277d7a71 : SUCCESS in 7m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b16360de5c8a45368dadf8efa09d5d22 : SUCCESS in 8m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6f90e251414a44a08812131b647c332f : SUCCESS in 9m 32s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dc3cc9cbeb874d379d806c02cad02e83 : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7da7f69eb6914a09a5f65dcf54029abd : SUCCESS in 9m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/458a4773484f46c7bf0f69b05e55713e : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/8ac86910b8d84b55bf1a3e2d65f84935 : SUCCESS in 12m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/0bb77fa57e8b4de68ddf50870b2dd97d : SUCCESS in 23m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f62f600c69284586ae0c01bb53c96547 : SUCCESS in 11m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/56eaa96860e449c28df35260e489a007 : SUCCESS in 11m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cbb9de720a364e4eb25d573f0805b261 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c99e566a6f7f46e6b0f75c3c7a9c73e4 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a809431cb484988992304687831f814 : SUCCESS in 22m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0e3a771589ad4332b3cd910502bd9328 : SUCCESS in 4m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a354ba353e0b40d18c325ce1579a7c3e : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/522acd80a2e24fe8999eabfcd8d644a9 : SUCCESS in 6m 24s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5b0418d2a42e4c56b49fb914cb2f99a7 : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5df415b5aca45e587e345d7c634b9c5 : SUCCESS in 48m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ca1ac5d281942ec97427eaad5aea35a : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3e28121d21cab82fdae0b37643e9684b253c5032","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 17:36:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34116817af21ad2acd95ecbdb601ec1228a87813","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 17:36:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8078258a15b9ddd0b80df1730d60679261a7e61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 18:47:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02f7abe807ee4a4d85e7c494dd8b7b95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5b4c6ee39264e9e984dc7b42d24e95c : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5babeb0aef404a1aa28bcb9521a069ff : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81459d95b27744678a9272e38237e481 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1e0c824be204c5080d7a377021c746b : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/418d874c327d4188b0b146dd1a2dcd1c : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/44ff232592e34391832510025e471bf8 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/014721ec8bed4ecfb4784568a0aaad57 : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/98b002d3c64c466783f956589e899a19 : SUCCESS in 9m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a7a1e3a2be674c9694c52e52db85f423 : SUCCESS in 15m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5fe30fc3b3d346dcbcd75205383a5995 : SUCCESS in 8m 08s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4accfbd500444c3f95c10b2808c41df4 : SUCCESS in 11m 11s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f15809edfac148b1ad10779be983220e : SUCCESS in 5m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/14b8301251364fc9b243b8758fcf0ce4 : SUCCESS in 6m 40s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/521fc98ea9d14e688ceb3f62db561f2c : SUCCESS in 8m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e0774d92a0eb41818b54e41a19edf133 : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88f0f2e0d3de4516a451536e62138c3f : SUCCESS in 7m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a3dc51ef48b54f85a3094b6fa5e160b7 : SUCCESS in 8m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/85483abc5a8e4934800fa86f4f3336a6 : SUCCESS in 11m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/448841f43c58440f9c8d6fb6a21e8bab : SUCCESS in 3m 52s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9d08c4b5f38d4816ac5430fd74939292 : SUCCESS in 13m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2a93096818d2407688e47d8fa68ea923 : POST_FAILURE in 5m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9405707d4387497f9c4fedfcae94816d : SUCCESS in 20m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bb8383ee7bcb4d819b36ee91e01659b6 : SUCCESS in 10m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/54571863f5894836ac68868c5c461b45 : SUCCESS in 14m 28s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0dcc5c5f946c4bde857cc751043a0321 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/abb545ab6ed145e4ae3c8f0b422ac1ec : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70e1d2b66a6c40b493c6a3f6b563852e : SUCCESS in 19m 58s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/126ff378f81240658414e24a44a805e8 : SUCCESS in 7m 46s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3e7ef94d3bd84532b48b63e2b7e5e4df : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46a4a0694eb24f498200088debaa87e0 : SUCCESS in 7m 30s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ff793e90a3b64a3bb9846e644eca3bcd : SUCCESS in 13m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e70890de7e147569e8eae2accc31327 : SUCCESS in 1h 04m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae5b5a63ef0c4798a18ea0bc711d0db9 : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":1},{"id":"70e4bb4528854c329ee7709f5269cd7ad92251d0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-22 22:34:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0bd85af4d48e85b43eddb39cd112c5b935c7880a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 00:24:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ee7cf130e1e484fa3432a5b7dcd8a63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83b99fb03114ed882d017870dc8d670 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3e68b4441e64a2f8c685f25c5dcdc3b : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf6d88de51114059afb753e7ee13c337 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87bff7b8a9ad4af38182f31833e8fdf2 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a210535dd610492ea2687503e134025c : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b00927faad58478f9ae2452c7c4428df : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e6dbaee0c5c49879217f9605fb9dbca : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf0f4fc396514d1aa6d19e2c53bf7c47 : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e65e78aba62145aa9238cf24aaabab79 : SUCCESS in 6m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d7912c48d43b42e093d1913dfcea5d57 : SUCCESS in 11m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cd7cb0c451c349d0b205af03bcb8154f : SUCCESS in 9m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/420ae87ed7aa4f86b73d35ded4a3284a : SUCCESS in 12m 18s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/858da26ca3ab45bda83261bc04480a45 : SUCCESS in 8m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/491dfd1cd7284baca08425be810de927 : SUCCESS in 7m 48s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/7469858b76124d13a6cbf02735189720 : SUCCESS in 4m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7c98ab6c9c9a4b6eba3cfc1eaaaa32df : SUCCESS in 9m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f37e383cdc81401c8615a975884b4711 : SUCCESS in 5m 07s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/857c0074e0fb4883b6dd444ef8fe6bea : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/961cbd22f6c049eeab127e4c7d58b634 : SUCCESS in 8m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3540bb74e4134c2299b727746cdca559 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88f9b7d58603421a8e90ec0753e30116 : SUCCESS in 8m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ce019877094c4aceaaacca00b6312d22 : SUCCESS in 8m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/21b059634de7495e91dcdc89c4e13b39 : SUCCESS in 13m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/12efefc73ec343ef8221714cba4b96d8 : SUCCESS in 4m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e72c4ca22602407382de948396088d31 : SUCCESS in 9m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4885e2d298474fb19f26a4cefb02250f : SUCCESS in 4m 47s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4880469b792049f1a98ae1a3e84abaa9 : SUCCESS in 9m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/469bd7fc71784984b5c328242372016a : SUCCESS in 20m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3ba233bc49a0425884dac0af1a8c7cb7 : SUCCESS in 9m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/340109fa595e4c62a8597616cf4e0957 : SUCCESS in 10m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ce82d4cfd70a4df5be1f07a840a29f14 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/463e54f9367c4c9d8f272a06192c639f : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94e76fdf4d934baeae0389844219facb : SUCCESS in 25m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b22a1aab13c947b69c32429be1c59212 : SUCCESS in 5m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b1cbc3216d1e45649a23f51aa4471de2 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb40687dc34e48689e9365509a9055b3 : SUCCESS in 5m 24s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ca0efa2f65bb483c825a8c8dc9ae9f1e : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/071c1213902941fca21a0c868bcc968c : SUCCESS in 42m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4ceaeb4009745b592d60ffc2b5d7782 : SUCCESS in 1h 41m 37s","accounts_in_message":[],"_revision_number":1},{"id":"77720a16ebfd6fddf9b452573603e2b750455743","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 00:24:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e86cc9cb3cd6877097f5d6e5e71b5d54f16a40ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 01:46:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2798f93a4b2e43d28c13f4506730875d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c203ee8a813f46599269734c43e46996 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5612a93fe9b4a6fafa4ea4cda3fee37 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7da3a2a0353a485b966a5cb06984e442 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b4cfc871b2a4fb4b7fa1cc1702fd379 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56460e1848ed4ba8baa613ca0c79d1b8 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0bef99214eb416d9a3643f0f6f3b27a : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9ae6206a9d34d1da31f5b8c0a0a9456 : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e94cbc86635c45cf9e4578a4eec198fd : SUCCESS in 8m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/efd84ff0d66c42bebbbff5bd1b9db5b8 : SUCCESS in 12m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7a67e3a61d274f0c8c743bf35c84d053 : SUCCESS in 9m 33s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/eacba21bb3b849d9867d11f16e80b7fa : SUCCESS in 10m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e9174745856942e6843e2d79c989c200 : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/9c71f1989569492ba013d89e319171d0 : SUCCESS in 4m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/273b81cd1a584cf6b1fa732107761006 : SUCCESS in 8m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/791524b3821e4ff080cd096bbeafa222 : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/826e90dd429b42e8bf8a466acab35d0a : SUCCESS in 7m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a3e3e25dcc794b758fb83f696c059c78 : SUCCESS in 8m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/630367eb0689411b93868824475cbb55 : SUCCESS in 10m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/897cd83151304817964adb013a5fb05a : SUCCESS in 3m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f1e5b901a87540408496729c20c4daaa : SUCCESS in 7m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0be0af17550a435792831f9d3250f4b5 : SUCCESS in 4m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7e87e947ebe746b1a12c10758bfba026 : SUCCESS in 19m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d5afb5aa5c2f44f2bff1e9b2770ce4e7 : SUCCESS in 10m 18s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/66234d5c8d96497f878d931de155b624 : SUCCESS in 14m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/67640f7ed7a54affbc94f8b22e590f00 : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eb19091fc28a4b6099366ff7702066eb : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/225d61f250a74569b2b9c608e98d0bfa : SUCCESS in 18m 22s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3fa176c514794ab9a0f912971a23c61f : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/598efb40e9694c0886945817ea553ceb : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/958fa95e28c5434f9fb0fdf69255b80d : SUCCESS in 6m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/c030ee12bdf44e2a960c641c5b93061a : SUCCESS in 14m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bc940151bd84530935d181b9fc920bb : SUCCESS in 41m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/431ff3bd10c24ac8a7e9f33a78af2e61 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a9c7347f1c08c6952325879bd8157797fcebba69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 01:46:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dac5c215821eebc95241c2ea3cb4ebc44c45e206","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 01:48:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a2cec74551948dba484510da3ea2101\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f25f6f724454614afc4e2b5eeffa6ba : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42fbeb8b5118215d1d9e4b72a6d9cd8805078582","revisions":{"42fbeb8b5118215d1d9e4b72a6d9cd8805078582":{"kind":"REWORK","_number":1,"created":"2023-07-26 12:07:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/889702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/889702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/889702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/889702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/889702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/889702/1"}}},"commit":{"parents":[{"commit":"875bd7fc31e874567cb9b2db037068c9bf508d4c","subject":"Merge \"Bump XStatic-jQuery version to 3.3.1.1 in upper-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/875bd7fc31e874567cb9b2db037068c9bf508d4c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-26 12:07:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-07-26 12:07:20.000000000","tz":0},"subject":"update constraint for tooz to new release 4.2.0","message":"update constraint for tooz to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: ricolin \u003crlin@vexxhost.com\u003e\nmeta: release:Commit: ricolin \u003crlin@vexxhost.com\u003e\nmeta: release:Change-Id: I844094fba98071e0153dd3e4e79e780a2d37aefd\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I5596f70aa052a0f3718defbadf4d7b5b0997b4ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42fbeb8b5118215d1d9e4b72a6d9cd8805078582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42fbeb8b5118215d1d9e4b72a6d9cd8805078582"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
