)]}'
{"id":"openstack%2Frequirements~889808","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I705ace770c74439c04a34eee10b3d1450b55d79c","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release-stable/2023.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-07-28 22:10:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I705ace770c74439c04a34eee10b3d1450b55d79c","subject":"update constraint for oslo.log to new release 5.2.0","status":"MERGED","created":"2023-07-28 09:21:41.000000000","updated":"2023-07-28 22:10:19.000000000","submitted":"2023-07-28 22:10:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889808-new-release-stable/2023.1","cherry_pick_of_change":878130,"cherry_pick_of_patch_set":1,"meta_rev_id":"bfa00657d2ce87988b50e4f49c04483c4db4301f","_number":889808,"virtual_id_number":889808,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-28 22:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-28 19:07:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-07-28 19:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-28 09:21:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-28 09:36:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-28 10:43:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-28 19:07:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bb07cc0a10fc7ccb81349bd2d9fe9791509a6a7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-28 09:21:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9058bfe09adc4a7d1523787a9286d16c62b23880","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 09:36:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5282ebda2894d76bc7b175e321f1d3c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4c8e2af94fdd4169a1c9fa4e3231f78b : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7c95d66aabb1d0149caf97cb43883beb070a6d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 10:43:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b129c0714e46478bf415794e91185e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd6a29d48fd4282b2aabe3552ac8dbc : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bc66a453b3f4449b7bb66b32eb132ef : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0f90edd95214c70a8bc955286ec4412 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de04e1e20d49408fb6c39dc8891e9546 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0f1b45f89598413d8e59cbb37d4cff25 : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da1342eadadb424ba2d14c36c5fa5173 : SUCCESS in 8m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fac8322eb2fc408c96a7d57d6e71e20c : SUCCESS in 11m 30s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fcd7b217eb7b402cae54c08c614cd532 : SUCCESS in 7m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/94f449ea7e104c1bbd16abce38112f6b : SUCCESS in 6m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/87a9856109564364996a2c12f89a977c : SUCCESS in 6m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cb28d033cfa94d0192c1b1bbb906ead0 : SUCCESS in 13m 26s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/848509e620f24b6387373a947063c924 : SUCCESS in 4m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a2c06a98ea80429190398ba006fb152d : SUCCESS in 5m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/777dd030159c4b2f86aff3af501d86b2 : SUCCESS in 7m 51s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8b711177d46d4ab2b1c8872e9dcf244d : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa63e9eeac504d94af65d16a28cc0104 : SUCCESS in 8m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a89037bcb9644ecb9a3763534de25a93 : SUCCESS in 10m 14s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f55e03a8108c4c058b7f647702535957 : SUCCESS in 12m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/39317b7666754a1f862706cca27bfcaf : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/567d6879a3654a7ebca938752141ada0 : SUCCESS in 15m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1e050660568942b9879024212e872dfb : SUCCESS in 5m 43s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/085dc238e0684dfbb357b9824ef2a9d4 : SUCCESS in 11m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5a5e16e2e5054d5286ff40e01c4843e4 : SUCCESS in 25m 11s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/6f0db8bdc7504474b37b41b07ef94943 : SUCCESS in 14m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/381779f277714a0587493ee74de68ae5 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/390f63c0383746e09f95c55113568786 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/102867beb9ce4b699321b09cc0440578 : SUCCESS in 28m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/034b90dd5e0b4c58876c0262a5d939eb : SUCCESS in 6m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1a11a4d0d33e42a8913747dde1625b2d : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8608b5d912c4fb783f05ded5e96c031 : SUCCESS in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/45284ef39007467f8276ca48d5370a5d : SUCCESS in 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc034e47dbbe44b89df1fcabcde00bf4 : SUCCESS in 51m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5df3649c46f140dc92f0a08c699978cb : SUCCESS in 1h 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"77e8e616fdcf40207a9b2c4c2ab25af7478d4650","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-28 19:07:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ace217a310d2f4276ef22b68cc9e5dc1c2d16742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 19:08:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b00fa9da1ce3c8be1cca67cd59ce35d6fffeb69b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 22:10:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11989c0ec78a419da24f7c3e19eb2fb8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926a8b7cd39644f6abdc78867d1ea7dc : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79f02358e7cf4b65a65ae793e68f96d7 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6762888f793840909cc3b1d8c22becf5 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05d3fffbdcd14fa5a19a83c07bd0e55e : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53ad1f25f1744026be54c5fb8cd8c780 : SUCCESS in 4m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aa96c4e5c13e4f28b8737b9b4ecea167 : SUCCESS in 8m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9cb9fb0292e24f2ca9bcedf7845d29bc : SUCCESS in 12m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/03bd68dcc057480a8425d64a5f2cf401 : SUCCESS in 11m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/602b891087f14ad79da59b41e65bae01 : SUCCESS in 6m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/131e6ad9ffd147ac8195244cee56eee7 : SUCCESS in 5m 15s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ca9fc78783474959a15c093802cae8e8 : SUCCESS in 9m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e1cffb08d4214b9bb1ae77723ba946de : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3d942160583499d8dbe5167868b1a4c : SUCCESS in 8m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/05dafc48744e43eb96110ce848043298 : SUCCESS in 8m 26s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/443836e2748c41fe96ed3fa819915cf6 : SUCCESS in 9m 59s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9df2867f01b64a4f87a8eed708a3976e : SUCCESS in 3m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f1c0595f309448e4a3479c5fe46334c7 : SUCCESS in 10m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b62baaba33814f3a8d350d0f07a263fd : SUCCESS in 4m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/199f27c4f4f14318b82ffd6096c0d8c0 : SUCCESS in 23m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/90431468059a4f9fbc7b42f5f8191a45 : SUCCESS in 13m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/835c4a3fc9264a668dc698decdfaf998 : SUCCESS in 6m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0cfe51b2b08f4f72b29f303a54f77adf : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc43943aacde4ad7b5cc57cde2ab8db3 : SUCCESS in 20m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e34fbb07c76445f69cacb52e458e5dc4 : SUCCESS in 7m 08s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ed177ed4c9874b82a1294f3ce758ed9b : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a9fa7ccc5484f2c987a0ce3628e38af : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5582131ac7f74dd2b680ab71be8b3888 : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae1ad84267c84e55aacffeb2c6da736e : SUCCESS in 48m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2cf667932364451a382e8a375a61c9a : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"bfa00657d2ce87988b50e4f49c04483c4db4301f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 22:10:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7e44103553fbe81a31f980417d9e8d83d746c25","revisions":{"c7e44103553fbe81a31f980417d9e8d83d746c25":{"kind":"REWORK","_number":1,"created":"2023-07-28 09:21:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/08/889808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/889808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/889808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/889808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/889808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/889808/1"}}},"commit":{"parents":[{"commit":"d450d25d953cc53e902f83c6a0123efa2536faeb","subject":"update constraint for oslo.messaging to new release 14.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d450d25d953cc53e902f83c6a0123efa2536faeb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-03-21 16:30:39.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-07-28 09:21:41.000000000","tz":0},"subject":"update constraint for oslo.log to new release 5.2.0","message":"update constraint for oslo.log to new release 5.2.0\n\nmeta: version: 5.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I53885de1f413495d843a63019a419833fe8e9066\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I705ace770c74439c04a34eee10b3d1450b55d79c\n(cherry picked from commit 1c7b61027a4271cab60d092408db3202f26a4e62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7e44103553fbe81a31f980417d9e8d83d746c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7e44103553fbe81a31f980417d9e8d83d746c25"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
