)]}'
{"id":"openstack%2Frequirements~890081","triplet_id":"openstack%2Frequirements~master~I88346d2cc0ed8303223717c88f1aa19d77876bb8","project":"openstack/requirements","branch":"master","topic":"bug/1955556","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-07-31 21:22:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88346d2cc0ed8303223717c88f1aa19d77876bb8","subject":"Bump XStatic-jQuery version to 3.5.1.1","status":"MERGED","created":"2023-07-31 11:18:59.000000000","updated":"2023-07-31 21:23:46.000000000","submitted":"2023-07-31 21:22:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890081-bug/1955556","meta_rev_id":"f36a0ee688b9bd1eb4ed12961a4a7c74d33e3f5f","_number":890081,"virtual_id_number":890081,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-31 21:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-07-31 17:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-07-31 19:51:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-07-31 19:51:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-31 12:35:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-31 17:02:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-07-31 19:51:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"41beed62e8a9e11807c96b95137b7fc77ff12b20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-31 11:18:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0088df450dc60a06e02aa87b8d93b78ec052299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 12:35:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4dac627db2e4748b5a6912fee9fe56e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47bf4618a8394fb29ec3d315bac57dd8 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea5b98875d5c4bfabd8ba13e80e892d8 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a7a7ba1375d4ec99b5967f24fdb2f33 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0223b8c3bba046bfac6362e07f803f4a : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a228fdce153e4522b95649d9bfed0ed0 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d75387030e34657aaf574f6027375c9 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99def37eedc54e7099a5fbfe85dc3d6d : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/51399755f00a45859305296cee58ba0c : SUCCESS in 9m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b51b35e67ba1402a880d728106d95b68 : SUCCESS in 12m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/be608f03ac6441c4bc99ca7a7f416c98 : SUCCESS in 8m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0b637b64db4d4864b021fb3ee3b91672 : SUCCESS in 11m 38s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1e1726bb8f0f4fbca15c0f7b5f0632fa : SUCCESS in 10m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/736b4747de74475ab8b7bcfc609b8dac : SUCCESS in 4m 26s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e32e2902689043d2ab4246d97ec68ef9 : FAILURE in 4m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1637307b54184a57bf7a217bb3d95f99 : SUCCESS in 10m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/41512a2611f24a4fb3e4ac01a5d2ebe6 : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4be73dd525834ef1acc426241d579bfa : SUCCESS in 4m 29s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/628959b9237047cd97fb789c80944f06 : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f3355a6c55c94cd3998fc5d67da0d8df : SUCCESS in 7m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73a6e785fdb4464e943c9aa0ae4be4c1 : SUCCESS in 7m 05s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b84224e1ea7944abb5021a168eaa3488 : SUCCESS in 7m 03s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2aec2cc953834d1cba90b7620fd3270f : SUCCESS in 20m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d8f2cb12ced74f42a55a0caf79dc7ebf : SUCCESS in 3m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/aea694fc441d47c3895357d24aaf2db0 : SUCCESS in 9m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cacc62d9f2894d349714f8cb28503cab : SUCCESS in 4m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/e2dc34a5558b410fa8cd88ed4bf0b06b : SUCCESS in 9m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/50bf2c4dc38744ab9b0b4528918b1b43 : SUCCESS in 23m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/568d7d5188aa4a52804d68e186c1fab4 : SUCCESS in 12m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2ef3be05886042b08da7501e2bd5ec51 : SUCCESS in 13m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9a6f7b361488499cb5bce5f09c6b3077 : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/19684a9d52024e58af7545d336a96bf9 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a5c59c61b444b8a94934edf8a5d902b : SUCCESS in 25m 40s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/494f1f3bc9ab4096a699ef95e709b515 : SUCCESS in 5m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e933d5a0e132493789dcf8b06b4d4fa9 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e950fea6db134f01bcec8a4ef9307d73 : SUCCESS in 6m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/46a0a53ebfa34aa6a1f9c59eb1f70373 : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e5e603e97e14dbb9956f6723ac7bd2a : SUCCESS in 1h 07m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a27b1f2fcd7b48d7b6cc2391983052b2 : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"09e9f9d5b99ccac78043eb65b336a36dbc4d2a6b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-31 14:27:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"058db3beac2d4643d14afe5b422134d336f01cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 16:07:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4e8d65bc02e4baeaece00a1b66786c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac5396a197b4309a6d67aa203546531 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d355591f7e948ff98226f844ad2284b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2de5d18a962e4f6b836e47388be5ce66 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b79540cb55b44fa8c60ecd9aa16f210 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bee821f0e16e452095e1769c0357507d : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/659956bf4b5941f393643c7af0b25d7e : SUCCESS in 6m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/466f8f87db1840159cb8eca24db2eead : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/13cf86a8e03c4114b38b4af922a57b7f : SUCCESS in 9m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3b0b10d87e914f9ab519e072917e12da : SUCCESS in 11m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/203ac1c734684467b087a5918b030c63 : SUCCESS in 8m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b8d4476526be45cbbb09011091a3b348 : SUCCESS in 10m 52s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c41264b5e37142439ec60c2f48683ac5 : SUCCESS in 8m 50s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/17a97e315f284c0884865c61d31668ee : SUCCESS in 7m 11s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b021a62f1b0743cb85610ea75cd5a04c : SUCCESS in 5m 04s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9664fd8511014753aa139e308403c386 : SUCCESS in 13m 15s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/89df431d7ba744708cb8ceae2b7b1fae : SUCCESS in 5m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/442367831bc1406ebed4e7bd95322743 : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c242c9696c694506937569fdde87de12 : SUCCESS in 9m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a9a9c4ba61d14b68ac875e1e6b3d8eb8 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e100adb5868c419aa101786701d6ca79 : SUCCESS in 8m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/905c410068b844329b3b71b396f87f09 : SUCCESS in 7m 49s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bf4becd56c0a462db549a695e10e8a27 : SUCCESS in 10m 41s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/5799e02d23a44df3bf9a05fac8ddbb48 : SUCCESS in 3m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b8cda3e52e684d7e930398342c66b9a4 : SUCCESS in 9m 38s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bee5019bb26a42c3a21bc0d9163c209c : SUCCESS in 4m 52s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/441d0d29683c4c39a6c6e8cfea9cd91d : SUCCESS in 11m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6faabd18ce3d4d22b61fe75e41b9f7e9 : SUCCESS in 30m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9358bcd88c3d47068f4ad44b0f2f0e50 : SUCCESS in 12m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3c629892f4be489f888629e3d8c218c5 : SUCCESS in 17m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/100fb57248d643b0bcd8b4b5a25161b9 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/118fb850074c431cb910a06342a1c2f8 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/166d824383234eec96dc939e9ebaf66d : SUCCESS in 25m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/25d80ac2561048f38db8de2e2d52e102 : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e202d734484649448568d8f859b26263 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b68e8fd3fb5b49128d31d9051eb50b46 : SUCCESS in 7m 01s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/05a9674133ee44b5a489ddd94c20102f : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1f33d696c814e5eba297f4e6da094cf : SUCCESS in 1h 06m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b648e6e03934e1ca9bbb0e2db98ea3c : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"045b20231f55ca708ddafb2864f81f644f2796de","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-07-31 17:02:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"268d7d92cd5ddd66fdd5ebc391c8c13ff93b37d1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-07-31 19:51:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a472ae1db20647822ddb2a086e91310443e19147","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 19:52:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bbb14fbb8c12a01491838c4710718305b16fbfe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 21:22:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd7f3af3baae41f6a185aab3cf0511b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc0affa2dd64e669ec0783d6261722f : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6677d0118e2b4d01848286871909bffa : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acc8960f37dc4b5c9d48c19033a1f3c0 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3636adbcc9d045bcaeede842f1f9957d : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec969170a3784ba0981e6dfcbf820b7a : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db859bd2e108488a9c199a0e3a013b39 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b346a5304544103980655949fd1d812 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f45eb887d554dd7b904fce49e5ad241 : SUCCESS in 8m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/770b52266d5c496daa08e586a952ad68 : SUCCESS in 12m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dfaed2346cf44d7db703a3c11d32d693 : SUCCESS in 9m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5f213f0b458646908ec6779ede192584 : SUCCESS in 11m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1f117e5eeb2c419da6bc0cbeb4a2982b : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/1107c3a377b344c8a931467da0cf6607 : SUCCESS in 4m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bb5bbb37dd764d5fb269d6426a91f095 : SUCCESS in 8m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fc9baad0002d48dfbf32897b62b6ad6c : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b587839f2b694389aedba85774fe46f4 : SUCCESS in 6m 56s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/84150c31a60f41468cfdae13ab954041 : SUCCESS in 7m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3a5a166895ca4788aee92db792f03f1d : SUCCESS in 9m 44s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/559fac2414db4b8f83de82be1f066d72 : SUCCESS in 4m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/abe283383efc4b29b39a01adbdd00814 : SUCCESS in 8m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8d74028ea47a46efa9a311d6dcc980f3 : SUCCESS in 4m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e3e1b34b09774aec85bde8e1ae235bcf : SUCCESS in 20m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7a33b4f2beb44806baaec07f0b391245 : SUCCESS in 8m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/df90625fe61b40f3bea02fdf0401c4ce : SUCCESS in 10m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/efd92455a71149389d61b4aba928a649 : SUCCESS in 4m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8b09b2cc023a4e8691aabd3cfeed56f5 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/657367ca13c048ed8ba53279242cb620 : SUCCESS in 19m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2b31d250e3c549e589b715e196fa0232 : SUCCESS in 5m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/675081828d6847bc8feb600a1db48b7e : SUCCESS in 4m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00058e25269e46c087fd561a5fc53be4 : SUCCESS in 4m 57s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e2ab324eab9041b398379cc5d4b8e3ea : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf1259a1ce8646e589bb76dd51eded65 : SUCCESS in 37m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c956b7566df4ce18b081298723e2df4 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b79a394763338d3abdab11c05166b71882de4bdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 21:22:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f36a0ee688b9bd1eb4ed12961a4a7c74d33e3f5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 21:23:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/930860b444f747d79b7cfbec915399cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70663cbd29e74d7fbe6dad78a243f23f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a","revisions":{"ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a":{"kind":"REWORK","_number":1,"created":"2023-07-31 11:18:59.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/81/890081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/890081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/890081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/890081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/890081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/890081/1"}}},"commit":{"parents":[{"commit":"a800e1446a29012861cebc8f9b6519c343986ab3","subject":"Remove Capping for Xstatic-jQuery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a800e1446a29012861cebc8f9b6519c343986ab3"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-31 11:14:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-07-31 11:14:19.000000000","tz":330},"subject":"Bump XStatic-jQuery version to 3.5.1.1","message":"Bump XStatic-jQuery version to 3.5.1.1\n\nThis patch update XStatic-jQuery version to\n3.5.1.1 in the ``upper-constraints.txt`` file\nbecuase it include some security fixes.\n\nRelated-Bug: #1955556\nChange-Id: I88346d2cc0ed8303223717c88f1aa19d77876bb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
