)]}'
{"id":"openstack%2Frequirements~890191","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I1565e7313cebb5e2c380471e81b8d3b05b290705","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"I1565e7313cebb5e2c380471e81b8d3b05b290705","subject":"update constraint for horizon to new release 22.1.1","status":"MERGED","created":"2023-08-01 10:54:02.000000000","updated":"2023-08-01 21:53:02.000000000","submitted":"2023-08-01 21:53:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890191-new-release","meta_rev_id":"10f7af89e0f320968c440c6c2557c0bab7cfce4f","_number":890191,"virtual_id_number":890191,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 21:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-08-01 14:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-08-01 14:56:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-08-01 20:13:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 11:02:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-01 12:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-01 14:02:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-08-01 14:56:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"9866647c780025780bc1a398b28aec73f6000663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-01 10:54:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3328a83f16b11054c15fdfb84911c77231927181","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 11:02:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6a333338ea04ca3987e113997fc4506\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/788ffd20e7d543d59cd2a6474dc038bc : FAILURE in 7m 52s","accounts_in_message":[],"_revision_number":1},{"id":"05f1f2a148ea60e59ec16e309661b7fa9942710c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 12:22:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e69d11f9af04ebf909e296dfba3c1b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a9430055f74335982a4baabf10298f : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b83eed42f028423b8e4cbd8d5761fa60 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ef82910e5a8471e85c3e830414458e7 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5869115c7cfb4f02aa8d1758cacff3d6 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e44709e9041b4389879db29ebd7f64d0 : SUCCESS in 2m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/35033999b1e94ab19c308e27bb6f95c0 : SUCCESS in 11m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17689d5ef3984db8af3aac314e322360 : SUCCESS in 8m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d6f0bcb1442c40b992bb5ca62422a3a2 : SUCCESS in 4m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/41d2df7368fb47eeb8d68e98ff6f99a7 : SUCCESS in 9m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/df2ef9a07e2049bb8f4409613ccc1973 : SUCCESS in 4m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/aa4360e6232e4729ae0c0f1e9b1332ed : SUCCESS in 6m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/de73038ccc204a64acfd6f6a3a7d4781 : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9460e9ed9c754755a3c5a0330ce42937 : SUCCESS in 8m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9522f4f8864d43cdb9561f8d4695016d : SUCCESS in 7m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3ea0f45c874d4a068219f1a338e9e593 : SUCCESS in 10m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5c8c989a12e7400994e1f418f8c28c67 : SUCCESS in 3m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d62b3690a8dd40618b02bd138dcd40ce : SUCCESS in 12m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b3e00eab089f4776afc3acfa47f5e94b : SUCCESS in 4m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a32ad3c1c7ea4bb394fbc1db054bcb97 : SUCCESS in 21m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cc292dbcab294018a4f7f33c0c00ae3c : SUCCESS in 11m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8112e567eac54188b1f2eb4f3c324836 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1b22949cd5549f9a10a28093271ce12 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a6c229ccad941e89d1b6b79be4f105e : FAILURE in 22m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d05c4067c04442f290bbbe69420d6267 : SUCCESS in 4m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d2783827c4ea4563bad0c45c203b27e5 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2913ec31525a4afd9d49f5d025566343 : SUCCESS in 5m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/64d7b988a2934d21b98c6293b334d765 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c12acffa2cd4a26a8494978a4ddef33 : FAILURE in 40m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f07a25d6b16d4e33bd1d016af5e937d4 : SUCCESS in 1h 21m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a88c19278244a9d6239cde170cc4c8ab1f11ce47","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-08-01 14:02:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73d7f34c63fcefc8edc96f078c9ecb87cb5b2885","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 14:10:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7a6db5e83d249e9acccf83f765f4592\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a4f62fb685c74c8394ab976fd1dab3d4 : FAILURE in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f121b7c8d2c2da1bf32686fca4ebe729737015fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-01 14:56:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"120609e16ed0e0dc6ef6dbe770b37daf0ee92373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 15:09:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4805fb1a0aa04ded9882bb6e24a17e62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48e9e09290b049fda3ecca2961659448 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e33c05f43014cc68d8a80a5cd27a86f : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70336215e9a44210b1635535f7912b26 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d89a881809142b6a81baedf48a0b1da : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3a96f735da04f8da9de271c04bd5de1 : SUCCESS in 3m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/49e429e3effc4992a30e0c92b516e5a0 : SUCCESS in 8m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/76cc09a31a2c4d5296bb6dcbf2ab0e37 : SUCCESS in 9m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7afa87118aec42ddbdacadd36c1bc590 : SUCCESS in 4m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e7db7bb8eaa348989516857fbaf75e4c : SUCCESS in 11m 06s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/321b6c96db0a4f66a8e0f10310ce6f61 : SUCCESS in 4m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f898f68535d34bfc896944d901524a37 : SUCCESS in 4m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a4d6636258bf4a57a1f6d1a0ff10f59a : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c128928b1cf443ddacf17676d4e75a2d : SUCCESS in 9m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/55aa2bdab7d040e5be2e8bda9ad54172 : SUCCESS in 6m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e841c479de6c4f04bac51d389e77ee42 : SUCCESS in 10m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dffcd6dd3f03426c899fa1ee5842a13a : SUCCESS in 3m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/31052cb6624a4a6e945dcbdab7785370 : SUCCESS in 15m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6f7347adfdb945b2add7f5fc4b3f613e : SUCCESS in 5m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5b5191f509cd49af93648bcbda483af6 : SUCCESS in 26m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2667f0e5c57c449a9bdd63bd23ad6867 : SUCCESS in 9m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c44c8db2e51c43d8a3468e927af27682 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d43dcda28684f77a8494345d6fff5f2 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb20f37f6f464bf1a89e81d2e9b4d9ce : SUCCESS in 20m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/94f912753ed041f394e2b3ba63f96f9b : SUCCESS in 6m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/721894df2dc2481db419a99cdf693cf4 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd46b857f69f42e8aa0c901339d8c7c2 : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e3de735920fa41888138ba87f3cb6d93 : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/870cbff6799f453bb3e363906a67ecf9 : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ed3465a9f7045838bf76e18b365ab87 : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1ac2595a4ca91fb8df96fd98c33a1d33c8cb41e5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-08-01 20:13:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f747a3a37d1b1d24d9b96d64d2a26df1642bb14a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 20:14:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d1b51f34523977c2329fb5a6e5064d5f4c1d32d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 21:53:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c1c667e6ad8440aa78a7b5e6b55afbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da9cb40cdb69461a86168f11a55e72a3 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e8797db56d846dbb27ac8fb8b3e0c0f : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6e96b03dea54b5ba557c6e657175def : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/571f3c1a58594552801e616cb71eebdb : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec6eb9c6058b4f389889040611e5ed4b : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/659cb3a4a3674369b51503b78c508b00 : SUCCESS in 11m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d74881e3c3a84a68bdcb25d7a29c1107 : SUCCESS in 8m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/70fdbbcee95f44a198dae80c1b05948d : SUCCESS in 10m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bf3b079a0f744957950a2336a8b040aa : SUCCESS in 5m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ce2bb918cc2348bbbf62c44795d7c6ef : SUCCESS in 5m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4560581cf0834e3b91b776274c1932e4 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6dd1054aaa93499690f9f466214a6257 : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9baa4d12bab04542af9f741b617e4abb : SUCCESS in 7m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3c666bde873f41218ec75250c45899d8 : SUCCESS in 10m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/363131c09c2a4ef1a714af6b2dcf3a0a : SUCCESS in 3m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/61e59cf2772941f79dcc6dabfab7ef7c : SUCCESS in 8m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/42a8fb61ef2a4d909605749340eafccd : SUCCESS in 4m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0309337e639840439b1e6c0ee4138172 : SUCCESS in 24m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0993afd5bb604735adb7f1559819baeb : SUCCESS in 11m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e66ad845335d41d69f7db999d9db1df0 : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b61bc519537d4262badacef9d7c38180 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f128272f37e34d83ac497df9ad90cac3 : SUCCESS in 20m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/df59271fcd77476ebdcc3c8bf9280464 : SUCCESS in 5m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/150203cde5794b7b9ed90f0aa119d4a5 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fee73390ee0b4c72a0f86880ec690c11 : SUCCESS in 6m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6a1e4b6e23714f55bda359fc8b7ef478 : SUCCESS in 8m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c65f71a310a8490ba270841a1eaf6b9e : SUCCESS in 42m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/feccdeb4af3547cea4fa15664f01a68c : SUCCESS in 1h 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"10f7af89e0f320968c440c6c2557c0bab7cfce4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 21:53:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0095b87fd7d19deb5fed14bb4d539f51e7d6ac88","revisions":{"0095b87fd7d19deb5fed14bb4d539f51e7d6ac88":{"kind":"REWORK","_number":1,"created":"2023-08-01 10:54:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/890191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/890191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/890191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/890191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/890191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/890191/1"}}},"commit":{"parents":[{"commit":"c01bb0304aa58b5b1022d5736d90a74a496d6b26","subject":"update constraint for python-tackerclient to new release 1.10.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c01bb0304aa58b5b1022d5736d90a74a496d6b26"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-01 10:53:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-01 10:53:59.000000000","tz":0},"subject":"update constraint for horizon to new release 22.1.1","message":"update constraint for horizon to new release 22.1.1\n\nmeta: version: 22.1.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I6a22f638513ff7f71a0f2ab2cc3e6cceeeff46c5\nmeta: release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1565e7313cebb5e2c380471e81b8d3b05b290705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0095b87fd7d19deb5fed14bb4d539f51e7d6ac88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0095b87fd7d19deb5fed14bb4d539f51e7d6ac88"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
