)]}'
{"id":"openstack%2Frequirements~890414","triplet_id":"openstack%2Frequirements~master~I72ad00b27555872aac22c6023dbf4a5adc720a75","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2023-08-25 14:25:17.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I72ad00b27555872aac22c6023dbf4a5adc720a75","subject":"update constraint for neutron-lib to new release 3.8.0","status":"MERGED","created":"2023-08-03 12:50:15.000000000","updated":"2023-08-25 16:29:47.000000000","submitted":"2023-08-25 16:28:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890414-new-release","meta_rev_id":"6c436b079ad07bd70dec698ff21be9bf0123f93e","_number":890414,"virtual_id_number":890414,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-25 16:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-08-25 09:47:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-08-25 14:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-08-07 08:47:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-08-03 13:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-08-25 14:25:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-03 13:06:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-08-03 14:19:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-07 08:45:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-08-07 08:47:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-08-25 09:47:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-08-25 09:48:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bd18484505f42e82f826f858da2efbc2dccb6f9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-03 12:50:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c7fde48e95f3e4ff1304df5e97f9a01f6358c01","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-03 13:06:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5963f96a6469431f458f96fa4b7110d21754308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 14:19:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80e6533f9ca84490aafcb0d5e63977c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd092eeb1d1b4f609850890d7174759c : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f3f9bf9ddc493f9024aa7e4ebb254d : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0be92d983ab442ebf68720e31d4939d : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2f7f05c12754545a156604bdc66cc5f : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b4326312e284f549ec3669016d4fbaf : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21e501bfaa48487d870f558be9d43e4a : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff3954d2ea73436f9dbd5eabc363adcb : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f1547d143e184529a74f544a2f22f6ae : SUCCESS in 11m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a4c80d3575144c8889ceda4bd35b6229 : TIMED_OUT in 40m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/39a35c20eb9840998054a7bf880a8e35 : SUCCESS in 11m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fcbcf402ea4c47e7986297ffbca11ee4 : SUCCESS in 14m 39s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/1bb9f3978c6e41d2b0076922f46eb9da : SUCCESS in 7m 25s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cc3334a7a67e419dba85d44725b5925d : SUCCESS in 4m 35s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/4457d1a7583a43c8bdb074e75b200a58 : SUCCESS in 4m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cc7ca1db506a440582c64b0054bf2d4f : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e503266851f04ec7830a5e38b5889539 : SUCCESS in 4m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/752d74f0781c403c9a1cf6a84a3b131e : SUCCESS in 5m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d8cdd4d19fef46e7a14f1a73aefccaec : SUCCESS in 9m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/127f8351272a42ffb978de0ff7da285c : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fdb7a683a74540bc9af99b2fb3fc374b : SUCCESS in 8m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/49be509b5bf149ca869c57b30c447c3f : SUCCESS in 8m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dbf693826ec04c0fa86d5ab2c678f7f2 : SUCCESS in 16m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dfe8c07cbbba4760b890273a88c0f819 : SUCCESS in 4m 34s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/82cac8ab95db4fc18e6cf139f2600b48 : SUCCESS in 14m 35s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/09d8fc42f69b4c1aa82ffeb77986310f : SUCCESS in 5m 49s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/859f601fba0f4a86a9df16c8921ab90a : SUCCESS in 9m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f8f448b156fd4221b103a78a11c3dc75 : SUCCESS in 37m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5f6deebc1da543b8abc05de85517edae : SUCCESS in 8m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4fce8a9db13d4e08a02909d966f3b313 : SUCCESS in 12m 43s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/f304be5887814ddfaf378fa857f0f087 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/753553b834264a77b2b5b545b6741ece : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65b3d3bb6a3b400497fc50a743faecc6 : SUCCESS in 23m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8331f793122c4cbda6dbc0bcd2028556 : SUCCESS in 7m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0004cbcb75f14f37bbf68ff987d94d20 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f146256c25b4a379234826758927646 : SUCCESS in 6m 21s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/740149eaef16498dad1b104ae82e5d1c : SUCCESS in 15m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e45ca986fa7546208f670c872bf0ad7c : SUCCESS in 56m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/159bd934089c4a6baab3d77c92146161 : SUCCESS in 1h 27m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f559330ecb2d53b0f28761c760d3b59b68b81101","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-07 08:45:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f08165c0520be8de971aaf2674c47c8022785bf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-07 08:47:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b37678e5fe8e0f94b707722a3c125dd8c7450e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 10:47:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0aca84e5a224a7381170801307eea75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24caaa403aed4178b6e8f6203e67194b : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d9d2ad8132a4548a7b2cba39a20cde9 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c1156f07a914927b9edf2529aefd2c2 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acaabd41bcbf422d9e83f6b4f9b30d9b : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31b319581b45448db1792005982d49c6 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1be7ea547a9489eb7cb2691d95859db : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c36c8b2d517a4922b4f4a3bbf8509866 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b7e7c8669799430a8a6dc6d7d2d9a7d1 : SUCCESS in 8m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/922ef1815716430bb7e88079a2a87084 : SUCCESS in 10m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9d778788827c4958b2d7d2259a99523c : SUCCESS in 9m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c8c72f3eb04544e7b5d8267803c0a52a : SUCCESS in 15m 09s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/397ecf1b35764e8fb3340e0879e1c42d : SUCCESS in 7m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/07f73650554c47929eba448ab686c92c : SUCCESS in 5m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/ec08251cd53e4a9992c97a07c26e0ea1 : SUCCESS in 6m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4576aa0116464f1fb8ed17dcfbd23582 : SUCCESS in 10m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/39b727e21fc44d17a5fe8e1007c2d694 : SUCCESS in 4m 57s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b4f75828063e4663aa6d18f1cabac7a1 : SUCCESS in 4m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/12f0dc9a323f4fb185d20b85bb9fda2b : SUCCESS in 7m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/14c3b1e02c4b498eaf257ff029d19d67 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79e72080280b4a989347544790c5a88b : SUCCESS in 9m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6deebbb86928403786146ed8951c3ec7 : SUCCESS in 6m 44s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e80fb2ef6c2f44948f38209868cf7b71 : SUCCESS in 16m 08s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4ef97e12af3b429aab48f6240bdae6ee : SUCCESS in 4m 10s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6296bdf129ba419884d12bd390f72a27 : SUCCESS in 7m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/50dd4d8c43d14c6691ea434e4dca9038 : SUCCESS in 5m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/83b9e1de68d149de999ecf035b8842b2 : SUCCESS in 8m 31s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/985f9c54e4e74b779e28eb7595bafa8f : SUCCESS in 32m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3e3ef97b92714bb1b30cff29daf16c17 : SUCCESS in 9m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f212ca1f943349b1ad3a70768135e15a : SUCCESS in 14m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/79c33cc5f5f643e28b2789d95a4a65c0 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4bdaeed53a5a41d5895f25ab95a09bc3 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/337d11e9cf5e43c29894ef994bc161af : SUCCESS in 26m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/281cb55a5bd74db6a70c3f2e0a06dd1a : SUCCESS in 4m 58s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c16b75e31310479cb1f2cd6817c26e12 : SUCCESS in 4m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/904b641ed38244ea9134ec344ecf9544 : SUCCESS in 7m 07s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/fa5563244afb42559f4348b55224417d : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2f485ff6ca74e47bf32053563409ea1 : SUCCESS in 1h 11m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da283b4eebf34ad58ed641697e565480 : FAILURE in 1h 35m 16s","accounts_in_message":[],"_revision_number":1},{"id":"359961bea6baac04a70d09b3fcd7078bf3b44e8e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-21 07:57:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bf681438f8d394cb65d7654107d4da15ec83811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 09:49:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a77bb226b1d4d30a16feb11864a2480\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202b995a8f0744d1bdcb48d38bfe3ee3 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8fbf5b38204461b8ca9508e59229579 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bdd85ad13174a1c80dac963b44b4668 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d639f2ac44d4944a631729515dfb8d5 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4221549dbe5b43ca909453433ed30d5a : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bd1119e1bda42a8b1af249ff6bf32bd : RETRY_LIMIT in 1m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2bb43b58aa442e28e191d0aa241e7a1 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d45cb17d04014bb29720e2aaa04cc004 : SUCCESS in 4m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e3759d642ee14f8490af3ba52308cc42 : SUCCESS in 9m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/63f4cf12621847d5bb1223faf39d7f55 : SUCCESS in 13m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e630f3bba287494b92269390f3845eb3 : SUCCESS in 10m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1d07e0dc58714044b2b5066b49120ca4 : SUCCESS in 11m 16s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ddcc61cd9aa242999fdefcfef82a65f1 : SUCCESS in 7m 12s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/eed5dd3a07cc4813a737de817e104228 : SUCCESS in 4m 55s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9998fee775414cfeb314f76983aa55f3 : SUCCESS in 4m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9e68b5194f394909972ec16ccb7f2b0e : SUCCESS in 11m 10s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5c8981f9e3a94aecb0cb00970f5ec15b : SUCCESS in 4m 47s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e58aad913dc84abc95da26291728d7ee : SUCCESS in 5m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bb26885e84684dc6b380f2b476ee1d9e : SUCCESS in 7m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/89f1fa98efed4319885facef8e908407 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ddf8fccd02834f96bf6ffe6faddbe042 : SUCCESS in 7m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fd256d125f754d3480f50d0cad718899 : SUCCESS in 7m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/af220ce3056c4c00acac115e9f5ec836 : SUCCESS in 11m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/04eeba4babbc49039707677cafdcf556 : SUCCESS in 3m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/567bacf106c4429b9f3469569cef008a : SUCCESS in 9m 41s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bddee3da258645a7b668b0bff93e07ee : SUCCESS in 5m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9eaf43bd2017445e89766cca73474f19 : SUCCESS in 12m 15s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7035e6dd62f94dc793e64e79fbdb12bf : SUCCESS in 26m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/71ec455b2d344c38aac52891c16a7612 : SUCCESS in 12m 27s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b69dbd0c583641f5a41fc01d13ade7e3 : SUCCESS in 12m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e07f771522674364ba1e3d37799983d2 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1968d33bc9754db1a7e1478b6bd5aace : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf41feb454274adab87339cff894c986 : SUCCESS in 34m 15s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b6c19c0803b2434b81079e229b3aa89e : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8b6d9451d4f243ebb68b9a1161d6d346 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d325814c23341d7a5255ce878ca8e2f : SUCCESS in 7m 12s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/91d77c0815ce4151b3a37e2f79bc3b3b : SUCCESS in 16m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab8c01c4cbf04eedaf8e462c013ec049 : SUCCESS in 1h 04m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ec174d8a0a94bec928c271608bee9b0 : SUCCESS in 1h 50m 34s","accounts_in_message":[],"_revision_number":1},{"id":"636f75fd53291f74481a17467b6e759c3aa3c7b6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-25 09:47:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a65018e329ff9ece15d1c4f616f6ff299bb8e849","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-25 14:25:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0764afe40eff041b9a65dd0447fecaf7674361c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 14:26:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f72fcd2575c3bc073c4d4c1d46cd71768af85ac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 16:28:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/839ed4fdb82549229ae869ab00044ce5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737b49ad236e43f6a5c335fb5884bff4 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10710b3d3d464f72be88767058d4bcda : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/646a91de1f8246079983b63505c3205f : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab0b3c9272a04c8b8b854b129b008cb9 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77625c58194e4a98a0872cd6f213dc4a : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a5f985f18324252955b3c8bd1177e95 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2de8391b97d448deb5d545f54c7023b2 : SUCCESS in 4m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c780ccd1a4c4dbdb8f29a37d14ba932 : SUCCESS in 13m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a1a356e105e54b23ae00a06aed41c70b : SUCCESS in 15m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b6d715579d524e8893e573461e2c3ed1 : SUCCESS in 9m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d9bbac5c7bfb4a5b930c58f1511aee7c : SUCCESS in 11m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9087cc7189e54fdbad53a38a3659bc6d : SUCCESS in 4m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/6ff6594d8db84bcaa6e2f5c74a37e410 : SUCCESS in 5m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5233a27061b041fb8c11d8c64f318dd2 : SUCCESS in 7m 52s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/439d967517794d4f872d26033714fa8e : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6beef544855742578b589a737a308a92 : SUCCESS in 9m 17s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8394ea5c39e34922a04c724b24195ff2 : SUCCESS in 9m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2a7c1533437b400f921a19baefc524a4 : SUCCESS in 11m 22s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cc834a80aabf4e7880080a2ed2181dd0 : SUCCESS in 3m 53s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/927df8ae49ac45a4a13d7a40a69c3fee : SUCCESS in 11m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1a8f0e376f7d4f34b9c01a5c1d0cb8e8 : SUCCESS in 5m 14s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/89a56eb87485459fb0e192f7359323c0 : SUCCESS in 26m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8b4fab703591420992376759a6f7c534 : SUCCESS in 11m 05s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ceee1fddc0cd466e991a515048ef16d4 : SUCCESS in 14m 44s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/e70c6dd36f614f7ab4896d4351e57614 : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f0013e55840a4129b5e9fe63cac418a3 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0857a6ec6c3541d199d7a25c2931b1c5 : SUCCESS in 23m 28s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/63b2b24b65564e9c9bb0114e66a4529b : SUCCESS in 5m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/11282839e5044670a3f1c332e171c4ff : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3ad8aac2a674d8d95d661f5cadbf778 : SUCCESS in 7m 30s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/bd7fee4125a64326afac4d76b6f78bc3 : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aac64a6256764251bef1e0c54bb8239e : SUCCESS in 1h 02m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ed515e2a31c4e778551c36b9a1a94b0 : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"88f9969f4bbea0db78a9c21bfd7a010c6862fd26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 16:28:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c436b079ad07bd70dec698ff21be9bf0123f93e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 16:29:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c701d2aac0aa4363ac9eb75825bde628\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c48052c2f454708a62eb1385491792f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c10f0d04dcc7dbc598ef30f6350ec9286d8053f","revisions":{"3c10f0d04dcc7dbc598ef30f6350ec9286d8053f":{"kind":"REWORK","_number":1,"created":"2023-08-03 12:50:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/890414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/890414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/890414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/890414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/890414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/890414/1"}}},"commit":{"parents":[{"commit":"ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a","subject":"Bump XStatic-jQuery version to 3.5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-03 12:50:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-03 12:50:12.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.8.0","message":"update constraint for neutron-lib to new release 3.8.0\n\nmeta: version: 3.8.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Change-Id: If36b717303db92c334d0555082c2ddae05196094\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I72ad00b27555872aac22c6023dbf4a5adc720a75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c10f0d04dcc7dbc598ef30f6350ec9286d8053f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c10f0d04dcc7dbc598ef30f6350ec9286d8053f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
