)]}'
{"id":"openstack%2Frequirements~890415","triplet_id":"openstack%2Frequirements~master~I5e7e69bc650541129910648c34778140273ca87c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5e7e69bc650541129910648c34778140273ca87c","subject":"update constraint for openstacksdk to new release 1.4.0","status":"MERGED","created":"2023-08-03 12:52:14.000000000","updated":"2023-08-25 16:28:26.000000000","submitted":"2023-08-25 16:28:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890415-new-release","meta_rev_id":"d5810a5e8490be779be24682a33f252815afd9d7","_number":890415,"virtual_id_number":890415,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-25 16:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-25 14:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-03 13:06:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-25 14:25:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-03 13:06:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-08-03 14:37:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-21 15:13:39.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-08-22 08:50:46.000000000","updated_by":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"reviewer":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"state":"CC"},{"updated":"2023-08-25 11:23:38.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2023-08-25 14:25:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d156a3890e1238e6b0bd87d1a00838144d55340f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-03 12:52:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38fb57446e7b774bcaadee22e46894b86fb7911b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-03 13:06:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"097512f95bc69f8a07e0f54c00970856fe383c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 14:37:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cfc3385ebf6430c84c5d997a61bdbce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380bb96c98bc45e7a66d633d14ea6f6d : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c695843d80d4d2b9281f815ba151f05 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83e993c89504448ca4991467e73a820d : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fd58b2647f44f80ab6479db96cdbcb6 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec537be648364e7491672dbac6d0879f : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50720c141b994ebeb4d0878db44a12eb : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1113285c6c58414681d3defa795d326e : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d05088ebf28c48f0afa7761e34734ba3 : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b8ed289ff33e441a8e42209fb854801b : SUCCESS in 12m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/179393a48f7b44918e45c6a5ca1d32ec : SUCCESS in 8m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/72359f02eea248d09e3825fa47fbf1ab : SUCCESS in 12m 10s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c9b7942eb4264f13873f7b8f9035cb72 : SUCCESS in 8m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/4a4c33c61d9840419fd40c0c9c774f7a : SUCCESS in 4m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/c462e294ceca4e10a7858d48cb5daa2d : FAILURE in 2m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/749fe1a53a3741c2b71879ed2d830991 : SUCCESS in 10m 27s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1bd303f5bb5643b89c2cfd9d39197ea7 : SUCCESS in 4m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d51856724f414d72890be9c21e22b012 : FAILURE in 3m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3fa5f27ef329491bae0d3f5f946ab08b : FAILURE in 3m 10s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c8cf15ed6e2042c887a2254542b1398f : FAILURE in 2m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39a96b5166dc4f4bbc66e56cc70cec15 : FAILURE in 6m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7419d65d4a2e4e2fb500d99706b13eed : SUCCESS in 8m 09s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b5fa74c34a4f4f019c5b3b4b44137af1 : SUCCESS in 15m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4c682bbcb258459bb7be3dac3d420b15 : FAILURE in 2m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/aeab1dd6a9cf4b27a49a36ed90afdbc4 : SUCCESS in 10m 38s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5fe181ec2e8e409b91c16ccc572b6399 : SUCCESS in 4m 29s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/c94581abfc0a4c48921e83e5a12d540d : SUCCESS in 10m 51s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7094d0489b2c4a1c8c55072bb1d65b33 : SUCCESS in 27m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b0101224e10141d7aca97a849a3bb1fb : SUCCESS in 12m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ded1e584b7794868924459c10bb72365 : SUCCESS in 14m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9051cb1dcc3041e3b0bf9dc33cbbf302 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b83daa22be984688b5660135e9b6c97d : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3723a1235fec4c2784d64e611ecb427e : SUCCESS in 26m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b81ac3f180004a8bbffa42de214bdbed : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/03c9456318534fabbc39113bdf7da5b6 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39441badc0cf4f2cbd82912d0626c6c5 : SUCCESS in 7m 05s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6dd42312186242249d9788df7307de8f : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6d8507d29fd4c1d8d3c274535207bea : FAILURE in 1h 17m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc615cd2aa2240b388b18e8016e7644c : SUCCESS in 1h 37m 09s","accounts_in_message":[],"_revision_number":1},{"id":"321e101494f81f2efa20d7746aeda035ba6d165a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-25 11:23:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca158ebdf622a89413e9186618a7f758b073637f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:36:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8965856714499691ce240e3df84765\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe27d0abd0c4e148cd78f983244ab9c : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f771b4a6b5240beb5866562da7dfeaf : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2045fc1dab1949d7b8020c3c03396497 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a2d702577044b2396d3f6e3f8f62c4d : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60d23851bd5a405a9332dcc2ceaea352 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11a01e201591418ea204c228a59b11b7 : RETRY_LIMIT in 1m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa7a9eb3236640069e5aabf37e765e7c : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a1ea88bbf5041dba135ac78cc01a2e0 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c6555188d92470cadb24681c671f14d : SUCCESS in 8m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8301e9049b81415ca850f7a0aefc8659 : SUCCESS in 14m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d32a870012e04654984aafe8416346f3 : SUCCESS in 8m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/38b9072d79024772a7b79568020d5be6 : SUCCESS in 10m 31s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/89eca53166654ecd8f9fe7020aadd6d8 : SUCCESS in 8m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/5e90da55abef4e50b1a7ba05b36a52c3 : SUCCESS in 4m 18s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9616dffc6ce24fdda89640332ff5d4d4 : SUCCESS in 6m 18s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/12fc2a46b5c440a0a82df697c3f95507 : SUCCESS in 11m 48s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a0e1acc6466c44c2812a76a4f1342cec : SUCCESS in 4m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/46ab2c0caadb4b6aaf0630257bb24e9f : SUCCESS in 5m 20s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/622b024260dc4b6b98daba64a752bd64 : SUCCESS in 6m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b373547ccfee479e805fec9ce7ab563a : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7723bd69aaf8479fa18b4b8437867a71 : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1ce425649b364efc9b46060d64fc3883 : SUCCESS in 8m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4268f065e98b4f219f209c748536242f : SUCCESS in 9m 03s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6015369767634739afe2f1d2294f852e : SUCCESS in 3m 25s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/32c5a752aeb84de8a8da0993894f5f2d : SUCCESS in 11m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e9d2dccd094f4959b9b8c8f6209a7e56 : SUCCESS in 4m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/411b2b89a1754cf897f7b6b341e9c69f : SUCCESS in 9m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/31606ca50f104ef28ff06b9be3c87abb : SUCCESS in 28m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a63c183835874117bf9339380dc3152e : SUCCESS in 11m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/03d2cbad1889470eb4a0d89aa9149f2a : SUCCESS in 11m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/78f82ca4369d4ef88a6dae1722a34cb2 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c63beda1eb94de989c2f8c5ab1badae : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b74aa74ec48e4ebeb3ad81c3fad54015 : SUCCESS in 20m 45s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/62fd16bee5624000be674c0834fd4ca2 : SUCCESS in 5m 42s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1a2973d9d0a9485c9f0295936747e4f7 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b49d3ba62b548c189a175aa59f7e5e5 : SUCCESS in 7m 24s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/14f7f7478e2b409facd70e4e9a8b0409 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5d3db6c2419446bbe9c372506905406 : SUCCESS in 46m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9130758098ae47ce9585d691a32f35de : SUCCESS in 1h 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a7e8acf164ded64441ca15ce136910b08aa30a9c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-25 14:25:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b397cf5bed2e114b781f5b2b61ee761da493b33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 14:26:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"817866bae987c7bae0c17c9567d6ce8c106646d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 16:28:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5607c4acfd78490aa1c9e61c5bd2f641\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c78dab1c7074c7ebe1c00c0d5b9804f : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c1c06481b844669b7817f4c03d6b78 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02b9f96e135e43128b1680b2b4dd3dd2 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5df75772104d4344b798f208e8aa2488 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ca6e861454e485b97a336855d1b15c6 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4759bfbeacc44ab48eb2af5c361d5d7c : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96f00de136214d5fa8616f8a1ee42f32 : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4dacd6902f3f466395108318f53c4dad : SUCCESS in 10m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/398682652ca94dd993a0efd71ef4200a : SUCCESS in 14m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d2721527d2174f64a7dcc44cddca83b4 : SUCCESS in 9m 16s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/0c8cc253541e4a2d8bb303595a016724 : SUCCESS in 14m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7618ed3e750d4ef59d5b4d18e345ea2d : SUCCESS in 6m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/494968ba93da44769083e85db9da5384 : SUCCESS in 6m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/36dd26d2ba504348822b58b543e88d50 : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/193be9aea15a40a0815bba2b4546faeb : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7f87921da8b484fb624d9458932e9d7 : SUCCESS in 10m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/524cdce46d924a21a30bef796b31dfc4 : SUCCESS in 8m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/618c2d8a35064e8189cc158d15e2e757 : SUCCESS in 11m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c0dda0413f3f4d308e974f4bffbd7ce9 : SUCCESS in 4m 21s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/cf383ddd014a43cf843bc2790a87959c : SUCCESS in 11m 09s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d868ae4f7a8b4d5dba366687fcf8b330 : SUCCESS in 5m 27s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/31bdc43c98c14d779bd961f841f0c967 : SUCCESS in 27m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/384cfa2e8bad4f028df0823a6a0514cf : SUCCESS in 11m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/49973ea3b6cf4c5ba446e265060d4ce5 : SUCCESS in 16m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/45915bf3a1ca4e33a47920bc696dea5c : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b1fc7fa08024487a849c364a6d88483 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a5224053f8a471ebd90b1bd8d027e54 : SUCCESS in 24m 41s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8cba086a3eeb465a9a7bea7048027dba : SUCCESS in 6m 25s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/181e6d539a4549c1b99c2428297130bd : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bcf3f153144640fea9c3974ab7594144 : SUCCESS in 8m 08s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/ef4aa7f574f44a55a5c29d42ee2652ba : SUCCESS in 14m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccbff2150fc74d7f94d33df1a5c47a78 : SUCCESS in 1h 02m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1de20faa7762454799cd95134edc785f : SUCCESS in 1h 35m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d5810a5e8490be779be24682a33f252815afd9d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 16:28:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f3a05d1478f56067ecad0b47694ca73f47c3fa6","revisions":{"4f3a05d1478f56067ecad0b47694ca73f47c3fa6":{"kind":"REWORK","_number":1,"created":"2023-08-03 12:52:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/890415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/890415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/890415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/890415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/890415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/890415/1"}}},"commit":{"parents":[{"commit":"ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a","subject":"Bump XStatic-jQuery version to 3.5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-03 12:52:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-03 12:52:13.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 1.4.0","message":"update constraint for openstacksdk to new release 1.4.0\n\nmeta: version: 1.4.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I19d6cc472e80b6a0024612598684cba92618b1ef\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nChange-Id: I5e7e69bc650541129910648c34778140273ca87c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f3a05d1478f56067ecad0b47694ca73f47c3fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f3a05d1478f56067ecad0b47694ca73f47c3fa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
