)]}'
{"id":"openstack%2Frequirements~890416","triplet_id":"openstack%2Frequirements~master~I48ffe31f6a7e533abd2a78f40db6ee3a67125f28","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I48ffe31f6a7e533abd2a78f40db6ee3a67125f28","subject":"update constraint for XStatic-Angular-FileUpload to new release 12.2.13.0","status":"MERGED","created":"2023-08-03 12:59:05.000000000","updated":"2023-08-04 21:21:28.000000000","submitted":"2023-08-04 21:20:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890416-new-release","meta_rev_id":"e98dd24a6379eb1790548bb0a062eb200bb2e590","_number":890416,"virtual_id_number":890416,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-04 21:20:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-04 18:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-08-03 15:28:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2023-08-03 13:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-04 18:41:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-03 13:06:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-08-03 15:28:56.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2023-08-03 15:44:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-04 18:41:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c94c4dc6a56f29dc0e5b5ec2c7958be35ebc8dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-03 12:59:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b027b9b3d0833f4f3a211ac8bd8c3a719cf6e6d8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-03 13:06:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"835e8ebda3ed88ceef2b6e09f7e68eaad80ae4e2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-03 15:28:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8e422f96e5cc61dacd5af91206894acb05f0ba5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 15:44:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ade206dc95e1463da363d74588048f07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deae932ad3264462870af50a90806311 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f55bc6ee392140adb5c174a5292dfaad : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b0848a04bff4c5890fe007f3401510b : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afe822fe9f6042828c3d9d7a6a405a1e : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01cfac77e7a344c4bfea5568b806d01f : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94300cc5f8d948a8bbc0e4fcff69d669 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/43a334c9d4db4488a7bcc197de56553f : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/22a1315cb19f4f27887437e4df210424 : SUCCESS in 9m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/94c712b3110f4f1d8e8b75276d0bd74b : SUCCESS in 12m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/692bdf90fd7d43c9918e8483d4384965 : SUCCESS in 7m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2d510e2f9b9c4de994a3770e86ab6bbc : SUCCESS in 10m 48s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b1915fbb0aa54654a659e079cf315fbc : SUCCESS in 8m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/2bd2955eb2fa47c78493502b044aaca0 : SUCCESS in 5m 22s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5a436fbf0424410c8d19c231ea86ca7e : SUCCESS in 5m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6684a81e4651495296428c3725d19f2e : SUCCESS in 10m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/81522296b5e242d9b1e70dfcd9ca5659 : SUCCESS in 4m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4ced36f7c45a4c2b9ccab644575c011e : SUCCESS in 4m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0817d3fd13fb42ce9646b435aeaa601f : SUCCESS in 8m 08s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/76ab2d08f4c841d1b7770ded366879f1 : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6eb7af38f68949d3a9073cb268e96d6b : SUCCESS in 7m 08s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6b82ca2081f8418ca80ae5a7bd8149fa : SUCCESS in 9m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/93bfa37ee6a14711a1efeb78debbd203 : SUCCESS in 13m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7ac378ba0bea4d0895fa59d0322f4c23 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7e043adca26c4320ab67ef70515aaf4e : SUCCESS in 8m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/7b89f93de44e4657be0324caaefa79a4 : SUCCESS in 4m 23s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d006036365d145d7a3f6024ba7fad973 : SUCCESS in 10m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cd2203e1ba2340df9efd989b01925c76 : SUCCESS in 27m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/114ee7f4e5b34d5e907109aade1c9894 : SUCCESS in 8m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/28fadadb0ad344ef86488321686f7c5b : SUCCESS in 13m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c1b55411c09f466baba069484d72dcb9 : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0306c459f5304fd68439ccf71a8fe4a6 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a583a87b9e23460bb823aa06e4c14787 : SUCCESS in 25m 14s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/21e5d670ab9449279959230b815d3beb : SUCCESS in 6m 04s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/50ebb07f538047d8856a9e97d4c4a896 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e8d3a4bcc3a47eda7c7eff9534acc1a : SUCCESS in 5m 33s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/887f6de04c6e4a7aab3ad31571b22edb : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75ffbd93183741b588b9ce923a2f7681 : TIMED_OUT in 2h 41m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c6ea2ab823d43df87c3ca97ee6637f2 : FAILURE in 1h 22m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2597257d30598cb90338fc288962c42dd4b14da9","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-04 04:53:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d38ccd2bb11003f08047dcf3141facd0e6f91cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 06:24:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38ded71a23de4c4e887daade62151750\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf9908797ce649c8bd5f83655a25f674 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4cf85f5f97e41558d976a8f4144aa50 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a6a1bca79034e27b3b242f4a39d6edd : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a418474a92e24409a370abd817755079 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0f095bfdcf0427c918a18796e887bc1 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2e8f89903184f6f940c31af5c130adb : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/defd4ee661d24078938b1fa8d5209a6a : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8dc6ffd9542a4f2cb1ae245bb1ba94b3 : SUCCESS in 11m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aa16a77737f84e708e87a88eacc60d44 : SUCCESS in 16m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2b27b94077e84258bd613e6aa2890da7 : SUCCESS in 13m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/389e9c49a173466b82c5c75353cf275a : SUCCESS in 17m 10s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3ef00295270443ce876fdc8059c1e540 : SUCCESS in 17m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/91f70337eb0647c694f5ab677339b522 : SUCCESS in 6m 22s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e29441a5d4484ca69facf48fdf7cf22b : SUCCESS in 5m 25s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f4a4965903eb414489e9435b6e52c927 : SUCCESS in 11m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/74422d74242d4d819d307fcac42a2309 : SUCCESS in 6m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/cf54f9200cb3406e825e911d3f83e337 : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/26b87870117843d6af04c890bfa82f98 : SUCCESS in 7m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e41ad13009e24f8184b1326056972e64 : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58e8ec829d344c87adb8d98835db4b50 : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7028f7b465ba48e7a6b7bbd2a074ba11 : SUCCESS in 9m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f520c64713ff47e18ecdf26b79a98296 : SUCCESS in 12m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6c2a8ac026304edab8ac0c0ed6a2a250 : SUCCESS in 4m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/35c9ccadd1004d7393bb32eae36e1416 : SUCCESS in 9m 29s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f20e753f942c4dbf938efbaf196c0d7c : SUCCESS in 4m 52s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/be42f41e9c27498caffb3caee1d21c09 : SUCCESS in 11m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fedaff7314494207be9f9f123a62f7bc : SUCCESS in 23m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/80ad9e1cb307411a98e5aaf61f72bb87 : SUCCESS in 12m 28s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3f52a305125a4417902ebf6ead8455f6 : SUCCESS in 15m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/aaa915c4c927498695a523f226f04df0 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c547cf71035f424ea0450832eee09f96 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93fd48d0aef041c5aa674745232d5200 : SUCCESS in 24m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/237b5abd1eac4eb4a7e7d29c2bd538a8 : SUCCESS in 6m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/902080fa512d4f609d51419110b7038d : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c6956795a3548ba9aeebae944d8d209 : SUCCESS in 8m 35s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/57a7f7b8b2124436a7b6a16d1819decc : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c894c02ab0914c94b8031b214144fefc : SUCCESS in 1h 01m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fb48e7329514f4bafea95136bc25592 : SUCCESS in 1h 23m 55s","accounts_in_message":[],"_revision_number":1},{"id":"642476e22324ca1bccbfe474fdc9195d172d0ebf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-04 18:41:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50cab278bf5d7b5b60ffac8f4c0be3d8e8cc8ed5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 18:41:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a39d32204ff9c70a0f7f556bf81e4f7f2b76a32d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 21:20:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3b7cc8023d14412b06306e5890b33c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1119c650544477f95ea180dd85513bf : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07455601a5f844afbcf0162e482489f9 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e2b80b5f8564eb0a7c4b0f0e0d59790 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/578be285f9094f1c809fa7fc7bb2c744 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff4fe2a6ae5c40d7a31a5466eea15dce : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f9694b13c764c0a94d86f30849d3124 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/331bc54c2f054c28b7e44c943bf4dbcc : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/126b2eaca17f424b82f81d6126004c83 : SUCCESS in 9m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3f32852ae15e45a895a785b60541df99 : SUCCESS in 14m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/dd0fdfa408ba4df8877a7d21905f516d : SUCCESS in 8m 15s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2f59f2f32ec94066a92dbfaeeeb1c5e9 : SUCCESS in 10m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c3314f25e2414b7bbddfa57843bb963b : SUCCESS in 4m 22s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/38dcf4620f2447a8b0e8aa10704cf54c : SUCCESS in 5m 13s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e2509a126d084c148fdfd94e07c3676c : SUCCESS in 8m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8098662f5f024410a4d0eed023219c7a : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e49df9fbeb44351931f45436f5e0582 : SUCCESS in 7m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/457b0f4549b141bbbb4f6453ff805f43 : SUCCESS in 7m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/746843b22442412e8dbecf8102bfe1df : SUCCESS in 11m 42s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/66b008e6333641b085eac360f76ca5db : SUCCESS in 4m 15s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/30dbed7aba344a3abb4e5bb4a92e0034 : SUCCESS in 11m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/becbe40ec95942bc96e2f450e2602c02 : SUCCESS in 4m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b9e3ca63f1b241a59365c01dbb729766 : SUCCESS in 27m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/547af230aa124f5ca8cf9e5881dd6cdf : SUCCESS in 8m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/94fb42f48e154f638bf248d5af55f000 : SUCCESS in 11m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/bc0341cca3b14f88b6af59ccb7caaea1 : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c9c2eadb213341f992456ccf792c2279 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/71075ffbaabc4faabbb4f6cbdfb95006 : SUCCESS in 19m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/fb981d63affe45c4b11fe3d5256ff871 : SUCCESS in 6m 10s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4972116b16e344e786dde2e570688bb8 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/44a86d6aea084f4a9aafdb146a839308 : SUCCESS in 5m 59s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/cdea7a0a972e47e2ac43e2bdcd290c01 : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e4c929931d64ba9bd42fe1779d767a0 : SUCCESS in 1h 11m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f23ca3764934166a1a6dcdd5a4224a7 : SUCCESS in 1h 32m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c7eebb25327d499405babeede7fb7639026d31b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 21:20:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e98dd24a6379eb1790548bb0a062eb200bb2e590","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 21:21:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c83d5d38974a81acf7d5b3cd5b91ed\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7066bd1fdb41497a976d18e1a1d3a1c6 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0bfe5e58439845f49bc67668ca14c6ee1127be3b","revisions":{"0bfe5e58439845f49bc67668ca14c6ee1127be3b":{"kind":"REWORK","_number":1,"created":"2023-08-03 12:59:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/890416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/890416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/890416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/890416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/890416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/890416/1"}}},"commit":{"parents":[{"commit":"ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a","subject":"Bump XStatic-jQuery version to 3.5.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ffc2e87f9d796743a6d4e8d8c7622dde1f18f26a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-03 12:59:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-03 12:59:03.000000000","tz":0},"subject":"update constraint for XStatic-Angular-FileUpload to new release 12.2.13.0","message":"update constraint for XStatic-Angular-FileUpload to new release 12.2.13.0\n\nmeta: version: 12.2.13.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Co-Author-By: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I2c1828604dbcef988a12751fae92197865999915\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nChange-Id: I48ffe31f6a7e533abd2a78f40db6ee3a67125f28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0bfe5e58439845f49bc67668ca14c6ee1127be3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0bfe5e58439845f49bc67668ca14c6ee1127be3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
