)]}'
{"id":"openstack%2Frequirements~891671","triplet_id":"openstack%2Frequirements~master~I0c979fb5363cb2e5cd937f77085d2bb52a322ac8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0c979fb5363cb2e5cd937f77085d2bb52a322ac8","subject":"update constraint for python-manilaclient to new release 4.5.1","status":"MERGED","created":"2023-08-17 09:55:39.000000000","updated":"2023-08-17 21:51:19.000000000","submitted":"2023-08-17 21:49:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891671-new-release","meta_rev_id":"8757f55015a8f57e2e6c1b1d25176de0f4dcc0c7","_number":891671,"virtual_id_number":891671,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 21:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-17 17:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-17 17:07:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 11:23:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-17 14:04:47.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"CC"},{"updated":"2023-08-17 17:07:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"47c30093f43985c49f12dbbddd2fd7415fe2472f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-17 09:55:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74c9a3e713eb7ff4514875f8aa4400fbf43965a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 11:23:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c4d1e9349e346dc9f95cbb74df83c76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e27aa971954fa8a82eeea702a5d254 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76149c04c8e941f4aafa8354499f750d : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a94bd2185ba41ef90f6cef75c410e6c : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0959f0fc41a41cb8069ddd492c4236d : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b483448de97f4dea9e7f071a46e7fec2 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1811ad1b90640b795bc50ce195a7d45 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdae8244ad214abaa427dd49a04e265e : SUCCESS in 3m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/86dc515bce23498abdcec071884def60 : SUCCESS in 7m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/10180833fb5a4cd3a2834e20357fe0cf : SUCCESS in 13m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a544f319b6a14b0692d98982e9b0440e : SUCCESS in 8m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7cf312ae25154633a9a4993c87b2b5e8 : SUCCESS in 12m 06s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/76fd25f33bae4bc6ab78c06fa44821dd : SUCCESS in 7m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/c80830eb477e492db63a61fff730a616 : SUCCESS in 4m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b1976d80d009433aaa390acf2708f98e : SUCCESS in 5m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/14f918d17fb541019f61f90b6c587d95 : SUCCESS in 13m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9a5bcb3f38834260b99bbe8b201f35f6 : SUCCESS in 5m 10s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/063afe8d3cfe46e581bffd27be5d41cf : SUCCESS in 5m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ea08039e49664fbe8f3f479bcfa4fd71 : SUCCESS in 8m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a7272ae8826b4928bf20becdd4573589 : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6467232be7c4d3fa0e62a70c3f2a22e : SUCCESS in 8m 02s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/cdcef4585e7343bea8b0f1705fff70bf : SUCCESS in 8m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a961ba57536f465a87c746c153fdb6b7 : SUCCESS in 11m 11s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/202bf67480104e3d91f72358efe9e4f8 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/d010eb10f36447c49b846c8c17fcb03b : SUCCESS in 10m 15s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/2c9ba7f8f2af4eb8bf481dcef583cec0 : SUCCESS in 4m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ab50da84c42d4c8eaf0c5b8985f8fdbb : SUCCESS in 10m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/99180fa639f045b49e008bc82d5e04c1 : SUCCESS in 27m 46s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4974bf998af24849a8b18444d0057792 : SUCCESS in 10m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/a10cc742ee834dc5821845b88759d89a : SUCCESS in 14m 48s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/735d1fb796e0472886165643e34adf8d : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54452757e0124a159fda9c4195e42ef6 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e42bbf4f61744f2483346ab8c28a4dd3 : SUCCESS in 24m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/16d9e5001ea44b599f64a9d15b3c8cad : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f08d844bb5084b18980e218df57842eb : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de362951f83241b2bdab276144f2aa65 : SUCCESS in 7m 44s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6cc65b4874824a7c8d801ce125a19e80 : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b7a5eb3dd5f4617b837e275c9bff729 : SUCCESS in 50m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d9d3ef66c5648b4ba075a94c1d777ab : SUCCESS in 1h 25m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0a1224470bbf2e429b81859304d0036c53290ad6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-17 17:07:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1b9aa3d61a598eac619a237a6c068b4f69d9372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 17:08:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf53117482e960612c65408ea4b8fca6fe64d031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 21:49:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d80a31540e34042b5b45f602a2ce009\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4337961c67a4cce9f225b6349110121 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a89b78e39542ffbf27cfe0dbad832f : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efc831494207479eb0c68feed86cf80a : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7027adbd06b4b59a88cc1eb35980658 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8673779ca0e943f284e058d235045864 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d884bf87bfa9416198796d6d2d2abbe2 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61d755dbd8e541539903364008947ed7 : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/01e963cd8eca46d3a74e19bec5efdca7 : SUCCESS in 9m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/68a7230cdbd44125bb7fdf7f502577e3 : SUCCESS in 14m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/845dd326caf0448c88636d4038c44a3f : SUCCESS in 8m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b867507558ad43a291436bf40ea53808 : SUCCESS in 12m 56s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e00c1ba9cc0e4506b48eb94abf89b679 : SUCCESS in 6m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/233bd3fa1d6045e980889fba59850a0c : SUCCESS in 6m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/03ef56864ec249cc8f7b75029b8a14d7 : SUCCESS in 8m 01s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f9c6184d7bdb49debcea058abace6236 : SUCCESS in 8m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/89dcde8c6d7e40c9ab8eb6689a276ac9 : SUCCESS in 7m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a329a7bf5daa48438938c98c78b383c7 : SUCCESS in 9m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/192346a60a1146e0b0579ff9921f7a3b : SUCCESS in 17m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a671788f2c8c4d7ba0b04104c404b423 : SUCCESS in 4m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ef9920499a0f4cd4bdd8e4cf7a18d046 : SUCCESS in 9m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/88988d1b487c4588ba512f7746563615 : SUCCESS in 5m 05s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/12573581c8e04e29a6cba31b40e0bae2 : SUCCESS in 29m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6714fe7beafa499692e47e104449fb19 : SUCCESS in 10m 46s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5ff03eb1078042a68e1f8b663212fdfb : SUCCESS in 19m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a4bfbcd4efd642169830db3d321cf18e : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a7a0b7c1170e4adcb68a3eb4ca836d3c : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c02694f11b7843dbbde338f288e2c70a : SUCCESS in 24m 07s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/246bbbe0ee8b4c32ab90ea1924cddcb7 : SUCCESS in 5m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/43efd6236b304edca0717d8dbc6858d1 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/645b381e3262491d958f8e46f6e03c5a : SUCCESS in 6m 56s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3a993bcb7d5b448c8f67e49877aada2c : SUCCESS in 19m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53a67c2d852f4563ae393a7f3b671140 : SUCCESS in 59m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a526b73afa1943eabd060d2f05d960e3 : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fd60aafc8943f5c01533d75d7001a2f2b8110006","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 21:49:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8757f55015a8f57e2e6c1b1d25176de0f4dcc0c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 21:51:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66f22e8781e642d78421bd1fff449792\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873171358a154d749c8cc25126509588 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f593d2f6c71ac29c3ff462e2c8266405e4c6fc5d","revisions":{"f593d2f6c71ac29c3ff462e2c8266405e4c6fc5d":{"kind":"REWORK","_number":1,"created":"2023-08-17 09:55:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/891671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/891671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/891671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/891671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/891671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/891671/1"}}},"commit":{"parents":[{"commit":"720635b93384f91f11b19fb420ae9412e75a1c71","subject":"Update constraint for tosca-parser to 2.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/720635b93384f91f11b19fb420ae9412e75a1c71"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-17 09:55:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-17 09:55:37.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 4.5.1","message":"update constraint for python-manilaclient to new release 4.5.1\n\nmeta: version: 4.5.1\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Commit: Liron Kuchlani \u003clkuchlan@redhat.com\u003e\nmeta: release:Change-Id: Iedea77d21e489334ec48de8464c8a21b43ccf8e7\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Code-Review+1: Vida Haririan \u003cvhariria@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I0c979fb5363cb2e5cd937f77085d2bb52a322ac8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f593d2f6c71ac29c3ff462e2c8266405e4c6fc5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f593d2f6c71ac29c3ff462e2c8266405e4c6fc5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
