)]}'
{"id":"openstack%2Frequirements~891674","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Ibcc24b5a82b92725718764381dd6ba5573af5501","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"Ibcc24b5a82b92725718764381dd6ba5573af5501","subject":"update constraint for python-solumclient to new release 3.5.1","status":"MERGED","created":"2023-08-17 10:00:00.000000000","updated":"2023-08-17 21:49:23.000000000","submitted":"2023-08-17 21:49:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891674-new-release","meta_rev_id":"010c3feb2e8cf49df47f93550eaddce08baaa53d","_number":891674,"virtual_id_number":891674,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 21:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-17 17:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-17 17:07:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 10:18:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-17 11:13:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-17 17:07:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"149bdf5541cba80ad31c8da2ad81fa13c204a678","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-17 10:00:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3cb3a3c569447046fc02a3d4c0b95e646464006","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 10:18:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0ea1d019e1a48199b7dc9f6680841dc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3f97d8a06da44ff6836d07f46bc131a2 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"97f26aae1e829fc51eb5435415fbc166baeb452e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 11:13:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5625452229140e198d3a100e0111dc9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7f327a072141c096356a94c9f3bc31 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77a32b3985354bd8af448660ee35fcb7 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab34c0178c86432c880b97dc7987cf63 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc040e6d41a9497ab94e2a7ec3878a40 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db6b6d3b04b8425f85eb7a1c2a0cdab4 : SUCCESS in 3m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fa8c4077d4024e3386c71b7fb9ab4309 : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b6d1ffb25a3e4029a86d57ba654ac341 : SUCCESS in 7m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ef93f17835c1443787b8271f10cc195f : SUCCESS in 4m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7455ea1331bb4332939ce02a4e891ff3 : SUCCESS in 11m 24s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/83d274a1a4c849afa48961b8b1add196 : SUCCESS in 5m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dbc5ff1a52224709b4433a498f930906 : SUCCESS in 4m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c82d0a5403f34304bd625da737d6dfb1 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0697ed2a2c6345d386681c90433ce9e9 : SUCCESS in 6m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e05e74e9ce1f4a098a6e867848f8b1f1 : SUCCESS in 8m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/118bfc12cd394dc19c8a72a87585446e : SUCCESS in 11m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/27796ab23d8343e68a00bfff4d8bb986 : SUCCESS in 4m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f9a32d8c9241437a87308496c5619cf5 : SUCCESS in 9m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f7b0da63c4d46ae942126db12a9d76b : SUCCESS in 5m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e0caff7623934e5494e489b49ac14396 : SUCCESS in 21m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5d13005d749d4621a8924ddb6c5dc7e1 : SUCCESS in 12m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/893f0853e102441dbe9e8c0854b0a703 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f36d696179d43dbbdd1ca04b1eb8850 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0885d5f91f39476696547b3ac9b3d97f : SUCCESS in 21m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/799daf7be11947cd9adbb979bf756e65 : SUCCESS in 5m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/868bd015274d4265945d1c70b1eec8b5 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0238cc219f164ea5bb7dce007b631d31 : SUCCESS in 7m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/16095bf1f35240a2acac453855677f77 : SUCCESS in 9m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a0eac9ca90a44eda01aa1c2a7987907 : SUCCESS in 45m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb7dac1fc91640a69b5eb443c23ffa97 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"204fa0360ea7054736fcee1eb55e5f362a212890","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-17 17:07:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2174bee3542e35067573e715baf01fc274833bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 17:08:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d89dd562ad45c4d3f55b0d5e93a2810f3ec90b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 21:49:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14d4f46c51b941d78bc06b7dc2c816a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31f1b777ce2044daa71379e2ce97004f : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd063bd797ca4d4e8ae25dda69c1973a : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79760409f02a4be28b22bdb250543af3 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7dd8756beeb44893babd44a1f2cb1f0d : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c75725f65c90402abe4af4f93e5a7b33 : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f03e5b5675841248b305388fb2e27df : SUCCESS in 17m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9970691d6d7e4d6c9edee8a41d6fafcb : SUCCESS in 9m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/139c4024a24648cda00d336f8dcdef84 : SUCCESS in 11m 51s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c7573ae4461f42bda6cd2081d751fc99 : SUCCESS in 5m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b215f5d5485e488f9bc93cdd93e4b337 : SUCCESS in 5m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/58cb05be6ffa4693b9ab955415f6ca93 : SUCCESS in 12m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b9ab36beafc4af984e9cf98c2461881 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5170a108f6d446da94d45c495e873a79 : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/11d2c9cf27374ba8ba15f6483d4fb240 : SUCCESS in 14m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2548a489cf164e139b7f2bb40a93de2b : SUCCESS in 4m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/37857885213f416683e7d20b9f205b68 : SUCCESS in 10m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7f8ba0beabfb464ab5827f211abbd97b : SUCCESS in 6m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3c0f39e2aa3f486fbe4aa1e487254c7f : SUCCESS in 25m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2c2ca8d2b6a444c5b77b308daa3977f1 : SUCCESS in 13m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2e673135b7b64f2eb75c9436cc65226f : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b315480c48dd47dcbc03e988103a3f71 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba31946209b54ab189605e3cd2f716cd : SUCCESS in 19m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d83d9bf34125436da420e510b39c701b : SUCCESS in 9m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ba913a926138432f9a95c52939996b1a : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5eeef38357df42a0adfce2d3184a032a : SUCCESS in 7m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8f12705a9875471499790c2506321d66 : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2d958fe06464444b679745dc15043da : SUCCESS in 1h 07m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/416aed9f9a244222b1a95af3bc888e84 : SUCCESS in 1h 32m 28s","accounts_in_message":[],"_revision_number":1},{"id":"010c3feb2e8cf49df47f93550eaddce08baaa53d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 21:49:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9076d9f06d2b58ef3bd259d9b5214e38047b0c42","revisions":{"9076d9f06d2b58ef3bd259d9b5214e38047b0c42":{"kind":"REWORK","_number":1,"created":"2023-08-17 10:00:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/891674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/891674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/891674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/891674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/891674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/891674/1"}}},"commit":{"parents":[{"commit":"0095b87fd7d19deb5fed14bb4d539f51e7d6ac88","subject":"update constraint for horizon to new release 22.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0095b87fd7d19deb5fed14bb4d539f51e7d6ac88"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-17 09:59:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-17 09:59:59.000000000","tz":0},"subject":"update constraint for python-solumclient to new release 3.5.1","message":"update constraint for python-solumclient to new release 3.5.1\n\nmeta: version: 3.5.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: branch: stable/yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4aef88a03463eb3a618958d477928625ad682c15\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ibcc24b5a82b92725718764381dd6ba5573af5501\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9076d9f06d2b58ef3bd259d9b5214e38047b0c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9076d9f06d2b58ef3bd259d9b5214e38047b0c42"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
