)]}'
{"id":"openstack%2Frequirements~892657","triplet_id":"openstack%2Frequirements~master~I04025615740283e098fa623b7ac7413c1d9b876d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I04025615740283e098fa623b7ac7413c1d9b876d","subject":"update constraint for os-ken to new release 2.7.0","status":"MERGED","created":"2023-08-24 09:10:40.000000000","updated":"2023-08-24 18:32:04.000000000","submitted":"2023-08-24 18:32:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892657-new-release","meta_rev_id":"fae98a80870b70ef73ca1f32db493de2d1ed18d2","_number":892657,"virtual_id_number":892657,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-24 18:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-24 14:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2023-08-24 15:01:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-24 14:34:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-24 10:45:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-24 14:34:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-24 15:01:44.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"1f76bc7b3affa3edf53ea5268040fddd54becf82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-24 09:10:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb267661a6efacf985d1c59fb21b5b3ddb363556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 10:45:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5df5c582c2b4c8babecf677c78f8e15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d1d0c6c1214fb9ad868d6edf7f267f : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c02d9e4164d744c8bb3ef018a345d476 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e56ad6c1a3d2457aaeed9d84d4408d68 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9430c94a9b24f0f9c969b17e33ade2c : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1695bcbe03ac483a95482069f5935031 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8997e3c24d0445c8be76ba69d6b3722 : RETRY_LIMIT in 1m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9726aca74944451b01bf1c43feb1c6c : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46d2df21b1fd40ba8fb5606d2ff3bae3 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/80c93966052c4fd5a059dd003e6f96c2 : SUCCESS in 9m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d140a0b03e284626a46ca9f5f24b9502 : SUCCESS in 13m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3d1a7dd66a724254878900df7cfadf47 : SUCCESS in 7m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6f04ac7cc7ca45c48fa1bf0781186746 : SUCCESS in 12m 49s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5071d120323b4aefb50ce2003c180966 : SUCCESS in 8m 23s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/753062415c03498ca22e7f7d3f33b3c7 : SUCCESS in 4m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/30f19a78dc734007afa312fbc21415ee : SUCCESS in 5m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a8fd3a502cec40fb91c0b814547a60c1 : SUCCESS in 11m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/155a42ec1b134da48aad0ec0d3477416 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e03109b1bd9e437690b1a84225578f10 : SUCCESS in 5m 17s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/29a638ea576948039ca0b4da4a336cef : SUCCESS in 8m 32s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/fd5b0e177ffb44198206ed43d1ec9b8a : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13149fe7b9e4473f98cda0f3f6e11cbd : SUCCESS in 7m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ac4e322312864be1b1d718ab35f8ad68 : SUCCESS in 8m 24s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c9fe2feed5914224be1863e72d327fbd : SUCCESS in 11m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0c15240280c442cfa578042264ebf076 : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b13eb431c1514a77807e826172ed87e0 : SUCCESS in 8m 45s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/0efb6e84a70d49aab0eeb30ee14acd2c : SUCCESS in 4m 10s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a3abe8c74a2d4609ae1afddd04f81719 : SUCCESS in 9m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/16ad273ac4d045ee8effc3d45535604e : SUCCESS in 29m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c99c7e48d259406dbe3c0830c6aa1754 : SUCCESS in 11m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8e036b3a8aed4aaa8ee2ffb57aae33ba : SUCCESS in 13m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8a5e256af36042f38f6ee41824edf85b : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/834c048170eb404b88d960941f7b33b5 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9042584a55543c08673a42b468ec45a : SUCCESS in 26m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/281d2f33cb064a27adf2e1467d8016c2 : SUCCESS in 6m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8ea0d1bb235d48418ce530bb0bc75c49 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e1ef77afdbc43c0842a124aa7055da2 : SUCCESS in 6m 46s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5f5415f1cde74086a2cb95bf39debcb0 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8e80e922da64e79a7545d6dcfba5924 : SUCCESS in 1h 01m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f96fd3642b9c45eb8dfb1a6af57c1c6d : SUCCESS in 1h 29m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f3fe0f4302da044b6a66e7fdbe1bf3093f36c204","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-24 14:34:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72f2e031204df79cdbff9b2c7142685b3a535ff6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:35:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7e83952440eb537a17e6b37f93aa5bed14d0326","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-08-24 15:01:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"329ef1da5d6211a77aad692a89bc19e5adc092ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 18:32:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fc6a7c7bab64b89b6923e31bc8354f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f7d70aada4448db088948b7a21e9a6 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee15952fc8244e8aa7d2f44a3c0f1e7f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1d81dcb65834b0aa1f74a5abb08e12a : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e64d7bb0062495f92b712401413c92e : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/906d4e2e61414325b844d10d664173a9 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b4a73a0b77647fb9782bc23b7e7f4e9 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f8d3c962c95488883907b7df9750878 : SUCCESS in 3m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c65a37ffb7fc4c9993cfa2964983432a : SUCCESS in 8m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b9333037fe214daa80576400b792a012 : SUCCESS in 13m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/03eb901e0acf42d4abff5f0e8ebe5cff : SUCCESS in 8m 46s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/97bb2721d73b4e92bfdbfed9dc7089b3 : SUCCESS in 11m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f1d07f72cd524ef19e288191eded85b5 : SUCCESS in 5m 17s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c48968622adc4f198822c3d322574105 : SUCCESS in 4m 45s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c0a24a92b2bd4257ae0a6da77f7ba9a1 : SUCCESS in 7m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c910269195b545229b28aa16aa831ed1 : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0970cd19d2946709d37f174185ecfc9 : SUCCESS in 9m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/2e76f5abf70b41259e85f786137112b0 : SUCCESS in 8m 16s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a0395dbbddd64285a10fdb0a9104f2c7 : SUCCESS in 11m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/81f5b14c6d84407c86bff80401d4ee1b : SUCCESS in 5m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/8ef21a5e40804175ab203de16e94d463 : SUCCESS in 11m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/207a04542753479aa506749e0ebbc042 : SUCCESS in 4m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/84c16982a25945dc8821065a10a0249f : SUCCESS in 26m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a27e210c582b44409f967b1ebb9ab7e3 : SUCCESS in 11m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/55f5ff1a58dc4828a093144cfd32c631 : SUCCESS in 14m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/19bdabb6fadf46b3baf9a69ae6dfacf8 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9630560760b442d2bf42aa294dff976d : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e9c3b3ac5dfa41b595c7eb41cacd8bf1 : SUCCESS in 23m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9b5d7bc17bd04793b7a41a15e4cc9428 : SUCCESS in 6m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/deb6e3614cf247b98dc188ea677152ad : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56e2e7f218ac426bae39252c3ec15a99 : SUCCESS in 6m 47s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/85e2cd6c68534e749bd62bd06412a030 : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ab496a1399143ddad00d7ca9e774f3f : SUCCESS in 51m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bb083c187f44660816217a6b17d216b : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fae98a80870b70ef73ca1f32db493de2d1ed18d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 18:32:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b20088ec1e54bf563378d0e708a580fe9a7cbce1","revisions":{"b20088ec1e54bf563378d0e708a580fe9a7cbce1":{"kind":"REWORK","_number":1,"created":"2023-08-24 09:10:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/892657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/892657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/892657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/892657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/892657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/892657/1"}}},"commit":{"parents":[{"commit":"1b1719fe98146f53d299f67a42fe35f51cbb2101","subject":"Merge \"update constraint for python-manilaclient to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b1719fe98146f53d299f67a42fe35f51cbb2101"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-24 09:10:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-24 09:10:39.000000000","tz":0},"subject":"update constraint for os-ken to new release 2.7.0","message":"update constraint for os-ken to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iafd184cc230836d152342e3e48db8bc05f1486b9\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nChange-Id: I04025615740283e098fa623b7ac7413c1d9b876d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b20088ec1e54bf563378d0e708a580fe9a7cbce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b20088ec1e54bf563378d0e708a580fe9a7cbce1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
