)]}'
{"id":"openstack%2Frequirements~892660","triplet_id":"openstack%2Frequirements~master~I4f56b86722875e8297e60fcc60a40fbe3972a535","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4f56b86722875e8297e60fcc60a40fbe3972a535","subject":"update constraint for osc-lib to new release 2.8.1","status":"MERGED","created":"2023-08-24 09:31:48.000000000","updated":"2023-08-24 18:33:53.000000000","submitted":"2023-08-24 18:32:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892660-new-release","meta_rev_id":"e0f31fb63633e8460aae1b7a42eb94f17cdb1d64","_number":892660,"virtual_id_number":892660,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-24 18:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-24 14:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-24 14:34:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-24 11:18:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-24 14:34:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4f2c2e0fd36239d728ec9b3d799d638da372422e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-24 09:31:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1a1467457e4f8a64facbcb5b5361ec758f17faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 11:18:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d11b40e9ee3840faae67439f954f356f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e60f64d5ede496686cf9b4659b3d1b9 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17864ae5c10c4d269b8fbfa02edc1360 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e26430c680144f89b3c8f7a5681cf07d : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0b3d9315b164052887ebe59e848a156 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61f05ec302f54ca8836deab7db314d4f : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36f04c0d728b45c3b214b3048cccc017 : RETRY_LIMIT in 2m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82ada620ae9a49c28795e1b336d728f8 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15332a4e3f084c58a8de4ef0e07d6aa6 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/81c69c29e0884505aa792de0b51fcb22 : SUCCESS in 8m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ae7e17c6ba46413aad225a1db86f8e1e : SUCCESS in 11m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/93141f5a290d4b2a96a649b71d4c3345 : SUCCESS in 7m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6e8c369a6f694f56b1b30d3f5ab260ae : SUCCESS in 12m 46s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5425ef273dbe4026aaf2e504130ad81d : SUCCESS in 8m 06s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/93963e2de02c43b6a0713c4be6424971 : SUCCESS in 5m 53s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/adc3a068fd2e40d7a7dae7a0d1e31c83 : SUCCESS in 4m 44s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5313a688d445493889f6fadfbb436dd2 : SUCCESS in 10m 06s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5e08df9e2ad0445ebe17fc7d3a8ceaa1 : SUCCESS in 4m 32s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/00c3a8160800412386fccae74f24dfe4 : SUCCESS in 5m 10s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e62083e706ff42488ab22d650e6e2f36 : SUCCESS in 7m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/abff1764485f4ab686d7d844ff35f6fc : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03b3c7ea7f434ea581fc160645d1dc00 : SUCCESS in 7m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/335634343e6d4a818ada44c420002c9f : SUCCESS in 8m 15s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/77891ff97d04495082436fd0115cedce : SUCCESS in 10m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cd17b4d73c754587ac9b4b888dab226c : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/79c7862f0d704ff5a1c939319a655951 : SUCCESS in 8m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e272aff9b7904ab5999afe18976c1fd3 : SUCCESS in 5m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/71653606d3784ed0bd2304aa1ccd9fd9 : SUCCESS in 11m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2a0d084f402649689c74a124771e9e56 : SUCCESS in 27m 21s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d3ff90103b924a179b8ce6d81245765d : SUCCESS in 11m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/66012faa70f442a08f77086673088329 : SUCCESS in 14m 16s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8415f95b86a64998bf1f2d2c697dbd51 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dfb6dd7387cf402eb65cde7ec32e09a5 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/289aab34dd414528a0ec3459a232061b : SUCCESS in 22m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/444f9949ed0b43388ea31e3219dd51a4 : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4750812219044a26a3abefb4de66e728 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ae747778869456494bf1b28bb0e6193 : SUCCESS in 7m 26s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e1769e436f4d4f6c8983252996948a15 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28ee92ca89214818b71bd709a8daed27 : SUCCESS in 54m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d40ea0b0ab1644b68a9e5c1c2a74ea6e : SUCCESS in 1h 38m 25s","accounts_in_message":[],"_revision_number":1},{"id":"96f0acd9abd600cef603c6877fcb9341bce2a7d3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-24 14:34:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60de910c952ffa286f1b3dad1762671152a6afbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:35:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"330dff12a3ced3495ce29d58cf7c819a9d0aafa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 18:32:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb00765ec7c4b009cc94140f49fca70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03f02d43102400096cd80875b75e743 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac90ee44c9747388b9d734a032abbf2 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/223019005d874088b6ed94493607c368 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4157123c593c443f8803d74800170566 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41d4a79265df4facab6d36e8003da799 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85cc14e538ee4098910b1896200554fa : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49a34ca075f94a6285d9012d8987f944 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/19114edb8b2748c4ae6830e266095e8a : SUCCESS in 8m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0c24f57cf0594610a0dfc59056b73fe7 : SUCCESS in 13m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/be921565e0124a1fbe0e8454fa380a73 : SUCCESS in 6m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/46a6701bb4ad4c05bf66fd56e7b0c19c : SUCCESS in 13m 14s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4a050127e909467ca6299caf13fe7a00 : SUCCESS in 5m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/76cc2218445243c48f0d6e6bc83304be : SUCCESS in 4m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bcb6ed9289ce4971a372108faeb42215 : SUCCESS in 9m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/23bef0e369174f74a2ca0d219f5b602c : SUCCESS in 8m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56e2573d73ed4f07935478fe696d2f44 : SUCCESS in 8m 25s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1a51c50913c9410c8b0a31210579b2d1 : SUCCESS in 7m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/447e1e7443e04a87b727236ca4c1cb18 : SUCCESS in 12m 55s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a71a077070834410a1010a5b68486811 : SUCCESS in 4m 00s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a4c84307d69b4cd488a82238b6795f30 : SUCCESS in 10m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/18d78ab1efca4a2fa03dba9ff9bde23b : SUCCESS in 5m 01s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6ab6c1e3f0f841e2911f1e1f90be26a6 : SUCCESS in 24m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7049ba245f1b4a4cbd9f953a5d149464 : SUCCESS in 10m 17s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/42e73638bb6a4119aabb93a26aae75b9 : SUCCESS in 14m 34s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ce1bfa5546ca47cead39db585c43b9ab : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8721edf7f67c48afaea485926e915b19 : SUCCESS in 5m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/353cfd92e17044a68e5d64f3ae9d40cf : SUCCESS in 25m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a6ee0462b10244b3825da3eabc044b6c : SUCCESS in 5m 30s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/349e65d377a84ac4b4976103fd4d688d : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b264bb5afa848b4bd20f18c5274458a : SUCCESS in 6m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/7494301ca51f4d588b581693ff321ab6 : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e603cc48493041f9a171198d25ac5389 : SUCCESS in 55m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f522f4d0e72430b924f362381405efc : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c187b815624645740cff49f8e409fc4da62e3928","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 18:32:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e0f31fb63633e8460aae1b7a42eb94f17cdb1d64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 18:33:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af7f5c8683ea48109309fcaa59810dcf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7445e3a8fa4485b6be832336419d45 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c019d16aaa0b3045215bae0b262b12120ff38b74","revisions":{"c019d16aaa0b3045215bae0b262b12120ff38b74":{"kind":"REWORK","_number":1,"created":"2023-08-24 09:31:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/892660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/892660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/892660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/892660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/892660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/892660/1"}}},"commit":{"parents":[{"commit":"1b1719fe98146f53d299f67a42fe35f51cbb2101","subject":"Merge \"update constraint for python-manilaclient to new release 4.5.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b1719fe98146f53d299f67a42fe35f51cbb2101"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-24 09:31:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-24 09:31:46.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.8.1","message":"update constraint for osc-lib to new release 2.8.1\n\nmeta: version: 2.8.1\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: I29351ded20afac0575279766a0f2d622c28b9496\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I4f56b86722875e8297e60fcc60a40fbe3972a535\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c019d16aaa0b3045215bae0b262b12120ff38b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c019d16aaa0b3045215bae0b262b12120ff38b74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
