)]}'
{"id":"openstack%2Frequirements~892857","triplet_id":"openstack%2Frequirements~master~I198da441e3b5fd225b61bf49900d15a7dec97464","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I198da441e3b5fd225b61bf49900d15a7dec97464","subject":"update constraint for openstackdocstheme to new release 3.2.0","status":"MERGED","created":"2023-08-28 08:25:44.000000000","updated":"2023-09-01 01:04:50.000000000","submitted":"2023-09-01 01:03:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892857-new-release","meta_rev_id":"3a6f39c782c43729e22ec3c8112f47a92c0622ee","_number":892857,"virtual_id_number":892857,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 01:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-29 19:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-29 19:33:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 10:12:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 19:33:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-31 20:33:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"755c74e31c3b7f62da833c8caa223e50b0324aff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 08:25:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec2cd057664c750704c1bccb2cda24cc6725bd39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 10:12:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b74257331945618f77693e275d34dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45dad082432842b5936cbd0e8712d8a2 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77406f7b37994187868f4fac388eb0d8 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a146f301cfa341c881c65f2e4078cba6 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ef056979d7942929c6f75b652671b90 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83391a38355b4d42b86475d7d5529c44 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19af199f416146e5b7a017c100ac1929 : RETRY_LIMIT in 1m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8403014b7a2147b9911249f11245b960 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39ba55fadd6846a8a00e60fd16fb7d87 : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2a82e14352214bd2ba7a6acd8ed3d555 : SUCCESS in 8m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4e39f27f25724e88960eaa0a0e72e36d : SUCCESS in 15m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c4fdb5f5156840b38668325fc8c1a617 : SUCCESS in 8m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fc477089b914492e823557b762b469a3 : SUCCESS in 11m 11s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ec68af4d9c174b90a2a253f6d83ca366 : SUCCESS in 6m 52s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d71311f1b2114d7490ad2fd1dd698499 : SUCCESS in 4m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/98dc3f206061461db6ba7efe61f532f8 : SUCCESS in 4m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/610ece98e59647828e754e3d1eb5bdfa : SUCCESS in 11m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b80f7c76265c4a069fd53e53a757965d : SUCCESS in 4m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0ace9926297d4912a448e6c7233b0a9d : SUCCESS in 4m 44s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/e6a81890166649c7a7a139bb1908f84d : SUCCESS in 11m 26s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/aad956bca9de43b79c5164acdd0cf29b : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e85a1b14106d4fd3a4e4de221cfc529d : SUCCESS in 8m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d552895de8e94f699b33565c0bc360e5 : SUCCESS in 8m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5067823533854c529373ba25b1ad8f6d : SUCCESS in 11m 58s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/673322dba36a49a0b4491c8d44bf62ef : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/36db75a7ec9d46ffbf7db3d28c15872d : SUCCESS in 8m 38s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/76b2f2d544574d9c93f603ccb8ba0f39 : SUCCESS in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2a6ddc6bcf3f44e1bec8483cc422f7d8 : SUCCESS in 9m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/a265414a77264ad4a694a31231803482 : SUCCESS in 30m 31s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ac6f7fc0938e40b587b14482061b0670 : SUCCESS in 10m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/b8db6ea9eb6a4d6d8f829add6a8e0281 : SUCCESS in 12m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8c258f43a8804d0b905acc874ac0b2fb : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/acb9925c23ed45298f564e18f8286571 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c81aff1025cb44a4995d427647f3eec1 : SUCCESS in 21m 49s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/750589cd14b64946a017e23e56c5fb68 : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fcb85df52928465caba8333bd9dbf333 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f7b7644c14824ea788f30f92354e1c71 : SUCCESS in 6m 47s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/89bba5f799a346c2bb132fb99f538d97 : SUCCESS in 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0892a2674d1e4a8e8c7dc1b1bce49a60 : SUCCESS in 1h 02m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab1289f91120461e9026ff39fb57bf2e : SUCCESS in 1h 41m 17s","accounts_in_message":[],"_revision_number":1},{"id":"87b745717c3e9a68a9c6c8b7014acdf8d6a34aa8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:33:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff1fff1289ca3e76cf935b63762cccab324ec0bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:34:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d38f7455efe46bd521ecb9926dde22654f78364","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:27:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/effc75e1a27e43398fb503f148bbdd45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc81ec6d10d0440290d052cd1c784e6a : FAILURE in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4847bfde0a494b748ac7fd7d0ffb44f9 : FAILURE in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01800d051c064d11a84ca0229dad0b78 : FAILURE in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c280086ab164ab98a1c4325aa045041 : FAILURE in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab65a4a345874a95abfd6fd4bb6c7662 : FAILURE in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/287efa60fbf34a8d81b99975dddcf692 : FAILURE in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/977f67b0d94a4d32a7cebecec0400afc : FAILURE in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5e933fe652943aeb506c029758b14fd : FAILURE in 8m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b792039695404933811dd2a8133e1f3d : FAILURE in 11m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b26ec7a8664c4c0088359eb5cefba3e1 : FAILURE in 7m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/78502206586c401c87493d33adc05f31 : SUCCESS in 11m 30s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f0e30af9aea84a34aa4b7fe547017d4c : SUCCESS in 6m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/dc3fa62b847f4986a2e64d017a9e0f73 : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/be527d1138c64e37a50c1fd5905da2cd : SUCCESS in 8m 34s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8ad3f7080fb44f43899eebe00c6b3cea : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c137266df014634a6403c05b546bc44 : SUCCESS in 7m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/db56625620b54f7ab82c92a02cda043a : SUCCESS in 7m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6a6731c153d2400cbd68c43722e08e28 : SUCCESS in 15m 32s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3c19a43149a440b284fa92b2cdfed9d2 : SUCCESS in 4m 08s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5f3e6c9deca04b84bcd9c82a1c747612 : SUCCESS in 10m 57s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b9829149fe604d499ef4cee884462699 : SUCCESS in 4m 28s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/036952a41fea4109b10a42ffb2856589 : SUCCESS in 22m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bbf086d32f5b49889ac91cc4643800d7 : SUCCESS in 11m 29s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/22fd1331c75e4e72a207dfde93cd942c : SUCCESS in 14m 18s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c90104571ffb456eb8cc04c2337dbaed : SUCCESS in 4m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8477e9ced10d4db28e61813e09ea0b12 : SUCCESS in 5m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/adf83f884a1c4b4f809104f98ebd414e : SUCCESS in 23m 51s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/30e2a79cf8814c779aed70627e67e637 : SUCCESS in 7m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/ea2ad859722e4190a0184b2ab732048b : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29bf775fa8f34aac9e6be65b1b4e3b7e : SUCCESS in 6m 50s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e9639a755ae946da81568fc96b65b0fe : SUCCESS in 15m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb1429fde58c4ae5828b975e22e6e63e : SUCCESS in 1h 04m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b6765db06f346e7a80c012daefa1c34 : SUCCESS in 1h 19m 19s","accounts_in_message":[],"_revision_number":1},{"id":"655a55934deae14eeb487a9cd7e7ec214c3f60bc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:33:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35eb112cecf080df90e0d156bab7443f2f3be176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:29:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc700187fd8e417aa288a6f287a1c8f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/823d3db6437945adacd834e0478f471f : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c5219a9cc3e470c9087470d2c024596 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87f833c9d8af426a926cdf6ed6a68d23 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e68867a37f640aa8d3bf76421635f14 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb4e0bb418204adebd24a7bbee2bf16f : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fadfcf801c940358efbc1cf847c3349 : RETRY_LIMIT in 1m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2c7eb3902ae4f18981bca04690f268e : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71f94d29af2541bfa147caadb8242c49 : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48ff11dbfb7044bcb48123477dffbb30 : SUCCESS in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d09c8e72da1b42959da7265c6e0bb2f0 : SUCCESS in 12m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2c3ac76e8a154b59a4a615db35468316 : SUCCESS in 8m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/681cbb6b4c0b4d5e9e8af175a610ee2c : SUCCESS in 13m 19s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2a9dc1f3028546cfa43cb70b6c348a60 : SUCCESS in 7m 42s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/77b8585df171498d86d95f968b75606a : SUCCESS in 6m 11s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fee0e9c98ac14c54a1eca2fc54198a3d : SUCCESS in 5m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7beba3adec1041f08509e4e572d860cb : SUCCESS in 13m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ff299a66a839422fbc55ede822aabf1d : SUCCESS in 5m 00s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/02518cc093e34e029abaaa7276c57d45 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/17fdae9d0a4d4ee2b91f402423e409d8 : SUCCESS in 8m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/95f00a3a41b34e6ea466462e46a5b3d8 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f638d420e75744fb9562048dbbb517ea : SUCCESS in 8m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5d76fe1a19c449dc8cd37b5e9137e784 : SUCCESS in 9m 46s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5e88e2a58afc4e4e9035b36ae2121cd9 : SUCCESS in 13m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/103ddebb328247c2acc49c1f5bab3ecf : SUCCESS in 4m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e8f6a1f3784b4c7086fac73659ba4595 : SUCCESS in 11m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/dda6a0841b384fafb3557ee5f8dbe239 : SUCCESS in 5m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/94ccbb6255b842a6ab132eb91b0b6029 : SUCCESS in 10m 26s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/d696c725161c45c8a1a19dca4b33eabb : SUCCESS in 26m 28s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/eaaaced56821484d9eab1780f6fe0cc8 : SUCCESS in 14m 04s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ae4df4fa56da46db8fdee0935dd5fd3f : SUCCESS in 14m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ad22c5b011f84cd4a73a9253c997bb79 : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2855b290681d4767b3e3f235c8296e0d : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4374eec013eb4e95938f062b99e5ab18 : SUCCESS in 24m 35s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/3b144aa392534a10b4d7668213dec188 : SUCCESS in 5m 29s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b6512d5330834c3c908246e783bb10c6 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6829866b74b54819a6a59eb6742c8e72 : SUCCESS in 6m 25s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/fcdb6d88c9d64232a6346b27a4cb9be4 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c09392f4b7484ee78446994fa2a86b3d : SUCCESS in 59m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62eb6c55401c4035908d11a8841268ff : SUCCESS in 1h 43m 06s","accounts_in_message":[],"_revision_number":1},{"id":"857aa6e63db23340422bc29ad8b175ec1d8e56f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:31:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4108cc4b82efc90f7f0ec0fb98916747c7400bab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 01:03:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f4a3bbd5a554f3899d31827af730f6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/549075eede0c463d8374a67b758efe8e : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/083c1877211041cda4dd578090fbec18 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2636e9e7bb4f4f72b8ee782228df7aac : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57443f27a14346419051fc8f410324ef : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9d486bb7d904525a63d6b7b8d0f56ce : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee32a36dfcc44a1cb79afcfde0e1a833 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/235784f6843f43d187f739faf41d64b9 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bad7058a0b1241bb82132ee42da8e6d1 : SUCCESS in 7m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0259c07811094f8ba4bc7121b74ee4d8 : SUCCESS in 18m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/510c36acfbc941999b28fd12b7db4dac : SUCCESS in 9m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/36bdf4ca7187418ca1aa243edd0d92b1 : SUCCESS in 11m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9bf78090cb76459f810dc4329d5357dc : SUCCESS in 5m 11s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5e45a83f0a9c4e3d89ead17bb50e59bb : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c435f4f78b024c60aecc237d60b75697 : SUCCESS in 7m 56s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0ab7fd731869458795f6e696736540f0 : SUCCESS in 14m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e04a8b2bf7242f59deadcf3d1541e7a : SUCCESS in 6m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4a1fd533aa6145289f3a91d8430af814 : SUCCESS in 8m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5a25de59870649aa86895bd7f5c44095 : SUCCESS in 11m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2d2b327dbc9b4706875579b78999ec1c : SUCCESS in 4m 33s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e8fcb9898c5d43298d823c04452f6097 : SUCCESS in 14m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bb011442522a4daba1f005f9beb5c6c0 : SUCCESS in 4m 37s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e73de8edb6334dcb9849ce82026da38e : SUCCESS in 18m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/74237fb7b167466083276e050b760bdc : SUCCESS in 9m 30s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3453b1d0db5d4b2c8be2e7098c74c324 : SUCCESS in 11m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/834ffb0146014752bcaa95664b09f170 : SUCCESS in 7m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f70dcda7519d4f488cb1c21192780560 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e86eb08f50448a9b99915b31b11e1f3 : SUCCESS in 24m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e5f1ef87e630426eb38d5efada5fdccd : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cfcff819fc5e4da8b4438f45032d5312 : SUCCESS in 10m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b7cba0faa9174072af94d1fec40203a8 : SUCCESS in 5m 29s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9c34a2119dc4444bacf4f1452ae5b271 : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69fa653b1d33452e80c407a23cf77a44 : SUCCESS in 42m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e10a6ea3e05f4a1aafd2ea696a733b90 : SUCCESS in 1h 32m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6b06a83a809e1cb42649195cf309aeac40fda580","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 01:03:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a6f39c782c43729e22ec3c8112f47a92c0622ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 01:04:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6314d1ded94509953ca87646a6091c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8813c5a85b8145cbaaf5177fbf17ccdf : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdc967ad286938698acdd3e481bf095369a92565","revisions":{"bdc967ad286938698acdd3e481bf095369a92565":{"kind":"REWORK","_number":1,"created":"2023-08-28 08:25:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/892857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/892857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/892857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/892857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/892857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/892857/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 08:25:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 08:25:40.000000000","tz":0},"subject":"update constraint for openstackdocstheme to new release 3.2.0","message":"update constraint for openstackdocstheme to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I666db7404214b292e02f09b3f83f99c108bec179\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I198da441e3b5fd225b61bf49900d15a7dec97464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdc967ad286938698acdd3e481bf095369a92565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdc967ad286938698acdd3e481bf095369a92565"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
