)]}'
{"id":"openstack%2Frequirements~892882","triplet_id":"openstack%2Frequirements~master~Ic6cc1afbeacfeb6c4060d2493a38015cec1620d5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic6cc1afbeacfeb6c4060d2493a38015cec1620d5","subject":"update constraint for os-brick to new release 6.4.0","status":"MERGED","created":"2023-08-28 08:49:32.000000000","updated":"2023-09-01 00:06:05.000000000","submitted":"2023-09-01 00:06:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892882-new-release","meta_rev_id":"6a1be32490c4b3bae3664b336898a77e67d946f4","_number":892882,"virtual_id_number":892882,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 00:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-29 19:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-29 19:33:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 10:56:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 19:33:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-31 20:30:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"c44b022600d45a0d6d514a1d05c30c704bebc021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 08:49:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02a50eebcc68428d21b6298afeb2938d43654726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 10:56:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56cd9af1ac34c2eb7c0385177ad67db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/096d3b0ca351429d8f171bce42be4938 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb51a49bf3849129b5fb5c5314d1a53 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bac6b7ceba1424fb836a7d905946cd5 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56402b79318b478e9b65930039b8ac36 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01b13322b86a4e4383a187f5d3243f10 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b647141389547d7b5cc67f186952b86 : RETRY_LIMIT in 2m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0811720bea35413898fbccf649a05715 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6fcddd14c9b44d086d623549efa5479 : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f0004ac0e1b4391a12484570fab3928 : SUCCESS in 8m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8fd567757f144fbc96518248040f6881 : SUCCESS in 10m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/eadea2b748c0421caa7a8c49018528c2 : SUCCESS in 8m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3b46436606be461793686bc0ba3e5db1 : SUCCESS in 11m 56s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/6a5956ad6ba54feab915cea8d5c4f08e : SUCCESS in 7m 51s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e18919bf9c4848f581b0b89edd214155 : SUCCESS in 5m 13s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/1a8e078233524bbb8c3cf93269c37bed : SUCCESS in 4m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/00b5774e3aa94efaa32953201ea7ee4f : SUCCESS in 10m 46s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cba08b1b627c479196c48e6f6f2a0b3c : SUCCESS in 4m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b62d5f576f4943feb7e39ac99a26ca8a : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c812b462b167453dbf1920f602b3ce85 : SUCCESS in 8m 53s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d9454716da0b44cc867101e0b5748ec6 : SUCCESS in 8m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e856525074245bf88b1a586de2c1ee1 : SUCCESS in 6m 55s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b5401ea154cc40d7bc8225a2b431c6c6 : SUCCESS in 7m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/70569cb8732a4e0ab7cc898804151b12 : SUCCESS in 11m 24s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d6d5bafabc574c889e4cca3d63d2c675 : SUCCESS in 4m 29s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3e11179d2c934a6293d6cea990682e15 : SUCCESS in 12m 02s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/923d296bd24c40e5b02eb1ec841723e5 : SUCCESS in 3m 54s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2d7cc9d97ceb41e2ab431685241a802f : SUCCESS in 12m 20s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6cf08a163bd74ebfb952471b4d5e8685 : SUCCESS in 20m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/972eab01e3584f8095c8ee25a7fae4f3 : SUCCESS in 11m 33s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/63024a03b2f043fd8dd9ec040572d954 : SUCCESS in 13m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/70a2ba47208d4cfbbf004da045e4a5ad : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4347bc6d3324c949bc6a8c009b3ad53 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9727746d65194d17bc3340beb587193e : SUCCESS in 23m 59s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e01f62207e3d44b9a6ade306f3fc3060 : SUCCESS in 5m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/389a8449e27d410aa48cd08d12f49031 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b966c7620734489b395ea0a1eacdf8b : SUCCESS in 7m 17s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/44e9a93a715a4ec0ab168a065b4af066 : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0e63f6545514cf38bd61d374f2f3d75 : SUCCESS in 1h 12m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3bfd75ca1054a92b997185565b677e6 : SUCCESS in 1h 22m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7ca40b31023f422fa77f07b926299b1448667e7a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:33:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81eb07b6caff70020d32483307999267d0b60191","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:34:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e7caab7ca5343acd1584649c1776d2687504c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:30:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/875cbb0ad60946229ae6591ad7c933cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fee4deec30a4f6ca5531f075cd500a8 : FAILURE in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ced1e6f6447480a807cecd4a132a457 : FAILURE in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd68203243064d52b9077c0a22a51272 : FAILURE in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d26b759b1c72491e89d47326f9a001a3 : FAILURE in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19e6900141b84fb5b6051e1547e2b9f7 : FAILURE in 2m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6bb6204c2bb743fab364a24f24a0f68c : FAILURE in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd3c926fc03446d3a11c64688cfb212c : FAILURE in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/919a5aac9c8d4108886ab0a06ea4a397 : FAILURE in 7m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/818337fa515b4a5a84e3cd67fac5278b : FAILURE in 9m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5f5476d22b784b18935d4d13b2c232b2 : FAILURE in 7m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/d069bccccd9d46fcb933604e190be390 : SUCCESS in 12m 00s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/6101b304161d42d9af880a37f4c5aeb0 : SUCCESS in 4m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/589a1ef61cc647858b85fd78eca1a684 : SUCCESS in 6m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d8883cb024d547feac781c95c599e7f3 : SUCCESS in 7m 53s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a5401ab44a914d7cb2cd7538b3e0b31c : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad6160f45a4f4646be8a337e687ead33 : SUCCESS in 7m 44s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/80a107bd3f8c4d3eb63b3a414d063a7c : SUCCESS in 9m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b32f4cc82d7b40628619b4c73c436abc : SUCCESS in 11m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e8890bc1296644ac91d5c9a3b29ae216 : SUCCESS in 3m 44s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6439d876ddfc4d6698dcd6be4cd0554b : SUCCESS in 9m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/efc78d48b3af4a9ab8770643bec4ea4c : SUCCESS in 4m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4410b997083a41e199fceb46e7a44ca4 : SUCCESS in 25m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d76694a622144f38bd0fe6b1850a41fb : SUCCESS in 11m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/33fb2c981d6445f29c8cd306dbb08249 : SUCCESS in 11m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/350ad56a4f664c1eb4af6ca804c28faf : SUCCESS in 4m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/15049c33660148399f6422dc5f776da6 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4b6347dcfcf4a4fae32c3abffda231e : SUCCESS in 23m 40s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/f3442714e98e43b5ae0142b2b2980642 : SUCCESS in 5m 33s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3b1b868bbb954516a9d68de639a06b86 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/359016aecf104ec4ac8ac1d629874104 : SUCCESS in 7m 04s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/83d012fec335411a94d0bb477c2a8a67 : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e96bcf8fad1045fe90457793fc2fdf60 : SUCCESS in 48m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a29130cf56a40e2a096b0e18f5d2820 : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0548c00f55fa31fbc4aad4fcb03481db99bf48f9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:30:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd98f1f4a6c04fc816baf06647369d81a53ca590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:18:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a01212882d6549c4bad7f74ef7a2fb23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa12e24ab964b2bb938eed5267fc545 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b4b228b70341b8a4d55af9f0605167 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32f3075ad6c34f7fa5b89831a7752fb3 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7717fb4ef69e449c90961f71987d1e88 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31a93308580f4b8b9a6be513e31fdc3f : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d05f65fd49a5405082c8d2a24c3d69ec : RETRY_LIMIT in 1m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07f48709b7ab46fd83aba8b5a6d25600 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f19c15f3a2ec483695e04496d5e8099c : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/29586530349d4ccd868fb29fba036492 : SUCCESS in 9m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4358dd0297754262b86d28445c048428 : SUCCESS in 15m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aa01131e5d1646eeaacc36e9390c9ced : SUCCESS in 11m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6f6e975eecec47af91aa0799646cea22 : SUCCESS in 11m 45s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/60208e75de6e433e9ac4033414a7078f : SUCCESS in 6m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/482f648ffdda4e8c9fc55255c1aba8dc : SUCCESS in 5m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5e36831146e546ed820ae444b05cb9b3 : SUCCESS in 7m 58s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/37cb948e53f34f2daf50cb11ddeea140 : SUCCESS in 10m 59s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2dc96d4022034930858583f4f4b93d57 : SUCCESS in 5m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/85da89378b2d48959756026cbbad8a55 : SUCCESS in 5m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/282da2189b504f618ce756642add78a3 : SUCCESS in 7m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ec067d3304244b0394fae4a759f04160 : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a98bfc0a40d4dc8badc2a1064cf59bd : SUCCESS in 8m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/958b16883637429eb99b906f6434a25c : SUCCESS in 8m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4a829c7e21f2432bbd555e1a03f2e973 : SUCCESS in 11m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d611e4cda90348caa9930766fc4f66b7 : SUCCESS in 4m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1931d62ede614afdb59c26f48d44fdd8 : SUCCESS in 11m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5acfcbeb1eaf4c0fa78934f1dc199b64 : SUCCESS in 5m 04s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/06a08ba540dc483781efe5607f1784b0 : SUCCESS in 9m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7679864ad46c4a1884ea7c08dedc3216 : SUCCESS in 27m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/27e4cd1cca6e4971b91f951ea0ef4161 : SUCCESS in 12m 49s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/18577057553748a1ade393ec59e6dbe7 : SUCCESS in 17m 50s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b9a40a9c0f074f178413c90892313373 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/822854f50d794039afb70302003aafc3 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/efc33eaf42774b59996ee5f3a9e7ba47 : SUCCESS in 21m 08s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2cfe895b0e2b4af38c699de3ac325b16 : SUCCESS in 7m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0145429d0063415497393dd1e87662a1 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e8693f8a58f4d69b699f087faee32f9 : SUCCESS in 6m 54s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/718faf7f75854abf897402e82c5c4b56 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a741fd9ab92e4b11b467990552aa371a : SUCCESS in 1h 09m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/576091077fd14ae8ac32326c2f1472ea : SUCCESS in 1h 41m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9c8c864435302079a2a5547a7d5a5836c83603f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 22:19:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ac49a6bc76213251d98a87e342588477a4735c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 00:06:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62f06108765846ebbc779d585cb7b75c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6562b65cd747c08cc089b12ad1f23b : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971be35c52c74e83838682eb9120139e : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96b7139927fd43a39c4196c033444391 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da6e5f7a38f64fab8783e7278be7e001 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/648ac4a25dd048afb8d63e3481f60163 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78496245ad6a4212b0bf71a4557af4d0 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4f1906483e140ef98dc780df2fce4ab : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8f76576f7ebd41eca06c818f012178a6 : SUCCESS in 8m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6622fe363afb4eebb51cceb680f51a3c : SUCCESS in 10m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9e91a993e6fb43d8947f1d242ab1f5c2 : SUCCESS in 7m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6faa569c81794151a833af419aa55209 : SUCCESS in 9m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d6fe5c7d24e04c7499497aba4a512466 : SUCCESS in 5m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/97439a2a208c4c9d99bbd778a1516dfe : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c14e0e6ae4af43a58c1790ea7aa03434 : SUCCESS in 8m 05s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/94bc32817a10419a81144fe19c902722 : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0286a5d83a73485bba0c8c5691cc8a6e : SUCCESS in 7m 59s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c959e0c18848450dab91fc0aad9858fc : SUCCESS in 8m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/24878208bb75494d9aab9d5edd4fa942 : SUCCESS in 12m 09s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0e4cf502d45d4619ac6c9a72998c652c : SUCCESS in 3m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4f55c1bdae6e45d99b9aa8be348e6144 : SUCCESS in 9m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4bb5703d87e741009a38b55fe830ea02 : SUCCESS in 4m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/275f4bd3017b420283cafd8c1e61bbc0 : SUCCESS in 35m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a06847081c5a4635ab837abc3f86567a : SUCCESS in 12m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/de8a34f27c4243219b2af768361ed8e8 : SUCCESS in 16m 03s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/36b42971aa26474893ffb7bebf60f6f0 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d333952c9f284c698c12271bcfc31f33 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd97953aa00244feba79a82c4a0f28f2 : SUCCESS in 25m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e0e17f5e0cdf45eebd5f6ea058c79441 : SUCCESS in 4m 36s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/af374223395148f5bd0a3db58f600e83 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/523a9839ac244a09b07b7da169d3473b : SUCCESS in 6m 31s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3a95f927d25f40e4a97ccb84016b9341 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33787679e0424bcf952db67ef33cffbe : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b8f1752ff6f4c33b06d712fa9eb8608 : SUCCESS in 1h 41m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6a1be32490c4b3bae3664b336898a77e67d946f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 00:06:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18268c5dc108550b87f3f7a16019d670d8788ddc","revisions":{"18268c5dc108550b87f3f7a16019d670d8788ddc":{"kind":"REWORK","_number":1,"created":"2023-08-28 08:49:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/892882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/892882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/892882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/892882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/892882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/892882/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 08:49:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 08:49:29.000000000","tz":0},"subject":"update constraint for os-brick to new release 6.4.0","message":"update constraint for os-brick to new release 6.4.0\n\nmeta: version: 6.4.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Icd4ca8c37cd958b7468c1afc6087e32464d21732\nmeta: release:Code-Review+1: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic6cc1afbeacfeb6c4060d2493a38015cec1620d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18268c5dc108550b87f3f7a16019d670d8788ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18268c5dc108550b87f3f7a16019d670d8788ddc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
