)]}'
{"id":"openstack%2Frequirements~892891","triplet_id":"openstack%2Frequirements~master~Iaf4abfcfa78933e923173714f3387e3b8757f409","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-04 15:43:14.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-01-04 15:43:14.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-01-04 15:43:14.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-09-01 07:05:35.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-01-04 15:43:11.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2024-01-04 15:43:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaf4abfcfa78933e923173714f3387e3b8757f409","subject":"Update OVS version to 3.1.2","status":"ABANDONED","created":"2023-08-28 09:50:18.000000000","updated":"2024-01-04 15:43:14.000000000","total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"fca51b94f9e70392456211f8ddacd86123b37683","_number":892891,"virtual_id_number":892891,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-02 19:40:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2023-09-02 17:56:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-2,"date":"2023-09-02 09:44:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-11-15 18:24:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-09-01 07:05:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-08-28 13:44:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2023-08-29 15:15:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2023-09-02 17:56:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2023-09-02 09:44:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 11:43:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 13:44:18.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-08-29 15:15:04.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2023-08-29 19:33:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-31 20:36:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-08-31 20:36:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-31 20:36:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-08-31 20:36:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-01 07:07:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-09-01 07:25:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2023-09-02 09:44:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-11-15 18:24:07.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"}],"messages":[{"id":"45e5fa11ed7a32804ec47e09bde0ce0e45e2b90f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-28 09:50:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72640f64f5173a071136bb733505d9ab9645ee9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 11:43:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/facb911dc5d24450a8779b8f225678d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc7047c8ce64a64aed952da6fea4fb8 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26837944efed40428b44b2d3d6df208f : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8de1b7cbc0314fb28be8face96f070f3 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c723a57dcb90410fb417e603b06fff03 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/862598f4f2324d7894e4735254537cdb : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/088b3fb85f984a9ab17a224fabfbee5b : RETRY_LIMIT in 1m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee67ace21eb844bcace32266bef320c2 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ef4be6d951b4fc894558d1160dde7d5 : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/130081ba3af34888bcd2338e8db4c255 : SUCCESS in 8m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bb218475f22a4cb9a2c6182111f167dd : SUCCESS in 12m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/77307a44c5294122b5c92fee180c2a98 : SUCCESS in 9m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e212ce68daf54897b8bf6ad999518d52 : SUCCESS in 11m 16s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/943407c4e6734c13a33f328c4a764ff5 : SUCCESS in 7m 30s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f098f02c0b594b548f802f86856a59ae : SUCCESS in 4m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9bd9a0f0a56345288da9440d48126e2f : SUCCESS in 5m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/85e4053e736b40be95d184efcef6c9bd : SUCCESS in 11m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fb42948907334fd09890c19049983c8a : SUCCESS in 5m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/731f545697584e56bd48eec597a47bc6 : SUCCESS in 5m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b39fbd024ed641ac8fdae6e1d5857355 : SUCCESS in 6m 33s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e0e6d2aa3f784de687557dacbc63b19e : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/600a5d4f865541b7bb58e05c905cfa72 : SUCCESS in 8m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/71e9e56d668a46fbb386e86769cb9782 : SUCCESS in 8m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/240638b1254541de82ef8960c16f1957 : SUCCESS in 10m 48s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a0aa56d3907a40c0be1aba3a725b06d3 : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/77135c75bdff4abeaaf694ed4cd5c8b5 : SUCCESS in 15m 18s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/22c8d3bbe0c0427e8d2773617eafec3b : SUCCESS in 4m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/24c0f27f718c457384ec1113fca4cb50 : SUCCESS in 9m 45s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/340d7c22a3ca410bbecfd5d11905b4fc : SUCCESS in 30m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/71485c654650491e9e2cf7256bd41cc9 : SUCCESS in 9m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ccf5cca1317f4cd6993e0d86178c48ed : SUCCESS in 13m 24s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ac5a541ecbd34cd48bcebd67cd9fb1d4 : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60fb67508b7642288ac9d9f6e9db95e1 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b8bf2f5ebc8440e819b974849db7871 : SUCCESS in 26m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6189d19923f64c88959644e4fd1b1596 : SUCCESS in 6m 02s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4b906e289ee44d66a5cf1a1b753fe7fc : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d70b81952fe8425e8c3490156b455894 : SUCCESS in 6m 55s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/dc8564b2201c4384ab24ec3c0fdf58af : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/734334260be44e778246be093bd4cc29 : SUCCESS in 57m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b88b169d01204df6932249118e7528fc : SUCCESS in 1h 39m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6fbb93be0ed91e5258cc861135eba6a92398da53","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-08-28 13:44:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"42ff9678e31ec8b6659e8cdff1587ea88dfb557c","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-08-29 15:15:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f4f1da9588e06d5e1f5c13ee56b055820a8bf05f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:33:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ef8cf9f86e9a4d0249ddbaa0b176b2b18f359ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:34:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9c2e70c483405b6436bb2a620638dbb65ce0416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:42:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a7bd28c929c4213bf5e83cc2f2e3679\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85226f83dd34ed98d53d86ae8fa9575 : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9967ef85c4483b8f37add86d1526b3 : FAILURE in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b495a45bf56a4f57ab8997589753d0ed : FAILURE in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09d4a6cd8d114bd78c2392b86a0201e7 : FAILURE in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e935012b771144a69f3c9afd5beb3419 : FAILURE in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/957768f4ee0e40368bf695965673323e : FAILURE in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d93e05137404edaadc10c6d47a6cbf4 : FAILURE in 3m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1694390496eb449096ba4938a3cb9bc9 : FAILURE in 7m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3fd9916db9824aeb9bc77a439ba455ca : FAILURE in 11m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c065496f809a472e98c2d0ccd377790b : FAILURE in 7m 06s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9acebb9bed24420a90889e7e9120e0df : SUCCESS in 10m 40s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/bb1b9bbf59d94a3a9d2536e9879aae71 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3979c5cfb53342e199cdaa4750d5cc53 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/fee0c707d3fe48258b046203d89c819d : SUCCESS in 8m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4a44959e4b0b4633938d6ae892772502 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72a1f59e995f4f8e85496d2a9c759091 : SUCCESS in 7m 04s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/49543fd5d7bb4889b71e1fb7c796ebec : SUCCESS in 10m 35s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5009cdc9169c4b2ea4a490cc26a30b0f : SUCCESS in 11m 05s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3ff40b808d2e4d3788edc639a90fe25b : SUCCESS in 3m 38s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/6f38f010ec0547cca74fadfdc00452ca : SUCCESS in 10m 01s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8c64b3629b2142cf811e6080568fe972 : SUCCESS in 4m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4f01386e79bd4db6ac2385cdc19fb886 : SUCCESS in 30m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/42ef3ed1d0d84447bcc3d5c015ceb671 : SUCCESS in 11m 19s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2724b49bac964a0888120a773f059cbc : SUCCESS in 15m 27s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c7f9442c399c4110b248facfb7f9f2a8 : SUCCESS in 4m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2fdb04023bd94c12a38515d73982dd43 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f339a9034184773a882a67cdd1fed6b : SUCCESS in 21m 37s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/06505e064f224e0e83752deccf688ac1 : SUCCESS in 5m 34s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9b28b9131d2049e1ad738c40ce388aea : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca59161126ad4c7f914235b0da43b858 : SUCCESS in 6m 34s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a049f95cef304c6088eef9bb5a7d282e : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fe1b14f3c8b43caacf94b8d107ce5fc : SUCCESS in 46m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/199531353049456d8ac8c5cc73015ccb : SUCCESS in 1h 29m 28s","accounts_in_message":[],"_revision_number":1},{"id":"80a5c5cc59054b4b91aa969013d0f7a56db44a76","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-31 20:36:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59e6d338879f7f9f4f47f493eea4092dec1793d8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-01 07:05:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"073422213e20b678d92b5a6da6c9fdf6298cc66f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-01 07:25:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d877db601e4a3d18ec02672814f7042fb1d206d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-01 07:33:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"974d6ff8e2259f8c7c97a73ad891c98739399144","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-02 09:44:08.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83c5e8d68364cc60871916b9d49342b829b63839","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:09.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eebe6bb75081df17f73e2d28bdf02de4f2b54f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 19:40:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c68e8075815d47b092aad25e3d3df125\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b963dfe56c4aecb8953087495fb672 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88967b223a304b529bc517b7405ddf3e : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b8b611691c447baa2e5ffcc5aac47e9 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/860c387e50364bf692044ff34a2f75fc : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2093f429403c4d6ab8e97f5fab8ac222 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1950388e80f494ab2ad0dd3b74c7acb : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/219522de71e643beb14ca9c96f59cdaa : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8a9d1c9eb4c47e0995e44b35935db22 : SUCCESS in 4m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/58142856a7db4e2faaa6f1f5c23835c8 : SUCCESS in 9m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9df81820df1c43a9947de58e9f3bd83f : SUCCESS in 13m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cbc44cd7b48d4235beffb9704fd2fa36 : SUCCESS in 9m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1f33d953a73c430a96a96bc20ed6a4d3 : SUCCESS in 11m 30s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2233b4870e37435f8b8cf9cd2bdbbc20 : SUCCESS in 7m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/20d2d282e7b34665a4f16767f1c685f0 : SUCCESS in 4m 23s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9cc4fdfcf59d44edadc50f93e5b79b53 : SUCCESS in 5m 54s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ad177fa4bdcf4e6fbc8d642b097ee35f : SUCCESS in 10m 39s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/69d50d613ce74dc89fce9af1adb3db4f : SUCCESS in 5m 33s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/b09b36c0d48848159ab285e9e105a305 : SUCCESS in 5m 12s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eb105e7bbd1f4736a880c603307b89bb : SUCCESS in 8m 50s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/707292d66c074d548ea19b3e3724c793 : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18848aac32024c51a8c3a3f1ff7cb89b : SUCCESS in 7m 45s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c3a9469d4d324a10b26b07f742a80fa0 : SUCCESS in 7m 58s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/487e3e5ca29542e6873a49c578b82651 : SUCCESS in 10m 20s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fba7f74dc3444dec8fbb23762161554f : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f05a4117ba014101b4af9e2b523ae019 : SUCCESS in 9m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4cca43fbd2554ddb9c35ce14bf16e85d : SUCCESS in 4m 32s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fb0c223aa6d94fb88ada23d1784fc112 : SUCCESS in 10m 34s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7d7af90c5c6e436fb426bf29b5c0ffcf : SUCCESS in 23m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b3f8de4ffe224684a2ef435d5cb9be56 : SUCCESS in 10m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3be208a2eaa140e9821536647bf92212 : SUCCESS in 14m 56s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fa1004b8634c47dfa0f8ca54e3a29078 : SUCCESS in 6m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b40a19708db94a108f1d734cdfdd1c5b : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b11cd25dd3d485bb47b62e0760dca9c : SUCCESS in 25m 12s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/769ef4b39e5249f38fa0330327928a68 : SUCCESS in 5m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6173a4b4eb4b452bb3aff09056d205c2 : SUCCESS in 8m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2f291bcf2d54339b9e78a2ccabe8c04 : SUCCESS in 7m 49s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/6cc64e1a85094c4e800c85200d765f09 : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/979f0d4315df4ca9a2d413d3f76bbc6e : SUCCESS in 57m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/469c3bb11a3749b4aa692b50c09d2d91 : SUCCESS in 1h 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7d48fdcefbea39bfa64e9da4c9d3964a6feafa3d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-06 11:14:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c2cd31134e6ddb405b28844759cccaf9a41e593","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2023-11-15 18:24:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"43394bd94750a073693633816d182074b576468a","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2023-11-15 18:24:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71f698dc02de812c95ea2c5947650f1e0ea85be5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-04 15:43:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fca51b94f9e70392456211f8ddacd86123b37683","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-01-04 15:43:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efa35bd5178f49486baf13333a0e245c3e98882a","revisions":{"efa35bd5178f49486baf13333a0e245c3e98882a":{"kind":"REWORK","_number":1,"created":"2023-08-28 09:50:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/892891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/892891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/892891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/892891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/892891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/892891/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-28 09:50:10.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-28 09:50:10.000000000","tz":120},"subject":"Update OVS version to 3.1.2","message":"Update OVS version to 3.1.2\n\nSince 2023.1 (Antelope) release distributions in PTI are shipped\nwith OVS 3.1. With that it makes sense to update ovs python bindings\nto corresponsive release as well.\n\nChange-Id: Iaf4abfcfa78933e923173714f3387e3b8757f409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efa35bd5178f49486baf13333a0e245c3e98882a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/efa35bd5178f49486baf13333a0e245c3e98882a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
