)]}'
{"id":"openstack%2Frequirements~892892","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I3596993880af2ad61274af30503f639199f76b9d","project":"openstack/requirements","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-01-25 07:15:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3596993880af2ad61274af30503f639199f76b9d","subject":"Update OVS to the latest version of 2.17","status":"MERGED","created":"2023-08-28 09:53:06.000000000","updated":"2024-01-25 07:15:58.000000000","submitted":"2024-01-25 07:15:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892892","meta_rev_id":"45b7fa1c0a2ae39a79b786f066f91758c37ce9ad","_number":892892,"virtual_id_number":892892,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 07:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-29 19:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2024-01-25 01:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2023-08-28 13:44:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2023-10-13 14:23:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-01-25 01:46:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 09:59:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-28 11:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 13:44:15.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-08-29 19:33:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-10-13 14:23:20.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2023-11-09 17:20:12.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2024-01-24 21:31:54.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-01-25 01:46:46.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"e5934a8b25ab4398db9670223d2437a8d5038d57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-28 09:53:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f4927fb0e1ab81106be5e40f0bf7b37a5a926a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 09:59:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd7ac853c0714ffda7e6abae5ae012f0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/795edcf01fbb43ba8df3861bd0c2f91f : FAILURE in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6ae8893d8d8a3031508c718f5a18f38391a74384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 11:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa112f7d4f0f4fae8503f497771b8d55\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d716566d70a4932b9d326912ce96101 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e06e4589dcbe443abdfd8dddd3309e4b : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e392e28f78bb4d54987277d4fe820b28 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3525d7511f694b08bef0b16cc405fa5c : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1929e8084af049bc98c718e662e783e5 : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b216f3e182f94e01a060b15e321f3e89 : SUCCESS in 10m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b95bfff5269f494894c81d7300f19897 : SUCCESS in 11m 16s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/73bbf3f83b2848eba89e95ee2b341cd4 : SUCCESS in 7m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9a13f8c8aa3741daa4c25efa83fd3d59 : SUCCESS in 5m 45s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/fa1e58075fd74f58a3c8e7cbcce4c592 : SUCCESS in 6m 32s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4f4ef1edd8684130bcdbf0faaf1147fe : SUCCESS in 9m 50s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ef7c3cc0519843a4a0f41eefb73e0d3a : SUCCESS in 5m 18s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8177fb23bfd043d79cb8713802ce4b76 : SUCCESS in 4m 43s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/f2275a0628984dc789383888b0cf8152 : SUCCESS in 7m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3f54da5090d94843b9b9d58bb4c2fdc0 : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4dc2716680084dddad85dc32df14884e : SUCCESS in 6m 50s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/65e971682eeb4e0d8bae707d484089cc : SUCCESS in 9m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/346dee629fe94f3d810b543c46014b0f : SUCCESS in 12m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7cf23cebe34e44d6bfe594119ae15b31 : SUCCESS in 4m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/565fc03e63994bb288c0da29102916d5 : SUCCESS in 11m 30s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d76a0fbba2b84c5b802ab3f794f48640 : SUCCESS in 5m 16s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2f907894c84745faae02ef4f2c88bd96 : SUCCESS in 11m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4fc2762f479f42c88a55822afff9709a : SUCCESS in 32m 16s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/62736ec50dba4b309e912fe8de9c7618 : SUCCESS in 17m 41s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/b7ab8beda8c64b5eb208fa59574593c4 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/68711a10d0544025a7b81bdbe63756b0 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46fc0d3be8c543a3af66c6e32591ec73 : SUCCESS in 29m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c0ae9a89884a4492978730cbd7a7859e : SUCCESS in 6m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/eba28299420940e093e58d733b5da96e : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/939ccd818d014cb59882243c0d1f9cdb : SUCCESS in 6m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ba0531307e394e65a5ce064298382ab6 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/521d4ba8135742b2b5d528ee16a4c93a : SUCCESS in 53m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fc67a97a961435c9ba26d5f3fc553af : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":1},{"id":"51a2a726004b9fe8252824e9e298f706e5f0096a","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-08-28 13:44:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6eff4060528d79292179aad83a4bab4e46d05b8f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:33:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"17ea0ecfcfa3cec93cbb7057f52436d735a39a3e","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2023-10-13 14:23:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf91e8d20b6298c6abfd18b157b628da331ef3aa","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-09 17:20:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b18a618d0d206e2d73b1b761f0146f4815ac989","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:25:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3463b3ecef134917b251a0e60337423b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ad404a0b5c9243b3b36d47f60e84b511 : FAILURE in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8cf8845fd020e2a30198e18b54836d741e0de91e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 18:49:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25787da3d78a412886de2edfc9112c99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fc1f437164748f8bb3768a1fe5c2c43 : FAILURE in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dda8fbef5cc42189b855913459a33aa : FAILURE in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb585e130435430faf2c9e39c743cb42 : FAILURE in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b3f05da10464cc4947d1419a1b8f136 : FAILURE in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1ae24bc6dfb143128337ac98b8661397 : FAILURE in 8m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fcc85d1e37be4281aafda715f22c6008 : FAILURE in 11m 11s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b9a754c3538c45179db47d444945afd6 : SUCCESS in 8m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/dd457dae604a406c81a4b7555b02773d : SUCCESS in 5m 54s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d79c6dde91a543778bf02309a638b117 : SUCCESS in 5m 43s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f7ce9dc3dd8b4da8930e62fac83adb17 : SUCCESS in 12m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/66207d6209574e68a6df4dff3994a3e2 : SUCCESS in 5m 08s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/e03612a75aeb405d8b55a025d19b4881 : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4d50205c134843d997d91f23f826aa9b : SUCCESS in 9m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b651c2196c644cec94652b4b39ddf1e6 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3020ada8378b48359a5b0d6ffa6c4f05 : SUCCESS in 7m 49s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6bcc9d445b2a471ca2acdd7436d9271c : SUCCESS in 9m 10s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bad578faa5ad41b791166315213b18a3 : SUCCESS in 12m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/889ea6f207d8417ea8290852d0815a52 : SUCCESS in 4m 07s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1bf810da0ee24fe997441bfc44bbecbe : SUCCESS in 9m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/71e575d38ef7453392b0d38fb61cad78 : SUCCESS in 4m 02s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d5f58ec72d3f468bba476a5f0acb7971 : SUCCESS in 9m 29s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6e7371300aef474799e6242c28de3ae7 : SUCCESS in 22m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/cac9b19020a64b53869233ab390fb967 : SUCCESS in 15m 40s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/93e8d9ed021d433cbeb876ff5c86ff0f : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a72df3e65aa4e5191c5ff6a2839f9b1 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58aa99ada061452cbaf39f3d7f4f14af : SUCCESS in 28m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/7b838d81676444c0b1a861cd96042eb5 : SUCCESS in 5m 45s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/27d3e4a566a74091b08acb477b43c560 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03aac68517754e27bc9b78ddf58b1bb7 : SUCCESS in 8m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/03744c0ec7054a30bc1216638d041b25 : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8beeb986e5124c8397af6b75af6eca16 : SUCCESS in 1h 01m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73cc57eeb963488c860e8b0734c7f0f4 : SUCCESS in 1h 22m 59s","accounts_in_message":[],"_revision_number":1},{"id":"72f120194365c34c5e127f975ebde47bfda8efce","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-01-24 21:31:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fef3670b483720d822c8acae9c80e5c9da1bc18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:44:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ce6d09b2fb745ffb6726c0e168c8ac7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/13e87d85473947a49821f09fc682dd1c : FAILURE in 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f461f52ce4359bbd4c113e7a0e4afc1753c8e222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 23:08:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9c6501bc37046d1af5a08aa980e16c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c7d4a8daad47f180f778966a8d2b9e : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba0f54b965c54c98bc1cd9b6cc0b1794 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/904a0daace264f10a30cf3386f72aa7f : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7b08633f915448a9b51d97503c634d7 : SUCCESS in 4m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/24b4af29e004432f8a44e9fef3b65d59 : SUCCESS in 14m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c33906eaac74f7ba33925c6dfde573b : SUCCESS in 13m 40s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7c224dbded5c40528aa352d25164fcd0 : SUCCESS in 7m 20s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/75c2461cf96149368070e1eb82b3c487 : FAILURE in 5m 17s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f524c83c4bbb4217bc3d05f483379175 : SUCCESS in 5m 11s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/37b3af1c21754e9b850e6d1c68515d2c : SUCCESS in 12m 05s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fbe352b6b8ef4c3db2abae9d2906eb3f : SUCCESS in 4m 54s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5e49952332a640aa8b411cff19906a4b : SUCCESS in 5m 22s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/198f8b2d373342d6bbbdf9b25cd0e20f : SUCCESS in 7m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5ab0d74a3e894889bb3a82f1808ec478 : SUCCESS in 7m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4111bcd7e3954f05aa86488620b19415 : SUCCESS in 6m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/beed46ad6534417ebb413d64df50f343 : SUCCESS in 8m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/52c71c427c094d579c760e9f3d0b5b37 : SUCCESS in 22m 25s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6ece5d1c2edc4c4dbe9865345205d4c4 : SUCCESS in 3m 47s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/befc9e84eef841b8a38a1f9920204702 : SUCCESS in 12m 00s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bc7e1f9218664700a014b8ccf146f9c7 : SUCCESS in 4m 54s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6e878a6b1d6b4424ae0164d33cdde4c0 : SUCCESS in 9m 36s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7f6123647798460a9180aaf481b12cea : SUCCESS in 21m 52s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9c5ff9d814f44ace8a3db4922e475a9a : SUCCESS in 12m 26s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/521dbbacc11e429283aa8ccdd3435e19 : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bb2a765c863d4827b1ab2120d4f1be54 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7cd48b85fd5e40668695844bb73f0017 : SUCCESS in 27m 55s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0533585fde7b431dad1ae6ec6555807d : SUCCESS in 5m 49s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d21affe6ac8747a6b440d6ccc6ad985c : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d79d00a1aa449ee9900008e79912fc2 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/225279ce30dd406fba9d6fefb6fb18c2 : SUCCESS in 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e922d3f0c1e64b5bb1bc54aa4b483fc6 : SUCCESS in 51m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5204d4682aad4b4c8f21b191f28499c3 : SUCCESS in 1h 29m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f02d365b82cb5403a6f70ba5d9de92882092f5fd","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-01-25 01:46:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aba2faf252370a20b14c8b5c4b7bf205d818143d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 01:51:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f63600ebffa42f5b161a6db0123a852\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/defe0002095f4339a7c8f2205be80517 : FAILURE in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8558e6f33619a9b5c6f558fa01471d2917fe8dbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 03:13:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecfb31bb715a48a6ab1c83f09f4146c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7f08435f2134c8187a2dd53ccad96d4 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b48426c68c148bc97d08ae7e9902084 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6251a598938e4742b43adedf379af2d4 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0bb76b257f99430f8ea0da7cb25b2aa2 : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/469b2d6792a1457b8a4decc011e1396c : SUCCESS in 9m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fdcb25fe6ec24e92b5071fd005ae7fed : SUCCESS in 14m 28s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e780e4a5fa1e49e88b5899dc5782f7fb : SUCCESS in 7m 00s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/bc1feff6e0a64267895a65fc8e055e59 : SUCCESS in 5m 30s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8072fea2ed91412db23fde3772b0d51a : SUCCESS in 4m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/bcf87ee43d734d548f84cc00aa0c366c : SUCCESS in 11m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/b3fcd2879a84405da12be9cee9ea1f9a : SUCCESS in 4m 53s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/fbcf086a31824537b2240b37827e7aa0 : SUCCESS in 5m 04s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c618ef30d0ea44ccb4ff85f95a8762d8 : SUCCESS in 7m 43s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/84cf129f567442b5bd6f03aa394536b1 : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3da670faa53b42378c1236ee61afea7e : SUCCESS in 9m 16s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/05a50ff1aaea4211acf2764d517c281c : SUCCESS in 9m 19s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/e3770a9425ea403e962827a628bbffd6 : SUCCESS in 14m 36s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6e73131fb85d4b46b8872dc643246234 : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/84e954ec4d6742d492276b158d1ff4c1 : SUCCESS in 11m 36s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/d7fd1dccd0a840c7ba8f060e3fce8c5a : SUCCESS in 4m 28s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/cd0a56cf9964471c9b68419a45f6506b : SUCCESS in 9m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b78058ee013e4f3ba388ee776ab581c5 : SUCCESS in 22m 45s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/118c83323160432b8643fd09a6e311c2 : SUCCESS in 12m 20s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3ee3d50a9fbc43c7b097da28294fe29e : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac2b7990f9ff42b3aaf1d0665f904f98 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/642f49bc198d4e3286ef3366a49b8586 : SUCCESS in 32m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/24c050edbba64211b12cc6226bb0900b : SUCCESS in 6m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/fd957c81aca84f8fb1cbc68f24a34e9e : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d26a32bcfd0420dadeee11d01c635d7 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c83fcd7660c3454087f4741bd33db778 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4405e3f1b0b4013a81051aa37e52ce3 : SUCCESS in 54m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbd37e226ed047b2b5677c3b56dbf151 : SUCCESS in 1h 23m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b55c150d2e2b44de489d7aae43c16bc059f542f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 03:14:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05fab3658b9ae6bc6e8377384e7a2fd00cd5f822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 07:15:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41fb2887818e46b28b955b5c32c77373\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf36ba4254a54c8dafcbe2fc17482be2 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0631214e3d64490dae487ce80f9a567b : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1957585aa2ca401d96cd6bb9e9aba9aa : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f24f4a2ef214f918c4dbe9b700469aa : SUCCESS in 4m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/95fe35b02fa54e20ac75fcc21d4958ec : SUCCESS in 14m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/43e58c184e2346efafd78c3fd0ac75ea : SUCCESS in 16m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/53144ac4e77e424dbcbc86d098275aea : SUCCESS in 11m 34s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4a160d4424ad4e35a20287639206b90e : SUCCESS in 6m 46s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/abe1d1645f3c492c84e4cf7c9b28b449 : SUCCESS in 6m 07s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/b16e9f5a09e14117938636c132518e8b : SUCCESS in 7m 47s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/6c79423dc8a0491099e289b09a444932 : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6aaa87e813f24e01902394713c232873 : SUCCESS in 7m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5ebe329cb9f74f2ab88c274098ef4a53 : SUCCESS in 9m 36s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/23f9d6e709da4698834e27f648f443a2 : SUCCESS in 10m 29s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3eacd70e108a40668e27484005302659 : SUCCESS in 4m 40s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b24ddcbb952442fabc6d12379cadf9ca : SUCCESS in 12m 31s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/5d4bfdb2234641cd82687af7066f35d6 : SUCCESS in 6m 58s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/dd43d6bb178f470db88bc01157b61047 : SUCCESS in 27m 58s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3bfa448171db4f04b639098d9c8b6cff : SUCCESS in 14m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9fdfcc6a7896416f9984059a8b8e4229 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a7d8394693449c9803caa289fa79820 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d91c63d0d3bd49b3836e436b0c2a7431 : SUCCESS in 25m 47s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2d123b3b695a48aeaffcf119566b2f8d : SUCCESS in 7m 03s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/20ddf3e83f204a148ed396a3fc28f8e4 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f8fc6d669f44d0b82c673cfa1a2af15 : SUCCESS in 8m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a2a259a62e1843ac963a2f64152d2db3 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ef6edf860574f4ca387dab1803096ad : SUCCESS in 1h 08m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14358870d3d64fbf872670b388344ca4 : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":1},{"id":"45b7fa1c0a2ae39a79b786f066f91758c37ce9ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 07:15:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a075dc69f63b82d09ddacb9f0de105402f36d1f","revisions":{"5a075dc69f63b82d09ddacb9f0de105402f36d1f":{"kind":"REWORK","_number":1,"created":"2023-08-28 09:53:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/92/892892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/892892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/892892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/892892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/892892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/892892/1"}}},"commit":{"parents":[{"commit":"2710414b4663b85b48ecbbba21f4f0ea42dc3df1","subject":"Merge \"update constraint for python-manilaclient to new release 4.4.0\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2710414b4663b85b48ecbbba21f4f0ea42dc3df1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-28 09:52:58.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-28 09:52:58.000000000","tz":120},"subject":"Update OVS to the latest version of 2.17","message":"Update OVS to the latest version of 2.17\n\nLatest 2.17 OVS python binding contain various bugfixes and\nvaluable improvements. It\u0027s worth to bump version of OVS to the latest\nminor release in series.\n\nChange-Id: I3596993880af2ad61274af30503f639199f76b9d\nRelated-Bug: #2027854\nRelated-Bug: #1985062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5a075dc69f63b82d09ddacb9f0de105402f36d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5a075dc69f63b82d09ddacb9f0de105402f36d1f"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
