)]}'
{"id":"openstack%2Frequirements~892919","triplet_id":"openstack%2Frequirements~master~I76465fb0bfd30d473b293ee26570d32b2ea6bef8","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-26 13:52:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I76465fb0bfd30d473b293ee26570d32b2ea6bef8","subject":"update constraint for oslo.messaging to new release 14.4.0","status":"ABANDONED","created":"2023-08-28 12:53:14.000000000","updated":"2023-09-26 13:52:55.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c5a2cbf8ff1220b1980e5da656d54dd5cebf1f8","_number":892919,"virtual_id_number":892919,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-28 15:09:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2023-09-02 17:56:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2023-09-20 05:40:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2023-09-02 17:56:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 15:09:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2023-08-29 12:23:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-09-01 14:36:31.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2023-09-02 01:55:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-09-02 15:34:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-09-19 13:34:01.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-09-25 16:31:15.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"}],"messages":[{"id":"e8ef8d627e09ea82b268cd6707d0ade94bff6777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 12:53:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"833a3f039e9beb0819b37c9afc531e155dfc29fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:09:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d60646e50e04c188ff9c96665114207\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/724c8e88d93f46878e28aa6d1e95ef86 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/413b0f8f55ac4a04ac7c690912d713e5 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94e1b29a39b945b9a3a9156e6ad3737c : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2385abb87af143c798cb5d175a8924fd : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a96fe7d4a90a4fcfa7b0ac9aef3131eb : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bd81ca62a774b9b9328e4665743744f : RETRY_LIMIT in 2m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d298de6ab1d240fc9b0804bd882fd801 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8bc10eaa93c54d0988f851aa548544b5 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/040cea25e5a34b0c8036129b2811236f : SUCCESS in 9m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/56f7955a8e8048c5be3fc5517eec80de : SUCCESS in 14m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d9f1a2417d544534af84cf27330c61c1 : SUCCESS in 9m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/428ebdee99664bb5817d52eb8e03557b : SUCCESS in 14m 29s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/cc7ad8223a2d4092acd0830829e83365 : SUCCESS in 7m 43s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/9eda25934fc64d698be2d55635c19339 : SUCCESS in 5m 00s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5f8462afd4904dd8942cc0edc9fe30b3 : SUCCESS in 6m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/4f00cfb9b9ec4a308c160bbd98043542 : SUCCESS in 13m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7cd09d1ee98f4e8daa574452b80c2310 : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/42ff04f718894730bc3020d211767305 : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d6bb582ecf944e508a46123e236f5351 : FAILURE in 5m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/658c42df757d4fca9bccc25de74f2b98 : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/00ea59c588af4caeb0b5a52fd5f93bd9 : SUCCESS in 7m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/48db5ac30be449c98583c4d67706414e : SUCCESS in 8m 34s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3d81b8a5713a407db9e8b96c2a281ea4 : SUCCESS in 10m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/724cd4ff16eb4cd9ba257e810222d306 : SUCCESS in 4m 02s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3009ddf0bb0d44faaca9c7829ce9e201 : SUCCESS in 9m 54s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/351e615ad91b47fa9658bcfd2914a9b3 : FAILURE in 5m 07s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f378d02199124c14893b116c2b2ef897 : SUCCESS in 10m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c10df765c034494bae9cdabd9a798b56 : SUCCESS in 36m 03s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/27904fb7dc6149bbae796b5861118565 : SUCCESS in 11m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/92a2ce0a96034c2495f3e8e76cf58432 : SUCCESS in 18m 58s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/712b8ad6dabe4a9ab2f0989f086b889f : SUCCESS in 5m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7510c116eb034fa0935b37b2b489c135 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a548964551e941fba793a7dfd4438338 : SUCCESS in 24m 42s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d46ae921d4f34611a43f35b0f7133c63 : SUCCESS in 5m 23s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e594459d00134840a28724e8da595dfd : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15c73a727c704c379b0d70d1bf8702b1 : SUCCESS in 7m 22s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/e061f5cd406d4b0592256329fdcdf834 : SUCCESS in 19m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e57d3c87f2d4fac82ef69d4b9f9e9c5 : SUCCESS in 1h 12m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a753c3d7f5a4bbe96114912ac3ae774 : TIMED_OUT in 2h 04m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d2eb180eac05866b86f8243e736c3dcd7de2d0f5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5f945ad4652522745cda6e4704b02a0ac79eec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:21:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50e71635cd37496bb2472c4911877f22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9995a68f0914cd3a9f34dc59a38f8bb : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86736ce7eb90438bad79976178e61d51 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f02b3f41baf5457ab2ed20a4c4cc8aec : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee8ff9d768e24aca9d5c545181dd1740 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e67ba87d64e42e38852d26549c4751f : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42aac488d8c3463299117943dfd9bcc2 : RETRY_LIMIT in 2m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/639edc1960dd4da6b282424cbc209e28 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92606c89f46044c18ad956ce2a2c291f : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/860d4d069a7d4680a3e78a7865fdaa10 : SUCCESS in 11m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c781a7ebcc944ff6bed5283fe39db51e : SUCCESS in 15m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8f0cb0929c1449069b158fe51e8886f1 : SUCCESS in 10m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7bd38cdf874f4b299af69a1ce2f98402 : SUCCESS in 14m 26s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/e60f604f82674c50b5286ff0a10f802a : SUCCESS in 7m 45s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/596f146522d94f4b99085dfc10ea2097 : SUCCESS in 6m 11s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/558d39bf9f754901993ba21d4cad83b6 : SUCCESS in 6m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/222c54d97c86410fb12c1903be2ec9f9 : SUCCESS in 10m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/da460a0747004295b165ec42b3ade83a : SUCCESS in 5m 19s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/26d6900e7cb74933893c387541681564 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/33fa342415714c998ce21877e473b208 : FAILURE in 5m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/58154f484f524e17803019318e9a243d : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cb22a64063d44cf8a888e1209173d70 : SUCCESS in 7m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/056e7d39d3b14642844a3d5431ee4cf6 : SUCCESS in 8m 32s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9314469912454ac097ff21d48cd04045 : SUCCESS in 13m 12s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f2c2f681b28d4617a544c1223e7cfd19 : SUCCESS in 4m 22s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/098e3df26f9940e8ba82df1bfeaee1d8 : SUCCESS in 12m 06s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/60fc6d12afe1497d920ae440ff1c212d : FAILURE in 5m 10s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fea6ab55a49543458aecf2ec7f2de841 : SUCCESS in 10m 43s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/bafbc4b85fb741f9b389f9a4f5d439de : SUCCESS in 26m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c9e2cc0c509d48bf92003d88bea255ce : SUCCESS in 11m 41s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/18ac5852e18e4a26950629743b15a470 : SUCCESS in 13m 51s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cba24e5013da4ac892ecfeba755d93a0 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59ad5c59d8fc43cbbf8df9433e477997 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c92c6544ce143e989adcba695b2be63 : SUCCESS in 24m 13s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/005107e599d44750b0663e1956b84951 : SUCCESS in 5m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c18f8e9384944b58a173cc9b745fabfb : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64fdb9eda39b457b8e15654743230ad3 : SUCCESS in 5m 56s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b68d4634c3654da0a5a6528e58aa9959 : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87d06c46fb994550871d12d01040494f : SUCCESS in 1h 08m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc662063082a45a38c2d42745c860a3d : SUCCESS in 1h 33m 07s","accounts_in_message":[],"_revision_number":1},{"id":"72ee3849b82595d8427143940ff2bb5791b6bad0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:23:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35d149177f0d42e7b7c0f49614e5daca769a7b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:09:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35a1cf0b642441a6946136e5f3a48144\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b33f046e2e41d9a00587f341a59112 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc172020231447f6ac66868f02531209 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/726c5bbedd2a40b9ade434abbc9ac4d7 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f45d446cb384f3c905b93618a648760 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a8a087687cf4e30b12ecbbdd30dd78c : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/156ce92fc45c4d63aca3ad7fce7d6071 : RETRY_LIMIT in 3m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e25f735d5604090b5d033e018a4c1df : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/837ab9f54e6c4d18beb3c1f72ea31a75 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c57544e6b7154b0ca3c6a00d8ea1f96f : SUCCESS in 8m 51s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c4f4fb0f36045e497737f67642f494a : SUCCESS in 9m 44s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/59f25b93f4744556bf20d96a60e52ce4 : SUCCESS in 9m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cafb72801c574297850a60d40147f1e6 : SUCCESS in 10m 08s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/242a3643021f41e6be74e454c4f15d86 : SUCCESS in 7m 38s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/31759e1494624277bb8010e5f3d0d0f1 : SUCCESS in 5m 08s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/df2d1d5fcad249848f111ff67bc25c3e : SUCCESS in 5m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c32e04411a1c4620bf7d97d8dbdff19d : SUCCESS in 10m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cd789504aeef4da79a9d555d935f8600 : SUCCESS in 5m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/46a4dcb3bbee402c9be7148912d5bc0d : SUCCESS in 5m 11s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/722886ccc6f5425f87b15962c2acdb2e : FAILURE in 6m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a0a7d3582c1a4df7af3d045544fbc56a : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/372c2f40fecc4ec8b13eaf58e6ef343a : SUCCESS in 7m 39s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/121f3d39395d4ac6b541db487eb9b186 : SUCCESS in 7m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6deb1baa801d4422b8e0976e104a6145 : SUCCESS in 9m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/7fda6c99c31f4719b1376352b38daa3e : SUCCESS in 3m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/e55a2e06915e4414820e45518605dd04 : SUCCESS in 10m 47s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b824d3fb1a5843e8b3af3874c1c9d89f : FAILURE in 4m 19s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/66d7c63b19c74da785bac38b66871c1e : SUCCESS in 9m 17s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/c9e46ff8f2344d7da5188caa882ba676 : SUCCESS in 33m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fb7e15e8d9de4f7ab7b27bd7ce5f2246 : SUCCESS in 12m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c3a558dc3af54c8e9d54d1c34cf60247 : SUCCESS in 14m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/25c864457bf349a79f61a34c8aee5c84 : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f521a9a8a48641038c89116b6027c3a9 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcaafa7325a04e38bd55c1912a45cb47 : SUCCESS in 23m 38s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/a90ae53ad9c94df28fe36b81a1d44597 : SUCCESS in 5m 38s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3b6a50e5337242ef892e21f217aebc2d : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e454425f93074c98835a68eec17d8154 : SUCCESS in 6m 57s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/b914b13484d345fb850e0f375debb76a : SUCCESS in 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/074b0b3f226842289668bfd2ad610183 : SUCCESS in 58m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b12bdc3deba14435a9f2452c412d86c3 : SUCCESS in 1h 39m 27s","accounts_in_message":[],"_revision_number":1},{"id":"05d2eed3ce04f54a38c01ec1cbbd14127ce83696","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-29 19:32:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d53a01a44cb65e69d626cf643ef5d63fe5617faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 20:46:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7779d98a3bc443b7b06c345df27feff9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a929a4e9d27e411bbbe933231d4fc92c : FAILURE in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19901ad48b3a4ad382bd984c31d36ad6 : FAILURE in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b69ba7e724d43f9ab25fe0f228dc936 : FAILURE in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fc6934a8dfd4c2a91bb9f8ae806e166 : FAILURE in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29f2a8f9cc8b420599046c5148b296ac : FAILURE in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1353b938ae614e9f909d94b8a9e027eb : RETRY_LIMIT in 2m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05000d09f80b4ddaab256927c8bfc2da : FAILURE in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79d52bd290e943b0a625d52634e336f7 : FAILURE in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bddb1ccfe20a43f1ba473465af865412 : FAILURE in 8m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6a08459e94514a369915a2491516317b : FAILURE in 14m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/909c82db156241f59de2c587014f7786 : FAILURE in 10m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7f2e8e523b1841bc885f24a056cc32da : FAILURE in 10m 20s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9fa5a8f5ff654ec5868feda3bce4ea43 : SUCCESS in 6m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cb1c98581e5841129144e673fa742815 : SUCCESS in 4m 32s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/618c824a96684ea5b46f0ebfff0bd800 : SUCCESS in 5m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b915b76d2e1143a98f5004b93bcbf10c : SUCCESS in 11m 40s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7da28b08a942442da6967e46bc7e3898 : SUCCESS in 4m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/dd888b93b9154b70822de4dd02fd07e3 : SUCCESS in 5m 14s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/98d62e99c1b34b7a85591c57a26b30aa : FAILURE in 5m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/70b16d15f19942b3b9c05933ec6ea5bf : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63878ac68a8145d887b5c08d56016838 : SUCCESS in 7m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/235ce03e551e4d899bd182f6d43720d3 : SUCCESS in 8m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/54bd2cb99cf2482985e87b14b1ce2c32 : SUCCESS in 15m 14s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e65fd7ba6fb24e779f526b0648a8fe07 : SUCCESS in 4m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7a9d826db35a4d908b28ae7e5d69c61c : SUCCESS in 10m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/07bc9357ff4c4d42aeae5d0844337508 : FAILURE in 5m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/734e9284fb20440a90acec0df7e3d391 : SUCCESS in 11m 03s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/87052457b8a34bd98a233a4f27c6e4f2 : SUCCESS in 32m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/36ff867a901b47ef8bb9ba7036583cb7 : SUCCESS in 11m 53s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7c5779cbfd084dc18506ab559a061a07 : SUCCESS in 17m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/504fc5bd65b247538af9596008bf2772 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bb6dfcdc8d7942aebc616c259fe0036d : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07c4de64a13942a0bcd11d2777d3d270 : SUCCESS in 22m 28s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/05875b6cf7f94171b16c2262b3807511 : SUCCESS in 6m 22s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/f79354309c9842e095d40d40b44dccf9 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cd3ef420474f4d6283cee47b34766c56 : SUCCESS in 6m 34s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f8fd82c76410484a9525ff948208a039 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e36f76722b14e539b5323bd9cebe61c : SUCCESS in 58m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c6e293229a049afa0a92dea8de3b7f6 : SUCCESS in 1h 05m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b1f848871def96a9810d0e605d5e5485979cacfc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-01 14:36:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ba58b88d854ea570dd757402c3baee9d7bae933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 16:11:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e964cea37bdc47efa07eefe8bc47f865\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007861435ef44aa4ba2f30640208d3c3 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f12bd1d82743c7a58eb9be0810f5ee : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34f5135a9faf450a9ce6daeb7272a8c2 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d88fea090954c2aaa96306f23a49429 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2aee0f8dbae4b01bb5fbfb9b265033d : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/990aafb2d66e4104ac6a0b24d7cbab1b : RETRY_LIMIT in 2m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/219d3755f3954422bf76cd00d1a40fe4 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4685f0b45d146619fb39571c774989d : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85c8a1f901a3493384de55df26a010ed : SUCCESS in 8m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8f685895d18149ec9aa256c70451388e : SUCCESS in 17m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/979b398b315d4877ae7e3cb4df5a40b3 : SUCCESS in 9m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/dc7daffe81fc471d85d6a314f9149faf : SUCCESS in 14m 48s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/5696c0c962784de199bb4ecd72d66cb6 : SUCCESS in 6m 57s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b6049bddaad641f5b8f0dbc37dbf7530 : SUCCESS in 8m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3f4cc2e658ca47e8902efbb6a871d132 : SUCCESS in 6m 12s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5f25a315c6cb4d32b186a1c99ea4e0eb : SUCCESS in 12m 25s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9054955ee1b1410eb3487070fd1902b2 : SUCCESS in 5m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/15cbe28b6caa4d1db35746e097256e69 : SUCCESS in 5m 49s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/24d5d2b6a9964801b709f7d729de0723 : FAILURE in 5m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/5bd19e4ebf8b4cb2acb5d3fd4cf3ab6f : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec1e1c0caf18486bac1b38342d59dddf : SUCCESS in 7m 34s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/342398ff15b0457081a99b7a2a5c3454 : SUCCESS in 11m 40s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/631475ab25b04d80a2e61a89e01935ba : SUCCESS in 12m 41s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1d28e14a9e684a1382a215cee314d961 : SUCCESS in 3m 15s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4bcf6e33a8a441bf95d0a3945f806f5c : SUCCESS in 8m 32s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/44619702db1e413680e168e245a1b8bb : FAILURE in 4m 45s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/ffc8313f514c402b83c09026cc035cfa : SUCCESS in 11m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/cc850ae6da524fe3b43d6919a7643ddf : SUCCESS in 24m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7cda0eebe3ee4f7a9ea8fa7153b8f4b3 : SUCCESS in 11m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c1593f99f0a548ada2f2ec0ca62f9566 : SUCCESS in 15m 22s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ad0c1b0a00a640dd9fcf3e9d9f19357f : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ac5b090392c41b2b428fd13e8a44921 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/157b89ff6cbf457584292cb7f12a049f : SUCCESS in 21m 52s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b28b3be649834a95ba7ca848bc8f0ab8 : SUCCESS in 6m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/89ace4f0f0d64df4bf88d578d23e36ec : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab5e738cfb514a0bb2e676ee72b959c8 : SUCCESS in 5m 55s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/88d4e9ca182441fda707500692052f4c : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea741783ff994aaf8478a3af5900b138 : SUCCESS in 54m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1a03a04ba3743ec9780c90d339ee700 : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a4513fc36feb0b198d0b2af4b4dc5ed58200475c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 01:55:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50f843d1bb1faf6a9b16ef3c53e17517fbaad3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:11:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4146acdac5b4b8c94453ae698389107\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/697e064517314ea4849a3071aef9ea97 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6aba597569c4c18962cf3aa6d60a079 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aa66ce9537c4cc2b888afa5708fcbda : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4027e1e77dd44d28a7b55316d401bf2 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0905274f89d41019a189253d4166cd4 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4323881734944692a6e594d4d7afc0a1 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37b1215c3a0242669705942d9452175a : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7882051f3c9b40b283833cf42a55ffcc : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eed9201616a64f3caa9fd0287aed8150 : SUCCESS in 10m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/abcb0241de934bd8ae31ffd86b483036 : SUCCESS in 11m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5ea87e4c09844e0da1a7d49d4aa56a3d : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/093f32b133404fa485fc70e16e73236a : SUCCESS in 13m 18s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fe784a6a5097448395730465b02b6122 : SUCCESS in 11m 31s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/a7f2bdc71a504850a06c75e88ca7adcc : SUCCESS in 4m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/260422ac330544b68c87ee851742d5b4 : SUCCESS in 5m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ea41b89dc73e4b4780dd5f9220eaf69f : SUCCESS in 16m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0da8f29cb23b49458ddefc1fa118eedc : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/47f9a608e26f4baba6ae620687c14017 : SUCCESS in 5m 15s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/45b08ce042d142b8bc9068a8020382d9 : FAILURE in 6m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/946747d83bd74ce0b693a8fafb2170f6 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3cefff3fcb64eae816a05226bef9023 : SUCCESS in 14m 00s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/e38d4c33bab54b559e9386f19f6199b5 : SUCCESS in 8m 08s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/c855221364674b63a27b0fdd51a539e8 : SUCCESS in 16m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3702a977aacb4486b1721f50874e96c7 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/202f0ef74b384827ac71173e77e58001 : SUCCESS in 10m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c450973a69604abd9172180cec732af5 : FAILURE in 4m 24s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/bdbd2f4752c64e8fb4359883ddb11397 : SUCCESS in 11m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b297e6345f6542ec9d53a7e73f4e5d59 : SUCCESS in 30m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c39c08a8f4044e99a97c58e478dde4a6 : SUCCESS in 9m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/aeb6a76dd0094799a7e8464791282b4b : SUCCESS in 12m 55s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d66cc408db0f4a9ba7d8a7166f237644 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63b39e4bc74e48bc8ba254d9e170d0e1 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d86c0d80d384b368154e5f13ef136df : SUCCESS in 23m 32s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/52cd6950d2c44df0934504a88710e028 : SUCCESS in 5m 47s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7cd8124022f8422ebc2cfb6756d48fcb : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b42b03d1bac4321a7850979175f6910 : SUCCESS in 6m 05s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f2c6c2f496f146dc85307397168f7c07 : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e93bea1b59f14015a4c780e95baa9d6c : SUCCESS in 50m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5ace2bb946d44e1acc4a2264a50d4f8 : SUCCESS in 1h 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4568b9767fcaeda7351ac7f60fe0d5e440e5194a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-02 15:34:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b430d253f4fd97a301e96c3531f476a23b00df7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-09-02 17:56:09.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a51dbf3e40280bcea807b5b6a8422b0d7f1c5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 19:36:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3d7c3dde0844fd3bad4bd3a08ecd9c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca324b023034a448f4df484a6e4cf4e : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3991e767920742ba9b88a376662f7338 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80de3c279de5407c90a209b7e3b748a3 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8a1a106787d4a878a8b11d432316400 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/438eb1a77b284346921d7d139e36bd73 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96ed95dd6ee54f438e7b89b62373e4c8 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a721026b011e491ca0f5c564c34c2df2 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd2cf71afbc94a3db2a40a2630d63059 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b0734bce59674edb968ed70bf906ea85 : SUCCESS in 7m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6182e87144c94d87816ec70bd44bdc9f : SUCCESS in 12m 28s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6db61d8dcbd2428bb54aab5f3e7e7377 : SUCCESS in 9m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7fc2d479d7b34c99870ce93a5125f17f : SUCCESS in 11m 56s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/340f9dc2b87f42db95c33609c2b09a8a : SUCCESS in 6m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6d2e538db659481da16520d154f06780 : SUCCESS in 7m 44s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/41b2736183fc4f7e98d6da23c1379af6 : SUCCESS in 5m 38s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/be0d2497d1b34103a71598bc9476d964 : SUCCESS in 10m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c16e3f5896de497a9990546e962af9bf : SUCCESS in 4m 28s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c148e6990f4548778723d752a63caf55 : SUCCESS in 5m 27s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4c6c3af9827c47ff939d33e312835c3b : FAILURE in 5m 36s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c7608460a4e54d04b6b89ec4d1175834 : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e1a0333842c4135b032207155748df2 : SUCCESS in 6m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/882b13cca4124d04ae407f8eb2f965bb : SUCCESS in 7m 54s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a2dd6fce03fd44b99642b26346836a21 : SUCCESS in 11m 43s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ff9bbaf1adca4748a4bce4665b6d787b : SUCCESS in 4m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0e5ddd408f1446f194e8aab621c9e0c9 : SUCCESS in 8m 50s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3ebe95be70ce4534aa3fe6a430e06e22 : FAILURE in 5m 02s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/f98c9dd03b014f38b0871c4876fb362e : POST_FAILURE in 11m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4bc16f3871684a98b51e19e9a77427d3 : SUCCESS in 27m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/80776bf0b943418da86fa9a2b09954d2 : SUCCESS in 11m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8b3fab87f0e74a5392d4c1a5d7c1bef6 : SUCCESS in 14m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/dc6c0acf87464ffe9af9159ce22c04da : SUCCESS in 6m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f176dd7830546e7b05aa30ef631784a : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a50df53e51db44858145b64881a103b8 : SUCCESS in 22m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/988a8ed821814d379b90df2e5be4e629 : SUCCESS in 5m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/b7ee0bcc09df4634a640668337a239b0 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf6c24f39d584bb785531ff0ed2cc881 : SUCCESS in 6m 57s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a2665cd1caf5443aaf9a7adc857550e5 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfcb0a8c3891407eac0dcd69c3d79214 : SUCCESS in 1h 13m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82b6b6d616d74cc0868d4dae9eb26654 : SUCCESS in 1h 33m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a2c48e12d87f3f3361a998f6d06f222f3fab494d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-19 13:34:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6448dd27885a97be585a8cb329d6df1a1c275c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 14:47:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09738b716130460aba850fd6323fa6ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b61e37e19af3457db4d88e99b01f9b91 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a648d1f70974a1491bd4de8f0bad811 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd2c3fb9cb6146e2805c4bde6f262dae : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f4b25934f74d9db350b2e9a1c3d76d : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cd27d7ff55b40fa8d27d98604b19271 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0305ceaca1724a77bb417613e2749f12 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f393a7556a334f8d90e6e58acd338bcb : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6011b6a570bd403ba4192fde01d35467 : SUCCESS in 3m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28d2e2f989954e19a2fd8fec75f4d12a : SUCCESS in 8m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b60053b26e414dc2b7e413c04b5926be : SUCCESS in 15m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/da20ffa8079b49f6b93d3461a1c9a3fb : SUCCESS in 9m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ac084d9bce9349e0ada1a93fe33bb627 : SUCCESS in 15m 17s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fb1432f613de48fdbcdd7603768c607e : SUCCESS in 7m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/cf2c0e8bda734d3892652fcf3087bec6 : SUCCESS in 5m 36s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3d59f775f48d466eb8ac577ee93ac1e2 : SUCCESS in 5m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/606119188181455990a8c5e4a7ad15e4 : SUCCESS in 11m 51s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/cf6f96073f6d4271a8ec10da873e8531 : SUCCESS in 5m 05s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/5db84ce4eb524a3d9f856c4bee20f013 : SUCCESS in 5m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1467c5e56a5a49c4b3428789ed300cd4 : FAILURE in 6m 20s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9690ccfbde844a068de643285f806089 : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b6a7ea382db408eb6451c4e5f4bccad : SUCCESS in 7m 29s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/fd442c43f2464e748b15eebcd5ef004c : SUCCESS in 8m 27s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ed2dc78777d14a33803b3d8489189311 : SUCCESS in 10m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/12f910c11d5d416f9c44458294a89a9b : SUCCESS in 4m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/edda02d5c8ef4e6c95e054e5ba456ede : SUCCESS in 9m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b5de886a2607452c940736d788b2f9cb : FAILURE in 4m 03s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/6730461be6224472b816759e9b7dbcfb : SUCCESS in 10m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/fdc56153b51e499e95e76c826506a5fc : SUCCESS in 24m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/de5c58372d7b4594ba957b59fb013ed6 : SUCCESS in 12m 12s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0bb56b302e6142bf9f12593f680abc37 : SUCCESS in 15m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cfb3a76313a84e1884aa3989f43ea6b1 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c64fcee1c311429fb01faa40ab611221 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/66510e274e3e470787a3bdf318c2f8bd : SUCCESS in 23m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d5bc3d7938fd48b4bc12f2a031a5fed1 : SUCCESS in 5m 41s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/5d3d2a64b59b4e4f81e27f0c5098215f : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/133d0b03a76b468e83e136f77b05e4ad : SUCCESS in 7m 34s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/f6707df246af47e49b7e88f508fdfe8c : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7543f816a5d64ba7987bef174bb6e53f : SUCCESS in 53m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c9f3bbef7314b75a3221721440a24ef : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e49747ec663a8a8f41877cb688aa67ae3a76342d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-20 05:40:10.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f33ba6e202332ded6ce9b8e40bf86c991115cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 07:06:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0734a1d7e5d4404f851c38d49662a8b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dad292bfed2e4a4482c3592f45761bcc : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35af10462ded446bbb6b10c1823d9096 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38d13d1defc14bafa2059b479cb3ab37 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1e83cb9e297440bade8a3110c7b99f7 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcc2e26cd8b44a2a96af8ad954858c22 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/833a2a879a9544dea8ccc0bac8170236 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e7d328863074f41acffd2d3c4f9db62 : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13f2c3ad3a1a48f9aa1ca93663f41c86 : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad7c1dddc4244cbb96be3da95a594deb : SUCCESS in 7m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1afea55596f6455e86907ea469d5acba : SUCCESS in 11m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/06edfacd5b214f16b57f90349e740f9a : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c7b65c935d5a4df1aa776738e85085fa : SUCCESS in 14m 51s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/9f3c445ee81544798c867eac10290db5 : SUCCESS in 7m 09s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/caa86c97edea421aa8b78a19b2cf3e73 : SUCCESS in 6m 13s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/408de3b3dadc46d5be9f77dbd6b66578 : SUCCESS in 5m 34s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/03dbfd170e964c74a06cc7d54bbbcb77 : SUCCESS in 9m 07s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/9c50bb2eac4841c19307d6fbd59af1b7 : SUCCESS in 4m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/0752f9d6a82c4bd9b858976b17eec7fd : SUCCESS in 5m 25s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2a8f2104cc3742919a9e9aacf97f4fb0 : FAILURE in 4m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/b4fafef8fa34465abdcdd3869e198a3e : SUCCESS in 8m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c5ef445550be4f29843d511f99ae0c86 : SUCCESS in 7m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/dc7e9ac699804805a5857d5a636a2c0a : SUCCESS in 5m 56s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/69c90f6371b64e8aa308d46b99d6470e : SUCCESS in 10m 06s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/9f37be3035824e3497dda156b47fc84b : SUCCESS in 3m 46s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/56dcaf2c3e3a4af3b430e2260ba2f903 : SUCCESS in 9m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/31ea35032d144c0c9e6137bdd52e1332 : SUCCESS in 5m 35s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9ba62f7d6e974a5e8849c942fcd67775 : SUCCESS in 12m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/05cda4e652fd41b9bb7f95e1f1f76571 : SUCCESS in 17m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ec7401c0c9344ff4ab520ea1d488e2b5 : SUCCESS in 11m 34s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f49d27a7afe146c8ae4d5190e7ac6a75 : SUCCESS in 13m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7b1de4b6082848cbbd46e930c101cbf2 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d827af87f89460a86fff59424992b1b : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf7d6e8099bf4522bfb490e58512b11b : SUCCESS in 20m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/74fca45c112d4acab3154c7c2143f1da : SUCCESS in 7m 35s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0a061d138e2f4e2eb87f7238c08b77ac : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68229638580a4c03a323d2a4d28a85e5 : SUCCESS in 4m 59s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5bec10e7bdfc47d1be48900a45b51241 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/102568cfb0104323bde1701889dbbb79 : SUCCESS in 1h 14m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/956343f95a6f44919bf07b9b618b8715 : SUCCESS in 1h 24m 21s","accounts_in_message":[],"_revision_number":1},{"id":"41d1c0026fa9aee424749ec823abb94b1aad245f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-25 08:15:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ceb3544180830293ebf7a3635cb43f88ff63904","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-25 08:19:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"870b4d24b00b4249ca344e6782d247287b47698a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 09:29:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb899ed6ce6140f082645f7e3b40b089\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ed7044d3b547c784a5f78a9090c020 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb12e7039d349a5b4f69556f998705f : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc72ca8576884f658339e5793de99d09 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6010782ca7de4aeebbc90821718d2c6e : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/680f4f0c05d142f9ab6f16138dab8bd3 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa9451d7a5ce42e58226775f35a0eeaa : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3aef250d2d024c1eb8ac1c3a6779f3f1 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef4c8aaafb844530980b43ebf10ad499 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c439ab80b21441d984af92786203252 : SUCCESS in 10m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c42a62247b8d4b5c92ee03a5d5f9eaeb : SUCCESS in 13m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/76e32ecd9c6741dda7df3d8a57330c6b : SUCCESS in 11m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/aa180644634f4a8488e57ef343586cf3 : SUCCESS in 13m 07s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0664ed6e17e748959f2d3c1fdd588daf : SUCCESS in 7m 07s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/7905d8a2f30b4da6a780b8d8a367a9fc : SUCCESS in 5m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9cfcf921f65143a386ce05e1fd37f494 : SUCCESS in 5m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/e96b67a2fc584e03bde63c2f50c60873 : SUCCESS in 11m 31s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f0a4b772ebac47afbe965b10952918b2 : SUCCESS in 5m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/527391077af8430a8c54974cfaea85d8 : SUCCESS in 5m 26s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c8778f99d68b4a5da6e6f4b0de941837 : FAILURE in 6m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/d206ee180ab94bc9beba46e167254c39 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69be3254c34f4f5483798df98944ef6f : SUCCESS in 8m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5440e8c3dab945548d3fbc9e9e42d700 : SUCCESS in 8m 38s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/408f1fab989d4f09a1270625a5c1ba7b : SUCCESS in 10m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/cf8938eed0c246428682f117465ddf76 : SUCCESS in 3m 50s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/86c13b2ad5414356a0f4063591eb3a36 : SUCCESS in 14m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/56d51c7b7c234d27a70a5d4e67afaf92 : SUCCESS in 4m 20s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/1609719cb77f4fe3937cc45becc51e94 : SUCCESS in 10m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6ce4bf32d1b848b5b3a4934176fadabe : SUCCESS in 24m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6a39ac60c1374200aa3331e10603e72f : SUCCESS in 11m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4727832060a5445bb1500cadf9d094d1 : SUCCESS in 10m 08s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/1cf11fed229c4d4bad24e76f64e8554a : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/080fe61dd7424e7285f57dea804c2c35 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d590a446a798476bb68a7bb6680c346f : SUCCESS in 25m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/98765f93aded451b93fee48a4b784695 : SUCCESS in 4m 42s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/4806cc823d1141f6bf5c7246e98f9a3b : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10976098bd96497992e2c888ce34a490 : SUCCESS in 7m 04s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/13cf3920efab48a995a14b2b34aa9166 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38ec5d8e0a854df1b699966bc007409b : SUCCESS in 59m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4ab61c0004b44ffb545a32ca2449e5e : SUCCESS in 1h 07m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9221159e4efc3b426455efb08708a78b1a1679c5","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 16:31:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4714b8afa86ff0a748351df5b385400e44175f35","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 17:32:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c5a2cbf8ff1220b1980e5da656d54dd5cebf1f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-26 13:52:55.000000000","message":"Abandoned\n\nsuperceded by 14.4.1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bdb735397b1aa9519a95d7fd22c1379658aae3d","revisions":{"6bdb735397b1aa9519a95d7fd22c1379658aae3d":{"kind":"REWORK","_number":1,"created":"2023-08-28 12:53:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/892919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/892919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/892919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/892919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/892919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/892919/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 12:53:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 12:53:13.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.4.0","message":"update constraint for oslo.messaging to new release 14.4.0\n\nmeta: version: 14.4.0\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Commit: Jay Faulkner \u003cjay@jvf.cc\u003e\nmeta: release:Change-Id: I5520e36f1863648920aefe3f4a88bd69fe9953e8\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I76465fb0bfd30d473b293ee26570d32b2ea6bef8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bdb735397b1aa9519a95d7fd22c1379658aae3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bdb735397b1aa9519a95d7fd22c1379658aae3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
