)]}'
{"id":"openstack%2Frequirements~892923","triplet_id":"openstack%2Frequirements~master~I2d07adc29b930a509f425940728d0ff1b0810054","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2d07adc29b930a509f425940728d0ff1b0810054","subject":"update constraint for oslo.policy to new release 4.2.1","status":"MERGED","created":"2023-08-28 13:02:29.000000000","updated":"2023-08-29 18:09:33.000000000","submitted":"2023-08-29 18:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892923-new-release","meta_rev_id":"4f23633f2322be258fd4d8564e85abdf3a6d4a81","_number":892923,"virtual_id_number":892923,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 18:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2023-08-28 16:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2023-08-29 12:23:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2023-08-28 16:35:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 14:28:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:35:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2023-08-29 12:23:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"abe648f6651cef41d767d2ad4d83a57db0ea8dd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-08-28 13:02:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c2c5fcce23e6c9fd2cfda4c195d1ee4c40b6bf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:28:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/983153d7c33e41bbb571266429a5f24a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a57c9e0e6cb947ffbd8cc5ef34eb19e3 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/789f252f3721459cad81f86e43167b54 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13feb415157042d1a1ba1b731a2ad30a : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72edff8972b14dfbbf3dc9a43bf02454 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97b7363baf164805abb15f944872ff4c : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3949c0138df44957b24b39dc85a195c3 : RETRY_LIMIT in 1m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76bda480e53246a4a71ef8062885dec1 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f040c1e986e54d9f81d1a47b4a11ebf4 : SUCCESS in 4m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/20619969d6d1422c85da1fb39e75d90b : SUCCESS in 7m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7b8b1973d06e4b7aa9e31477960c327b : SUCCESS in 12m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9590a722ff4040a8898c6f4ed9e15060 : SUCCESS in 8m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/329c018fc1eb441f924531acc8ebbd84 : SUCCESS in 13m 10s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3ad0c69e1be74cfca804a8a8b603d3cb : SUCCESS in 7m 25s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/208282496f924663a18b28154b5d575f : SUCCESS in 4m 38s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/af646b1715a24effa9a34de9cebf296f : SUCCESS in 5m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1ae18ba08f09441b8ce453c4c000ad8b : SUCCESS in 10m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/661240771f3c44d08baa9a21c201e071 : SUCCESS in 4m 51s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/219553cfee8e469aae7d44ba004c099c : SUCCESS in 5m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8b1aca2d73c54d4ebd7b25e5a7813a6d : SUCCESS in 8m 09s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/31ec70cdcb494434aa987ef3dd57924f : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1055165df1f34e99907229d923f2b218 : SUCCESS in 7m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/028490dd1757462a87610d39a287bab9 : SUCCESS in 10m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/b809667123bc4a08b1e3fb15c6ec2205 : SUCCESS in 12m 01s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/85cd393bc04e4ab19b9d7005efd197bf : SUCCESS in 5m 18s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/fa5465489f074c809524f1c3fc238aa3 : SUCCESS in 11m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/67cbaab4d69549988c184af5b244e088 : SUCCESS in 5m 26s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/fb7be335eb8b44408e0cdec9ca5f0c3f : SUCCESS in 11m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/7fd34e9f49b84c3b9f72ec8432fca162 : SUCCESS in 25m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/836157bc786d4965a22cac69d5205eeb : SUCCESS in 10m 39s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c455ee8215f34e3b952c14d1a5e3a665 : SUCCESS in 14m 29s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/62fe48403153484a808eeac39499df19 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0c83b995df1940fc95b61595861bfdce : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b74dc45c24be41ce88f903191b8e3e0a : SUCCESS in 24m 24s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9a5f574ee0cf498b99ec34c3616612c7 : SUCCESS in 5m 53s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/e0dc6b08806449789301843b948ee3d2 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3f2171a478540e9a7232ced135a5f5b : SUCCESS in 6m 59s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/3f8cf5e8014f477190813efdf9a3b72a : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/288eec951dc84e5bb6d871943de3cf96 : SUCCESS in 1h 00m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f2fa488fa6a46908cb0941dc4de16d6 : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2e97df8dcb578b3a0f23ae81e52c048e3cc3f606","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2023-08-28 16:35:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f30daa8462a0ba9f4d5f5d1b5bebdb2a26044821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:36:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"487ad66ed15a9b5076d51dfee80c5ba6b4a9e5bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:42:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37e157cf86e743bc8cedf446d130b332\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f3de97245e44530913526fefbf7a8cb : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d949d17d2347d4b84a839afa196b1b : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44daea61185d4121a7659803c250e901 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff10a9414402432a89095021aa239ba1 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65459f7608144f55b765bc8c0bcff6b2 : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36c15b10d387425299561066fb4950ad : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f61aa91ff01146c89fa172e406e26398 : SUCCESS in 5m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/70309835aa6e499c82a8399ad0c17824 : SUCCESS in 10m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/86c54b1f11764c609cdb2f73342c9cbc : SUCCESS in 13m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/54d0232f9c5a4b86be667e8975f9c8a5 : SUCCESS in 9m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/23a3013552d64006ace4409c769883eb : SUCCESS in 12m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/e192552d404d4059ac2f771459d3861a : SUCCESS in 6m 31s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f58af82ef17a49b29421eb161a627594 : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/426267e5240f4a25a1368ec5dd19c1c1 : SUCCESS in 8m 19s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c8eb3c154e7c48d9838cf769910780ff : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d42044c81cd4fb68f98bc55ef794ea7 : SUCCESS in 8m 41s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1b3bd59ef2e040ada3ea517931b06983 : SUCCESS in 8m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4a195135234b40cc95cbd1a935e1f733 : SUCCESS in 14m 13s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/6d7dabc947484fdaaeebd11daace7658 : SUCCESS in 3m 58s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2b4cf52d5d824f0685c0caec808e1a83 : SUCCESS in 11m 40s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/f908cb77d83c44859774781e97d697c2 : SUCCESS in 4m 23s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/690e5831eea846a0ae2de93a957a2813 : SUCCESS in 27m 24s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2336b8907e6e4e428cfd296a60762eab : SUCCESS in 11m 50s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4eb9d22940344ade98a416b2852cbe12 : SUCCESS in 13m 31s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/d8e483f0bc4d47f8b48a480a25992e23 : SUCCESS in 5m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b570291f8a344768f835860e330dc63 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/773eb9be99854256a9900a3283ace293 : SUCCESS in 26m 31s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/03fc9afef4964fc99733583a0867f30c : SUCCESS in 6m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d44cdd1e33654315954c7ef379035bb9 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdd996eef43e4dda8f332a15cf414180 : SUCCESS in 7m 48s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/9c349de021014fb6969f19531e2f6689 : SUCCESS in 15m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d27f6cebe8941c1ad3dff2ee402199d : SUCCESS in 1h 04m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59ba19ca02914433b2b22cb9e9a97a53 : TIMED_OUT in 2h 05m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f694e62b0df5c8e0abfaff87c211550a9b008def","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-29 12:23:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46f99620b89801784f1cdfd81708550ba3e18680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 13:57:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20cfc4fd7174e9cb6b894d02ee706b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde731b85d2e40a3891f8165f54c6aa5 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d56bfbd59d4795b48ae54cbd9fc6cd : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31350d3448374bc2aef15546e1ab2203 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93ad7e534bb849a889867511aa2f5279 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c0629b5231d483589e719ce7431f34d : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4182441773046c0a0dd43031b4062fc : RETRY_LIMIT in 1m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a2beb68503e4c3db3808845837f6afe : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5ec13b335174762a28337cf727d4903 : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e63ce287446455a9be2cfc802aff294 : SUCCESS in 8m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5f22990fb91c4e95bd28a09bc12afee5 : SUCCESS in 10m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1afccda7614d491f8945bd3d03cbb2f8 : SUCCESS in 7m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0fa4b5ed1a8148f591e21a828beaa3a8 : SUCCESS in 11m 26s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/987e8d4a194649d3a9eac394a161f5e7 : SUCCESS in 6m 42s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/1cf4dda1ff6a4dd1937295f847ae78a0 : SUCCESS in 5m 33s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/77337474e8df4387baca9098c7999ae0 : SUCCESS in 5m 19s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8e601a2ad0514ababae56613a9b16141 : SUCCESS in 12m 23s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/11114c6c1a1b43338af24682e19178b8 : SUCCESS in 5m 14s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/451aa42b1363429588b5541d73ed1d30 : SUCCESS in 7m 18s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/2b90e42e70d440058a1572c7a52e3144 : SUCCESS in 7m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a8222048653347e59914c5add35d23da : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/55f14c5e352f4aca81a7bbf2066813a2 : SUCCESS in 8m 28s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ada7f36002764b00aef041bcf23cac2c : SUCCESS in 8m 06s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dda0fe9fa41040b2b189469ba5cee111 : SUCCESS in 11m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2aca83d8643f4eb880a558d98796cb62 : SUCCESS in 6m 43s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2e29419a28ac4601b457a075703bff4e : SUCCESS in 11m 25s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c9dc7217c8fd45158580a32983780c1c : SUCCESS in 5m 28s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0e3515020a7d4ee0963f37a66a7419be : SUCCESS in 11m 10s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/04eb94dd4c2d46de94b44fb18904e460 : SUCCESS in 24m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c5ffb42c760644bc88d60492d83ea697 : SUCCESS in 8m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8fd7543abea74b559b9280258847fab9 : SUCCESS in 13m 13s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7bfa2ec4ab0d46188dca87ce025bbf73 : SUCCESS in 6m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59ed7d714abf4ff9990045883af0c98c : SUCCESS in 5m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e08f70a31b7248c888f8db27ba09965e : SUCCESS in 24m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/86d04c40f56b4f399c9319a38c683dd3 : SUCCESS in 5m 12s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0703aac811ab45cbab77b347f737df51 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04798f6ebeee42079fd246084a83ac35 : SUCCESS in 7m 50s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5eda2151317c4cd88392500b08db28d3 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd21769bc3af4641a994f173770ff95f : SUCCESS in 1h 03m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a974927ee0584963af706b12984de84c : SUCCESS in 1h 28m 56s","accounts_in_message":[],"_revision_number":1},{"id":"32e50490498a8d947d97a08e4653fa3dfdad7db4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 13:58:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88a5b4a6774a15a80ebc23de18cd212be3035d3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:08:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13336999a57f4c30b7ce6bf9446f6248\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342d06f1278e420bb58438006dc0be06 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3595b42d9c44b128b7346383e08bf98 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a47e16d6dddc4d90983aef77745a5daa : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e429976dd9c5455c8f0d5a5ce3e078c1 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb28ab9d7cd1445a8623ea87382bfbe2 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/726c54814d4f498185a675c1a5adac09 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb517eb630b343bd95481ef84617572a : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6b6f756fa42b44c8b6d63bb42d84a6ab : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a181e21337004829ba29d71bdfb79202 : SUCCESS in 11m 39s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bc98ffcf54ba44d0b8644ef7f17825a1 : SUCCESS in 8m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/001a77c476624b30bcd2f20be30623e0 : SUCCESS in 12m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/1f4ead50abba4e5888db4db8ab547aeb : SUCCESS in 5m 30s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/265664ec4adc4929975fe0f5f083e82a : SUCCESS in 6m 02s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c0b0a13f3e114721973de1d02ca4e900 : SUCCESS in 9m 06s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/781ea1a665f04e78b5e0ff79ca993d23 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5833e216b5fb46d3ade5df0448d2d57f : SUCCESS in 7m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/81aa5891c3934618a9084ca6def29c5e : SUCCESS in 9m 51s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/dcedbdb8003441e88dcd23cdb50768f8 : SUCCESS in 11m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/1979a0d4c37447b089a04a3637f94fda : SUCCESS in 3m 45s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2d621756eb794d0091552bc713f036ae : SUCCESS in 9m 12s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4ad59c4e57864748bca6e251ef814380 : SUCCESS in 4m 46s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/55f28af1e07c4b3fbab96da2749f4d80 : SUCCESS in 28m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a7b8105da9ed457eb956fc1560ba3fe7 : SUCCESS in 10m 56s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/9af0221183f84c6ea54221ea612382c7 : SUCCESS in 13m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/13a58ddb62bb47e09a7871118716018f : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63f215e930e84e46aaf230e3898ee5de : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/39a0cbd130a74fe0bcc828c8053cdccb : SUCCESS in 27m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/387b2516a9974100b1e9d710ed77082e : SUCCESS in 5m 11s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7a689cb70069411aa25d945b5dad21b5 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/442c11b8c41b49ab821b21a56583af0d : SUCCESS in 6m 56s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/5e92916c189f435db70129c6d4ece548 : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa9e6c09391644f79ce8e3bdc8a2a009 : SUCCESS in 1h 02m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a32874eab61434cbda43b2194809609 : SUCCESS in 1h 42m 55s","accounts_in_message":[],"_revision_number":1},{"id":"36b6a606e83686fd1ac9786f10bc68f48b1bdd10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:08:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f23633f2322be258fd4d8564e85abdf3a6d4a81","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:09:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c1c293003041d9861c3f6570547ccc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba1ed8c49184dc0a9d831294551c804 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6adb3e98a749e54b22c20045c67eff0d2604ccf","revisions":{"d6adb3e98a749e54b22c20045c67eff0d2604ccf":{"kind":"REWORK","_number":1,"created":"2023-08-28 13:02:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/892923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/892923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/892923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/892923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/892923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/892923/1"}}},"commit":{"parents":[{"commit":"eeed0e945a70565e135cd7859a77cee778cd788b","subject":"Merge \"update constraint for ovsdbapp to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eeed0e945a70565e135cd7859a77cee778cd788b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:02:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-08-28 13:02:28.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.2.1","message":"update constraint for oslo.policy to new release 4.2.1\n\nmeta: version: 4.2.1\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie2e555b07de58078495c6ee4ae1ea50b962ff289\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2d07adc29b930a509f425940728d0ff1b0810054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6adb3e98a749e54b22c20045c67eff0d2604ccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6adb3e98a749e54b22c20045c67eff0d2604ccf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
